)]}'
{"id":"openstack%2Fkolla~758553","triplet_id":"openstack%2Fkolla~master~I2a1e2b6661fd4b62648756416e6147b604dda686","project":"openstack/kolla","branch":"master","topic":"openstack_base_constraints","hashtags":[],"change_id":"I2a1e2b6661fd4b62648756416e6147b604dda686","subject":"Use upper constraints for virtualenv in openstack-base","status":"MERGED","created":"2020-10-16 11:11:27.000000000","updated":"2020-10-23 18:54:27.000000000","submitted":"2020-10-23 18:53:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758553-1603479190604-041edf42","meta_rev_id":"3e4529ec824b6fac4cfdd8e9460fb21b1932074c","_number":758553,"virtual_id_number":758553,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-23 18:53:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-10-21 09:29:50.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-10-21 10:44:32.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-10-23 17:24:01.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-16 11:49:36.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2020-10-21 09:29:50.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-10-23 17:24:01.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-10-23 18:53:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4c0bca9f1b758fee5d9d65386de20014b25d8e58","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-10-16 11:11:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9472bc7e28fa63aa1b88791ab8443dd50f166ca2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 11:50:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/12cb1d7f78724d0db2c25b05589dbb83 : FAILURE in 38m 34s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7d50e96c9f384ec0875bd0a7c3c106c3 : FAILURE in 27m 18s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c5d5269217a243c4b671262953d5f67e : FAILURE in 22m 50s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4769b38a1fa8450f807d79efcbd1b162 : SUCCESS in 26m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ca16731408f365007a9b4541c80ecdd8d583400b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 14:28:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/9fb2bb187a934932a78dd3fd28d70fca : SUCCESS in 24m 56s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d7cfd322e0184a1fa78ac4d9da903f25 : SUCCESS in 28m 30s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/415262fd2fc34e4bba365b084ce46647 : SUCCESS in 1h 03m 59s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/0693cfbb0a6a47fba800945379e1c179 : SUCCESS in 1h 14m 15s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/4b3c87b1c04c488d8c052c4e3d8906de : SUCCESS in 1h 18m 40s\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/6886ca6ecb5247b285836b40d5338ce8 : SUCCESS in 28m 06s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/f5506572f9ee462982b24fb94c7c7be2 : SUCCESS in 2h 11m 55s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/460f1c1f21ab4658a7a4ea197ae8adf5 : SUCCESS in 1h 46m 14s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/96dc63de1e0e430abd65b8e56cf50308 : SUCCESS in 1h 38m 18s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/8bd3c9a262984b2bb9a399e5cb510a40 : FAILURE in 1h 05m 16s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/6fc367f3c05a4be1a958544ed682cf92 : SUCCESS in 1h 05m 54s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/cb599819557f40fa867b796a9a0ca8db : FAILURE in 44m 44s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/27829cfa758640a18f3c5b0a4d48cec3 : SUCCESS in 37m 22s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7d0c58aba6a54a4486289cb9f390bcbc : SUCCESS in 38m 26s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/85ba0d992a114cc083b47fae0f94caf5 : SUCCESS in 55m 30s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/552b5c285cbc485cb0deb3061b0bfa87 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9baa614f49a8481fac088b887a611430 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51f732c0f830484382eddcc64331d8be : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b29391b023b41f184cc516bd2dab263 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef6ebfcf6bd645a3b054da8a8a33fb5f : SUCCESS in 7m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4c8e226e0cd9483cb302dec0581b934a : SUCCESS in 22m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bb2441696d5b4a9784b012a515dd5f71 : SUCCESS in 20m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/efb32b278dbf4cbea5e82a3cf5df05cc : SUCCESS in 58m 54s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5bf77de4568a4ef59fd4d8d638942b0d : SUCCESS in 50m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/aef69353fd464a30a6ab5acb3900edc6 : SUCCESS in 1h 08m 33s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/7fc01941ba774e6fade739c8d5c3fb49 : SUCCESS in 1h 21m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/a5f98cfeac434ab2bdbc46d6de756d2f : SUCCESS in 1h 37m 51s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/68e01502147243e1801ff9b78fec8e17 : FAILURE in 1h 16m 10s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/a2ceade11d304d4c8dac939b3ac08eae : SUCCESS in 1h 01m 17s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/009cc3927c7a40dbac4ab06b704d1db1 : FAILURE in 48m 07s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/acb9b021a8b941b29b0ca9a3fd0c13e6 : SUCCESS in 51m 12s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/d8d50c774c2a4645b2c1fb4d0a4a6e47 : SUCCESS in 48m 46s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/24be53cca1d64bbaa8287c20e7589498 : FAILURE in 47m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a53c0c80d133c0a8f5eb12df9451d4dc9191c581","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-10-19 10:13:20.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"69880995f0baa6a051a498ae082d5fc9f9160b17","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-10-19 10:30:24.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"3b7517a3eae7b0aaf9c7008a12737925b8197a11","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-10-19 10:38:49.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"7019ad09fa29c9a56edcfb4154d9c52f7abd1d3a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 11:32:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/5521eb3d40db4b76a93d78263b5d5ec0 : FAILURE in 47m 20s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a209e6da247f4d158a08a9d027ff28bf : FAILURE in 44m 04s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c0d2807f65e74f94ae8d67097253ecc9 : FAILURE in 32m 23s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b646fad8c25246c88091a6a8876c19c4 : SUCCESS in 38m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"642e37a58225a03d3a72503f83a83481115cfbfd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-10-19 14:21:29.000000000","message":"Patch Set 4:\n\nor let\u0027s drop virtualenv finally...","accounts_in_message":[],"_revision_number":4},{"id":"91f2f27f300aaa762aee50dc6d7f60fcefe5c023","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 14:58:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/46c9025eac4e4af28ecf9cdcdab5b4b5 : SUCCESS in 25m 51s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2c697266dbf14d4eac6c3d7cf8753f01 : SUCCESS in 28m 50s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f715d8bf244d430da1890e7518048945 : SUCCESS in 1h 09m 38s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/594e7d580c614f39bfeb9d5324c9d762 : SUCCESS in 1h 06m 13s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/23c7865ee70b4d668d4a4b9d71674dad : SUCCESS in 1h 08m 18s\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/63c29fa970ed45008e85c1f24b9b4219 : SUCCESS in 31m 37s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/f6ad9f49e4024b968c7d3f042694f417 : SUCCESS in 1h 36m 00s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/cf0c8e2cfb4f4768b3f87ce1373721a1 : FAILURE in 1h 47m 23s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/e78769f7888844afb4c91cc463b84dcb : SUCCESS in 1h 44m 32s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/9ae16dc55fb6464ca3658a6ee960abe6 : FAILURE in 1h 02m 33s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/13defd85e60141929ecd44f5013a54ea : SUCCESS in 1h 10m 20s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/6e588a014b924967a3fc083c0a3e70e5 : SUCCESS in 1h 35m 01s (non-voting)\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/1935004fcb774d99b072eed2aa944706 : FAILURE in 55m 45s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/dfc06086a34a4a8e9efc4c2d1f7e2cb5 : SUCCESS in 19m 44s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8269befb13d7412cb226b5a3e98ef845 : SUCCESS in 19m 11s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bf9b19c992c641d39e6a272ece6a54cf : SUCCESS in 56m 30s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a267b4a839140f88f95a40b2ad9651e : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f7e6ea1b8a9428184af8d51e1b5ada5 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e46fb67083b14ed5bb05f80d91bdc7ad : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1704b5d4c45f4b2b87e64c12c507f3ee : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30a6ca73223f4b4980821c9c2b37afca : SUCCESS in 9m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cf24e792e11c4c77bc99ed4473a20a7f : SUCCESS in 22m 51s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/48b5048c6ba643ebbe98588e5e79ba53 : SUCCESS in 22m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/807e6c3b27db4945a71490626a42725b : SUCCESS in 52m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/445c6fa2fbf64ad2ab9c59f53c9b09b2 : SUCCESS in 50m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5ead912630654c11a900b5045d34a1f6 : SUCCESS in 1h 12m 13s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/93e5a595c50441b9baa75fa5c99277a1 : SUCCESS in 1h 22m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/78a6545c0a5443409bf0399acc766bdb : SUCCESS in 1h 33m 46s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/48889cfa5f604423b784b2eb64f7cdaf : SUCCESS in 1h 23m 10s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/0b49ffd1eccc4c918cef791489621f02 : SUCCESS in 56m 23s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/dab7312897714c13801d9e87c9a01282 : SUCCESS in 1h 16m 57s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/69ad5d738f514f0f93ce4432242d21b4 : SUCCESS in 46m 30s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/c9bed1e4747440e7851067292275c6da : SUCCESS in 51m 22s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/ea6d77fd26624f1ebef7c0a68cb04c32 : FAILURE in 45m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"98d84b36540ec0f1430af0f66d73661f7d3812c3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-10-21 09:29:50.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"07945192ddf3a2c501f8abefc593ab821d4e1cb4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-10-21 10:44:32.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"c59da1062ae21a065d7c94c17455d9001ef7f207","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-10-21 10:44:43.000000000","message":"Patch Set 4:\n\ndo not approve - gate b0rken","accounts_in_message":[],"_revision_number":4},{"id":"4b044e78921e9643cfff389587dbc52281c01869","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-10-23 17:24:01.000000000","message":"Patch Set 4: Workflow+1\n\ngreen now","accounts_in_message":[],"_revision_number":4},{"id":"eaf7548f0703924df7283ce43d9e95c9b6db9b9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 17:24:14.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"253ad83508fc9a6483ae52a14b9a4f8d5e13d32a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 18:53:10.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/83d643c9e065408da59b03b4b16d36f4 : SUCCESS in 41m 27s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/96eb945697124e1b90558b78cb6dac65 : SUCCESS in 1h 15m 01s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/45217035c47b4090a2f71b7e9aa7f474 : SUCCESS in 1h 09m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d4a934dcd3e4516b64c35ff73461c5b : SUCCESS in 7m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/061cb395bc2b4398a2655599436b9218 : SUCCESS in 7m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c598364435794602bfeb3542d2920d4d : SUCCESS in 7m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ed73237b43441e09e0942d08fcec287 : SUCCESS in 6m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b29298a7e3134a2498c1332aabaad5b7 : SUCCESS in 8m 18s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c2d962fe44404c5f86c29844a2a8bf5e : SUCCESS in 24m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e6d61eded0604330a765a2e7afdd9f25 : SUCCESS in 55m 08s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d2cd505fd2c0498882a11c6e6d780b25 : SUCCESS in 1h 01m 45s","accounts_in_message":[],"_revision_number":4},{"id":"fd25dd6fc2375319d60ec2d3d7d9c596d63738df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 18:53:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"8fad776ac4f87bba6f8da509ccf4e035a4712cdc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 18:54:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01a9dd88bde14f76b9ad6ddc81c0f1f4 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b1709a87ed6bead417a3cd6848de068e06190370","revisions":{"c03058fea32921c7fa248496da45866330601c4f":{"kind":"REWORK","_number":1,"created":"2020-10-16 11:11:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/53/758553/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/53/758553/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/758553/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/758553/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/758553/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/53/758553/1"}}},"commit":{"parents":[{"commit":"c26086b1bea35784a10d9b9682b7301175d27603","subject":"Merge \"rally: Add rally-openstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c26086b1bea35784a10d9b9682b7301175d27603"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-10-16 11:10:03.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-10-16 11:11:24.000000000","tz":120},"subject":"WIP: use upper constraints in openstack-base","message":"WIP: use upper constraints in openstack-base\n\nSince Rally has virtualenv in requirements, and is using contraints - it\u0027s\ntrying to downgrade virtualenv from inside a virtualenv.\nLet\u0027s try to install proper version of packages in openstack-base.\n\nChange-Id: I2a1e2b6661fd4b62648756416e6147b604dda686\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c03058fea32921c7fa248496da45866330601c4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c03058fea32921c7fa248496da45866330601c4f"}]},"branch":"refs/heads/master"},"d195c1cee5930e3e895dac02369064bfb3bf30fa":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-10-19 10:13:20.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/53/758553/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/53/758553/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/758553/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/758553/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/758553/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/53/758553/2"}}},"commit":{"parents":[{"commit":"c26086b1bea35784a10d9b9682b7301175d27603","subject":"Merge \"rally: Add rally-openstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c26086b1bea35784a10d9b9682b7301175d27603"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-10-16 11:10:03.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-10-19 10:13:12.000000000","tz":0},"subject":"Use upper constraints in openstack-base","message":"Use upper constraints in openstack-base\n\nSince Rally has virtualenv in requirements, and is using contraints - it\u0027s\ntrying to downgrade virtualenv from inside a virtualenv.\nLet\u0027s try to install proper version of packages in openstack-base.\n\nChange-Id: I2a1e2b6661fd4b62648756416e6147b604dda686\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d195c1cee5930e3e895dac02369064bfb3bf30fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d195c1cee5930e3e895dac02369064bfb3bf30fa"}]},"branch":"refs/heads/master"},"322384f0b1ec7882675ea7e7f7c652e2ee728a12":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-10-19 10:30:24.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/53/758553/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/53/758553/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/758553/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/758553/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/758553/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/53/758553/3"}}},"commit":{"parents":[{"commit":"c26086b1bea35784a10d9b9682b7301175d27603","subject":"Merge \"rally: Add rally-openstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c26086b1bea35784a10d9b9682b7301175d27603"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-10-16 11:10:03.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-10-19 10:30:17.000000000","tz":0},"subject":"Use upper constraints in openstack-base","message":"Use upper constraints in openstack-base\n\nSince Rally has virtualenv in requirements, and is using contraints - it\u0027s\ntrying to downgrade virtualenv from inside a virtualenv.\nLet\u0027s try to install proper version of packages in openstack-base.\n\nCloses-Bug: #1900419\nChange-Id: I2a1e2b6661fd4b62648756416e6147b604dda686\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/322384f0b1ec7882675ea7e7f7c652e2ee728a12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/322384f0b1ec7882675ea7e7f7c652e2ee728a12"}]},"branch":"refs/heads/master"},"b1709a87ed6bead417a3cd6848de068e06190370":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-10-19 10:38:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/53/758553/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/53/758553/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/758553/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/758553/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/758553/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/53/758553/4"}}},"commit":{"parents":[{"commit":"c26086b1bea35784a10d9b9682b7301175d27603","subject":"Merge \"rally: Add rally-openstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c26086b1bea35784a10d9b9682b7301175d27603"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-10-16 11:10:03.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-10-19 10:38:46.000000000","tz":0},"subject":"Use upper constraints for virtualenv in openstack-base","message":"Use upper constraints for virtualenv in openstack-base\n\nSince Rally has virtualenv in requirements, and is using contraints - it\u0027s\ntrying to downgrade virtualenv from inside a virtualenv.\nLet\u0027s try to install proper version of packages in openstack-base.\n\nCloses-Bug: #1900419\nChange-Id: I2a1e2b6661fd4b62648756416e6147b604dda686\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b1709a87ed6bead417a3cd6848de068e06190370"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b1709a87ed6bead417a3cd6848de068e06190370"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
