)]}'
{"id":"openstack%2Fkolla~759387","triplet_id":"openstack%2Fkolla~master~Ie5ae119dcf31e05112405c8a307bcd2613d45de8","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Ie5ae119dcf31e05112405c8a307bcd2613d45de8","subject":"base: use Victoria repos for Debian/x86-64","status":"MERGED","created":"2020-10-23 09:49:48.000000000","updated":"2020-10-26 20:03:24.000000000","submitted":"2020-10-26 20:01:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"759387-1603742509200-ce1bb64b","meta_rev_id":"442f21911ddc2363cb6ba9718385038ca026255a","_number":759387,"virtual_id_number":759387,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-26 20:01:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"recommended":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":1,"date":"2020-10-26 09:50:37.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":2,"date":"2020-10-26 17:22:53.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":1,"date":"2020-10-26 17:22:53.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-26 09:50:37.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2020-10-26 17:22:53.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-10-26 20:01:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3f5c8e1fef1bf3d469c411485951ef23a94e5f44","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-10-23 09:49:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8170a4ded91700bd054b82304d596ade618cdb70","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 09:58:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"87947661b730d637fcb6b61d86ee6e52706a5a5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 13:36:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/86bc9fa7ce7d41b29a20673ccdaf299a : FAILURE in 31m 43s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/5780c80d727e411c9e7ec03d616a7c83 : FAILURE in 2h 38m 54s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/41158ba07d664660836d58de34ccb9f1 : FAILURE in 42m 00s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/00c4e8e4d64a43c5981cc87309875138 : SUCCESS in 1h 02m 32s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/94141b38dbfc4c26a0705ed64d4f6d49 : SUCCESS in 1h 13m 49s\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/ab2fb91e431b4e9da96c4f1f093b1610 : FAILURE in 21m 59s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/a6d09741009049a0882cc8a3ce133253 : SUCCESS in 1h 34m 17s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/0d6ab15960bb4e519f45e21ae25d0482 : FAILURE in 1h 56m 19s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/7e3d3b225cb94e82b53e4e734f845f8d : RETRY_LIMIT in 3m 55s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/f04a43a3f4a44ccb9115419bfbc5fe28 : FAILURE in 1h 02m 20s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/42208c2f38da4db49be3c3eb087506a8 : FAILURE in 34m 29s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/3d14ac91cc6b414aa41d5d7851c0c7e9 : SUCCESS in 1h 48m 35s (non-voting)\n- kolla-ansible-centos8-source-mariadb https://zuul.opendev.org/t/openstack/build/b1e61eb46c4e44069036860a8c2b4ac1 : SUCCESS in 32m 09s (non-voting)\n- kolla-ansible-centos8-source-masakari https://zuul.opendev.org/t/openstack/build/3bf924770cdf4653b74fc26853e078b3 : SUCCESS in 1h 17m 58s (non-voting)\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/1f15ddfcd3fd42738510dd9fd4a429a2 : FAILURE in 53m 35s (non-voting)\n- kolla-ansible-centos8-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/ff2fdb380220492a936c02a512ab66a2 : FAILURE in 26m 53s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/120c40183b0f45b08e4033887b76a1af : FAILURE in 18m 49s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/64507742f2ac4cf481d88478f4e29205 : FAILURE in 20m 24s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6d07944994df4769ac149a1ad9a636c0 : FAILURE in 37m 47s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72911d10a6ea45178d10f4e51099ead8 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7416dfabb13e426fb8a0c18f8a988631 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93b184d0b6f74b3fa7df2ab1551396a0 : RETRY_LIMIT in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ff7ee03ab54418c955161e953a29019 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fef7d8770fd74a17ac9b439b5465aedf : SUCCESS in 8m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8be2f25e61d34904b12b9bf513ca0423 : RETRY_LIMIT in 3m 48s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/67c9bd9e05124efebc1f71dfe5bfa381 : TIMED_OUT in 3h 03m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fa91b70936cb46918b99ed056331d3f1 : FAILURE in 35m 43s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a3264d7510d5418d9b68581af266e5d9 : SUCCESS in 50m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f389b7098efe46ef8d3315c2f3e9c5ae : SUCCESS in 1h 06m 29s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/a7f66c2c0bb14641a2ec2f9d540e86cd : DISK_FULL in 1h 19m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/d972a29257c54b3ab5a65ed14c43ae82 : POST_FAILURE in 1h 31m 07s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/46c808d0f4434db3be82e81d833bb09b : SUCCESS in 1h 21m 46s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/0ec03c9185d84e91bcc1976893f946bd : FAILURE in 33m 43s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/c872501837d54a3b95258449e30a933c : DISK_FULL in 1h 04m 14s (non-voting)\n- kolla-ansible-ubuntu-source-mariadb https://zuul.opendev.org/t/openstack/build/3f1a4324159149dbb9f7b47bc6096a1f : SUCCESS in 29m 47s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/e49799bcf1014d578131c8aa4e337fc1 : SUCCESS in 51m 37s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/75f802a4a8454f8b93b568bd7cb31f0c : SUCCESS in 54m 27s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/0bcc68f9445e4033be53765ede33468b : FAILURE in 43m 40s (non-voting)\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/ad6be6a5d71b46d1af56e58f86f777eb : FAILURE in 17m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2596dfc3f5639251b88a292ac7ec0f82691d13a2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-10-23 17:54:08.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"2444b203a3f92d8a04b02edd036f9fc1c858e6ca","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-10-23 18:13:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fb10aa09e8277cba5d8b7d203291b4527fe99669","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 18:30:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/47deb7aae00744e99085f1c2d9d37ceb : SUCCESS in 33m 36s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/03e7e54f2ef0486cbe7a6bf78598d668 : FAILURE in 26m 05s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fbd1eb9ec6364110ab8cc06a2608467a : FAILURE in 22m 04s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/07a54c97836d4967bc15d187f3d44c50 : SUCCESS in 31m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bf12eb7e809c59703d67f3efc2184b2a7f285ad0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 21:34:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/8c981ebdea0643439adcf391a02fc6ff : SUCCESS in 24m 01s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/71447468d5284accbe59e357505ae72d : SUCCESS in 24m 50s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/18ac302feb744d3fb2d21c9575c9631f : SUCCESS in 1h 09m 33s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/16ce18c88e6f4fc3ba92f74489c52869 : SUCCESS in 1h 09m 14s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/92d2734e3ad245de9fc05d665658a39e : SUCCESS in 1h 09m 04s\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/a529293ea9d94d91a8c770a451ff084f : SUCCESS in 31m 26s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/8c5b08ae00584ed99f3f5a47e5f10a8c : SUCCESS in 1h 38m 59s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/e679a17040f2421e8a3716e76c7855a5 : SUCCESS in 1h 45m 25s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/3e013b2c3245406c834644ad2ca80482 : SUCCESS in 2h 13m 39s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/3ea7a2195f394a3b8ed151d281c09285 : FAILURE in 1h 03m 21s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/eb01860f1f8841f7a66dcf295baf1276 : SUCCESS in 1h 08m 05s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/811563cc108e4cfebd00f4e080b0f927 : FAILURE in 1h 01m 25s (non-voting)\n- kolla-ansible-centos8-source-mariadb https://zuul.opendev.org/t/openstack/build/4966b2cf23f54ea9981a2fac3183673e : SUCCESS in 31m 59s (non-voting)\n- kolla-ansible-centos8-source-masakari https://zuul.opendev.org/t/openstack/build/0381152276c848f4bcb5944230e4688a : SUCCESS in 1h 28m 38s (non-voting)\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/5474c360eb4a4e2cbb2f43d826ff19fc : SUCCESS in 1h 48m 50s (non-voting)\n- kolla-ansible-centos8-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/8b61500b03754d29b4766153de04e0a2 : SUCCESS in 31m 40s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/930c95fee74b435095126af462654b61 : SUCCESS in 19m 37s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/39f152fecf2a45d48d9e18fb5d3edd3f : FAILURE in 51m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/268d8cae69c44b178ed28f743efc8800 : SUCCESS in 1h 19m 41s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40eaa1b283604165b5243beaa76e6e51 : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e64eab438c84c6ebdd625b1b24d057a : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53c7aad2c23c40f6a83bee4840aff735 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5c714e4744740a0abec6666036d8ac2 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d36347c125942ada556a815f928e334 : SUCCESS in 7m 53s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/00b49dd08da14579b029ce3ff21a7d78 : SUCCESS in 50m 30s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/60ddaf0894214227a54e79d7712dcd0a : SUCCESS in 21m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/24f8914fc525468eb94d0969a17f1ec7 : SUCCESS in 1h 23m 13s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/351bc13ba4af45d696f8dede51f58a25 : SUCCESS in 52m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9af4baa312114d1aab4f3906f8b60a08 : SUCCESS in 59m 06s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/e0b36bbb6c94461283e55f2820feb873 : FAILURE in 1h 29m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/7e7769c79dc34a93b9e55e557294458c : POST_FAILURE in 1h 32m 19s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/51e9b911c5e74e9ea67955bbc789c73c : DISK_FULL in 1h 05m 15s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/c5f29f4a6901469c9a548cdd004fdf1e : SUCCESS in 52m 57s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/ca2415fd8c4a4c248c1314b9b7796897 : SUCCESS in 1h 13m 50s (non-voting)\n- kolla-ansible-ubuntu-source-mariadb https://zuul.opendev.org/t/openstack/build/9e66eb166fa74cc0b217ee38bda0bdf7 : SUCCESS in 50m 40s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/18d4388727a94e948e2c1a236302e778 : SUCCESS in 1h 25m 43s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/522b322e3a744e21be5b9f5f3a09f375 : SUCCESS in 1h 54m 44s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/6f91f1dcf6144c0880a9df117ec27fab : DISK_FULL in 1h 27m 40s (non-voting)\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/995ea7e3b477461bb0eda2396259eb0d : SUCCESS in 27m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f2da519732085e0d12ecc48409bf64a4d6853f2d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-10-24 17:01:19.000000000","message":"Patch Set 2: -Code-Review\n\nINFO:kolla.common.utils.manila-base:install: cannot stat \u0027/usr/share/manila-common/policy.json\u0027: No such file or directory\nINFO:kolla.common.utils.manila-base:dpkg: error processing package manila-common (--configure):\nINFO:kolla.common.utils.manila-base: installed manila-common package post-installation script subprocess returned error exit status 1","accounts_in_message":[],"_revision_number":2},{"id":"6c00a41e5fabe09f47c74649b939866e3ece645a","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2020-10-26 09:50:37.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"bd7067e48f679df0dd140dba87f94eb9e607319e","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2020-10-26 09:51:10.000000000","message":"Patch Set 2:\n\n\u003e INFO:kolla.common.utils.manila-base:install: cannot stat\n \u003e \u0027/usr/share/manila-common/policy.json\u0027: No such file or directory\n \u003e INFO:kolla.common.utils.manila-base:dpkg: error processing package\n \u003e manila-common (--configure):\n \u003e INFO:kolla.common.utils.manila-base: installed manila-common\n \u003e package post-installation script subprocess returned error exit\n \u003e status 1\n\ngoing to check","accounts_in_message":[],"_revision_number":2},{"id":"2f31c045e123d48ce9aebcb6f874407f61339d1c","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2020-10-26 11:44:00.000000000","message":"Patch Set 2:\n\n\u003e INFO:kolla.common.utils.manila-base:install: cannot stat\n \u003e \u0027/usr/share/manila-common/policy.json\u0027: No such file or directory\n \u003e INFO:kolla.common.utils.manila-base:dpkg: error processing package\n \u003e manila-common (--configure):\n \u003e INFO:kolla.common.utils.manila-base: installed manila-common\n \u003e package post-installation script subprocess returned error exit\n \u003e status 1\n\nFixed, new revision in repo available Version: 1:11.0.0-2~bpo10+1","accounts_in_message":[],"_revision_number":2},{"id":"f2f71d637af47924fe73b6592aaa843e45d83909","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2020-10-26 12:17:54.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"614570795aef54a18e2605918754cb81a991fd02","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 12:51:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/5a9fe2ba34af4f80824f3e448dfac351 : SUCCESS in 29m 04s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8bec451081154773b000f93197f8f3d1 : FAILURE in 23m 53s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ea1e97c2c3c74e57b0340e399d3ab8b8 : FAILURE in 20m 50s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9472023362544fa1ba450c6303ea3d66 : SUCCESS in 27m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"30458cedc4bd787c892501d9623c1063a56dfcb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 16:14:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/13dd09c3a199420e8e050a51383afe9f : SUCCESS in 20m 42s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/963581133b34418a904b95b35797a5c6 : SUCCESS in 23m 45s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7be40d79b0cf4cca8815ee098657b64e : SUCCESS in 1h 07m 49s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/88da516fb20b4f998cceeab783dab779 : SUCCESS in 1h 39m 48s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/337a20558b9d4df799967b7927204a90 : SUCCESS in 1h 18m 35s\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/95c2d42c5bfd43e7be9a74a38f8efdb2 : SUCCESS in 27m 53s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/c380e7b9884b40af89f17bba7914b044 : FAILURE in 2h 03m 47s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/fa8f9c1993484ab68a7f911aef75891f : TIMED_OUT in 2h 43m 15s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/ac9e0c4f624c4ec887af8eeb0990174d : SUCCESS in 1h 44m 58s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/6179f25995ff4b2584e80a0bb25c9968 : FAILURE in 2h 02m 06s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/d39d1c05d79147bfa5a7452029986704 : SUCCESS in 1h 09m 43s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/3172cdd5198b49c485738d5f4add0201 : FAILURE in 1h 00m 56s (non-voting)\n- kolla-ansible-centos8-source-mariadb https://zuul.opendev.org/t/openstack/build/a285239d2f594e6cbf2552e8c62455bc : SUCCESS in 32m 27s (non-voting)\n- kolla-ansible-centos8-source-masakari https://zuul.opendev.org/t/openstack/build/9b1cc2958c824b84b89e0f357ead61d1 : SUCCESS in 1h 52m 50s (non-voting)\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/6327a7b509ad4ae5938b3fa1ee876389 : TIMED_OUT in 2h 11m 17s (non-voting)\n- kolla-ansible-centos8-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/631904a6b2cd4eb19ef7c7844d3d0eed : SUCCESS in 30m 34s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ddd26306260f45b48145949f1c0d8a62 : SUCCESS in 19m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/59416108e70d434f8a90760df6e2f1c9 : SUCCESS in 1h 55m 37s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a6d764cfda4e410783e8f1e8a882408e : SUCCESS in 57m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2aefe7e9bf214136af54a1a31b051720 : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/115c6dddfac641619d47014014ef240f : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c119ba0155ca4f5ba0a5425297474bf6 : SUCCESS in 7m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c8ea7ad13bf4a749eb175217d9c511c : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfc7b7e129b54dbb8ff1ea8e10b66461 : SUCCESS in 15m 22s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/22ab7d806b384abe8905d7e25e0e2552 : SUCCESS in 19m 51s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/72f914a3f58543189fa575bdade0f9a7 : SUCCESS in 1h 54m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9241bb8f94454ba1b4cbaacba46f238e : SUCCESS in 50m 56s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/162a9c44eefa4fe58bc79b01ae5d947e : SUCCESS in 1h 43m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/62d8ba325ad44592acf9afa6d38dba59 : SUCCESS in 1h 00m 34s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/0bbe339be3d84f16965f77c4252e5ef5 : FAILURE in 2h 04m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/8b7e5f5169d745a79f23552dfa4f05f4 : POST_FAILURE in 55m 38s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/584cbc15855f4581ac62c47e5dfcdfb5 : TIMED_OUT in 2h 10m 19s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/a2de59e4de43493a8f1d4539d008243c : SUCCESS in 55m 08s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/1ddb32e175b04d95a7475d6234af0e8f : SUCCESS in 1h 14m 36s (non-voting)\n- kolla-ansible-ubuntu-source-mariadb https://zuul.opendev.org/t/openstack/build/debaf1e088b84503b37efdf3a8e9c20e : SUCCESS in 59m 36s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/57ecb1b3c89845fba655ef678a7deb67 : SUCCESS in 43m 06s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/8f00382a0b1d4561b0ece28ee9330e87 : SUCCESS in 52m 18s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/a5b7d7668b8946b58d4aae915ede05b0 : DISK_FULL in 1h 04m 02s (non-voting)\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/2586fd16a0604d34b8f0dbe2018803af : SUCCESS in 1h 03m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a9ca85f860f9054d150cf08b5b7ff651e19cf5bd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-10-26 17:22:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nit builds, /me approves","accounts_in_message":[],"_revision_number":2},{"id":"d755e035b1ab0bb648d87f9180ce5ee5e73d0b56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 17:23:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"17614642b73605cb4e7011bbd27a028386fe92b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 20:01:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/9095b59c774e4de3bdad7ecccfbe4ce6 : SUCCESS in 26m 02s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/08ed3026bac34409b6b76605df0619f8 : SUCCESS in 1h 06m 21s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/fbd97d89e1f344a7ba15d0e05087f13a : SUCCESS in 1h 08m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/009dab79bd204eb88393a201d16836c3 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92f6f4b3960e4a4195e2103a95d1a6de : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53860068525f42c4bb35e3720fa72b3a : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa8bf05292d64e4b92dd45235766608c : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/802689a8a8ec4878978b270989eff3c0 : SUCCESS in 6m 41s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8270885b5aae497587632ba3f77ac7d1 : SUCCESS in 2h 35m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a4f85d8424ab4a18961cf0376789fb77 : SUCCESS in 51m 07s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/decfa93018734804b17ea0c48b99b205 : SUCCESS in 59m 26s","accounts_in_message":[],"_revision_number":2},{"id":"70e318ff89e6ef7263cc52d4e838e1975fe42808","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 20:01:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"442f21911ddc2363cb6ba9718385038ca026255a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 20:03:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daf4a41a2ab9458a802cb43be0445248 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6f1ff63ba9b8b09011d579dffe815c5a3634d6ab","revisions":{"80aca2c09df9940b1bbca5f5683e9f63aff6cc4f":{"kind":"REWORK","_number":1,"created":"2020-10-23 09:49:48.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/87/759387/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/759387/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/759387/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/759387/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/759387/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/759387/1"}}},"commit":{"parents":[{"commit":"83aaa9c40188c5badf50905bae3f84c19cc67121","subject":"Merge \"Finish off removing Congress\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/83aaa9c40188c5badf50905bae3f84c19cc67121"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-10-23 09:49:15.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-10-23 09:49:15.000000000","tz":120},"subject":"base: use Victoria repos for Debian/x86-64","message":"base: use Victoria repos for Debian/x86-64\n\nChange-Id: Ie5ae119dcf31e05112405c8a307bcd2613d45de8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/80aca2c09df9940b1bbca5f5683e9f63aff6cc4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/80aca2c09df9940b1bbca5f5683e9f63aff6cc4f"}]},"branch":"refs/heads/master"},"6f1ff63ba9b8b09011d579dffe815c5a3634d6ab":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-10-23 17:54:08.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/87/759387/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/759387/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/759387/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/759387/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/759387/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/759387/2"}}},"commit":{"parents":[{"commit":"78351df2ccfa7d4ccc20a41fc51837b124e18358","subject":"Switch to Victoria stable branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78351df2ccfa7d4ccc20a41fc51837b124e18358"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-10-23 09:49:15.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-10-23 17:54:08.000000000","tz":0},"subject":"base: use Victoria repos for Debian/x86-64","message":"base: use Victoria repos for Debian/x86-64\n\nChange-Id: Ie5ae119dcf31e05112405c8a307bcd2613d45de8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6f1ff63ba9b8b09011d579dffe815c5a3634d6ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6f1ff63ba9b8b09011d579dffe815c5a3634d6ab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
