)]}'
{"id":"openstack%2Fkolla~759855","triplet_id":"openstack%2Fkolla~master~Iedc782a3317c93a2af0da4696fd71c05e0e4a55e","project":"openstack/kolla","branch":"master","topic":"fluent-logger","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-04-21 22:32:00.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2022-04-21 22:32:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iedc782a3317c93a2af0da4696fd71c05e0e4a55e","subject":"Add fluent-logger to images","status":"MERGED","created":"2020-10-27 12:58:42.000000000","updated":"2022-04-22 04:07:30.000000000","submitted":"2022-04-21 22:32:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"759855-fluent-logger","meta_rev_id":"4a3c58a26f778e23cf060b04db44e787b89742e9","_number":759855,"virtual_id_number":759855,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-21 22:32:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"all":[{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-04-21 18:55:26.000000000","_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-04-22 04:07:30.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"all":[{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-04-21 18:55:26.000000000","_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"reviewers":{"REVIEWER":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-02 09:37:23.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-11-05 12:02:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-08 19:32:56.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-11-09 03:15:37.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2020-11-09 15:15:06.000000000","updated_by":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"reviewer":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"state":"REVIEWER"},{"updated":"2020-11-10 09:19:33.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-01-13 11:40:39.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-07-22 11:19:42.000000000","updated_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"REVIEWER"},{"updated":"2022-04-21 18:55:26.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"state":"REVIEWER"}],"messages":[{"id":"29dcd4e4a3da0962dbb66d339581218d3f6ef064","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2020-10-27 12:58:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a6daf311832563590ac89c886edc95b1cce83e1a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 13:41:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/a57d3e39c3734fed997e8b67fdd6e91c : SUCCESS in 31m 34s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e7fa4b7ff1494a1b997dc1cec74aef9b : FAILURE in 27m 30s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/57ee15f36c014377b27b77921bc97850 : FAILURE in 22m 03s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3e2cb7135e954e4fbb8899b2fbc95940 : SUCCESS in 26m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7727a6aa6dbc8c3a84d58fbe412d6d6381466914","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 15:48:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/a5820575b4a847a6bc49cf9aac88b1d5 : SUCCESS in 18m 35s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/cd2febc240b642e58a91bd8d29d9923c : SUCCESS in 26m 44s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/be26e8595e454a3aa7925b1fc379cb79 : SUCCESS in 50m 52s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/64146db5e8224dfa87b7629ddb77d18d : SUCCESS in 1h 10m 41s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/74fb8b1276a242bfa907780b2ff8fba1 : SUCCESS in 1h 02m 54s\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/df8fd8f3b7f44946b23ee577311caafe : SUCCESS in 31m 07s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/293c5f15b4924f8db5cb396b7994b87f : SUCCESS in 1h 42m 14s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/a1ec43989a534976ac38471503c763d5 : SUCCESS in 1h 43m 06s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/5763d3cf458445488b75b9c51f871e8f : TIMED_OUT in 2h 09m 50s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/c688a968681249cc9ceed8ba7b1ee87f : FAILURE in 2h 04m 56s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/cc385962c01849679424d868c0c0e675 : SUCCESS in 1h 06m 15s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/4a58b7afa947472a9861458a85b86a58 : SUCCESS in 1h 19m 27s (non-voting)\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/5f5656c0b2ab487aac6696c00a118c24 : SUCCESS in 2h 07m 25s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/371bf430efa640a8b230dcc977304fec : SUCCESS in 18m 50s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e7065319b3df46538aa1f1157672b019 : SUCCESS in 16m 23s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/76d88081df07469cb4165bda635a3481 : SUCCESS in 53m 01s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3fb69717b28425b83674649a2af464b : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f74158348e84cb0a742b5e5a82c0843 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55acf77e33a04d57a6124a640a1a6676 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/894402264f3b43988d8880d13fee6858 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e41eec2b40914b549861d2c6a325a621 : SUCCESS in 5m 53s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/57ef0ab5af384f1b9709ed346de8f879 : SUCCESS in 21m 51s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/75b52b9bb48b42f495da0bb6e4b92fa9 : SUCCESS in 1h 53m 58s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/76041961ca4f42a28f437e7de459f185 : SUCCESS in 54m 41s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e44aeed18be14aad91f5e6f27d0044af : SUCCESS in 1h 37m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a78e44e35ced454fbe2f7d3ab4cc036c : TIMED_OUT in 2h 31m 49s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/081adbd3492b4fe191aa5aaa77b96660 : SUCCESS in 1h 13m 31s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/7304243cfb424414941507b90463cae9 : SUCCESS in 1h 29m 49s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/a2482c1af9054719a17d2740555fffa9 : DISK_FULL in 1h 03m 47s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/986fc27b140d4b9a98a767a0aab15cdf : SUCCESS in 46m 46s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/5c518cfd00a24b8f884d020a0671b678 : SUCCESS in 2h 10m 39s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/ba8326a3508544e49386f0e110ff4d71 : SUCCESS in 1h 16m 31s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/20caab988271416f8eab21b8073d901b : SUCCESS in 1h 49m 42s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/d3dbad9ed8c149abb6b416a1220a49d7 : FAILURE in 39m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"084b54e39c4e9b7fb6ae375a11f6c7cc1dc7b967","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2020-10-28 12:37:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2557e1eecdfade20c5deee33a73b42d1d8b81db1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 13:12:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/f7f32f57d673405fb9c4e198a9329139 : SUCCESS in 32m 51s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/591ecb85d894442b9aa076c9eaee667f : FAILURE in 27m 37s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8f2a1935d7ba4fc487ce6690b37498c6 : FAILURE in 23m 10s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/5ffae607c59a4692b2bdcda6d6ba46d7 : SUCCESS in 27m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"56ccf761b51597df84e0d619b3c28f98a51b639d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 14:52:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/d92268a92b104129858ae9f6a61da899 : SUCCESS in 22m 30s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/0d47f46eee1b4498ac225248ece4fa3d : SUCCESS in 25m 48s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/e0931b866ddd47f590d8cdf8b00845f0 : SUCCESS in 1h 05m 16s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/d4bfd0cc31c247a395f1a78c1cff58d4 : SUCCESS in 1h 03m 26s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/42f79bd4411945fe9700980b9b301c7f : SUCCESS in 1h 09m 38s\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/a93c2c8763114f9e9d1f4b02f8240c7c : SUCCESS in 29m 00s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/bde022acd95f4448bafebcea490ef151 : SUCCESS in 1h 34m 15s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/ffcd21c0ee394a0fa837667c8d4ec5bf : SUCCESS in 1h 52m 20s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/228d9b65b96f4733ab0301d2ddf60bba : SUCCESS in 1h 37m 48s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/51f7dc1f0fcb4f45bd65ad4df198ae40 : FAILURE in 1h 02m 25s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/f745b11f60cc475da0bbf0a75eabcea4 : SUCCESS in 1h 23m 08s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/807f120149da4538b4c8ed94bbda0372 : TIMED_OUT in 2h 05m 17s (non-voting)\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/6af7f829b8104e3e801145722fc14e65 : SUCCESS in 1h 20m 02s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1e438ea11dd44b7b81166615d8a7acc3 : SUCCESS in 16m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2cd39afc693a499a8819f7364d36aa49 : SUCCESS in 21m 05s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8259fb17edc04193904c918f6516a5d0 : SUCCESS in 1h 00m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab3bef03400947779d48c2debce5cf00 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46545d0ed6464b219755937d7a260824 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b2b46c64d4d4479be50b46717adad5f : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b10fe30cacb449c4bf4ec4439ffa534e : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b8b5354aba24a4e9a2d6299d60eb74c : SUCCESS in 5m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6cf3b578ea95426f9eddfa98c6223840 : SUCCESS in 19m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/867f1b7de834475591d90aca091c2953 : SUCCESS in 22m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6530246f86b745278b1b2d6856434a93 : SUCCESS in 54m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ebc499cc1029478c8830bd5e5a8d05a2 : SUCCESS in 1h 18m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/aa26bcafe1534e8d8665d080d1ecf049 : SUCCESS in 59m 01s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/2410c57ad5f64b70b58f3c5048455eec : SUCCESS in 1h 12m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/7246f7b928904dd1bd9db027581e9c3c : SUCCESS in 1h 47m 47s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/c40e6a3da0e14e62951dde73397306ca : SUCCESS in 1h 16m 45s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/e48586b2c0b3444db9613fa4efc06cf2 : SUCCESS in 1h 52m 56s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/e5d75320ca1d49b48ea13a4fe145d38f : DISK_FULL in 1h 18m 50s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/b1717f3fbf4a4977b1a1372441a2441c : SUCCESS in 1h 24m 10s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/a2dc43f4785b47b7ba80d95be2fb5675 : SUCCESS in 47m 35s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/7968871220e24d428e09a110e9037f2b : SUCCESS in 1h 10m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b0b928830f42e2bb75b4d4d5413decd1b0ea9d07","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-11-02 09:37:23.000000000","message":"Patch Set 1: Code-Review-1\n\nPlease describe in commit message why you do this change.","accounts_in_message":[],"_revision_number":1},{"id":"88ad70530122171c8c74fbc13bf49ac651e2ced7","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2020-11-02 14:46:06.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a625ed4be2cf516d3db25bad8a722f7671ff3898","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2020-11-02 14:47:16.000000000","message":"Patch Set 2:\n\nCommit message updated. I want to just note that kolla-ansible is not currently blocker for this change to be merged. This change only adding tiny python fluent library to images.","accounts_in_message":[],"_revision_number":2},{"id":"7a92f87fd60cd17e23e65def7e6f0e20c44be895","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 15:22:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/531dd28e051a43f6b3385ac5e624f54f : SUCCESS in 34m 55s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/17326260fa5f44198c165b2bda90f852 : FAILURE in 28m 46s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/46f7c788448a426fa01e1ffb3cad54c1 : FAILURE in 24m 31s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b1eaf7383fcf4604a37e538c2c3ee472 : SUCCESS in 26m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5feb72a1554dbc1aa4c0064624f325a6be431904","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 18:26:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e448678e94fb4c01b9bb78c967d4429c : SUCCESS in 41m 24s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7b100ce003284a57bfda3ac2306e42af : SUCCESS in 22m 40s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d9269e83833d4eeb923f212e7edc70e4 : FAILURE in 40m 14s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/26e374e6c8f141779e17a6f8d72bf246 : SUCCESS in 59m 40s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/26faddbe6569444ca0256f328e486ad7 : SUCCESS in 1h 04m 47s\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/7a301809af0c4962b638cafdc8e93740 : SUCCESS in 29m 11s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/d62eee9b3ca54da799df49b2700b4a11 : FAILURE in 1h 33m 21s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/7bdac65da0604aad88f887476b2d5625 : SUCCESS in 2h 05m 20s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/fce9e7c4e48149b495601b06008fe40b : SUCCESS in 1h 45m 16s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/8186620d42704e40865e978a4b4a1123 : TIMED_OUT in 2h 09m 34s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/5c47648adf074308a28fc1493b760125 : SUCCESS in 59m 42s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/f963d7b7287f49f89edcfa629e109c53 : SUCCESS in 1h 41m 10s (non-voting)\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/027d6102f6b64e28822fbeacf2a905be : SUCCESS in 1h 17m 38s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f30cf05ace2e48eda4353802ed29d25b : SUCCESS in 16m 58s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5973b906b8a143ce97c613b944a93557 : SUCCESS in 19m 47s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/96ea458826234c8abdcae79cdf1fd856 : FAILURE in 46m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c23c27bd8aa7431fb985d2adfdba9c2c : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51fb3d74e6214529a70a9d5e1af0197c : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16dfcadf99ae4c27923df9308d0e0d1d : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c304d110f974402bb4f0a51e522d8114 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc1a1c1423154f01a17945adab42efb0 : SUCCESS in 5m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cf025ae7f3c5461799f74d946f3da9ae : SUCCESS in 22m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/46efd2bb0df34353a10a88ca8bdfcbcd : SUCCESS in 18m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/48ce2d42167d4296bd07b977e395595f : FAILURE in 32m 22s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1ccda14d856f426789425c2d48c9c62a : SUCCESS in 54m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c8738cb703c145418dfb2361fcdd34c0 : SUCCESS in 59m 35s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/3c620cc9a49149bfb3fdf551f57c0771 : SUCCESS in 1h 20m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/854ed4b392f34c32a7cedc033153132b : SUCCESS in 1h 32m 41s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/f22446d4761d46a6bb3e0c7c2f8cc0c0 : FAILURE in 50m 37s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/dbaa3ad526f346399eb2c9cbf576052c : SUCCESS in 55m 52s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/1edf550daff74eb3a626854967e04778 : SUCCESS in 1h 15m 13s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/074ce8a57b0e464d8ca8197aad9720ef : SUCCESS in 1h 01m 55s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/751d26b36d604e9a8f44c16b3cc4fc9c : SUCCESS in 52m 32s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/f00a35f215c646d9b2634335d156f607 : SUCCESS in 1h 14m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a65a681b6c3289c42b218e09ebcf48f21493fe2d","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2020-11-02 18:42:13.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5fb2797fdc5ae83a7c434ee0269484622f633aba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 19:14:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/e82b4e79b40246eeb5092b3d5e69e8f5 : SUCCESS in 31m 18s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/00148349110a4836b92376a20e72cda3 : FAILURE in 26m 39s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5eb5187651e447d6920b4f76016b8129 : FAILURE in 21m 30s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/49bb8cac5a554cfdb96de5840e3882ed : SUCCESS in 24m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"91886002f8f77aa589c67e7f4b9fe83d613c1a30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 21:27:57.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/71881118b1104a1791232731e8bd039e : SUCCESS in 18m 07s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2a79c0a6fb0f4af3aa1d10d291b3f4ab : SUCCESS in 21m 51s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f810097690fe4e3ca8479ebb7d1d9927 : FAILURE in 40m 57s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/81bc71ffbbcd494f92d6e26e844a5f03 : SUCCESS in 1h 07m 49s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/8d495c9aaf524af2b38e01b57923a75e : SUCCESS in 1h 07m 04s\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/d2319b6063ed4d78b6581dec23ab8e07 : SUCCESS in 25m 47s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/89d2d5dad6d64706a4ade298ca7504a6 : TIMED_OUT in 2h 40m 11s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/3afaf57888cf4b24a2bc50e9b5a4d473 : SUCCESS in 1h 52m 10s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/8d2d0dc9dc884e3c8b1e1d1f01e1df80 : SUCCESS in 1h 43m 49s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/3a1246d71bdc41319f07e5475ae428af : FAILURE in 1h 03m 23s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/762dd979af9345ceab114fd59939f0a2 : SUCCESS in 1h 04m 07s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/a59e4a2cbacf4684a4a4a14f36f9c273 : SUCCESS in 1h 31m 32s (non-voting)\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/9ba4ee7e6cfc46eba35591b151c9dce3 : SUCCESS in 1h 25m 32s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/50f5d4315ba94ef682ef74e3d7c6651d : SUCCESS in 21m 03s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a17c6dbf29e445fb908342fee4760d16 : SUCCESS in 1h 49m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/615364f05b394036a6bb3985845a9896 : FAILURE in 37m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/485689178ee64025b78c26091b67b528 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfc60f4202a5455c92fab453f87491ed : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f88a3d49c154620be0a880158b32c76 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/232fe0e5a21f47eb82a2bd1c72084803 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90ed2bb0f7b846ffb411abe7a035b3e3 : SUCCESS in 6m 25s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c7dfb8f8dc184a08b617d0bebebbf2f6 : SUCCESS in 24m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/871c60c309d84667bc7c749bc7bcfa19 : SUCCESS in 24m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e8e6fb094f5c4e70b68388cf500f29e5 : FAILURE in 35m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1388098977414a8581061e4ddf7c88dc : SUCCESS in 55m 06s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e9153f93de77413c85ae135724e2ff2c : SUCCESS in 58m 10s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/0e8142f6608645f89c67727c053bbba5 : FAILURE in 52m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/84d7d08ae7184900b3cd536537bf324d : SUCCESS in 1h 38m 37s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/867a56d13d834bbf8230b524d0eedc0b : SUCCESS in 1h 23m 07s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/5a1171f32dc5487eb8ac21c69c57626a : SUCCESS in 55m 31s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/dd094a7a42eb4504ac739485f877a7fa : SUCCESS in 1h 13m 07s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/c6b4c2986fed408d88fe6854a5f64076 : SUCCESS in 47m 25s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/7fef4c46dff94af98ec785780c75185e : SUCCESS in 50m 36s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/bd7b1c8543484465adf457f127f605ef : SUCCESS in 1h 09m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"34256cb4c71bf84e78166c5fa98869537464dd2f","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2020-11-02 21:37:28.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ea774738bb61513d58bcacea9d9c1b4ee31b6c5a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 22:13:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/2e6a735fa4f8428b893122e866a7c1c1 : SUCCESS in 28m 56s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d36c515954234dcbbeb377ec83c5357b : FAILURE in 25m 02s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/abafd6ea7bd941bfacf1bcb5de6ae28a : FAILURE in 21m 19s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/5a6df29ffead49f6a892a2c730220904 : SUCCESS in 24m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ead401d5adeae5edf9eb9092e4022fa42461bb8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 01:42:33.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/26883f3444b148d98cfc89ba7a66c54c : SUCCESS in 2h 40m 07s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/77ed84b4b7c44633b12dac48cf85749e : SUCCESS in 2h 38m 58s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/27d46dc7633a4e568b85e2ab2a9a8105 : FAILURE in 44m 02s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/5934449ce9684886a3484566e40c9902 : SUCCESS in 1h 23m 14s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c6ab5f0965b24c79b5994adb7577aede : SUCCESS in 1h 14m 24s\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/c54bbd63802e4a32a3df2ae8f8812dcc : SUCCESS in 31m 53s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/a0861e12865c450a83fa73c338ccb454 : SUCCESS in 1h 37m 33s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/91282a41a0184402878e54494caadbb8 : POST_FAILURE in 46m 24s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/f8985828f5ef43529fcd221e63c64e05 : SUCCESS in 1h 41m 14s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/b388f76af3384710a5e843e28ac073a9 : FAILURE in 1h 05m 05s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/8801fdb07e7243f1928c0bd573f51e77 : SUCCESS in 1h 13m 35s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/b5c78cffe26349969dfe060d84717f75 : SUCCESS in 1h 35m 36s (non-voting)\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/12c50055a7ad46769679bd0fcb22c9b9 : SUCCESS in 1h 24m 25s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/57210f0e221e4830986e7cc8cc0a23ee : SUCCESS in 23m 16s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/40f049aaa461416caa7c196cc7a31e3c : SUCCESS in 18m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4b2b57659dcf4add83eafd9f194c3f85 : FAILURE in 39m 47s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b739a37e44c41e79138f283f44f02b3 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d671bf9fc704daba554f42889ce9c97 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a97eb4f719943b4968604709278e34c : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/918501e685bc421e94d7e02925c962d9 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45fa7a4c990644f498fe6e260aa50e6a : SUCCESS in 6m 44s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/17e724d7014f4afaad7f3cc44fcd97e2 : SUCCESS in 18m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c3c92b9d37354e68bdfe39c9592da51c : SUCCESS in 22m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fc5fdae84bbf49e3b2ee631789171135 : FAILURE in 33m 45s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4a4af0cc37864d5f9626e112f99f5528 : SUCCESS in 54m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/947606c7fa154bb58293b82a85d3d0f2 : SUCCESS in 1h 00m 08s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/021fd457af194b7e8e3a95e52fea8fe0 : FAILURE in 1h 08m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/ee68682582e4446ca007b3a1169841ca : SUCCESS in 1h 42m 18s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/6d2db7f141d84d4c8bda90a7a447ba40 : DISK_FULL in 1h 13m 58s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/2cb6b8d220554f1f81318f5fc5fa8e88 : SUCCESS in 52m 59s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/ab5a5361bbf34854be3b303d2d558b49 : DISK_FULL in 1h 00m 46s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/63da96f1df634353babefda1c2150b43 : SUCCESS in 51m 16s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/c0eca2ea03ec492cb1e43c4cb3be59a7 : SUCCESS in 53m 03s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/d65f8ded570c4a26ab0ee0be96eace7e : SUCCESS in 1h 04m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b191350262d848e8b5b0da9875d4a4130adca2d6","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2020-11-03 08:28:26.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3764c33c2cdf67c0793ba9386f4c7408b95e0295","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 09:03:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/23374c6f73a54c3caebc41f1b6b0ae46 : SUCCESS in 30m 57s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3942b806a1164545b2cd0602f04cf5aa : FAILURE in 26m 20s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/35149033d6654fe994851e10558a82bd : FAILURE in 20m 20s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d8299789afea4d41bbfc5bf9cf08b857 : SUCCESS in 26m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5d9bdbcb43a6559786ca7e2f5add2a16f61253c5","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2020-11-03 10:55:19.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1cce3f21c66413ef51b015e02e74fd5d25498945","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 11:34:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/2bf9b80db2cf40138968d088435e2f63 : SUCCESS in 37m 25s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1185c8bc2da74d5282014228a676d656 : FAILURE in 27m 12s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3f56195eab7f4f5187c41b4023e9a38b : FAILURE in 23m 43s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/7596c45dc7a949baa7bc93614dece6e0 : SUCCESS in 26m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9445aec5d32649502004e286f39e1a7ce4e5c023","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 11:41:55.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/2903c6121ef94c5ea4983f096dfe4dc6 : SUCCESS in 40m 02s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/df8b5828e0e7422a8d5c534259a929c5 : SUCCESS in 39m 26s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a1534a55fe144e02af128a169cfc55f8 : FAILURE in 1h 02m 45s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/dbda3b9c166a4c56867d56f2d50fedeb : SUCCESS in 1h 19m 54s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/1a5e3c25177a4622a356166136c10064 : SUCCESS in 1h 02m 42s\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/34ed6a3d5eb348ffa66bab87242e0989 : SUCCESS in 44m 38s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/8c3cff46d3cd4b8f95a323b45a6ddcd8 : SUCCESS in 1h 30m 30s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/a819e985174046d9bf2ae91165e4e033 : FAILURE in 57m 55s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/39ef7c0dbf1b4539891ebf50a73434d6 : SUCCESS in 1h 36m 59s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/a3e6ff6ec5f741f79dd51fcdee5ceb93 : TIMED_OUT in 2h 04m 18s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/fafad33936064d2ead67a6d2616f2338 : SUCCESS in 1h 26m 20s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/ccc36b7feef44c269a3e4d8f8d766f08 : SUCCESS in 1h 29m 52s (non-voting)\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/3a3846a841f34b18be94108763a91e0c : SUCCESS in 1h 19m 15s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/83b80fbbc9ec4e688b50946a37eab99c : SUCCESS in 18m 38s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/bf201be22529497cae882da134047416 : SUCCESS in 39m 32s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4879cda660f143959bc5a49c3f5cfafb : FAILURE in 1h 01m 42s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ffcbc86839140bdbe8604c898cc749e : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cefad27341b45d8bd00aa21240e7b59 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ecf4a607358f4f1eb0b2356a5934d9fe : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e535f04ccdfa4ebea817c5aa3038c7f5 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d36a69ea9b3940d6ad9474d029e7b5b6 : SUCCESS in 6m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4f3fa0c89fa4406aaf29c90d3fd28412 : SUCCESS in 40m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/70e2ed5b8b6e4690845f5ff151850e1a : SUCCESS in 2h 09m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/33bcd97f0e1947f28e3a316e64433500 : FAILURE in 40m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0b9403be56534a38964c639e5cbf3f73 : DISK_FULL in 2h 07m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/53546a812ec24d6e8e413a1149fa1491 : SUCCESS in 1h 07m 43s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/a7b0501015b9441a823dd1d4c1754eb0 : DISK_FULL in 1h 22m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/62d2425c8da24e128a51f6240a3762a0 : SUCCESS in 1h 37m 04s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/d33983dd48994b1dba5f2be5ef373253 : SUCCESS in 1h 20m 43s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/051aacbc949a4b02a9189a8a34dc2b48 : SUCCESS in 59m 35s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/f409b2fc1fa64d22a184139c3585d054 : DISK_FULL in 2h 05m 03s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/cea27619e515455f92b7bf73c197e611 : SUCCESS in 49m 58s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/e4140899998b4304b2b9ad0f93f35059 : SUCCESS in 48m 57s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/ec54c145e77049c6b08ec93cbcd55a51 : SUCCESS in 1h 07m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d362c6062fd79dc79defa9952d17d204a7589a08","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2020-11-04 09:54:12.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7872cb8567c97959f5b3df247e3dde42effd2eba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 10:58:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/4da48218de8b4762af3c735d7a30c716 : FAILURE in 56m 08s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5e61c9d6257f4e78b4f00624d22fe33b : FAILURE in 49m 45s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4f5a6fdc14cd464bba4333e6c0fccd60 : FAILURE in 37m 08s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/aab27c6ae5b94872a4e20db0d5252092 : SUCCESS in 50m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f0e7dc8f5313fc7d0ae69d50311df5d21bfd6565","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 13:00:40.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/db686f4f1a7148268891198866f401ce : FAILURE in 1h 47m 42s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/63cd28938ea5437b812588fed2b9d7da : FAILURE in 29m 38s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/74137b0b9420449ea3ce2d023bfce9d1 : FAILURE in 46m 51s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/d547dfbb4cbd468f816b94051d756c23 : SUCCESS in 1h 05m 45s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/10a3028f9c684da994eedb9b37c7ba06 : SUCCESS in 1h 21m 38s\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/ce5fa95568fd437ead8db3fbc939049c : SUCCESS in 28m 08s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/745e5dd789f140f4afc733d9dc5eba01 : SUCCESS in 1h 37m 13s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/f8a671135e29448682dd59b53a12ac41 : SUCCESS in 1h 44m 52s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/ddbd438449324c2893ec52c6fa50f7d2 : SUCCESS in 1h 42m 13s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/56f6c57be9664243aa3d95d1e92eba45 : FAILURE in 1h 08m 31s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/31809cd9accc4eaeabe33fc9761dd4c1 : SUCCESS in 1h 10m 04s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/fffae99f1bc349e6b19883e09002ec6c : SUCCESS in 1h 31m 48s (non-voting)\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/c582f6e27ee54252a99e57b6244d7381 : SUCCESS in 1h 27m 30s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c765641821a14cd1a8719d04c7c90ba4 : SUCCESS in 20m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/979323b510004e2aa942394f861c5f49 : SUCCESS in 17m 46s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/aefbc7a7059340cfadab80a0a29f1efb : FAILURE in 59m 58s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50ce6164a9284ea99da603f1be5e61e7 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/826c68c9d3d54e32b39250e93df9f5fc : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6beb0eff8376454dabb720b7a376668f : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7cd258c23c5b4986bbc8f72d3210fc93 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ceff5c428e5429099368f9d51f04c82 : SUCCESS in 5m 52s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3b5e6b59885f4a629c323d50f12783d8 : SUCCESS in 19m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/05ff1346c0bd4551bf6d891665507dc5 : SUCCESS in 19m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e766626264ea4f5d86be317a2b002177 : FAILURE in 31m 50s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a102a53eabcc4c7f9eba7e074079c6eb : SUCCESS in 1h 03m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6e1b739cf89940c2a7cf559eb4116426 : SUCCESS in 1h 06m 10s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/3529d5f8cd0c4d31896203e97ba7764f : SUCCESS in 1h 31m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/8193f753d37f4b85a0305bcacd981ec2 : POST_FAILURE in 56m 45s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/371fc69e48f34c18aa20d1b523478d61 : SUCCESS in 1h 18m 18s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/7966c01ae1fd486382a4c4225f58b82c : SUCCESS in 56m 16s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/f10a8ea50f9841cca5cae82f9a2f5b77 : SUCCESS in 1h 17m 58s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/92ea196d274b40f2a9be7ee2e75d3a92 : SUCCESS in 48m 55s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/2f5c065fe2044397becf22a15014b306 : SUCCESS in 50m 57s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/096dec55eee7490f8b955a55ab1166d8 : DISK_FULL in 1h 30m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bd734e173b9999e8c89f7ef3d4c920c4a8f864b9","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2020-11-05 08:35:32.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"be710e332a9ab55f1792290947a8ea6ac6da4a22","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 09:24:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/eb3ba9b3e89a4a5ba7a06ccef24d725a : RETRY_LIMIT in 2m 56s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/891b361cce844425b2b0cc5660df325e : RETRY_LIMIT in 2m 22s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6540de189cfe4c71ae75ece907f2bb7a : RETRY_LIMIT in 2m 30s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d8ad1803ce134ce68a7dae6858939315 : RETRY_LIMIT in 2m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ec29c128d92e5adc618a6aea4ce74f8cd653f757","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 12:02:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/9d65b5e4eb1f4b9481b0db8fe2d43a88 : SUCCESS in 24m 27s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8b41eec98d7341c2a0f3f9f8fc457c99 : SUCCESS in 35m 34s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/5936995b99c349f990bc7f8ef630ce0a : SUCCESS in 1h 10m 09s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/ad0696123b444b3db66611159e45b38c : SUCCESS in 1h 03m 13s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c86e8d23c3ff40858ac7504014819192 : SUCCESS in 1h 17m 15s\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/d734e511962d447f9fc375a83dc97198 : SUCCESS in 43m 07s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/83b37c2d57a54d97989c3b9596d2ed78 : SUCCESS in 1h 50m 58s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/692864b8869347739436fd3189131d3a : SUCCESS in 2h 18m 55s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/47d4a543c3f049ea9cd2ad0fc2c06aa4 : SUCCESS in 1h 37m 03s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/920f31c5266c4dee81e8792c124ee5c1 : FAILURE in 1h 16m 16s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/85eb0b41fc0b4d84a0a9ccbe54ca634b : SUCCESS in 1h 03m 35s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/17eee7f9a3bc45da9658267138d7c8a1 : SUCCESS in 1h 30m 07s (non-voting)\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/622b7cbcc85f4277a36fb9b0cb2fc50d : SUCCESS in 1h 23m 21s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1d9d255d3912407eb6556c7920884547 : SUCCESS in 26m 22s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9e25aa49eeba4bfb8d5b1b980f39eef9 : SUCCESS in 18m 34s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/aa80a0b33d1542e9b2846d0d4cc03a4b : SUCCESS in 59m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af670f7270064b7095c67f9c4222dc99 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1225e16d84884cb894a6a4b06b9b8b0b : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be75853eaec042ed87e30f970f5b2479 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20e59e9694524a71a34222c2a6b8783b : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9af5f29e176e4f3088f69d5196beb876 : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a45569e91c164f0f91254c07cba70843 : SUCCESS in 15m 01s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0fc1dd125dd740859dc16b49d2ea16f7 : SUCCESS in 21m 02s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2cd72905af5242c982ba8b4b5e6f83d8 : SUCCESS in 20m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bd575e7f869147ac8f812fb608df1f99 : SUCCESS in 53m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/06808e51a87e4e32b0c14710b7322711 : SUCCESS in 57m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3dd98b041c5a452f8d17c47e7aae88a9 : SUCCESS in 1h 04m 48s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/2fb2c4abc5fd403ca7759a3dc13db2a7 : DISK_FULL in 1h 18m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/938d503bbe7b4fd2b5a59f006657051e : POST_FAILURE in 52m 43s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/c300ac60a53b44c8bf1740523b835aed : DISK_FULL in 1h 24m 44s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/19030e669c044ebda9d9d95a1baf1a8a : SUCCESS in 50m 23s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/ffff1521f13040cd8fbeb88ed979655c : DISK_FULL in 1h 06m 01s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/a2d4af92eaca47b394d86c11d175feac : SUCCESS in 43m 26s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/17cb470a9ccc4241b99f245791a4aee2 : SUCCESS in 48m 30s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/d0ba8f351fc740db93c18310e27dce2c : SUCCESS in 1h 06m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"662f9bfcf9a1f6ff5795f50c422fa564d7fabd21","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2020-11-07 19:12:16.000000000","message":"Patch Set 2:\n\n\u003e Please describe in commit message why you do this change.\n\nDone, is it OK now ? Can be now merged ?","accounts_in_message":[],"_revision_number":2},{"id":"cdf2818db6e4b0d8da44308270629c6a964a0085","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-08 19:32:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"81ad8637613fdd1bdb03cda6fed0521170b53cf2","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-11-09 03:15:37.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f9a442c97d6f4cc66714f9bf88c701fb7e8e9795","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-11-09 15:15:06.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"cc0eb23d0a66299d68ece1ecc3835632bb6724af","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-11-10 09:19:33.000000000","message":"Patch Set 2:\n\nthat\u0027s only for Debian-binary and source - what\u0027s with other distros? Do they have this out of the box?","accounts_in_message":[],"_revision_number":2},{"id":"359544a208552e3561f940e1e4c6c81d7042af29","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2020-11-11 13:37:44.000000000","message":"Patch Set 2:\n\n\u003e that\u0027s only for Debian-binary and source - what\u0027s with other\n \u003e distros? Do they have this out of the box?\n\nWell, as I am DD i\u0027ve built a package and upload to debian archive, and to our openstack repos also. But I don\u0027t have a power to do this with RPM, but yes, it should be fine to have it also in centos.\n\nShould I ask someone to build and place to repo for centos ? Or, do you guys have some additional repository where you can add package if it is not included in official centos repos ?","accounts_in_message":[],"_revision_number":2},{"id":"d4baf1bb97234ca021a3d7f1c4e54eb7811a5431","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-13 11:40:39.000000000","message":"Patch Set 2: Code-Review+1\n\nI drop myself to +1 but I want this merged in general - perhaps we want to have features that are matrix-based - if CentOS and Ubuntu do not deliver this in their packages, then it is simply not to be used with them. Period. We should not limit ourselves in the source case.","accounts_in_message":[],"_revision_number":2},{"id":"527810b4aa47ba6a0330c7396fcc367173054556","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-13 12:26:41.000000000","message":"Patch Set 2: Code-Review-1\n\nI have discovered fluent requires msgpack\u003c1.0.0 while u-c require 1.0.0 in V and 1.0.2 in W now so we can\u0027t even do this. I wonder if images build now. They should not be able to.","accounts_in_message":[],"_revision_number":2},{"id":"259f6b90953c6b6fd91f58ee666555b44a2bdc7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-13 12:26:47.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"2d5b4eb8550486d957a649238d600efcb47a20cc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-13 12:27:16.000000000","message":"Patch Set 3:\n\nlet\u0027s see whether this fails (as it should; otherwise we are doing something wrong)","accounts_in_message":[],"_revision_number":3},{"id":"2578e90f0a8e3a6fcb1f255fcd117234e1590edc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 13:08:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/ca81044a405140dd9cb9d8015bb7ab24 : RETRY_LIMIT in 3m 30s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a28fdeb06ad0462ab8a7b0205a4d26ee : RETRY_LIMIT in 2m 52s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b35ae95f0056483da25b62e3590e688b : RETRY_LIMIT in 2m 48s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/cb999c74831c4184a0a6f4ec95dd529c : RETRY_LIMIT in 2m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7ba0042cd7d5de90017d7e3d98f01c65183d2eb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 15:25:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e2f7a0b53879463c8ff35e94d93dfdce : SUCCESS in 50m 25s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/38b9bf2f2adf4557932a3ba443044a95 : SUCCESS in 1h 06m 53s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a279b7f6639542f0a07eaf8df6ec5e61 : SUCCESS in 1h 37m 31s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/8eb4336e44904c04a7169e7269517876 : SUCCESS in 1h 02m 33s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/d467c84321e04c58b1629560fef35275 : SUCCESS in 1h 17m 39s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2c7d2f9cc2b8498a976114d42195f51d : SUCCESS in 31m 20s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2ce84b1116c649d79a5a6d44ad2ac4fe : SUCCESS in 21m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ff5e42ad0a95467f87b14b64f1fd22c9 : SUCCESS in 1h 24m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4f47edbd8444421bfccb4c1efc54fc1 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c4accbfac03433295187d75fe462b5a : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e28fedfb5174c6eb6e2dffab8ea0585 : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/88d3ff1e63ff4023ad82bb23b9fb806e : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e4ad298e75d4dee9667c5c0294798bf : SUCCESS in 6m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/682fc576cd45483fbf1549ae6a23385d : SUCCESS in 45m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/60591ccaad564594926cf1589646a45f : SUCCESS in 54m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1eb63823364a42adb98a7a3dac2e9153 : SUCCESS in 53m 03s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f9b89e1587874025adeb1d80b1fcff93 : SUCCESS in 1h 16m 26s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0a4b1dec5dc94494940882783e55bb4e : TIMED_OUT in 2h 36m 14s","accounts_in_message":[],"_revision_number":3},{"id":"36f5177f1742d791b9a29c652b24298f98b90cc6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-13 16:55:21.000000000","message":"Patch Set 3:\n\nuh-oh why is it happy?","accounts_in_message":[],"_revision_number":3},{"id":"9621073142769da6a417d09bd6040202c0921568","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-13 16:56:31.000000000","message":"Patch Set 3:\n\nINFO:kolla.common.utils.openstack-base:fluent-logger 0.9.6 requires msgpack\u003c1.0.0, but you\u0027ll have msgpack 1.0.2 which is incompatible.","accounts_in_message":[],"_revision_number":3},{"id":"9cc89a4621c7ac337d5feadb18de9092982580d6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-13 16:57:03.000000000","message":"Patch Set 3:\n\nit seems we did not update pip even for source builds...","accounts_in_message":[],"_revision_number":3},{"id":"89b6fbcac8c5e437e125706b57aa3e5d00bf8725","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-13 17:07:01.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"8755e7521c23a8efebb5b2f4956205d85900b662","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-13 17:09:01.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"070b92e95ce348085a4cdc45f21b525b56085633","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 19:15:44.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/0d79ec05d6704b8f93566002d085bf8c : SUCCESS in 23m 55s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/322e67bc4aeb4a92adc05dc11665be92 : FAILURE in 40m 18s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/06e6da08002d4a03b5ee5f5f0f5e0b52 : FAILURE in 24m 07s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/97648097f3e94388b4932121e99fbc5b : SUCCESS in 1h 04m 32s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/8c3adff2cb094e2d92c7f37474d9dbfc : FAILURE in 19m 59s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6d2d8fd94f0e455186b53ffbee185842 : FAILURE in 4m 01s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0678f9c11d6e4069bc1ebb6d37353e7b : FAILURE in 4m 24s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/11d4ac9c05784b5085c9430ff7bc8c12 : FAILURE in 6m 39s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c50eba7a156342fd94555950037156fe : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a61bbd25b2284e3ebb05e58ef3440933 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f054dba24dea4fe9b3853f09fda761e9 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f390ebb6457c42fabe0a614c49f4499f : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be0bec08ca8146b49cedea30464a44eb : SUCCESS in 6m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f227e82496a249b9ad2d59367ac41e45 : SUCCESS in 22m 37s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7e978949ac1146f78a05fd1847451ba4 : FAILURE in 34m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/17b667ce2aa1445a917fff5398e3f03c : FAILURE in 25m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bc56c9dac7574827bf1b0c528b1700fa : SUCCESS in 1h 38m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/986fee9d983a4ebe888958902445831b : FAILURE in 1h 09m 31s","accounts_in_message":[],"_revision_number":5},{"id":"534aae5cfc533f950cd300f17c3eaf0e8b5d7f13","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 19:20:17.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/1ba1feffc3574a0ebdf54242510cd571 : FAILURE in 1h 46m 18s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f700bc1600504a2a8f67b269b90e253c : FAILURE in 1h 52m 00s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8094f3f74c1340ae806d468df87c2286 : FAILURE in 54m 13s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d5cfb667f93844f2b6b4aa4be49d091e : FAILURE in 1h 47m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7a9e80551d7f65a5ee2f20ea0d127de4394c4636","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-13 19:24:13.000000000","message":"Patch Set 5:\n\nWhoaaa, a very curious behaviour of the new pip - it installed an older fluent-logger to satisfy the need to have newer msgpack (as older fluent-logger did not pin its dependency). Oh what a dependency hell!","accounts_in_message":[],"_revision_number":5},{"id":"64c3e4fa6e7a885a419fcece98e27801a648d3e0","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-01-15 18:25:11.000000000","message":"Patch Set 5:\n\n\u003e Patch Set 5:\n\u003e \n\u003e Whoaaa, a very curious behaviour of the new pip - it installed an older fluent-logger to satisfy the need to have newer msgpack (as older fluent-logger did not pin its dependency). Oh what a dependency hell!\n\nFound that there is a pull request for unpin msgpack with fix ... I asked author to check ..so we will see\n\nhttps://github.com/fluent/fluent-logger-python/pull/171/commits","accounts_in_message":[],"_revision_number":5},{"id":"ce8a6a4b0f95644adec02048fb99e4d5139aa2ac","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-01-26 19:00:17.000000000","message":"Patch Set 5:\n\nSome mail exchange with author to address msgpack \u003e 1.0.0\n\n---------- Forwarded message ---------\nOd: Arcadiy Ivanov \u003carcadiy@ivanov.biz\u003e\nDate: so 23. 1. 2021 v 15:50\nSubject: Re: python-fluent-logger - msgpack support\nTo: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n\n\n The issue with tests is that they are not real (yet) - unpinned version was also passing.  \n The failures in question occur in fluent-bit and fluentd unpacking of the messages, not the fluent-logger-python itself: https://github.com/fluent/fluent-logger-python/issues/157\n \n On 1/23/21 8:35 AM, Michal Arbet wrote:\n Hi, \n As I said , there is also pull request with change to be able bump msgpack ..  \n From tests it looks like it is too enough, isn\u0027t it ?  \n Thanks, Michal \n Dne so 23. 1. 2021 14:32 uživatel Arcadiy Ivanov \u003carcadiy@ivanov.biz\u003e napsal:\n Hi Michal,\n \n No worries at all contacting me directly. I\u0027m unfortunately swamped with \n paid work (14-16 hrs a day including weekends).\n Enabling msgpack \u003e 1 breaks fluent-logger last I checked, hence it was \n pinned in the first place.\n I am planning on releasing a new version but that plan depends on having \n free time to breathe. :)\n \n Regards,\n Arcadiy\n \n On 1/22/21 4:50 AM, Michal Arbet wrote:\n \u003e Hi Arcadiy,\n \u003e\n \u003e Let me firstly say big thank you for working on python-fluent-logger.\n \u003e Secondly let me ask some questions ..\n \u003e\n \u003e We would like to use python-fluent-logger in kolla-ansible ( Openstack \n \u003e project for deployment OS cloud).\n \u003e Currently I have review opened against master branch :\n \u003e https://review.opendev.org/c/openstack/kolla/+/759855 \n \u003e \u003chttps://review.opendev.org/c/openstack/kolla/+/759855\u003e\n \u003e https://review.opendev.org/c/openstack/kolla-ansible/+/755775 \n \u003e \u003chttps://review.opendev.org/c/openstack/kolla-ansible/+/755775\u003e\n \u003e\n \u003e But problem is that Openstack has lower-constraints  set to msgpack  \u003e \n \u003e 1.0.2 and python-fluent-logger does not meet this requirement.\n \u003e\n \u003e I\u0027m contacting you  directly ( my apologize ) as I\u0027ve seen that you \n \u003e are able to merge pull requests, and no reply from comments in pull \n \u003e requests :( .\n \u003e\n \u003e Please are you planning to release a new version of \n \u003e python-fluent-logger ?\n \u003e It seems there is already support for msgpack \u003e 1.0.0  here :\n \u003e\n \u003e https://github.com/fluent/fluent-logger-python/pull/171 \n \u003e \u003chttps://github.com/fluent/fluent-logger-python/pull/171\u003e\n \u003e\n \u003e It\u0027s just needed to merge.\n \u003e\n \u003e Thank you,\n \u003e\n \u003e Best regards,\n \u003e Michal Arbet ( kevko )\n \n \n -- \n Arcadiy Ivanov\n arcadiy@ivanov.biz | @arcivanov | https://ivanov.biz\n https://github.com/arcivanov\n \n \n \n -- Arcadiy Ivanovarcadiy@ivanov.biz | @arcivanov | https://ivanov.bizhttps://github.com/arcivanov","accounts_in_message":[],"_revision_number":5},{"id":"9258db9c9aa6c8d8b2338e4dc50036f82b5955af","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-07-09 10:35:15.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"c002bb306af85267a27e5ca83e1d67ecf67be416","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 12:00:16.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/267f5d027de14e33a974df7eef25ff40 : SUCCESS in 24m 14s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2144e5fa7db8467bb57a2b4951ad7a65 : SUCCESS in 38m 50s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/4ff7e4d421024297b133c386367cc806 : SUCCESS in 1h 01m 14s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/7f70d22492d444c690e47ae9d62c2736 : SUCCESS in 1h 14m 43s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/7254772a63bf46b194e844c4aa6334e0 : SUCCESS in 1h 16m 58s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f53b38bbea2a45279a7084576f66718c : SUCCESS in 27m 57s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1ea01b1f7ae2496082c66822562ac2ac : SUCCESS in 23m 05s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/474d03db50644f46927c69a09229f700 : SUCCESS in 50m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c75bd575f144232b0c67142c6e9dfd2 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65c714a9513c40b4a2d70d2bfc14c31e : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b85b67ca89f94d75b15cd6e2ccca29f1 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dfa5e62f72324b658e1558b4becc3c72 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06d485fbf4ef44f19a1b9076347d0a9e : SUCCESS in 6m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/122425455066442dba9ee35190feee8e : SUCCESS in 20m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d131b7d129af48fea82dced2bd464779 : SUCCESS in 34m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7526c6e38ddd4ba2998b99ba76f59224 : SUCCESS in 49m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5918e3a939ff4969aa4929636be7e16e : SUCCESS in 53m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8b77ebc695f14285a74c8d83c83b9f4a : SUCCESS in 1h 22m 37s","accounts_in_message":[],"_revision_number":5},{"id":"152a7880c341a433743feb16e8e466ede4080984","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 13:23:25.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/3321d2097ce94315bd85686acb0faef4 : SUCCESS in 2h 45m 36s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a85f49b47b3240d1942e0def55d48ca8 : SUCCESS in 2h 18m 38s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/24e17055dc6941f0a85dd2bb1ce637ad : SUCCESS in 2h 24m 33s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9d22292eaa6c4e21819083593f47f1eb : SUCCESS in 13m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ecbd001334ff4e9ca854e4afb7ccc24f : SUCCESS in 7m 09s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/bc05640e9fc044b6bc2bfe97cf87de35 : SUCCESS in 1h 33m 41s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"43f172b507b100d49f279a909098eb118825db55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-08-02 22:56:29.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c7f9195d713b26693ea27262631bfa1752c3551f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 00:22:43.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/2a48aa5d06ff4b969ca9b8d8c890b550 : SUCCESS in 46m 03s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b3bf267b6b554922a725637bf58cb2c4 : SUCCESS in 39m 36s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f8f5047c7c284e14a71e5caa6bd0ad84 : SUCCESS in 1h 20m 17s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2d04bded073c49eabeb86677520d1b61 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/517abd1bc70f4b10a181ec79c8b8e2dd : SUCCESS in 4m 42s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/dc92dd0b9ddf43e29c1df232a7f137e2 : SUCCESS in 43m 40s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1b0069ba65d27c2c41c3a9b8aba926f4e66709b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 00:48:39.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/bbd459f8458340cbad372f2747e70080 : SUCCESS in 24m 50s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/fbd3ca91bc8441499446c9cbb3404664 : SUCCESS in 38m 52s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a1c95646d68c4fdc96f88fcf183d19b2 : SUCCESS in 1h 48m 24s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/209a8a10a89a423c92fcf77b9e16ac59 : SUCCESS in 1h 02m 22s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ea3a6aee9d0346248f36fac7ede4daee : SUCCESS in 1h 20m 13s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a28af9c8139b4f518e9cc893ddaef444 : SUCCESS in 36m 32s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/39eff05c12094e72abefe200301a8500 : SUCCESS in 24m 25s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/cb16a0aa4a384f50a10bce716975f9ec : SUCCESS in 1h 00m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e06cef6ae8d1415c93c8811d1981f34e : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66ac6146c67845188b859f2471ca60bd : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4e319827b754a13ad54c5568761462a : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e58e4dc3462544fbb557aaa37d7cfa75 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/287c825162ce4fdfb85840c4884dd79b : SUCCESS in 7m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/031f6527a3cd481484fb2f79adb9aa66 : SUCCESS in 23m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9624529e12a14ea0b5581e2c5e70f768 : SUCCESS in 24m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3f614750160a4ba1a9524197c57824ee : SUCCESS in 47m 45s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/abeceb986f1c419985099bbfaff92f0b : SUCCESS in 46m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/72a51808891b4ffab16fd8438a1e11a1 : SUCCESS in 55m 58s","accounts_in_message":[],"_revision_number":6},{"id":"601dab34d9f414ff3448d704d489765af261cf2b","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-08-23 15:00:33.000000000","message":"Patch Set 6:\n\nWhat about this ? It seems msgpack,fluent-logger and upper constraints for openstack should be OK now .","accounts_in_message":[],"_revision_number":6},{"id":"86b768ed6a3b118f377d8fdfb63626c97da68e5b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-08 13:39:09.000000000","message":"Patch Set 6:\n\n(1 comment)\n\n\u003e Patch Set 6:\n\u003e \n\u003e What about this ? It seems msgpack,fluent-logger and upper constraints for openstack should be OK now .","accounts_in_message":[],"_revision_number":6},{"id":"a68ba9909f4c065e5a2c314bdda793e7c35ac2bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-09-16 11:55:44.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"5000b4b9f96e8b6985c8a52f495405fd079621ca","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-09-16 11:56:57.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e2eef65f2b7aeb7f707999546263e2bb47d8e743","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-09-16 11:57:24.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"632448bb2bfb5d8faf1fc101f2fdf8156ea64a32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-09-16 11:58:38.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"f0edb8dc93e6afe85c54aa327b09345d34c6724f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 13:20:56.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/81c37163963c4fe1a8e4044615967f42 : SUCCESS in 24m 26s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/77f928120da746bdba0c22bfd903679b : SUCCESS in 41m 22s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1ffec4f77cbb43d7acb6a0aff59cafdf : SUCCESS in 1h 17m 14s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/15b1c231094c44c4a7afc6c782238ba5 : SUCCESS in 1h 15m 34s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e898f78d4b5c4ab880ac3c1174a70ab9 : SUCCESS in 1h 09m 01s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6b951f464a4949a583f036eebcbfa3f9 : SUCCESS in 38m 30s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c7e2ae92a9a14110a24443b2182b9a2f : SUCCESS in 27m 00s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/46295971304b4777b8f3581ee19dea02 : SUCCESS in 1h 06m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3688b2b07b05477b96cf6aca05a0ac1c : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a06f94e7778e4d538c397aa2fe756cf8 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0db3b8a9689749e3b3f7e0c8f2a00ca4 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/59fcdc1c235043f4b9d5b46c9a93714b : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63ff7f1a0b844a45a50ae1a3fb87c348 : SUCCESS in 6m 49s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/97c607456aff4bdba507e80da239ca27 : SUCCESS in 18m 37s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/60bf36fb5cd048378ec2492484e01b5f : SUCCESS in 32m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1a62d04f74064083bb04290765575494 : SUCCESS in 49m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bb065ad4d81c4eeaa4e0fec804aff0a2 : SUCCESS in 57m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9e89177bb5cd42eca325bdd1bedec7f4 : SUCCESS in 55m 07s","accounts_in_message":[],"_revision_number":8},{"id":"2e01d8ecd97062f43e43fda7d2f117bf8f570ce3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 13:34:39.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/ac9186d398ac439ab643b2a4cab5f4cc : SUCCESS in 1h 29m 05s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d5eb58d54673468f87595deea9f0a14a : SUCCESS in 40m 29s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3037b6da14eb423383da02bb4a22e5d6 : SUCCESS in 1h 29m 09s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/32e2a4d1451842988e5717810ef67434 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/29bb41bfcccb431fa77330e1bb00309f : SUCCESS in 6m 02s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d0d7f8eedf094e328094b1d67ce75a84 : SUCCESS in 38m 52s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"0f800c13770efba5536b0b57c577bc8469c56bd3","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-03-09 15:22:51.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"e51c6da828aced157cce33a09a2847aef3766dbb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-24 11:01:10.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"cdb14042cb902447de8042f4697ae9ddad9a3d50","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-21 15:39:05.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"6e85ced3f9886cf4e2079a8140bc6a5f6d37cc06","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-21 15:39:30.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"6cc59d29be1a6d259cb8f2dce5582510293c4b93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-04-21 16:17:07.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"2d4d6be6b84d76be5cb1662d236de46b748f3dbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-04-21 16:17:39.000000000","message":"Uploaded patch set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"4882ae1b268b8f8168f68dd2e1a91a6e11843df2","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-04-21 16:17:53.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"f0461d9b59f5d8d4cff02f52deaa145bfc2c88c7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-21 16:20:04.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"ee2a49cd37984ff5802749315c94dba366fa5e36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-21 18:54:12.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"73f752be23e4e8dbdba690ff9abce0ec902def0c","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2022-04-21 18:55:26.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"94f8720abd5fad0456973cd464d61c23991f99dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 21:10:57.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/7cb131801ab64dafb94ee699665d052f : SUCCESS in 32m 13s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/229d2cf1b3424cb5a94cb77bc51d742b : SUCCESS in 1h 08m 32s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/6948ef93f48d44638618a7689edd6ae9 : SUCCESS in 1h 19m 15s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c0257e30500342f08bb1d3a9d5d29aa5 : SUCCESS in 22m 16s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3207520b8c6545b1b1943d792f4b44fe : SUCCESS in 59m 24s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/8e35e40e6c364ec1aca9bfb1cc531a86 : SUCCESS in 1h 01m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e571276fa0c14a41ba6559bb2f2cad3e : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60741ee59a1c477488022b53a6e8c28f : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aed4320725644706a0ca36b2e1147f00 : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bc3d627b645b403d80bc92bf58ebf812 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/139d0a7997b944e38bedfbd1e6d00434 : SUCCESS in 6m 11s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a13285d79d0b4ec0a2487f22c58ba311 : SUCCESS in 21m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d10a341e1b5a4fcf9fc0c5395df4240b : SUCCESS in 47m 05s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b591653dd6354350a04b3dadee1f3399 : SUCCESS in 59m 39s","accounts_in_message":[],"_revision_number":11},{"id":"6893873a6f72981e7a2bea520cf4ab097817ac00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 21:11:43.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"4540684716ef78f05594eda919b2f4c927d701f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 22:32:00.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/a173fe3c117c4466b09c023fbceca683 : SUCCESS in 34m 00s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b939a8f84d3b4400bca41342b65dece3 : SUCCESS in 1h 14m 32s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/42287fe4e6bc4514bc3e26ecdea35ef6 : SUCCESS in 1h 19m 24s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ca5af8dad7b94f978fd77053aac800f3 : SUCCESS in 19m 43s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/92dfc7c915ca41dcb42c1b782eacf674 : SUCCESS in 58m 41s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/a1b5042020e241e59d138764dbaac985 : SUCCESS in 1h 11m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c827d0b38a04ae8a240ae6d01ab4692 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/918a4515a2424ab5bff7c55359b919d6 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/274252c0eb1f4e38b7a928ac855a75a0 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56e8166cf2bc4c0984d6706a9534dc04 : SUCCESS in 6m 35s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ceac9346482c42bc9ba5734fea7ccc21 : SUCCESS in 23m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b1bbb2841be84eae97966f297974ea79 : SUCCESS in 47m 20s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0db8d77a6af44c088725e183e3e5145d : SUCCESS in 58m 02s","accounts_in_message":[],"_revision_number":11},{"id":"672ff41c2213f8bc69d55a9c013e9a26ec102230","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 22:32:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"da36307c47c41a6f5b9d91e0dac56a014b858384","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 22:33:32.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f93aa7171d784a82a18585af243d2477 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":11},{"id":"a0817192330c012b3a8e8864089caf510208fca2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 04:07:30.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/7d1d7b1e2af243168417743f33f7ecf9 : SUCCESS in 2h 18m 42s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a2a8c4a6477d4adbb90f8334bee4f063 : SUCCESS in 2h 01m 09s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b140b7f5ed084d8fa05c4f57a3ba70e9 : SUCCESS in 2h 15m 08s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a092a0fdc1bf4fbbbaadbca28e257e4a : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1eed72f0bc284aa4812eadf2c0891a77 : SUCCESS in 6m 49s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/db802ebf286e4fdba66c0c533df773af : SUCCESS in 1h 44m 22s (non-voting)","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"9dc0df2a998df52af80074623115e9a1c8361b0e","revisions":{"244af0996fdd13357aa0306741bd8995145a6b78":{"kind":"REWORK","_number":1,"created":"2020-10-27 12:58:42.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/55/759855/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/759855/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/759855/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/759855/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/759855/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/759855/1"}}},"commit":{"parents":[{"commit":"efaab1f3a568543a4a22d6b0789e8210d00eec4f","subject":"Merge \"base: use Victoria repos for Debian/x86-64\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/efaab1f3a568543a4a22d6b0789e8210d00eec4f"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2020-10-27 12:58:11.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2020-10-27 12:58:40.000000000","tz":60},"subject":"Add fluent-logger to images - source, binary-debian","message":"Add fluent-logger to images - source, binary-debian\n\nChange-Id: Iedc782a3317c93a2af0da4696fd71c05e0e4a55e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/244af0996fdd13357aa0306741bd8995145a6b78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/244af0996fdd13357aa0306741bd8995145a6b78"}]},"branch":"refs/heads/master"},"af92d0762e496d8aa7a8862dbf72d314a4a09d66":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-11-02 14:46:06.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/55/759855/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/759855/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/759855/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/759855/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/759855/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/759855/2"}}},"commit":{"parents":[{"commit":"efaab1f3a568543a4a22d6b0789e8210d00eec4f","subject":"Merge \"base: use Victoria repos for Debian/x86-64\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/efaab1f3a568543a4a22d6b0789e8210d00eec4f"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2020-10-27 12:58:11.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2020-11-02 14:41:13.000000000","tz":60},"subject":"Add fluent-logger to images - source, binary-debian","message":"Add fluent-logger to images - source, binary-debian\n\nThere is proposed change in kolla-ansible below\nwhich is changing the way how central-logging\nworks and adding native fluent logging.\n\nBecause of this, it\u0027s needed to add python\nfluent library to openstack images.\n\nProposed Change:\n - https://review.opendev.org/#/c/755775/\n\nChange-Id: Iedc782a3317c93a2af0da4696fd71c05e0e4a55e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af92d0762e496d8aa7a8862dbf72d314a4a09d66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af92d0762e496d8aa7a8862dbf72d314a4a09d66"}]},"branch":"refs/heads/master"},"a2acd352e9ece1d9c2bc525b4cdfd7b6ca5859d3":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-01-13 12:26:47.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/55/759855/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/759855/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/759855/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/759855/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/759855/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/759855/3"}}},"commit":{"parents":[{"commit":"93204ecfeca7ba394bd4c3872147357aa708f46e","subject":"Add a storm-base image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/93204ecfeca7ba394bd4c3872147357aa708f46e"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2020-10-27 12:58:11.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-01-13 12:26:47.000000000","tz":0},"subject":"Add fluent-logger to images - source, binary-debian","message":"Add fluent-logger to images - source, binary-debian\n\nThere is proposed change in kolla-ansible below\nwhich is changing the way how central-logging\nworks and adding native fluent logging.\n\nBecause of this, it\u0027s needed to add python\nfluent library to openstack images.\n\nProposed Change:\n - https://review.opendev.org/#/c/755775/\n\nChange-Id: Iedc782a3317c93a2af0da4696fd71c05e0e4a55e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a2acd352e9ece1d9c2bc525b4cdfd7b6ca5859d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a2acd352e9ece1d9c2bc525b4cdfd7b6ca5859d3"}]},"branch":"refs/heads/master","description":"Rebase"},"374138b0f125a4f6a22a3c0e4837db1bb6a6656c":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-01-13 17:07:01.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/55/759855/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/759855/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/759855/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/759855/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/759855/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/759855/4"}}},"commit":{"parents":[{"commit":"75b9c648b33dbf5f6ac59c51e47953007f7bf39c","subject":"Upgrade pip and friends in source images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/75b9c648b33dbf5f6ac59c51e47953007f7bf39c"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2020-10-27 12:58:11.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-01-13 17:07:01.000000000","tz":0},"subject":"Add fluent-logger to images - source, binary-debian","message":"Add fluent-logger to images - source, binary-debian\n\nThere is proposed change in kolla-ansible below\nwhich is changing the way how central-logging\nworks and adding native fluent logging.\n\nBecause of this, it\u0027s needed to add python\nfluent library to openstack images.\n\nProposed Change:\n - https://review.opendev.org/#/c/755775/\n\nChange-Id: Iedc782a3317c93a2af0da4696fd71c05e0e4a55e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/374138b0f125a4f6a22a3c0e4837db1bb6a6656c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/374138b0f125a4f6a22a3c0e4837db1bb6a6656c"}]},"branch":"refs/heads/master","description":"Rebase"},"6c2a8fa2c5ee610849e6dc80f0aa10a32b55eb45":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-01-13 17:09:01.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/55/759855/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/759855/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/759855/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/759855/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/759855/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/759855/5"}}},"commit":{"parents":[{"commit":"afbd99de86b54746f88b633f90c8a4ee576efc8d","subject":"Upgrade pip and friends in source images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/afbd99de86b54746f88b633f90c8a4ee576efc8d"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2020-10-27 12:58:11.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-01-13 17:09:01.000000000","tz":0},"subject":"Add fluent-logger to images - source, binary-debian","message":"Add fluent-logger to images - source, binary-debian\n\nThere is proposed change in kolla-ansible below\nwhich is changing the way how central-logging\nworks and adding native fluent logging.\n\nBecause of this, it\u0027s needed to add python\nfluent library to openstack images.\n\nProposed Change:\n - https://review.opendev.org/#/c/755775/\n\nChange-Id: Iedc782a3317c93a2af0da4696fd71c05e0e4a55e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c2a8fa2c5ee610849e6dc80f0aa10a32b55eb45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c2a8fa2c5ee610849e6dc80f0aa10a32b55eb45"}]},"branch":"refs/heads/master","description":"Rebase"},"73915bda10a749cfcca427ce7e40ece06aef8b5e":{"kind":"REWORK","_number":6,"created":"2021-08-02 22:56:29.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/55/759855/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/759855/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/759855/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/759855/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/759855/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/759855/6"}}},"commit":{"parents":[{"commit":"8eb172c25d7b6d00217b17a3c994eec59dd74011","subject":"Merge \"Fix missing pacemaker-cli-utils in Debian hacluster images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8eb172c25d7b6d00217b17a3c994eec59dd74011"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2020-10-27 12:58:11.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-08-02 22:56:26.000000000","tz":120},"subject":"Add fluent-logger to images - source, binary-debian","message":"Add fluent-logger to images - source, binary-debian\n\nThere is proposed change in kolla-ansible below\nwhich is changing the way how central-logging\nworks and adding native fluent logging.\n\nBecause of this, it\u0027s needed to add python\nfluent library to openstack images.\n\nProposed Change:\n - https://review.opendev.org/#/c/755775/\n\nChange-Id: Iedc782a3317c93a2af0da4696fd71c05e0e4a55e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/73915bda10a749cfcca427ce7e40ece06aef8b5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/73915bda10a749cfcca427ce7e40ece06aef8b5e"}]},"branch":"refs/heads/master"},"4875fdcc3bb86f4cd33607527e8044b1ed06ad2d":{"kind":"REWORK","_number":7,"created":"2021-09-16 11:55:44.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/55/759855/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/759855/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/759855/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/759855/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/759855/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/759855/7"}}},"commit":{"parents":[{"commit":"8eb172c25d7b6d00217b17a3c994eec59dd74011","subject":"Merge \"Fix missing pacemaker-cli-utils in Debian hacluster images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8eb172c25d7b6d00217b17a3c994eec59dd74011"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2020-10-27 12:58:11.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-09-16 11:55:42.000000000","tz":120},"subject":"Add fluent-logger to images - source, binary-debian","message":"Add fluent-logger to images - source, binary-debian\n\nThere is proposed change in kolla-ansible below\nwhich is changing the way how central-logging\nworks and adding native fluent logging.\n\nBecause of this, it\u0027s needed to add python\nfluent library to openstack images.\n\nProposed Change:\n - https://review.opendev.org/#/c/755775/\n\nChange-Id: Iedc782a3317c93a2af0da4696fd71c05e0e4a55e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4875fdcc3bb86f4cd33607527e8044b1ed06ad2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4875fdcc3bb86f4cd33607527e8044b1ed06ad2d"}]},"branch":"refs/heads/master"},"e39af35225d78748b16611804547710b5027adef":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-09-16 11:58:38.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/55/759855/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/759855/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/759855/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/759855/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/759855/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/759855/8"}}},"commit":{"parents":[{"commit":"f98ad9462b7bd70ba79e448f97c7d3b21cf5e14b","subject":"Merge \"toolbox: Move custom Ansible config to global location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f98ad9462b7bd70ba79e448f97c7d3b21cf5e14b"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2020-10-27 12:58:11.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-09-16 11:58:27.000000000","tz":120},"subject":"Add fluent-logger to images - source, binary-debian","message":"Add fluent-logger to images - source, binary-debian\n\nThere is proposed change in kolla-ansible below\nwhich is changing the way how central-logging\nworks and adding native fluent logging.\n\nBecause of this, it\u0027s needed to add python\nfluent library to openstack images.\n\nProposed Change:\n - https://review.opendev.org/#/c/755775/\n\nChange-Id: Iedc782a3317c93a2af0da4696fd71c05e0e4a55e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e39af35225d78748b16611804547710b5027adef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e39af35225d78748b16611804547710b5027adef"}]},"branch":"refs/heads/master"},"643c602729c7599836979941a32da71c61f3f9b6":{"kind":"REWORK","_number":9,"created":"2022-04-21 16:17:07.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/55/759855/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/759855/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/759855/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/759855/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/759855/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/759855/9"}}},"commit":{"parents":[{"commit":"e877556aafb2b571152567a8fc1f6ec640ffb2a4","subject":"add a way to not have \u0027install_type\u0027 in image name (temporary)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e877556aafb2b571152567a8fc1f6ec640ffb2a4"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2020-10-27 12:58:11.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-04-21 16:17:05.000000000","tz":120},"subject":"Add fluent-logger to images - source, binary-debian","message":"Add fluent-logger to images - source, binary-debian\n\nThere is proposed change in kolla-ansible below\nwhich is changing the way how central-logging\nworks and adding native fluent logging.\n\nBecause of this, it\u0027s needed to add python\nfluent library to openstack images.\n\nProposed Change:\n - https://review.opendev.org/#/c/755775/\n\nChange-Id: Iedc782a3317c93a2af0da4696fd71c05e0e4a55e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/643c602729c7599836979941a32da71c61f3f9b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/643c602729c7599836979941a32da71c61f3f9b6"}]},"branch":"refs/heads/master"},"61338ef885766c58411725e97835a931ee64c8ab":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2022-04-21 16:17:39.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/55/759855/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/759855/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/759855/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/759855/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/759855/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/759855/10"}}},"commit":{"parents":[{"commit":"e877556aafb2b571152567a8fc1f6ec640ffb2a4","subject":"add a way to not have \u0027install_type\u0027 in image name (temporary)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e877556aafb2b571152567a8fc1f6ec640ffb2a4"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2020-10-27 12:58:11.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-04-21 16:17:18.000000000","tz":120},"subject":"Add fluent-logger to images","message":"Add fluent-logger to images\n\nThere is proposed change in kolla-ansible below\nwhich is changing the way how central-logging\nworks and adding native fluent logging.\n\nBecause of this, it\u0027s needed to add python\nfluent library to openstack images.\n\nProposed Change:\n - https://review.opendev.org/#/c/755775/\n\nChange-Id: Iedc782a3317c93a2af0da4696fd71c05e0e4a55e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61338ef885766c58411725e97835a931ee64c8ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61338ef885766c58411725e97835a931ee64c8ab"}]},"branch":"refs/heads/master"},"9dc0df2a998df52af80074623115e9a1c8361b0e":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2022-04-21 18:54:12.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/55/759855/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/759855/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/759855/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/759855/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/759855/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/759855/11"}}},"commit":{"parents":[{"commit":"08b80e34dae0236e6b106e7debcdb755508fe1d9","subject":"switch to short image names by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08b80e34dae0236e6b106e7debcdb755508fe1d9"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2020-10-27 12:58:11.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-04-21 18:54:12.000000000","tz":0},"subject":"Add fluent-logger to images","message":"Add fluent-logger to images\n\nThere is proposed change in kolla-ansible below\nwhich is changing the way how central-logging\nworks and adding native fluent logging.\n\nBecause of this, it\u0027s needed to add python\nfluent library to openstack images.\n\nProposed Change:\n - https://review.opendev.org/#/c/755775/\n\nChange-Id: Iedc782a3317c93a2af0da4696fd71c05e0e4a55e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9dc0df2a998df52af80074623115e9a1c8361b0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9dc0df2a998df52af80074623115e9a1c8361b0e"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
