)]}'
{"id":"openstack%2Fkolla~760533","triplet_id":"openstack%2Fkolla~master~I7bc1a4e025799bf155b07881adaedf408193ef55","project":"openstack/kolla","branch":"master","topic":"dnm-1-thread-only","hashtags":[],"change_id":"I7bc1a4e025799bf155b07881adaedf408193ef55","subject":"[DNM] Use a single thread","status":"ABANDONED","created":"2020-10-30 10:09:48.000000000","updated":"2020-12-16 11:31:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ffe1dbcf4b840dcb96f49ef911c12c4ac3953c80","_number":760533,"virtual_id_number":760533,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-10-30 15:46:29.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-30 20:01:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-30 10:10:34.000000000","updated_by":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"reviewer":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"state":"REVIEWER"},{"updated":"2020-10-30 20:01:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-04 09:11:11.000000000","updated_by":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"reviewer":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"state":"REMOVED"}],"messages":[{"id":"101c0e2b22843648bdbf58f501005a74a7b48a5e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-10-30 10:09:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b4c1e05c76261914424751a555a242798524e27","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2020-10-30 10:10:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"695643b987b659c1e1a7c8e23d86afc379c73c5a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 11:58:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/74f0660b46f24d169d5ef38b1fcdda40 : SUCCESS in 1h 47m 37s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4156d9632fc644c19a52bcb30e896264 : FAILURE in 1h 30m 26s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b8fbd10cf35c41e2814cc8b310bb6b26 : FAILURE in 39m 13s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/fcb91c5ce92d4d01bac27614fb215789 : SUCCESS in 1h 29m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7fb558bc5ff2ede701f77250cf8d6e7bf4504abc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 13:30:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/19c13b1fc3654819b41f9c1663536362 : FAILURE in 1h 27m 10s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3141ff9e48e34f9fa68a901d406c849c : SUCCESS in 1h 36m 12s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/50e4ad4fc154457facb0e587081fdc84 : SUCCESS in 1h 26m 09s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/9c3dbff0591d4c3697d97f27b9cae5f1 : SUCCESS in 1h 40m 29s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/138031b9dd9f43cc8f09204a724fd9d2 : SUCCESS in 1h 59m 19s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f9a65ccfb5b4440884d290a7352437bd : SUCCESS in 1h 03m 37s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/59120088c832436bacad7859f83157c8 : SUCCESS in 1h 50m 06s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4f2f06c4537e4811bd1320c1253a8240 : SUCCESS in 1h 24m 41s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a6d79fcbfd5d40c685703b9f4a839783 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97c0120bc98243618015d06e19c22de8 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84b440ef2fe647418a97de77914eb8a4 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8827bc4ce2a8408eb68c5c2e7b389ae5 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/091365025fde4128a23b818024abe55f : SUCCESS in 11m 15s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/de1ff29a6ebb4fcb91a1eab14807db72 : SUCCESS in 2h 06m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/717fb45b5b2847d5a3633dc14f379d17 : SUCCESS in 1h 50m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/743b2a4d8e9245f9a3d7e38dafb98109 : SUCCESS in 1h 22m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/139152e75b2a490eb6f4eaac7a7eb255 : SUCCESS in 1h 29m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2a27b9fa6af0417bb913fb54a331fe03 : SUCCESS in 1h 20m 21s","accounts_in_message":[],"_revision_number":1},{"id":"48e822e3f4b5022bf40c9927c9b9a0d19107d3ee","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-10-30 15:46:29.000000000","message":"Patch Set 1:\n\nrecheck\n\nit sounds like we are getting like 3x speedup from 8x threads by default","accounts_in_message":[],"_revision_number":1},{"id":"dd635780cc919348b47d14354dbcd44b5c484dcd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 17:39:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/7febf6bf4e9a4b96a4dc4454464ca543 : SUCCESS in 1h 51m 39s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a5f05e7dcdf44c689478cce14d3dc7b0 : FAILURE in 1h 34m 14s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e21a27c8e81a41f7a9d6c557ce1b2ede : FAILURE in 40m 30s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a47ef87c03854c80b7b2b388569600bb : SUCCESS in 1h 37m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d3b7459253744de86bd4d14bd6d9bffc0a253c11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 20:01:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/28d975d8b4c04fb09084b5afa84004d1 : SUCCESS in 1h 32m 36s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/55dcc94ae673450dab2bb0321535a444 : TIMED_OUT in 3h 01m 50s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/5338eb03c8e24e91928276158006bf75 : SUCCESS in 1h 33m 31s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/4e94670121e6405cae890df5feb2fe98 : TIMED_OUT in 2h 06m 04s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/150ac8584d1c426d97ca0222c4a9c2e5 : SUCCESS in 1h 25m 43s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c1dc7de3549e4c4c93c4629759defc0e : SUCCESS in 1h 07m 57s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/130a59f632ab422cbb0cb8b18e331861 : SUCCESS in 1h 10m 09s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/972b9cbcc9af41eb81fd7152bb9cad6f : SUCCESS in 1h 27m 17s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a312677aa80d471fa3b74e370485adfe : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75f4f1de9c3e4d48a8a2e0f702234645 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96da78d046b14aeca0a45d0208708ade : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c922013603a5426297e0a45395c1cf30 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e970535a6909440090e382e61ce283b9 : SUCCESS in 5m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6180ea9240d14d61baa45ea185ef7fe6 : SUCCESS in 1h 19m 41s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9f3c0ed0e29e4c749cd54d8f79868528 : SUCCESS in 1h 25m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/67281bdca167472e8cedaa58bc8cdaa7 : SUCCESS in 1h 06m 53s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8c00140a9c7f418faba8beb744b9a269 : SUCCESS in 1h 31m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3469ac280de5451cb3b3930612fe60f1 : SUCCESS in 1h 20m 19s","accounts_in_message":[],"_revision_number":1},{"id":"5ac6768865deda3825775bcb541cd4e4a45ebecc","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2020-12-04 09:11:11.000000000","message":"Removed reviewer Gaël THEROND with the following votes:\n\n* Code-Review+1 by Gaël THEROND \u003cgael.therond@bitswalk.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"ffe1dbcf4b840dcb96f49ef911c12c4ac3953c80","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-16 11:31:46.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c98b503b1725ca5f1fa97ea4ed787e1b744140af","revisions":{"c98b503b1725ca5f1fa97ea4ed787e1b744140af":{"kind":"REWORK","_number":1,"created":"2020-10-30 10:09:48.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/33/760533/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/33/760533/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/760533/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/760533/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/760533/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/33/760533/1"}}},"commit":{"parents":[{"commit":"9d9ea0f63284d381062eedfa7a9bfc851cb4f273","subject":"Merge \"[CI] Use local RDO mirror\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d9ea0f63284d381062eedfa7a9bfc851cb4f273"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-10-30 10:09:16.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-10-30 10:09:16.000000000","tz":60},"subject":"[DNM] Use a single thread","message":"[DNM] Use a single thread\n\nto debug issues in the gate.\n\nChange-Id: I7bc1a4e025799bf155b07881adaedf408193ef55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c98b503b1725ca5f1fa97ea4ed787e1b744140af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c98b503b1725ca5f1fa97ea4ed787e1b744140af"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
