)]}'
{"id":"openstack%2Fkolla~760566","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I2a1e2b6661fd4b62648756416e6147b604dda686","project":"openstack/kolla","branch":"stable/ussuri","topic":"openstack_base_constraints","hashtags":[],"change_id":"I2a1e2b6661fd4b62648756416e6147b604dda686","subject":"Use upper constraints for virtualenv in openstack-base","status":"MERGED","created":"2020-10-30 14:25:50.000000000","updated":"2020-11-07 21:59:08.000000000","submitted":"2020-11-07 21:58:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"760566-1604786282578-40a8ff94","meta_rev_id":"fa18607e963419ac5e6cf78f541bacb361a494e8","_number":760566,"virtual_id_number":760566,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"date":"2020-11-06 10:15:26.000000000","_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-07 21:58:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-11-07 20:54:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-11-07 20:42:53.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2020-11-07 20:54:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-07 20:42:53.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-11-07 20:54:55.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-11-07 21:58:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1e2797fba908624b554bf9351463ddfc1dd3ce25","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2020-10-30 14:25:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"455de5020bce765f7584b79d0d09c60a63e0d189","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 15:55:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/f203c04d672345c7be841ad6a06fa056 : FAILURE in 35m 58s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f9ff34392002481e8331c9048eab1020 : FAILURE in 30m 00s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/55706770f72e498db8e40b2b79ca763b : SUCCESS in 1h 24m 41s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/dce8e1ba2ef74fb09201dada8c70b248 : FAILURE in 30m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f1c922c6091653ab0c7a7e7e918b46f40a016628","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 17:29:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/3b6b9789331a49939a3896e3880c8579 : FAILURE in 23m 33s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f8126c17860c4eb694619b58e709d1b5 : FAILURE in 26m 50s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/32c06b73cd51432d86f22c7f3161bfbe : SUCCESS in 1h 09m 35s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/2fc19dfae0b742de90053e7ec48bc3d4 : SUCCESS in 1h 16m 33s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/fcb34ea486874c5f9e4c0cbd49020569 : SUCCESS in 1h 10m 20s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/40a84f212c8449b0aae41d9edfb327ba : FAILURE in 28m 55s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/b15272cbae8844328f5a3505b2fc0a5e : SUCCESS in 37m 24s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/5a401b628c404b7188b93c3de59c4821 : SUCCESS in 1h 46m 05s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/5026e48abae74cf39533264420577ff5 : SUCCESS in 1h 53m 07s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/91205332b80c4e148e7d2fd09838779f : SUCCESS in 1h 35m 22s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/7523dfb60bbc403da2ef40f26cd4e327 : SUCCESS in 1h 42m 32s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/9fdfceea846d42a8bab7f795994d637f : SUCCESS in 1h 10m 33s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/aa490223509e4a43b106c376cba6e186 : SUCCESS in 1h 28m 52s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9217c603d1a848338cff0ae5af08e4f3 : FAILURE in 23m 23s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8909ca51e83e4f40afe676ea2e6ad572 : FAILURE in 19m 15s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e9abdaa61fba4b22a29f9b1e4d26f255 : SUCCESS in 1h 15m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ece9adc665e94a52bf25a123d3457661 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6c1369e9c2740c593faaa7c9144c6c0 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32a173fb979b4b3b92c1bbe09b1e71af : SUCCESS in 3m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/05e671bbb61c42caa0e39002f592dc86 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f45eee01580e41209b6c20c0139efc5c : SUCCESS in 6m 11s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b248470d44e244499b5d127968ed091c : FAILURE in 24m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cbb9c9b1a99d42b48ad8cc9a1e87fce5 : FAILURE in 29m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e49b459023bf4046af096e2a1f01d7bc : SUCCESS in 1h 06m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9e15ed39e2ce41c594baaa6c381d27c8 : SUCCESS in 1h 03m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7c36bbbb5ba64c63bb09cae7dccdb099 : SUCCESS in 1h 13m 11s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/dc63a89d12404b2f9314f01d3b54e531 : SUCCESS in 1h 30m 40s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/58e9f6da109548939f22d86cab90da0a : FAILURE in 57m 14s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/a94b568274f14cd7942f0598222c9382 : FAILURE in 1h 26m 51s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/3e186992db714c829eccad644b04dd85 : SUCCESS in 1h 02m 59s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/321368de6ed5481483522d11d59caec1 : SUCCESS in 1h 12m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4bffd2387954ba688cb621c3c193e5e5896ff590","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-01 08:59:56.000000000","message":"Patch Set 1:\n\nboto failing on escurator","accounts_in_message":[],"_revision_number":1},{"id":"b3b27bb2ac04d7957d2a3c731fd133514ffef1c4","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2020-11-06 10:15:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d30d57233ca0193606185f738d5b1eeead55e9ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 10:45:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/df735fdf949641c891f6b9ba1ad9f389 : RETRY_LIMIT in 2m 26s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d6321d92bd194b89ba4ff06210fd1c69 : RETRY_LIMIT in 2m 11s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1edf06e266154a6e8450b78419dec52d : RETRY_LIMIT in 1m 56s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/7310c52bf835447abbde650f4a201fbc : RETRY_LIMIT in 2m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"151a9674b9073097f8ffe33a9daf38934a9247b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 14:43:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/da95cccec54d4b9fafa1d696d0063190 : SUCCESS in 23m 17s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/c9f48dfcb69d4c7cb5307cf8cf7a451c : SUCCESS in 28m 08s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/29741aded1634a04978fd723f3938de9 : SUCCESS in 1h 02m 28s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/f6acb59a1c9c456994e47c95f961ea48 : SUCCESS in 57m 26s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/3255c4207eba4ec5a84cf816fc3d3023 : SUCCESS in 1h 04m 53s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/09717ac26d7d4da7b59015433d9f01ee : FAILURE in 26m 19s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/29f599f1d25a4bc08dc6c87248bc4861 : SUCCESS in 27m 42s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/8698329f0fbe442ba5e6c29e65023724 : SUCCESS in 1h 34m 50s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/8347463c843d4f68b686af2abbd7c524 : SUCCESS in 1h 45m 10s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/77b4d877e7e948c88652a2ab4dfbb816 : SUCCESS in 1h 34m 36s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/c358b8d2f19040f89a43aade017f1981 : SUCCESS in 1h 39m 06s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/f489491a151e485d80c1d27617ce6f7a : SUCCESS in 1h 11m 28s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/29f2b4840233455bb0c9b2e85bd9fc76 : SUCCESS in 1h 42m 10s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c235d661585140798b434cf99c06334b : SUCCESS in 20m 56s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c466f29ade5545a1bb1fd692bddc9ff9 : SUCCESS in 1h 16m 23s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2130fae049f54293a719e6ce887c8c02 : SUCCESS in 54m 28s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f311038ebc3241af97ca48919b718a24 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d516be364e54b52b5c8440f54bf557a : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a5e9af8794543ad91f091e23d7be282 : SUCCESS in 3m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4344bc2376ff434f939be52a2354b3a6 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b88bad2034374ff3952ca13835d2d221 : SUCCESS in 6m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0104ccfdbd51437881cf140dc63ac84c : SUCCESS in 23m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1aaf9a5079c344199132dc641a8eea45 : SUCCESS in 30m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/44c1b7ebe0434f569f935e592a6452c8 : SUCCESS in 1h 03m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fd29f3f9fe5642d1bb0d0e2d93479f93 : SUCCESS in 57m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ff246d3b40b347b0b7e0618f3aa28230 : SUCCESS in 59m 44s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/95b6acdfe7b44c69afa6653bf7b26c66 : TIMED_OUT in 2h 40m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/7c5100864fb2432ca75a406cce24e10b : SUCCESS in 2h 21m 23s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/d872987774484e04903342162d8e11b3 : SUCCESS in 1h 34m 11s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/7f94a0a4712045a1bc0ea50036068f0a : SUCCESS in 54m 11s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/b5ea93b27537405c892ab6f420fee886 : SUCCESS in 1h 20m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8a5ae2006ff8d385345db877d861d6364f5c40c2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-07 20:42:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"14fad0f0ee962d60d661c22ffe38ef1fddccc8d6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-11-07 20:54:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"819065e9c80e58ded4aaf554e12318dfb7adc075","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-07 20:55:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d9d31b953e9c24a972e647e9c41c9afb8f67be63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-07 21:58:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d2cf79072fc742098479dc8ec2dbb918 : SUCCESS in 25m 14s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/5550c17b8f5449b28b0d95d01e0c885f : SUCCESS in 55m 29s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/bd88c2d0317b411e85de94519bb45430 : SUCCESS in 59m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/839cef1cba58456d94f10ba5c08054dd : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b404558342a4fdcbe94bf2bf5197165 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe2bfe1c7bf04e949c2dd9881747ef70 : SUCCESS in 3m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a7a8abd3196c44ebaba8dd4021216721 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f3397b6a86944b2a17bc23db6ab1999 : SUCCESS in 4m 44s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/42c8dd88afb0412a80d757fe928628e4 : SUCCESS in 21m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/01c72c4adb204220b3fd8fe6bde03c96 : SUCCESS in 48m 02s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/542224d1ffd449329c8c97621880782e : SUCCESS in 54m 48s","accounts_in_message":[],"_revision_number":1},{"id":"6c01a3f22fe5f9937fc7911d0ced4707db14939c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-07 21:58:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0cb11f4ad0447ad70478090e3e52e46c7daca676","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-07 21:59:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a2f138192b74e788340b6f2379ac9db : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bebad0409ea1f4a6b9de8051821f6b3fa444aca3","revisions":{"bebad0409ea1f4a6b9de8051821f6b3fa444aca3":{"kind":"REWORK","_number":1,"created":"2020-10-30 14:25:50.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/66/760566/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/66/760566/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/760566/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/760566/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/760566/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/66/760566/1"}}},"commit":{"parents":[{"commit":"ca699a830c6cd95a2dce07548f8608dca4b25be7","subject":"Merge \"Fix kolla_set_configs --check with a directory\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ca699a830c6cd95a2dce07548f8608dca4b25be7"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-10-16 11:10:03.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2020-10-30 14:25:49.000000000","tz":60},"subject":"Use upper constraints for virtualenv in openstack-base","message":"Use upper constraints for virtualenv in openstack-base\n\nSince Rally has virtualenv in requirements, and is using contraints - it\u0027s\ntrying to downgrade virtualenv from inside a virtualenv.\nLet\u0027s try to install proper version of packages in openstack-base.\n\nCloses-Bug: #1900419\nChange-Id: I2a1e2b6661fd4b62648756416e6147b604dda686\n(cherry picked from commit b1709a87ed6bead417a3cd6848de068e06190370)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bebad0409ea1f4a6b9de8051821f6b3fa444aca3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bebad0409ea1f4a6b9de8051821f6b3fa444aca3"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
