)]}'
{"id":"openstack%2Fkolla~761182","triplet_id":"openstack%2Fkolla~master~I151420d3036410dd5ce6d90907d72dd3c20643b5","project":"openstack/kolla","branch":"master","topic":"neutron_update_alternatives","hashtags":[],"change_id":"I151420d3036410dd5ce6d90907d72dd3c20643b5","subject":"neutron: Fix update-alternatives on CentOS","status":"MERGED","created":"2020-11-03 15:26:58.000000000","updated":"2021-10-11 12:58:01.000000000","submitted":"2020-11-08 20:42:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"761182-1604868176937-422172d5","meta_rev_id":"76dca890a48fdd8ef5951ecd1c2faf4c3b7536dc","_number":761182,"virtual_id_number":761182,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-11-05 12:54:00.000000000","_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-08 20:42:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-11-06 08:39:15.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-11-08 19:30:33.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-11-08 19:30:33.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-11-08 19:30:49.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-06 08:39:15.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-11-08 19:30:33.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-11-08 20:42:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b747dfa3921e414ae99a9d43ceb60e9f8dd227cf","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-11-03 15:26:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d653d007a05a83cf4188c4899c53a9dafc02c963","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 16:17:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/6364ca2126ec4b0e884b4786c0c4b6e7 : SUCCESS in 41m 28s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ed1b0c5b4260424aa876a5f03c4951a5 : FAILURE in 33m 03s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a8d94cc0250842daa5d67351d8d8c1d3 : FAILURE in 28m 32s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/0ad7c3f22df44e14add0751720df3980 : SUCCESS in 38m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bb7deb08b46dd6e137bb8b83f5b2076ed3b7c5ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 16:51:57.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/faf3da751beb4e1ca55e1af466c7a919 : SUCCESS in 20m 39s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/48af8d4c4ba1483589694c458692a11e : SUCCESS in 19m 23s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/35d34a389e97490192a78adb46ff05fb : FAILURE in 41m 06s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/e6148c95989147f9850a3a25cc37016e : FAILURE in 39m 55s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/73be9ff09a6b4e1c82c91f78a9404a40 : FAILURE in 1h 05m 12s\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/68436d3a0ba047af9b02a4cf5f62dead : FAILURE in 58m 08s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/81745e3b8b594b26bc8e074ce2a581f6 : SUCCESS in 29m 47s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6fbee97edc7a4c5f9b00c8c303a916fa : SUCCESS in 33m 21s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3dd7c7e0a48d47cabdff16f69cf8be7d : FAILURE in 55m 21s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aba60b8ffc9242cb893f32305a9f6d26 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6d7b43e202f46d9aa1767656309727f : FAILURE in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63e08b5c31d4498fb385d02a5c1fa1b7 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e07f3f9086940658885e7298ae2f873 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba54d6c3bab4415e936a27b8600740f0 : SUCCESS in 5m 22s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6046b485723849d183eddba8c97b24be : SUCCESS in 1h 14m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/136f1102f0794b6c8e848755f4121fe5 : SUCCESS in 30m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bcf7783c949b4acf8777de97d94af706 : FAILURE in 52m 49s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1f7e8d8e2e2b4c9190a65b99ad4610e1 : FAILURE in 46m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c900306100e54293ad61237a658b0d90 : FAILURE in 58m 32s\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/c6988667f9e44d1b8d7dbc025df8b5ff : FAILURE in 35m 43s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/206f1a9bb5f643238df0b2d572440981 : FAILURE in 55m 09s (non-voting)\n\nWarning:\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: file docker/neutron/neutron-base/extend_start.sh has only 31 lines\u0027","accounts_in_message":[],"_revision_number":1},{"id":"908a23f002e88ff36a15bf4df04da6ef37ce747e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-04 09:41:24.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e2f7eb8b88bafc979eed629a3cf1ea8f661896d6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-11-04 09:51:16.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"4e9db5c64837417e85842f86222f8892c9de1224","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 10:53:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/f2b3034620394b3ca70b8ca0ef898f06 : FAILURE in 57m 27s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f5d96c416e4a4aa59be53bbb7f63e1e0 : FAILURE in 52m 06s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b3e3d28f35d54f469ecdbb4717a53d9a : FAILURE in 40m 36s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/40832131f6514732a270c2963e9a4e34 : SUCCESS in 51m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b7b6813d92a316ce50defd4340f84a8e1b01b02d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 11:11:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/ba7ff12e589e418293f135f229247f27 : FAILURE in 23m 57s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/5287507fff794bf8852d6654756bcdf8 : FAILURE in 24m 38s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/3e7e8e1c1a134e8aae556868dc96f0d0 : FAILURE in 42m 09s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/88e9e1e40b6c4f8c91441aab30c75bee : SUCCESS in 1h 06m 07s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/d7ce57c1fe234e2c945609fa633223b5 : SUCCESS in 1h 11m 15s\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/c4e1c06d9c18484fb4d2c2f84117f78f : SUCCESS in 1h 15m 00s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d52ed820122941228a746e2449acdb25 : SUCCESS in 19m 49s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/193e458a01504b1baa0ea1e39287728c : SUCCESS in 15m 21s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6c47f0cb8eff49978e4b3a3fbad4b4b8 : FAILURE in 38m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6bc984eeace47de84acfcb69bc21c2b : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30baa69ff16f4a54ad56dc894aae0c12 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1eaeddcd616f47f18ed26c35250a60c6 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8e452cf529c47d4a666e29e9c9202d8 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb3be02d62f54f4e975a67a60dc9d31a : SUCCESS in 5m 49s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e7bf6783d1f145f89facbc84a05ce0a9 : SUCCESS in 21m 01s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/aa7ae271ec714ca8afd27345d31b7f59 : SUCCESS in 19m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1e353f8837a3485faf737d5cc1d2ed0d : FAILURE in 32m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e7bc3a1e04864547baff4614cebee968 : SUCCESS in 54m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/256c80a39e8544dc89fcfd0fce3ced18 : SUCCESS in 1h 02m 16s\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/3ed4d326b6a14a12bedf10f4f9d6c444 : SUCCESS in 52m 14s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/783b0fbc22054349a0fcf5ba3cab98ed : SUCCESS in 1h 05m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"370e90c006e1ee1aa63766f7aaf12a6952168b8e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-11-04 16:29:59.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"2f9db5f086f087f245b0110de1c21e3ff7061e48","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 17:22:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/9cb1a8c441d743b7a3b08f8f99fa1b1e : SUCCESS in 45m 47s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/827ab8866748424aa45b8095daf0fefa : FAILURE in 36m 08s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c667f90493894160b74806ad9269c826 : FAILURE in 29m 51s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9a1b572f7f3247dcb832ee14e2e4299a : SUCCESS in 36m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"22cd422ca41acbfc646733641ec1be4d4531ff77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 18:46:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/6919345469844b49bb0201d22479d6ec : SUCCESS in 22m 19s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/5f9cac371a4849fda766302bf097bc97 : SUCCESS in 24m 21s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/96d2ff7afca54b66bcb64a7f7fe4a75f : FAILURE in 41m 08s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/594ffddfe9234c3ab9ad393fb8bec7b2 : SUCCESS in 1h 06m 26s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c09b704d977e4c73af68141eb4269330 : SUCCESS in 1h 15m 47s\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/5f12668d5c094b3296a66b205eb1ab0e : TIMED_OUT in 2h 06m 32s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4c8a220976814584b3cde0663089ce28 : SUCCESS in 18m 08s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a255d986ca544375aa999523d6937399 : SUCCESS in 29m 10s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/035aeeec652d4658b4a732b5d8cf0afc : FAILURE in 46m 55s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b267155cbd734fb2a090d94561929535 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d019f5550c2f4ee5a7bfa378ce24a076 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef30f43b924143fda173c81325ba8149 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a486b7368be47f985747b958229af3d : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37f81664451347d98ff00150d1349293 : SUCCESS in 7m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/48fe4cba446e4593a297d425b61c11f7 : SUCCESS in 30m 55s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a167e358a1d5422491557764421a4065 : SUCCESS in 28m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fe5895b6f126454dba30ace4c8d60765 : FAILURE in 39m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/54ff06ea50c64101a6f9fa7ce8faeca5 : SUCCESS in 54m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bb0b9d2e23ba4782b793291433bea1a4 : FAILURE in 31m 58s\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/60ddb25ca81540f9b8afb829f052c50e : SUCCESS in 46m 46s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/e4bce03eee2e416cbac07c853e3f7922 : SUCCESS in 1h 08m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"836d8c2eb798e42eda2fe077ff90be85ae5638af","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-11-05 06:56:48.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"6590c9c9209fac05cbc9c998fb0414af00409503","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 07:34:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/adcfba0059b64075a42b2f33300b8374 : FAILURE in 36m 19s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6b81c31c04a74ab68149adea5b010e26 : FAILURE in 30m 02s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/95dcd1a797be47a5bd98eabe65087086 : FAILURE in 11m 06s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/8f383a765235490f8b05ddf0e08540e2 : FAILURE in 10m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"66af254c4377a76d70188327a6d29e4e83b51a11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 09:33:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/79d5a9c1e32b4515a4a44552f7c304b8 : SUCCESS in 21m 59s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f2e7f93130db41f7bf6168ef3ccde886 : SUCCESS in 48m 16s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/4acf49c4005d41c284b11894953c505e : SUCCESS in 1h 02m 17s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/676bec028a7a44cda77d3df648280727 : TIMED_OUT in 2h 03m 37s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/a0a782def27a41488d8e72dc41aecbd4 : SUCCESS in 1h 17m 09s\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/9975425b8eeb4fb493372283b3b03534 : SUCCESS in 1h 24m 08s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4c58c1245be146d9aefeb3cfad829e9f : SUCCESS in 17m 48s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/42a9e3b015804dd9be10526dddf17f28 : SUCCESS in 18m 12s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0133e37dbf6645958c876a7d0f0bd869 : TIMED_OUT in 2h 04m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a5093b260c045e182ce04ae8bc183b8 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c017d476f664decabe2d76834f2d10f : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4cd5f2bffff3442aa5a599c14a71cff6 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a2c92d262674468ac678bf8e0e84c05 : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e781f75692e4815826365b839c93fd3 : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f0ee4706513445ea4ae90c063d2206b : SUCCESS in 6m 51s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/021ee740e94d40cd951ea4d2394d1d9c : SUCCESS in 21m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6bace80412a84caeaf188a89c53ef40c : SUCCESS in 19m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/77d05c78b53e4e33a052b58e5b44690f : SUCCESS in 1h 16m 50s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3b26fdab33714975a80fb2c893b9ff17 : SUCCESS in 1h 11m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8af1cebd11f84010afe7204e662dcb39 : SUCCESS in 1h 10m 02s\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/e49f9f822a2e41e8a815ae26d957bbc9 : SUCCESS in 51m 03s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/e3be8fddefff41f485d40acd85836926 : SUCCESS in 1h 25m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"400d746176d81878884329dd872a31816d9a5c2c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-11-05 12:54:00.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"7ce5dfea800c5c743d841c616c96f3b5f79531a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 13:30:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/642a0660f001419dbfcbdfd05347bfd8 : RETRY_LIMIT in 2m 41s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/32eec9c7e0d245059e46ce078448d888 : RETRY_LIMIT in 2m 06s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3170a1b1eb384d5197826ab76222d2ec : RETRY_LIMIT in 2m 42s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/6191ea96ca354b38b7827db0319f12ca : RETRY_LIMIT in 2m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c274caee1684bd03a498d354c651e3e6570215ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 14:43:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/6cf106ddb065482d8fe0f9cc3dcca42e : SUCCESS in 21m 04s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7737a93ffba74ba7b7fabd9885367827 : SUCCESS in 23m 49s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/79e9ef8d8b3e4afaa03ae18e95bb4117 : SUCCESS in 1h 05m 50s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/aa29b4552575489ab3bda20132c37e25 : SUCCESS in 1h 09m 44s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/7dee6a9fee834f41876e544bc495a8af : SUCCESS in 1h 19m 42s\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/f5adf8a028d94e4a87fca8df603b8f88 : SUCCESS in 1h 22m 24s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/76546a967c6e44c486634a961ef79e07 : SUCCESS in 18m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/48ea6f08c09944c3924c363fdc5b8ebd : SUCCESS in 16m 51s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/85e2f60226e6495695eb03f29b188a40 : SUCCESS in 56m 55s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b6f0d39c843b469c9ba7cbc3520f77ff : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44e50da583a340e6994beeb4400af150 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e084c09f4a194ed38e9e9abd0a8405a2 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd3c108af3a54dffb7c03d6ac6f9169c : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b276185ddfe5414e82f026cb97c69d53 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cb10266fe0a48078ea04f0d4deae110 : SUCCESS in 7m 49s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4ec75aae79fc4d0a9bd98a62fa6f7da4 : SUCCESS in 21m 42s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e0c6aaa8a27748f49b5a055962969839 : SUCCESS in 19m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d1f3e74734d94809a17a9467420fb1d1 : SUCCESS in 52m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cb013c5a8fd64c0c802de8f436167d62 : SUCCESS in 50m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2be4faa481fa4a4ba3d7a41634d7eb79 : SUCCESS in 1h 04m 36s\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/63db2728ceb342f29ecb2e9ecc0a6f58 : SUCCESS in 50m 38s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/9d4708bfcc8d4f0ab5f9e5e922528d6e : DISK_FULL in 1h 08m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8a82b8a7b32a3a6bbb9222771fe34c19ddaf126c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-06 08:39:15.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f6a0bef8ddfc12178d4cd7d2aa7e181f172f8a75","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-08 19:30:33.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5abc31a4cd8339d970dd466f34c42d6345a67591","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-08 19:30:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"155cb5030a0b3398d3585d648667ab834592fc4f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-08 19:30:49.000000000","message":"Patch Set 3: Backport-Candidate+1","accounts_in_message":[],"_revision_number":3},{"id":"45bd35b2c7e1f2a172406c07c2234d8e1cbc037b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-08 20:42:56.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b67ac6efb97c43fe865ba699adf7a126 : SUCCESS in 20m 15s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/6045821d330a479da05ce030a416340b : SUCCESS in 1h 10m 19s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/a0d29e629b514491ad24338eb896ef6d : SUCCESS in 1h 01m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa0c9b7f819c43a186298c36389296f3 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cc2628efadf4429a85555b8bacaeab5 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7af24991367946dcba37691a0b0f0738 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d2b05f51e9442a5b9105576d3e19e7c : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e733a50d9c6549348b13562623280f3d : SUCCESS in 6m 37s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e2c79c149c324e188458f0e4605051cb : SUCCESS in 18m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1f1fab7b7baa4d299666bb9baee24bea : SUCCESS in 50m 04s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7e51428eb6cf4cdb9b3c0237f43dd471 : SUCCESS in 1h 01m 20s","accounts_in_message":[],"_revision_number":3},{"id":"d944857bfd1e0447f855ebd8d91ef6ba3963f1d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-08 20:42:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"a7c1dd8fc2823e8a3fe5110352d9610582884ddf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-08 20:44:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2df6cabe414b4db186d65cd433c9a10e : SUCCESS in 46s","accounts_in_message":[],"_revision_number":3},{"id":"d1ea309bc9bf75d70715f17fbc533b8e8cf3b672","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-08 20:44:36.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/victoria as commit 4dfdde42cd422af6d8b72949e3947c1a006bc37a","accounts_in_message":[],"_revision_number":3},{"id":"fbc9d6a270d75001f9c6568759b2919e39abdeb3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-11 10:29:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"76dca890a48fdd8ef5951ecd1c2faf4c3b7536dc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-11 12:58:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7a4ba91fa667a295a7e1023abfaa15e0874a371a","revisions":{"88cdcbb280e3cbdd3ea6199878b670f66b24a363":{"kind":"REWORK","_number":1,"created":"2020-11-03 15:26:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/82/761182/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/761182/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/761182/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/761182/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/761182/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/761182/1"}}},"commit":{"parents":[{"commit":"f47f7b4b6cf5ec31b50aa9a9bd6f3815327ed131","subject":"Merge \"mark trove images as buildable for Debian/x86-64\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f47f7b4b6cf5ec31b50aa9a9bd6f3815327ed131"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-11-03 15:26:13.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-11-03 15:26:55.000000000","tz":60},"subject":"neutron: Fix update-alternatives on CentOS","message":"neutron: Fix update-alternatives on CentOS\n\nChange-Id: I151420d3036410dd5ce6d90907d72dd3c20643b5\nCloses-Bug: #1884778\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/88cdcbb280e3cbdd3ea6199878b670f66b24a363"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/88cdcbb280e3cbdd3ea6199878b670f66b24a363"}]},"branch":"refs/heads/master"},"e5a705e3571d846eeb5130a0a39c6c2515bf0203":{"kind":"REWORK","_number":2,"created":"2020-11-04 09:51:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/82/761182/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/761182/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/761182/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/761182/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/761182/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/761182/2"}}},"commit":{"parents":[{"commit":"f47f7b4b6cf5ec31b50aa9a9bd6f3815327ed131","subject":"Merge \"mark trove images as buildable for Debian/x86-64\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f47f7b4b6cf5ec31b50aa9a9bd6f3815327ed131"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-11-03 15:26:13.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-11-04 09:51:11.000000000","tz":0},"subject":"neutron: Fix update-alternatives on CentOS","message":"neutron: Fix update-alternatives on CentOS\n\nChange-Id: I151420d3036410dd5ce6d90907d72dd3c20643b5\nCloses-Bug: #1884778\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5a705e3571d846eeb5130a0a39c6c2515bf0203"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5a705e3571d846eeb5130a0a39c6c2515bf0203"}]},"branch":"refs/heads/master"},"7a4ba91fa667a295a7e1023abfaa15e0874a371a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-11-04 16:29:59.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/82/761182/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/761182/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/761182/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/761182/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/761182/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/761182/3"}}},"commit":{"parents":[{"commit":"34460d636e9e62f30a0d17eda2d743658ccb0b33","subject":"curator: Add wheel to elasticsearch-curator","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/34460d636e9e62f30a0d17eda2d743658ccb0b33"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-11-03 15:26:13.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-11-04 16:29:59.000000000","tz":0},"subject":"neutron: Fix update-alternatives on CentOS","message":"neutron: Fix update-alternatives on CentOS\n\nChange-Id: I151420d3036410dd5ce6d90907d72dd3c20643b5\nCloses-Bug: #1884778\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7a4ba91fa667a295a7e1023abfaa15e0874a371a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7a4ba91fa667a295a7e1023abfaa15e0874a371a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
