)]}'
{"id":"openstack%2Fkolla~761403","triplet_id":"openstack%2Fkolla~master~Ia98af905953d9a00be27e738a21b0b08bd7fa9ad","project":"openstack/kolla","branch":"master","topic":"sudoers_for_ovs","hashtags":[],"change_id":"Ia98af905953d9a00be27e738a21b0b08bd7fa9ad","subject":"toolbox: Add sudoers for kolla_toolbox become","status":"ABANDONED","created":"2020-11-04 12:48:26.000000000","updated":"2020-11-08 19:48:20.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"11aaaae7e86312a535737e96f7d5f1baaa816019","_number":761403,"virtual_id_number":761403,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-11-04 14:16:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-04 14:16:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2301ed04a5e90809b12e5aeb746c947544152488","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-11-04 12:48:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ece28230b92bb22600dbb4e7f0593a68783268fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 13:30:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/70d8b084b8f6449eb2ffc947aebb6104 : FAILURE in 41m 45s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/68f83056376c4b9d930329acfe12a246 : FAILURE in 31m 13s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/93ee7d2ebbc242c2a84b6b72c2049f14 : FAILURE in 25m 44s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/2f1f8953416d4a95816b3378df1a5ae7 : SUCCESS in 32m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7bb2d0953f95cdabf2dd700c8ac6a723b54fb58e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 14:16:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b201b879890d4605b682ae8241045be2 : FAILURE in 24m 57s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/656a7a836b2143e28ca0c87d82c39b55 : FAILURE in 26m 28s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/ec3d30da43924d8b8c34159396f0dd09 : FAILURE in 39m 51s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/82fcc3d32c60486caeeb93fe0dd8b921 : SUCCESS in 1h 09m 45s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/dac5acfabac743098ebd98bb9c12fd9d : SUCCESS in 1h 18m 29s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/12121c20cf6240f3adba69c3cb38472e : SUCCESS in 23m 34s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8d3fd0cfba2644e29524f33601fe68cc : SUCCESS in 17m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e84011bce510415d9460f932640ee150 : FAILURE in 42m 07s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/24c3f0c6784a43d694a5962ff46068ef : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d7bdb48a71e403da8537ea14568e482 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc2d8be3e4f546a8bdb4a401d5121295 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/795fc1b0ba9842f49a9b71c1d620fe96 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89fb20a0c7d44067983d3be252dff136 : SUCCESS in 5m 11s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a86317b8eb2e4902b0236c7b3df83fb8 : SUCCESS in 17m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6edf7e0ef2644d19a9a1312cc44a682c : SUCCESS in 19m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/af69d410a18b43e2a94a9ff4421f4145 : FAILURE in 34m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/09dbfb3ac3b94698b84207740dbd553a : SUCCESS in 54m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b67f0e8bcec24accbc9d97364c64b500 : SUCCESS in 1h 03m 47s","accounts_in_message":[],"_revision_number":1},{"id":"fe1aafc9fc954e231c0ea012cebf63970816570d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-11-04 16:09:10.000000000","message":"Abandoned\n\nanother approach","accounts_in_message":[],"_revision_number":1},{"id":"ab34ca141d5f37ea7e86410dcdd7a695259d2cb0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-08 19:48:20.000000000","message":"Patch Set 1:\n\n\u003e Abandoned\n \u003e \n \u003e another approach\n\ncare to link? /me missing context","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8b3ba6c43b91425fa577e84ed40a8e0747df44e9","revisions":{"8b3ba6c43b91425fa577e84ed40a8e0747df44e9":{"kind":"REWORK","_number":1,"created":"2020-11-04 12:48:26.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/03/761403/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/03/761403/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/761403/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/761403/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/761403/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/03/761403/1"}}},"commit":{"parents":[{"commit":"f47f7b4b6cf5ec31b50aa9a9bd6f3815327ed131","subject":"Merge \"mark trove images as buildable for Debian/x86-64\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f47f7b4b6cf5ec31b50aa9a9bd6f3815327ed131"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-11-04 12:48:07.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-11-04 12:48:23.000000000","tz":60},"subject":"toolbox: Add sudoers for kolla_toolbox become","message":"toolbox: Add sudoers for kolla_toolbox become\n\nChange-Id: Ia98af905953d9a00be27e738a21b0b08bd7fa9ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b3ba6c43b91425fa577e84ed40a8e0747df44e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b3ba6c43b91425fa577e84ed40a8e0747df44e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
