)]}'
{"id":"openstack%2Fkolla~761442","triplet_id":"openstack%2Fkolla~master~Idf980eedcd318b88101b3a82442b9b99838fd84c","project":"openstack/kolla","branch":"master","topic":"curator","hashtags":[],"change_id":"Idf980eedcd318b88101b3a82442b9b99838fd84c","subject":"curator: Add wheel to elasticsearch-curator","status":"MERGED","created":"2020-11-04 16:25:58.000000000","updated":"2020-11-05 09:07:24.000000000","submitted":"2020-11-04 23:39:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"761442-1604533148783-692fdd93","meta_rev_id":"3616b27da8e79619f883b57c5367abd9a3e86e9f","_number":761442,"virtual_id_number":761442,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-04 23:39:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2020-11-04 16:27:42.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-11-04 19:36:54.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-11-04 19:37:17.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-04 16:27:42.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-11-04 19:37:17.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-11-04 23:39:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4b6befd9abb27d73d85cf59c632e92e4cd838aea","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-11-04 16:25:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a76c9bbbc50a9de67775d741706e41eeaa4b4537","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-11-04 16:27:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f84be997f6310cf8474efd20c8638901f1ffad47","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 17:17:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/53efaeeff4c1458e827ce9139f272dba : SUCCESS in 45m 31s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7d362f6d212d495081076a9e216292b3 : FAILURE in 39m 01s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0919bde476e64aad9340a6e9e502da07 : FAILURE in 27m 48s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d2d78cc7536849b89f16a63e0a4d37b3 : SUCCESS in 37m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"46b0f3d1bd51fb201f452c3291eebc7bffa08c9a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-04 19:36:54.000000000","message":"Patch Set 1: Code-Review+2\n\nneed to fix k-a ci first (tls)","accounts_in_message":[],"_revision_number":1},{"id":"d57a4483b89ed9f37c7377cd3606f9c2018aca58","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-04 19:37:17.000000000","message":"Patch Set 1: Workflow+1\n\nok, it merged: https://review.opendev.org/761282","accounts_in_message":[],"_revision_number":1},{"id":"d8ca3dfbbb0e89dc2ed960af3bc1681e7bbbf9b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 19:38:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/061c188a82514c3e8ea26f8f6f81e9f9 : TIMED_OUT in 3h 04m 46s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b3549c94c3164de3ac46fb9585b1712d : SUCCESS in 26m 35s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/b09c96b1b6ad454e9ced04c28dd4a160 : FAILURE in 49m 00s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/7c44650c49b544eeb7c75210041a96eb : SUCCESS in 1h 07m 55s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/0de58ede19b14bc0b06797c39668dda8 : SUCCESS in 1h 09m 08s\n- kolla-ansible-centos8-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/1a0c0a98ade34b49bce1d96ff48d26b9 : SUCCESS in 32m 38s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d346ec820b9840cc894cd9cd921953b1 : SUCCESS in 20m 00s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/96a09776cea3434b8508ccee0615e161 : SUCCESS in 27m 48s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/56cd988995c14f55b4292a48b50ba756 : FAILURE in 46m 26s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9643cf08e7374847add8424f59e36c5c : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13cc9721edf047f4a36f0f516e750944 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a29070fdaff744da97d4ceb2aeb0bfb3 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ffaf7726e0a1454aa1366d96cebdf4ae : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c0c1c41b7104d57917a314a84f62d2c : SUCCESS in 5m 21s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/256e2df4ab2f48c98ab20aa75cc06f56 : SUCCESS in 28m 44s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8b5575b7a04544c29177e86cfc942adc : SUCCESS in 20m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b06ade0e76144f70b58a17d14f2b1ea6 : FAILURE in 34m 56s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cf5f5a40d3934899a14f13284ff36f06 : SUCCESS in 52m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7bdc0033b26a4005a716c5ed02ff6bf4 : SUCCESS in 59m 39s\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/e017377ab3d24cf1aeb5327238831361 : SUCCESS in 26m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8a15e93dedd3ca18c58ef6df46d2db743cafac2f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-04 20:51:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"55bd610995447d6002f120aecd55e40d1498c60d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 21:26:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/fbce1394e0f848f593eb4194ff61d40b : SUCCESS in 34m 41s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/50de951cbe444c9cb1b323b2f84b275c : FAILURE in 26m 02s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e3dc5742654645e7b4c4156d1815faa8 : FAILURE in 21m 13s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/71a547b184de4b8b9173a47e2870d8a1 : SUCCESS in 27m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"683bd913f2b484e54d12654ffbb5f8131ac27bad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 22:16:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/2c57a0811c244aebb743f91bacf65db1 : SUCCESS in 20m 19s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6ff5399d2d4649f7ab6b78f6a42a1cc8 : SUCCESS in 29m 02s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/28889e160d12403a866312587d711bd0 : SUCCESS in 1h 03m 21s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/7bd09908703c41ca8023c313144219d9 : SUCCESS in 1h 07m 16s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/540bea3ff01e4b1693621bd77c48b32b : SUCCESS in 1h 11m 22s\n- kolla-ansible-centos8-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/766a4b646d2442719753021fe6551166 : SUCCESS in 29m 44s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/beb64fe49b904049881e015f3e44ce9c : SUCCESS in 20m 30s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d32aec5f09c74a82a8fcbed19803a01c : SUCCESS in 16m 42s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/38cf935bb3444e59a364a2d768f98919 : SUCCESS in 55m 03s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8dfdce554b8742bc97cd1071b52dc369 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d32cc25f827740f5ae370dd3bc59c838 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66249755f0834252b72c4c3b821cc736 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41bfd9ecdfcb421c910b6365a32ebd04 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c042bd1c2bc44968d6734e4452d2f36 : SUCCESS in 5m 33s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5c66bb1b232b48ee8cf544c09178935d : SUCCESS in 1h 13m 26s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1e6ad75152ce401085895b61295b26e2 : SUCCESS in 22m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/028bd4066f134b3dafda8edc42598832 : SUCCESS in 50m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d43b559fb4b540009a1cc69ca87c794e : SUCCESS in 56m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d3e392e222c94ef9a9eea568a39eb41d : SUCCESS in 1h 01m 03s\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/77a269ba7dc64f14bb5fdcc6d988d580 : SUCCESS in 26m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7bec71fd6d9cc85ab97d2e8cda40aab61cdc7d7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 22:17:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"845cc4d433f6e709ea2c23369ecea7c2a3c38c07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 23:39:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/0d73a4a00cba468fb7d829530fb5e1a0 : SUCCESS in 25m 36s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1b8b26908a544a55b39c39450a6e5423 : SUCCESS in 1h 01m 46s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/77bd0d85e19a4a8b83608c814148cb16 : SUCCESS in 1h 21m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d4c57dd888b4cf5a13d323e8e73dac4 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/398ec881d7af44ec97d011f851d45605 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe064ee2ac6a4f75ae36013ba802b817 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc8cc014a0ee489b90dbec5ce15c6b9b : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/319acb211e334446a1f300981ab91f2f : SUCCESS in 5m 07s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3de26cae0bba489fb066c584aaa69bb0 : SUCCESS in 24m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/15af3284582e4507a1d08162721d37e8 : SUCCESS in 51m 55s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c343f8e0a5534bc6a9382b5ba4738373 : SUCCESS in 1h 00m 48s","accounts_in_message":[],"_revision_number":1},{"id":"3974a9da8bc3011578fb47a037620ab4184097d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 23:39:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7fe66435c30b30d34cb090bcd56dc0beb081c853","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 23:40:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1f0161079ed4bc28e65f3086b00edd6 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1},{"id":"f266e79acd8041dbef1bbc4df7d7b21480391d2c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-05 08:23:07.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 9f65b87f9e9dabc69f786d8c74646f51f20536ca","accounts_in_message":[],"_revision_number":1},{"id":"4229ca595f0c43d0f0526a481eed4c2f23d23536","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-05 08:23:54.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/victoria as commit 477e233d4a5ce2427d9e6b6da2675a9adde570e1","accounts_in_message":[],"_revision_number":1},{"id":"bc3f94b979c57b76e187e98c611115f0b3ca4932","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-05 09:06:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"68c1d9f64dca47f168a0aa47b520d4d4dfcb57e1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-11-05 09:07:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"34460d636e9e62f30a0d17eda2d743658ccb0b33","revisions":{"34460d636e9e62f30a0d17eda2d743658ccb0b33":{"kind":"REWORK","_number":1,"created":"2020-11-04 16:25:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/42/761442/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/761442/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/761442/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/761442/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/761442/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/761442/1"}}},"commit":{"parents":[{"commit":"f47f7b4b6cf5ec31b50aa9a9bd6f3815327ed131","subject":"Merge \"mark trove images as buildable for Debian/x86-64\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f47f7b4b6cf5ec31b50aa9a9bd6f3815327ed131"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-11-04 16:25:39.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-11-04 16:25:56.000000000","tz":0},"subject":"curator: Add wheel to elasticsearch-curator","message":"curator: Add wheel to elasticsearch-curator\n\nChange-Id: Idf980eedcd318b88101b3a82442b9b99838fd84c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/34460d636e9e62f30a0d17eda2d743658ccb0b33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/34460d636e9e62f30a0d17eda2d743658ccb0b33"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
