)]}'
{"id":"openstack%2Fkolla~761510","triplet_id":"openstack%2Fkolla~stable%2Fussuri~Idf980eedcd318b88101b3a82442b9b99838fd84c","project":"openstack/kolla","branch":"stable/ussuri","topic":"curator-stable/ussuri","hashtags":[],"change_id":"Idf980eedcd318b88101b3a82442b9b99838fd84c","subject":"curator: Add wheel to elasticsearch-curator","status":"MERGED","created":"2020-11-05 08:23:07.000000000","updated":"2020-11-06 10:15:04.000000000","submitted":"2020-11-06 10:13:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"761510-1604657622199-d96e2c5f","meta_rev_id":"c434ce153e23facfcab909917b8183599588398f","_number":761510,"virtual_id_number":761510,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-06 10:13:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-11-06 08:51:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-11-06 08:37:17.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2020-11-06 08:51:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-05 08:39:09.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2020-11-06 08:37:17.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-11-06 08:51:52.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-11-06 10:13:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c9ba1a4e4958b532f49b347e9dff42264dde963d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-05 08:23:07.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5a1629506703cdba04dc67882493305453c1f17b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-05 08:23:13.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit e7c774f344ac62cc86e2ce3e654d37bdb4df1752","accounts_in_message":[],"_revision_number":1},{"id":"b6adabeb0e01a43aa0a111f5603e7d31eb0b4a70","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 09:01:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/1500e4cc90034ae5b03c165703c564dd : RETRY_LIMIT in 2m 37s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bd41f1986ba640a1b320b47564dba3d8 : RETRY_LIMIT in 2m 18s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/830905883faf448883907516c4d3ec48 : RETRY_LIMIT in 2m 34s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/24280b2d86124fc6852bacf7b4b4103b : RETRY_LIMIT in 2m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fc93d269d06b34423f516d9cadfb46b808720af5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-05 09:21:00.000000000","message":"Patch Set 1:\n\nwe need workaround backport in here","accounts_in_message":[],"_revision_number":1},{"id":"1a77900fc5906ed494540035363134165431c49d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 10:42:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/a9856afc55c945b980ab0ba3058dbcdb : FAILURE in 30m 08s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/52336ef2ec4b4bf49ae65fe6d6e6ca6f : FAILURE in 22m 47s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/34aeb39c9344400398866c96e9ec86ad : SUCCESS in 1h 07m 53s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/24c3c58836284fc59178a66127466aa7 : TIMED_OUT in 2h 03m 43s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/a29e0933a3474ad99c5598a20aefc39f : SUCCESS in 1h 27m 23s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/2dbf36a1bbeb46fc9b15c7f9fcc72824 : FAILURE in 26m 43s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2036f89ab4234cd6bec8e76cc516eb36 : FAILURE in 36m 07s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a9e4e1b7c07d40edb1067eb940e9280e : FAILURE in 29m 29s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2337f4e8f5b644b3897442186d71d3d8 : SUCCESS in 1h 11m 19s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a49439c4e779442792f450451d4c1e41 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3b31fb097e54324a6c9f113577180aa : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fba3b1afca4344cca0c98bf870428891 : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7a2813c872b1441b9a2da1b189bb7a06 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b16a70182842440a8b394877f6b7fcd6 : SUCCESS in 6m 33s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/66ccd31f685b462e8caaeddf7cee052d : FAILURE in 37m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/dcddd75ff8524e5c8bc4aa2eb538baa8 : FAILURE in 21m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cf29c52547e345b9a494c88622889337 : SUCCESS in 51m 10s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/87554f08579d4cdfb1ea31f0e6e28cac : SUCCESS in 51m 04s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4175457c34934350a584917307847552 : SUCCESS in 1h 08m 20s","accounts_in_message":[],"_revision_number":1},{"id":"5aba89d915a8968b196e50d08bf86c7d2bc8a5c8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-11-05 13:12:25.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c82747319d63699d6fdc0fd91c1335b07491b43f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 13:49:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/ffa0b51ffec043b58d07b7345eef80d0 : RETRY_LIMIT in 2m 21s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c206b0c744134a6e91a41f4f7b8ec98b : RETRY_LIMIT in 2m 09s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c46ed50ede264149b5c9164c4f66c825 : RETRY_LIMIT in 1m 50s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/6632df071f5b47759ed26cf0d58cb8e7 : RETRY_LIMIT in 2m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"007dbf7d72ca79b43f438429020609490c939bb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 16:19:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/bdc9509c82b848df857f8623dcd50668 : TIMED_OUT in 3h 04m 14s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/09f3c1170c2949ec8f53c2445734a9db : TIMED_OUT in 3h 03m 04s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7eea4fe60a3d422a8bf3cfb757be484a : TIMED_OUT in 2h 03m 30s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/9e46d6c2d6c1432e9e678b8cc73834dd : SUCCESS in 59m 31s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/3fa9b5a4bbd3445b9dc6495cf161c985 : SUCCESS in 1h 10m 28s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/c2678431f08c44fa8288a58b1ab7e6e9 : FAILURE in 20m 36s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1b834605289b4d0f882f39e869dcd24b : SUCCESS in 25m 07s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/47e0e4559df94d55ba20def6f5127ff7 : SUCCESS in 18m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/abeba90733864ab49f07758c26d68089 : SUCCESS in 1h 01m 13s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37ff3fee1ae04ad9b9fb7d1d43206171 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a45dc349527f406d83d37979df458611 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b391ee8316c147a2995c0fc9d158bcb4 : SUCCESS in 6m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2d15957e165c43b08e678279c753b22b : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c472688a9a348c1b2fc1f71a511b7e3 : SUCCESS in 6m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/02439f1fe0fd4e30bf2629af8e221bc3 : SUCCESS in 19m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3e618eef43aa4988b7311b790236f8ff : SUCCESS in 22m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/78293ab13cba4b5dadaefe980fa5b12c : SUCCESS in 54m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b64bdd557e494111a498f035b297c151 : SUCCESS in 57m 06s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/14b6f2f955b94283b293f66e8d621b48 : SUCCESS in 55m 50s","accounts_in_message":[],"_revision_number":2},{"id":"5fb83933cc12ff9ebc2997824d65bd8903949902","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-11-05 17:17:36.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c483ffbeeab8ed768930cbd1bf50941d92f9e491","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 18:04:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/46ce2e6d7c7b4057a69c25dff8a7f9bc : RETRY_LIMIT in 2m 14s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f5a871ef77b64f56abc1566de2999161 : RETRY_LIMIT in 2m 31s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/361a3c2fd4174661bf5dff2932156ca5 : RETRY_LIMIT in 2m 36s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/be754cde0a6e4419bc551ea443a271c3 : RETRY_LIMIT in 2m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ff7e12a659389c11e1501a450a391b2fce6cfeb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 18:58:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/87b41e628a0348248d069964ec29e1b7 : SUCCESS in 41m 20s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/031f40743faa4765b33be58f923ca3bd : SUCCESS in 23m 11s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/efcb4b08826c4970ab49fc0b88f868aa : SUCCESS in 1h 27m 34s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/780fbd1fa385407488061f0fa9f96481 : SUCCESS in 1h 21m 16s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/220eacf503934816954de6432502a4b5 : SUCCESS in 1h 35m 06s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/1cab484e35e74852be3775e3dc71697f : FAILURE in 33m 10s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/88a62bd5ef4c4051b2cee0b5e4bb7e63 : SUCCESS in 44m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/168d19f3c06043f2b5b112a1f225a7c5 : SUCCESS in 43m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4522560533ca4f3db3da4b7395c294da : SUCCESS in 57m 43s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cdde12a3b719422faf4e69867a9e3e54 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/892de6c064c041ae8fb4389909eec680 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d160d5bf87364601aaab72b52e85b2c0 : SUCCESS in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fd3f70f0114b4aff864f5cb1d457a109 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30d7452a21d4457281d8b30a5249f725 : SUCCESS in 6m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/df5cfacb8af5470891a75eef7279d151 : SUCCESS in 20m 49s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ca5c7644fd2f4f0b8ef1f44faa1fa3a6 : SUCCESS in 22m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bb597e12fbb844faa184ced814021d71 : SUCCESS in 57m 12s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f4d6574cf8884e58b0bdaa83e58ce5e7 : SUCCESS in 55m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f993b9759b5341dcb4192887706ecd56 : SUCCESS in 1h 03m 38s","accounts_in_message":[],"_revision_number":2},{"id":"4970db714e4e71b12fe4cf5774f069e77b365684","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-06 08:37:17.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"86c882ff942acf1108983b8496b134e33b124911","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-11-06 08:51:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"54dce73d701e1650a8430d22fd0c16954f693d1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 08:52:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4f269f0731c19c9b2fe34cb2c70451ae4c572fa5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 10:13:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b7e8c1c90022463798a99eeb67015a2a : SUCCESS in 33m 56s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/fc547f1071c1466f9f962ed1898b3e7a : SUCCESS in 1h 05m 51s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/aa9c372f9b704a4295b22bb256c66d9a : SUCCESS in 1h 14m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28afb78d07bd4c548d861e998d4e8958 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3be8244cd12b499db39ebbb5146bd862 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16458e94a59a4f3c835263dd0f37b711 : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9f67d88259cc4a65b168ac93e7089752 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/475a9069e72c433a84bcb2230015d884 : SUCCESS in 5m 54s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f5a62e65f3404dc58cd248805eb978f6 : SUCCESS in 20m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6ee71016081c45f680369c1061a61a11 : SUCCESS in 58m 34s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/022730d19bdc40d2aa2848c461c15d8a : SUCCESS in 1h 12m 58s","accounts_in_message":[],"_revision_number":2},{"id":"e96c099abd7a0655517565c48a200d37a09bfb4d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 10:13:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"62fef736796baf661c2cecdbb92554c07f43270e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 10:15:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4be4bd38a66d4b05b47f96f04cdb521c : SUCCESS in 56s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f834290de4046681410e7b781d0c5648e45e9ed1","revisions":{"9f65b87f9e9dabc69f786d8c74646f51f20536ca":{"kind":"REWORK","_number":1,"created":"2020-11-05 08:23:07.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/10/761510/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/761510/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/761510/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/761510/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/761510/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/761510/1"}}},"commit":{"parents":[{"commit":"579a3ebde42e51ffaf656ebc0b8b1bcc25b7c2e6","subject":"Merge \"Improve pip install process for offline deployment.\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/579a3ebde42e51ffaf656ebc0b8b1bcc25b7c2e6"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-11-04 16:25:39.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-11-05 08:23:07.000000000","tz":0},"subject":"curator: Add wheel to elasticsearch-curator","message":"curator: Add wheel to elasticsearch-curator\n\nChange-Id: Idf980eedcd318b88101b3a82442b9b99838fd84c\n(cherry picked from commit 34460d636e9e62f30a0d17eda2d743658ccb0b33)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f65b87f9e9dabc69f786d8c74646f51f20536ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f65b87f9e9dabc69f786d8c74646f51f20536ca"}]},"branch":"refs/heads/stable/ussuri"},"f834290de4046681410e7b781d0c5648e45e9ed1":{"kind":"REWORK","_number":2,"created":"2020-11-05 13:12:25.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/10/761510/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/761510/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/761510/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/761510/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/761510/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/761510/2"}}},"commit":{"parents":[{"commit":"579a3ebde42e51ffaf656ebc0b8b1bcc25b7c2e6","subject":"Merge \"Improve pip install process for offline deployment.\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/579a3ebde42e51ffaf656ebc0b8b1bcc25b7c2e6"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-11-04 16:25:39.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-11-05 13:12:20.000000000","tz":0},"subject":"curator: Add wheel to elasticsearch-curator","message":"curator: Add wheel to elasticsearch-curator\n\nboto part cherry picked from change [1].\n\n[1]: https://review.opendev.org/755339/\n\nChange-Id: Idf980eedcd318b88101b3a82442b9b99838fd84c\n(cherry picked from commit 34460d636e9e62f30a0d17eda2d743658ccb0b33)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f834290de4046681410e7b781d0c5648e45e9ed1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f834290de4046681410e7b781d0c5648e45e9ed1"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
