)]}'
{"id":"openstack%2Fkolla~761511","triplet_id":"openstack%2Fkolla~stable%2Ftrain~Idf980eedcd318b88101b3a82442b9b99838fd84c","project":"openstack/kolla","branch":"stable/train","topic":"curator-stable/ussuri-stable/train","hashtags":[],"change_id":"Idf980eedcd318b88101b3a82442b9b99838fd84c","subject":"curator: Add wheel to elasticsearch-curator","status":"MERGED","created":"2020-11-05 08:23:13.000000000","updated":"2020-11-06 18:02:26.000000000","submitted":"2020-11-06 18:01:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"761511-1604685665888-b3c3e99b","meta_rev_id":"d1bc3817c4835cb504e1eac44d8306f0c7da7489","_number":761511,"virtual_id_number":761511,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-06 18:01:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-11-06 13:27:22.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-11-06 16:38:00.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-11-06 16:38:00.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-05 08:23:13.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-11-05 08:39:06.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2020-11-06 13:27:22.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-11-06 18:01:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2a3e87a678ab3fdea2f830961c82a35570b60cef","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-05 08:23:13.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"c70187a272e45be386cf28f5dafae8aefd03581d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-05 09:21:08.000000000","message":"Patch Set 1:\n\nwe need workaround backport in here","accounts_in_message":[],"_revision_number":1},{"id":"93635a05e4ef79e9d80bdaeafe3165b2dd8eccef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 10:35:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a43232b0ca474a8cb442499131517283 : FAILURE in 44m 43s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/42db3a4af8024e62abc1f701e0486a78 : FAILURE in 51m 07s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/6638f4d7d44844da9f30d138ae4a9b17 : FAILURE in 31m 33s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/42c547faa1a24bd5a839eb38dde1e143 : FAILURE in 40m 38s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/2e3fec3ba11847b3965be548ecc1d6d7 : SUCCESS in 1h 15m 11s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0efb724853ff44cab752039e5019048f : SUCCESS in 1h 00m 02s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/5e8d212e8a3c453b92dccceeb8241c68 : SUCCESS in 1h 09m 53s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/eeeb6e83287944738004c018257ce7fe : SUCCESS in 1h 09m 19s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b20aaa59ab1c48f9b7d1d0e18e60b8fc : SUCCESS in 1h 46m 50s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/27d005b57dba406aab185045620608fb : SUCCESS in 57m 47s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b9c1c4f33958451f8684f53e3189b3cb : FAILURE in 25m 35s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2a86a84ef3e54c1a8cae0fdc6da17783 : FAILURE in 20m 38s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a96335525e394fbc8e49ebed435a5273 : SUCCESS in 46m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bab78316efd24791baece6dfefed7f89 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8387291ebae2435da449e574c770055e : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a4e3e6f07c8b4b18b80150a48b59a664 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/156f5571dd364260bd043ea36718c209 : SUCCESS in 5m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/23ed120bca114cd7b2eb4ceea804761f : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf48686148ae4fc48f984ecb98ab562e : SUCCESS in 5m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6f76cc10f188453eb033a8a06ab77274 : FAILURE in 23m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/50278ecdd33d47f88299a8a55a4a8ecc : FAILURE in 25m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/afb74bf0c4e74f8188b08946eb435b1a : SUCCESS in 47m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5967a18f3285403b8bdbea2d3c0ea518 : SUCCESS in 49m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ecd8cc9229554220b8f0e1508b219a5e : SUCCESS in 57m 35s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"2720c4960e790ae8c57bbad49cb55427d4861c74","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-11-06 10:18:48.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"6e3b64eb83a9dd1cb22d3daad404d29efb1fbb01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 12:20:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/36c0ebd888c24f9cb6636ee34be3574a : FAILURE in 40m 15s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/fe61059c4d7449848fa42f523704cb97 : FAILURE in 49m 21s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e813f53084044ca5a61d4c0fe2f5e4e2 : FAILURE in 26m 54s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b4077b55f841485d844d043cd3ad4116 : FAILURE in 32m 32s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/23efb7ff314a403dbcc87afb17283119 : SUCCESS in 1h 19m 50s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/18ae5d348e93493c88411284b4323105 : SUCCESS in 56m 36s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/88fd76edeadc42938f04b16388420c1c : SUCCESS in 1h 15m 59s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b29c42b5432d4766ad49c0a93b2d8c0b : SUCCESS in 1h 02m 06s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/3c6bab09a56f448497c827e18f3a379f : SUCCESS in 1h 42m 20s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/91fa6267eaf741e59e4750a0eb54cf26 : SUCCESS in 1h 04m 56s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fd2c656a89704971ae022a495dbfafe8 : FAILURE in 22m 13s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6e0a246c20774bc9897c43ac6bfe6172 : FAILURE in 16m 46s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/debbf7c44dac4634ae3bb4c8ab9c122f : SUCCESS in 50m 38s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57f5dfc37f3d4018bc50dd11b4d41d5e : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f837472a95944a6ba6bd4d3f82e0034 : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ff0766bb7b9b4bc586c230c4f22adf2d : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38cc3e7c0a54470a8f166351a37b3ddc : SUCCESS in 3m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/14d7162c67b74da39572ee5e9f853bc8 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21a760c33d944e78b580a49b189a9558 : SUCCESS in 6m 18s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9e2cac90dfa54113bf520841abdfdf59 : FAILURE in 29m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a6e3812521ff4915b3eb2f521fca69ae : FAILURE in 30m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/846b89f74c02415da147bda887aa7958 : SUCCESS in 44m 13s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c7193465deec4342b908007aaa3df433 : SUCCESS in 50m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/37dd842deb6e4de1ac3a85b984cac768 : SUCCESS in 1h 01m 02s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"8ab030785720447f95ab18218ae32529543a1c96","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-11-06 13:20:27.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"569e91db368534e66f196e5f613ee3fea98f75fb","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-11-06 13:27:22.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b6ddd0bbb28b52bfe7ab085a8f08084710982df8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 15:54:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/b5724163e16b422b9cd9d0717ec784c0 : SUCCESS in 41m 24s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/330a04546f8e4f3987a5276a2204084b : SUCCESS in 38m 05s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/67a2512e1275492eb93737f31d35d7cf : SUCCESS in 40m 58s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2ebdf9fe9cce4d0db57fa21523734f85 : SUCCESS in 43m 33s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/6cb4ccf5f6824b8aa5adad57f41ca8cd : SUCCESS in 1h 05m 51s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/470f587629734d5cb9af18f63a41de05 : SUCCESS in 1h 15m 14s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/aeade99dfce14bae98cae1185f9767d9 : SUCCESS in 1h 12m 32s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/df12b95c34224dffbd8d4fd3dac4b913 : SUCCESS in 59m 09s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/fc045dc8af834b7894135ed4e8619c41 : SUCCESS in 1h 16m 40s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/5ac3c6c8ff864089b3bb7d958568bc3a : SUCCESS in 48m 23s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2cfa1f87491b455b9a2f7baa08bd9310 : SUCCESS in 20m 01s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6ad977d1f17f48a19021cfdb362330ef : SUCCESS in 2h 28m 21s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9791cd9b90f54f00bd80a667989332b9 : SUCCESS in 57m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7afb0bb0d25a49ff9ed16018f1609d4e : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faea3f00264143378e3d14df0067d71a : SUCCESS in 6m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/05b24b9f3d0542d2b93da20ada8b30f9 : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1eab15eefddd4bc49d2155cda1b889a2 : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/91e4366a159a40b99f038b5a0bed983e : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54bf87ec0b4e46b8b3b506a2ab710dbe : SUCCESS in 6m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1b7afa864468428f8da5e3fba44f3b71 : SUCCESS in 42m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/15a64282c20444dca7022aa63bb4d3e7 : SUCCESS in 42m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ad1fc85f2d2e441fa7a8e17f4f0b5788 : SUCCESS in 49m 37s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1feda0fd301b44038315d3a13faa7982 : SUCCESS in 56m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5f6dcacb6f5e488dac92771064a581ec : SUCCESS in 1h 35m 53s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":3},{"id":"8085d1bbfd81c183b735805dc1c2e5e3fb5a358d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-06 16:38:00.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"47e18c2cab946d3bde3621b2100a7edad7cccb7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 16:38:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e2c7794824bd3cfe702281cbf3918ed4787b9745","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 18:01:05.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/97b29162474941b99d4e15f0ebd616f3 : SUCCESS in 42m 55s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/887ad44b4f654f6a860b8124b7929e06 : SUCCESS in 30m 22s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a946abb1b9374cd597a639ccc2f43163 : SUCCESS in 1h 13m 42s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/6b25eb038f1f42eb8dbb054204ed742b : SUCCESS in 56m 13s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/149e1682ba6448c08b23a2306cc0e3d5 : SUCCESS in 1h 20m 06s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/fcf213a18f63452da318ab77caa9216d : SUCCESS in 44m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b681a09dc844470829f775b0cb2f76a : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7132556aa4494242ad3f2fa82fbe371f : SUCCESS in 4m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c0ef096f79664f89a317b508cbb12577 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c55745e54e2404ca84159e7b54716ef : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/21f85168c03b4aeeaf4380b0d30e00c5 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37293391dbc04d90bc189173b5869efb : SUCCESS in 5m 46s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2420a2b944ba4f27942630240c15d9da : SUCCESS in 26m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6e31537319fa4be1a47c5a1ff8c2dc58 : SUCCESS in 53m 23s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d4fb72bfe4b14e73bc015e6fe0b857d4 : SUCCESS in 1h 03m 37s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":3},{"id":"54b3b9267e24c6123a7c4910d55d63efa025acbd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 18:01:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"ce5a27b7ee7818cdfc1b6ab5b77f62a484ec860c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 18:02:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb304f09ceaf4d08b5e1856a4b8cb9e5 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b3462dee0583b76ba3ec83a579072cf6e452cac2","revisions":{"e7c774f344ac62cc86e2ce3e654d37bdb4df1752":{"kind":"REWORK","_number":1,"created":"2020-11-05 08:23:13.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/11/761511/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/761511/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/761511/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/761511/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/761511/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/761511/1"}}},"commit":{"parents":[{"commit":"4e9b17bf87e2efc205c31204c69248f3519602d8","subject":"Merge \"Add sysfsutils to nova-compute ubuntu source\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e9b17bf87e2efc205c31204c69248f3519602d8"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-11-04 16:25:39.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-11-05 08:23:13.000000000","tz":0},"subject":"curator: Add wheel to elasticsearch-curator","message":"curator: Add wheel to elasticsearch-curator\n\nChange-Id: Idf980eedcd318b88101b3a82442b9b99838fd84c\n(cherry picked from commit 34460d636e9e62f30a0d17eda2d743658ccb0b33)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e7c774f344ac62cc86e2ce3e654d37bdb4df1752"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e7c774f344ac62cc86e2ce3e654d37bdb4df1752"}]},"branch":"refs/heads/stable/train"},"c3932104bfed0451fd9f2e5e5425f49c20f65d83":{"kind":"REWORK","_number":2,"created":"2020-11-06 10:18:48.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/11/761511/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/761511/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/761511/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/761511/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/761511/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/761511/2"}}},"commit":{"parents":[{"commit":"4e9b17bf87e2efc205c31204c69248f3519602d8","subject":"Merge \"Add sysfsutils to nova-compute ubuntu source\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e9b17bf87e2efc205c31204c69248f3519602d8"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-11-04 16:25:39.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-11-06 10:18:46.000000000","tz":0},"subject":"curator: Add wheel to elasticsearch-curator","message":"curator: Add wheel to elasticsearch-curator\n\nboto part cherry picked from change [1].\n\n[1]: https://review.opendev.org/755339/\n\nChange-Id: Idf980eedcd318b88101b3a82442b9b99838fd84c\n(cherry picked from commit 34460d636e9e62f30a0d17eda2d743658ccb0b33)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3932104bfed0451fd9f2e5e5425f49c20f65d83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3932104bfed0451fd9f2e5e5425f49c20f65d83"}]},"branch":"refs/heads/stable/train"},"b3462dee0583b76ba3ec83a579072cf6e452cac2":{"kind":"REWORK","_number":3,"created":"2020-11-06 13:20:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/11/761511/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/761511/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/761511/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/761511/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/761511/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/761511/3"}}},"commit":{"parents":[{"commit":"4e9b17bf87e2efc205c31204c69248f3519602d8","subject":"Merge \"Add sysfsutils to nova-compute ubuntu source\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e9b17bf87e2efc205c31204c69248f3519602d8"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-11-04 16:25:39.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-11-06 13:19:58.000000000","tz":0},"subject":"curator: Add wheel to elasticsearch-curator","message":"curator: Add wheel to elasticsearch-curator\n\nboto part cherry picked from change [1].\n\nDisabling mongodb build due to missing packages in CentOS repos.\n\n[1]: https://review.opendev.org/755339/\n\nChange-Id: Idf980eedcd318b88101b3a82442b9b99838fd84c\n(cherry picked from commit 34460d636e9e62f30a0d17eda2d743658ccb0b33)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b3462dee0583b76ba3ec83a579072cf6e452cac2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b3462dee0583b76ba3ec83a579072cf6e452cac2"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
