)]}'
{"id":"openstack%2Fkolla~761723","triplet_id":"openstack%2Fkolla~stable%2Fstein~Ia6e81509379a77009d3d416522ec91ad51d6611b","project":"openstack/kolla","branch":"stable/stein","hashtags":[],"change_id":"Ia6e81509379a77009d3d416522ec91ad51d6611b","subject":"Add mongodb to skipped images","status":"MERGED","created":"2020-11-06 10:32:44.000000000","updated":"2020-11-07 21:51:46.000000000","submitted":"2020-11-07 21:50:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"761723-1604785838672-195fdec3","meta_rev_id":"7a55fe3a957c7d31382667e1f5d46796a84a54ae","_number":761723,"virtual_id_number":761723,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-07 21:50:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":2,"date":"2020-11-06 20:02:36.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":1,"date":"2020-11-07 20:54:23.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-07 20:54:23.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-11-07 21:50:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bde6fba95a2c17643a6f2c5312c5c990f556295b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-11-06 10:32:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6172dc8d4a9ac1b295d64c98d949dfd564c5986e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 13:39:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/4e3f666e0d66442caf0f229b29ccd819 : SUCCESS in 40m 56s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/cff9307f07c7480191f87f12c2f89988 : SUCCESS in 47m 43s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/0eda794cfc574434adb2c73f8c115b0f : SUCCESS in 1h 21m 24s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/811d05442f27406a96455aab6d6b286d : SUCCESS in 1h 11m 02s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b4b414a77a3445e89d984165a0281551 : SUCCESS in 1h 20m 53s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/e6fa20fe086444da88279330ebfa418a : SUCCESS in 52m 26s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/b50237deecd6421ca055d194e75903ef : TIMED_OUT in 3h 01m 12s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/05bc6694774d4be1b7569d75417b709d : SUCCESS in 46m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3f215587b2b4d15b0faaef3dacd49d4 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0c76432042f46139be879750068f9e0 : SUCCESS in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9e06986b153846cd9f0ba962d7aca794 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fb060a624e741368f991507d91b67aa : SUCCESS in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/37dc732a7fd846d9a71dabdb32f3f3df : SUCCESS in 7m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/087bcf5776c741af839f039de94e9c89 : SUCCESS in 3m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6065cd9c2c0848cf9165d16c405ea762 : SUCCESS in 36m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ef772fc571884ba8a60ec6326138606d : SUCCESS in 34m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d9fb6452a0064dbf910b315a7120e76d : SUCCESS in 1h 48m 31s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d126c27ef62741e9b8541e4ba37b205c : FAILURE in 33m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e596c01fe55b4bdf86c47b1cb9e0da64 : SUCCESS in 1h 03m 06s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"155f93e86214e1b2fa3883597ce74b56e8bd0f51","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-06 20:02:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4f417cb3ad0c40f6e859635df42337c152b18177","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-07 20:54:23.000000000","message":"Patch Set 1: Workflow+1\n\ngetting CI green","accounts_in_message":[],"_revision_number":1},{"id":"734d32bae90c1044ed79552344b6cfd0fd030a77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-07 20:54:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a49bf1992b7f817b186a7dea5248689aab52cd3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-07 21:50:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/9271f25941174fae9a221357433a4f65 : SUCCESS in 51m 12s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/75e9316b0b6641e191eac440eb8d51ae : SUCCESS in 44m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c4b25d05721948f393c8fb3f0e84648a : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2649d19fcb324359ab54947f4028aeae : SUCCESS in 3m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0c6f955611bd4344972a32688ad57961 : SUCCESS in 2m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68b7a0568ec44d5bbd93afa2b8862252 : SUCCESS in 3m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf2d609061334f65bacd026446a54c2d : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84d3434aeeb24c139f3c94cf47e0a551 : SUCCESS in 2m 48s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a761f36515c7429c8e7036f86c1acaf1 : SUCCESS in 23m 05s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"14cd33113f21dc81e764a7e5e18bd2ac6d5b3d2d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-07 21:50:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7a55fe3a957c7d31382667e1f5d46796a84a54ae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-07 21:51:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09e5bfaa9e5f45ba8af019d9fc71d547 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3a46b48e6e99d09d8229b64c240826338d9ce2bb","revisions":{"3a46b48e6e99d09d8229b64c240826338d9ce2bb":{"kind":"REWORK","_number":1,"created":"2020-11-06 10:32:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/23/761723/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/761723/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/761723/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/761723/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/761723/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/761723/1"}}},"commit":{"parents":[{"commit":"33bd3c5f0fd7fffa5eb6b4761631fb082f249c26","subject":"Merge \"Add sysfsutils to nova-compute ubuntu source\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/33bd3c5f0fd7fffa5eb6b4761631fb082f249c26"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-11-06 10:27:56.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-11-06 10:32:41.000000000","tz":60},"subject":"Add mongodb to skipped images","message":"Add mongodb to skipped images\n\nMongodb builds for CentOS and Oraclelinux are failing with missing mongodb\npackage.\n\n[0]: https://376645110508c67547f1-711d728cec9366cabf90489c3301c285.ssl.cf1.rackcdn.com/periodic/opendev.org/openstack/kolla/stable/stein/kolla-publish-centos-source/02c19f5/kolla/build/000_FAILED_mongodb.log\n[1]: https://be16c502bd559ea2be0a-e7edb05e06a1172316eb4206e173039e.ssl.cf5.rackcdn.com/periodic/opendev.org/openstack/kolla/stable/stein/kolla-publish-centos-binary/7906290/kolla/build/000_FAILED_mongodb.log\n[2]: https://c0896a76e4bfc649b00b-bb1c7e0df0c2b154f674f44976092bb9.ssl.cf5.rackcdn.com/periodic/opendev.org/openstack/kolla/stable/stein/kolla-publish-oraclelinux-source/97b0841/kolla/build/000_FAILED_mongodb.log\n[3]: https://130e95a9c5888f92af5c-5775ad363b5cad1640e005d0e8356875.ssl.cf2.rackcdn.com/periodic/opendev.org/openstack/kolla/stable/stein/kolla-publish-oraclelinux-binary/31aefed/kolla/build/mongodb.log\n\nChange-Id: Ia6e81509379a77009d3d416522ec91ad51d6611b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3a46b48e6e99d09d8229b64c240826338d9ce2bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3a46b48e6e99d09d8229b64c240826338d9ce2bb"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
