)]}'
{"id":"openstack%2Fkolla~761742","triplet_id":"openstack%2Fkolla~master~I408514c446c52f8dfa6ff9db764cc34d7f612795","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I408514c446c52f8dfa6ff9db764cc34d7f612795","subject":"Switch OPENSTACK_RELEASE back to master","status":"MERGED","created":"2020-11-06 13:37:55.000000000","updated":"2020-11-09 11:16:33.000000000","submitted":"2020-11-09 11:14:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"761742-1604920496061-ddfe7261","meta_rev_id":"284d24338a83e315693ab51cf712c50197704a40","_number":761742,"virtual_id_number":761742,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-09 11:14:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-11-09 09:32:09.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-11-07 21:19:53.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-11-09 09:32:09.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-07 21:19:53.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-11-09 09:32:09.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-11-09 11:14:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e674224cd7398658970f4cdcd369cbe97531753a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-11-06 13:37:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"722d66bd97d2c0add47842b2679a1f87059f9a29","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-11-06 13:43:42.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a1a734bf3f640a59122354994020cf688c163455","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 14:32:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/0b19c4876f1f4006808e06295d3e5165 : RETRY_LIMIT in 2m 33s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6d84543f352c47c28f32d372d6dd5727 : RETRY_LIMIT in 2m 03s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8b4bcd777f7941df9ce041cdc95b4805 : RETRY_LIMIT in 2m 12s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e07b2ff2963e4a68bbeb1635f021ca05 : RETRY_LIMIT in 2m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"af938994e94bbc89b39922c03e1b5b1a410a875f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 16:24:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/d8f518287b604e7f9eaf5ae708d973e0 : SUCCESS in 22m 49s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d93ec195ed534bb8933a1c92e48c579f : SUCCESS in 26m 00s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/34ed3ae38f534f78aeb435bfbf6efa27 : SUCCESS in 1h 23m 53s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/16a5ad3c935c41ada8277d9d0de41bbc : SUCCESS in 1h 07m 41s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/14d461275e474bb5a4796744be2bdde8 : SUCCESS in 1h 50m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b949279845904b7aac17ea04ad54ed22 : SUCCESS in 31m 04s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8eeebab4693746aa84dc776334309ee2 : SUCCESS in 16m 18s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/aaaff75e9b5d472ab62cdbc22a60bfd6 : SUCCESS in 55m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c4264b4aca94aa9b1668872416f74b5 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38f27373b58b4a81af6c231c24c59116 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85971e4ca7da45b1ae6f9be034336cd2 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d2a540389ab45819f0288101134875f : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16f49114ed2b4f368b86b807fab2f176 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b813c1e53d5e4a63bbef23c1d3d5cf69 : SUCCESS in 5m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/64f30a8e1f234b5ca2d2095872d314f4 : SUCCESS in 23m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ae1bdb5b232c4771b951ae13b41be73a : SUCCESS in 55m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/780ed6320e0e4970bbec38d3a1c4fedb : SUCCESS in 59m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7f032ecf07e840faa6cc99278637e92c : SUCCESS in 57m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6731f4a9689d4ac990a353d04922cd87 : SUCCESS in 1h 13m 18s","accounts_in_message":[],"_revision_number":2},{"id":"38e19c2c2318683ddbf149fb272f5d70670418b5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-07 21:19:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d0fdbc9e2258947f0fdbe5441612bf31ee2c57e8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-09 09:32:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"019507689cfcc4343508508f1d63c3ec7cfba1f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 09:32:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fbfabc2ea6325246dd19ec4c763765b33b65bfd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 11:14:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d7a15fec912f4d9087e9bf658a08ba8e : SUCCESS in 23m 06s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/80488fddd31e4ac7a297a817e01e1c93 : SUCCESS in 1h 16m 19s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/467e4f8f0f904eb1a71f7611eda5bb11 : SUCCESS in 1h 35m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd453681f2c64946a6ea8ff4f3e21952 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87fb9502c9f14aa6b05f80188c25774e : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef8c859c2f5e4fd692bf0e0370c040b7 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8005b189baa64dc296be4c08c56bac72 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8778b39270d4400a40527e38af0b9c5 : SUCCESS in 5m 33s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0aaa523f2ba140d78e573383312a98f8 : SUCCESS in 21m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7f6a71374fe94756a4b407724f8d697b : SUCCESS in 1h 00m 24s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f75c74d75fc2496e847f5d40db6d67ac : SUCCESS in 1h 28m 13s","accounts_in_message":[],"_revision_number":2},{"id":"a066745019c656f756b1d27dc02399484062b03a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 11:14:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"284d24338a83e315693ab51cf712c50197704a40","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 11:16:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9224be340fbb4456865f825e3a1290c8 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4df47d7c60534bc863c9aa9e3eaeaef6a6c3ee31","revisions":{"6a933f1858ef921feed123218b2aa20232d45564":{"kind":"REWORK","_number":1,"created":"2020-11-06 13:37:55.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/42/761742/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/761742/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/761742/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/761742/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/761742/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/761742/1"}}},"commit":{"parents":[{"commit":"8f73a83d2a8fefd5cb4ee1dc35ccb3d1501aec7d","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f73a83d2a8fefd5cb4ee1dc35ccb3d1501aec7d"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-11-06 13:34:34.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-11-06 13:36:55.000000000","tz":60},"subject":"Switch OPENSTACK_RELEASE to back to master","message":"Switch OPENSTACK_RELEASE to back to master\n\nThis reverts commit d3bada4aada3bf1e5eb3b7f039ca8c6454295668.\n\nChange-Id: I408514c446c52f8dfa6ff9db764cc34d7f612795\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6a933f1858ef921feed123218b2aa20232d45564"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6a933f1858ef921feed123218b2aa20232d45564"}]},"branch":"refs/heads/master"},"4df47d7c60534bc863c9aa9e3eaeaef6a6c3ee31":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-11-06 13:43:42.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/42/761742/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/761742/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/761742/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/761742/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/761742/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/761742/2"}}},"commit":{"parents":[{"commit":"8f73a83d2a8fefd5cb4ee1dc35ccb3d1501aec7d","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f73a83d2a8fefd5cb4ee1dc35ccb3d1501aec7d"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-11-06 13:34:34.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-11-06 13:43:39.000000000","tz":0},"subject":"Switch OPENSTACK_RELEASE back to master","message":"Switch OPENSTACK_RELEASE back to master\n\nThis reverts commit d3bada4aada3bf1e5eb3b7f039ca8c6454295668.\n\nChange-Id: I408514c446c52f8dfa6ff9db764cc34d7f612795\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4df47d7c60534bc863c9aa9e3eaeaef6a6c3ee31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4df47d7c60534bc863c9aa9e3eaeaef6a6c3ee31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
