)]}'
{"id":"openstack%2Fkolla~761743","triplet_id":"openstack%2Fkolla~master~Ic670f323a646c39a5caa56812cbd185020afa99b","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Ic670f323a646c39a5caa56812cbd185020afa99b","subject":"Revert \"Switch to Victoria stable branches\"","status":"MERGED","created":"2020-11-06 13:43:11.000000000","updated":"2020-11-12 19:27:27.000000000","submitted":"2020-11-12 19:25:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"761743-1605209149096-bced994f","meta_rev_id":"fa981762fa3a7db7be69c46fe62ebcf60d0631a6","_number":761743,"virtual_id_number":761743,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-12 19:25:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-11-12 18:07:31.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-11-12 18:07:31.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-09 09:33:44.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-11-12 18:07:31.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-11-12 19:25:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2d86f2aa2c744a132370dab5cfb21a6716ff83dc","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-11-06 13:43:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43ae7bc850408b53a7e3023de9cd4b5146a2fbf7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 14:32:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/321b4567d44e41f9982f4b7f2a1000af : RETRY_LIMIT in 2m 28s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b35d9d3c85c14f329dc4f5af61d22e23 : RETRY_LIMIT in 2m 10s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/68944abd7ee94e59ab2c0106c308c35f : RETRY_LIMIT in 2m 47s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/8c4b7c11322a4b038f57f8f9e4a6676d : RETRY_LIMIT in 2m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6cf2d55356ffc2516eae1a33b089a7c873c86bd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 15:33:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/5af1f961b5954f73978cacb786064ec6 : SUCCESS in 20m 39s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/5a39093c0edb4d128087e70b26d36d3f : SUCCESS in 25m 02s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/fb5c0d76aaf3453096eb38d37d42dc89 : SUCCESS in 1h 05m 16s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/3403875db9084ce4936c7d6e65e00c5b : SUCCESS in 1h 06m 07s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/feb37029e9c94b0eb3b0df98a28cd0c6 : SUCCESS in 1h 45m 42s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b007d003c74a41aeafa3cb6dd2089620 : SUCCESS in 16m 38s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7e3606cce7ec4abe95534a65fc75dffa : SUCCESS in 29m 24s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5ecddd86f64b424ebe4bd9ce3ff4e93f : SUCCESS in 50m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c2fb8d2e67874ff294959ce86b0e7308 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c0ed7a3ef874df3b6dabdf2a7c60420 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9834ad1503841c1a1e04bbca50f7d4a : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94989be4d2754942b4860cbeea139d01 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/277d94c85b4942728de4d10dd15db4f7 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a809df5819124b0abe1c80a6de616213 : SUCCESS in 7m 23s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/eb91b35316a74fd0a2c7ea6f3d9973b0 : SUCCESS in 35m 43s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/af4ffa5e723e48cfa0df11d5e065d925 : SUCCESS in 1h 07m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8bde9709bd9d4801bbc2cb1ce7bd1ecd : SUCCESS in 1h 14m 02s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/04ac8eaceb614d149b4770497df82eb3 : SUCCESS in 1h 00m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e88aaee1767442c9ac0497b128e15e47 : SUCCESS in 58m 55s","accounts_in_message":[],"_revision_number":1},{"id":"d3adb3bab585ecfab7467b451bd21bbda6770163","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-07 21:19:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9454b0a19d2852b8cfecc0db8c095ccfc07360a4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-09 09:33:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b6cb6b9de9a82bde33e1153e7eeb5a4347bfa870","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-09 09:37:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9dc93bba02e8eefb63f28764619483beb9d427ce","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-09 09:38:08.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"fe4dcd79a335a46257a2159785bb904228e16e64","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-09 09:38:38.000000000","message":"Patch Set 2:\n\nfor some reason it was considered non-mergable...","accounts_in_message":[],"_revision_number":2},{"id":"4823b808b76363b8c8b78a3330967b62ed21cd0c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-09 10:06:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4368afcda6db9f51513d5fa4c171ab42ff8f14af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 10:35:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/21d5dbcf0deb4fa2bfd88b13f7ebc55d : FAILURE in 56m 18s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fd965ee7c7ac414aaff1bebccb076327 : FAILURE in 40m 13s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/94e7a31f3a8145f68da2c02fb62a85c4 : FAILURE in 29m 19s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a86ed21b875b487d87673fd0880f442b : FAILURE in 48m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"581a1efcc7099168d487e7456feb0fc9c983e98d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 10:35:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3c98a009a1bd6eda64dc85da26cd108b496bd468","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 11:23:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b345a8284b1f4259979de7f85af4aa2a : SUCCESS in 40m 12s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/53f2d6eed79844d2bf4e9289265d2af0 : SUCCESS in 40m 09s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f296aac202e040638f873fa543b782f2 : SUCCESS in 1h 26m 56s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/123f76ecb2f046b7bb9f48053b117010 : SUCCESS in 1h 21m 30s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/35f563ba49a4424aa36fda3e7bfe2e4a : FAILURE in 1h 35m 06s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4490df2ba1c6483bb83294c348bb841f : SUCCESS in 37m 50s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/07ee7fc1fac940a3a6c6d732f76c1b6d : SUCCESS in 18m 45s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/dda70880a872406792257d9667e53b5f : SUCCESS in 52m 33s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2450dbdb7e09404f98f7e6782d4665c4 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf3319c9ba1f4ebc806e16b98a7d1c99 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ade5ef9cdc745ca8e4cb8dba4d089e4 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26c95f2134fa4ace8d9402e14b704d64 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f02aa815d224f1a80d438bd27ec08b4 : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9d7086b7d8d4baf97b1847d7deee75e : SUCCESS in 5m 20s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2532f724af3d432ba56bfb0102ff46c5 : SUCCESS in 20m 50s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8db0c22b77994c319ea30ed06e857caa : SUCCESS in 18m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8f7ab95e8bb64f4fba23f4c38ac0c793 : SUCCESS in 54m 57s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/767860496afe429b919b3597599fa5c0 : SUCCESS in 53m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b2b6b9e57a7248a49f7074b09601a9dc : FAILURE in 1h 04m 22s","accounts_in_message":[],"_revision_number":2},{"id":"f5275ccfa1623e977fac47ea6eb0af48b09eb1be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 12:01:42.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/0e6e564c3a1a433b9593180bf5b2da52 : SUCCESS in 23m 33s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/acd1fd84a4a84bc3a155065bc51de0b5 : SUCCESS in 1h 09m 58s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c43ac3a8f1de476cb7bb2d44c8406e15 : FAILURE in 1h 11m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47e26e6004e34100914dd9715b0577c6 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b4c74bd4e604dbcb07c60a5d3c83103 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2af97043afc04d79a8c7f4f03bc18184 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/559361a20c794416962e935ce5c5c01f : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/640d514c7689450eb2c80b3bef1643b4 : SUCCESS in 7m 09s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7d8735e07c3445228ca837fc1cf6a0c7 : SUCCESS in 18m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c54ad478770e4b98a6d8414139152031 : SUCCESS in 55m 11s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ece694b022b1450bab8b800a66c74f77 : FAILURE in 1h 00m 03s","accounts_in_message":[],"_revision_number":2},{"id":"1d5d2bfaba272fbbe5254e43ecef837e22e86235","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-09 12:09:04.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8aed76a51bb3c603db88a596fb0492b86e09109e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 12:58:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/6d293997bc4d486bb173aebf27d67cbd : FAILURE in 49m 03s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e1a400ce29b345a7a9bc1c94fb3c1418 : FAILURE in 30m 12s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8a12c97b516b4b5ca004470a6f8eede4 : FAILURE in 24m 51s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b58b3784d5fd4bcfa7972fb647413b82 : FAILURE in 40m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fc64a451e77ceae738b74c956a6b89d6c0f3b193","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-09 14:02:51.000000000","message":"Patch Set 2:\n\n2020-11-09 12:57:42.570 7 CRITICAL nova [req-a1118d7a-a870-4b06-830e-1787591f9377 - - - - -] Unhandled error: nova.exception.TooOldComputeService: Current Nova version does not support computes older than Victoria but the minimum compute service level in your system is 51 and the oldest supported service level is 52.","accounts_in_message":[],"_revision_number":2},{"id":"1099dc4ea22986820c2e1001eaa255076817ff74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 14:05:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/57afcae9872a4c70907305e6b888d4af : SUCCESS in 55m 09s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/db09e79daaaf4e67b94640a58cd58259 : SUCCESS in 32m 15s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/246716aaa9144e2aaf50cfebbd44a2d9 : SUCCESS in 1h 03m 50s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/2d24407d39bb411b9ccdaa6141f87f90 : SUCCESS in 1h 33m 24s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/43373fc0face4ebcab03d9d4ed0619c4 : FAILURE in 1h 48m 07s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a7c285441016428eb43dc2f15fbc659b : SUCCESS in 59m 55s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/310e93065e6249dbafc6b4c36e0f6590 : SUCCESS in 54m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6ee81102146247eaa8eb270ceb1f9d19 : SUCCESS in 1h 27m 42s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a0a23180787c42bc88c1b07639c2cf16 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/846972808fe541a2ab63d2ec10370f04 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/740a59a1f0ac4399967645eb248b04bc : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c2816d61ce94a59ae81ef0936e3a2f0 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/178dfb555e394a65b1afe73566ca620d : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64f3394e51c24a389872e2bf3aa79626 : SUCCESS in 7m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bf1c74ccaaa641de968d156ccb128fff : SUCCESS in 54m 05s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e2f2507242e54664a31b550951745b13 : SUCCESS in 50m 58s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5205721cfb4349c1aa32d5b4f752e305 : SUCCESS in 53m 43s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c293f371173b4bdca5a5073f63703526 : SUCCESS in 52m 31s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/56316295786d47df9eb6f323a85c988a : FAILURE in 55m 53s","accounts_in_message":[],"_revision_number":2},{"id":"7b818b100c5282134d00819b4c92849a317c116b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-10 10:59:19.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1c8b48986926fbc6784ddcfd5b6b323fb3b6ac14","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-10 11:00:12.000000000","message":"Patch Set 2:\n\nWe hit the nova upgrade issue in Kayobe, and found that we were upgrading from Ussuri to Wallaby, which Nova now blocks. Unsure what\u0027s going on here.","accounts_in_message":[],"_revision_number":2},{"id":"e486b059345d2a36dd3bca2673c1e28b54ec23db","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-10 11:01:29.000000000","message":"Patch Set 2:\n\n\u003e We hit the nova upgrade issue in Kayobe, and found that we were\n \u003e upgrading from Ussuri to Wallaby, which Nova now blocks. Unsure\n \u003e what\u0027s going on here.\n\nActually I think the dependency needs to go the other way. Upgrade from victoria, then revert sources to master.","accounts_in_message":[],"_revision_number":2},{"id":"2ab3989ad9ddd45bdf64f5d758c60d25b1281170","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-10 11:02:13.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"98c054d1da087c73420856230e5aeb82ef89113c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 12:09:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/7b8b69bb36e8402090bf257fadb3ed79 : FAILURE in 58m 13s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/172fec17a24c4c448f612eb49a3bd8dd : FAILURE in 39m 14s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/551e9ec38bd748a59a50a3423de6c66c : FAILURE in 28m 40s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/247df4d5f924488f849a7c741c71487a : FAILURE in 49m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"44ea8a5d1577a7dd13adf2591a96a46f7637704d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 13:14:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/5e0f5f60a3d84417a084f53f81b350d4 : SUCCESS in 20m 34s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/5693098a2c8f4d4ca607ac84fbe77915 : SUCCESS in 28m 23s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/814423a0e616457c8d74e2060be59dc7 : FAILURE in 56m 46s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/472e2c0d7c5f4591b63224a723f4d943 : SUCCESS in 1h 04m 49s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/5898538cd048424f83d6e4e1580703c3 : SUCCESS in 1h 12m 53s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8e3aba9d489a40c693effeac6bbedc4a : SUCCESS in 16m 58s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/43b9a6c7f4e146a7a70014d999ff2406 : SUCCESS in 37m 48s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fd9674c4d79b4f78b4a4a119ad36ed38 : TIMED_OUT in 2h 03m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33804c64ec1d46259e730ca3dd844536 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bae3b9fe47d54b8b94eef69fec3f8bb3 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1091050ae64447d59b8b2840edb6102c : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/554f88ea7cb44f34be4029cb0d9548c3 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21eae255ab1648e0b2af4a2e67a81f22 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9dc803b54384be5b8d9748663630eb1 : SUCCESS in 5m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/94586568445048ebb48b003a4b55b359 : SUCCESS in 20m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ba92e3216e814fa2a67aa42f35a8cfe4 : SUCCESS in 20m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/78674e2ff2684d0fb343a1122d3407e8 : SUCCESS in 51m 37s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d73877a01a614bc3904b4cafc2250ff1 : SUCCESS in 57m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0eb85253ec214efeaf1558c804646b38 : SUCCESS in 1h 06m 41s","accounts_in_message":[],"_revision_number":3},{"id":"e0a5a23d7f87afe28a2c0185dbc807dce18f5977","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-10 18:10:05.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"35c4a6d9cfa6f8fa55b6f6632ad5aa18357300aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 19:17:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/c9892d54a9584947ad5827dcb0eb1337 : FAILURE in 1h 05m 19s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e44660c4450c40c7b305c903f78b3c42 : FAILURE in 54m 04s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/853ea79174d04c32a0a2080ffdcce34c : FAILURE in 46m 36s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/94da032007c54f5481d9ffe051a09731 : FAILURE in 1h 04m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cf231728996c22dc955c2ab5e824a923c38f56ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 20:00:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/159d62e67ab648b2b08b5a2c4d305be7 : SUCCESS in 22m 29s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/692d211282da46dcb8d84dfe7a94d953 : SUCCESS in 22m 58s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/cd885858a0ad4afcac375c6dcbe5cc63 : SUCCESS in 1h 10m 14s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/879ca97646cd49a0817da6f24a0d5c44 : SUCCESS in 52m 24s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/941d630c52e84a7ba972e64d18053319 : SUCCESS in 1h 44m 16s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/74d0c8581ab74f3280eec10d18bd9506 : SUCCESS in 44m 54s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/065f2571b7e548c7a2146abd89784bb2 : SUCCESS in 17m 45s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a301989d63094ff8a38300e847954e6b : SUCCESS in 1h 01m 27s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db621ed9dcca4cbdb4b7a4dfd4d1ba4d : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46fd7d47d20645a593b1607bc25e2d3f : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b610df84f1c04e9789376e265c69bb68 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a0c794e683f44c4bceb78a6e266c2d0 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43498d694eb647f2929d138e4de00c17 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7b7abdeb6324d91afbdf0466a7f3a77 : SUCCESS in 5m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1833cd4aa0d14fbd9a7db3d5ab0de045 : SUCCESS in 21m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d2aee5d696ee420e9e9d290056e81ecb : SUCCESS in 20m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5257db83089f463f8a479977879d38b7 : SUCCESS in 1h 04m 28s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0ff872cf20c74f4181964140302a73e4 : SUCCESS in 1h 02m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5c512166d86a49bcb16daa4a022877ad : SUCCESS in 1h 02m 47s","accounts_in_message":[],"_revision_number":3},{"id":"65189872920afba00741f66f2bc8886f0cd3ca23","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-12 18:07:31.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nnot letting it rot","accounts_in_message":[],"_revision_number":3},{"id":"105a2674d88d8749dd0f35669cd944339146ac77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 18:07:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1c49e0f04d9bc6450eb80e813dbbc37a99f07eda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 19:25:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/9055ad315aba42b0b3c65ef1fa87330f : SUCCESS in 25m 37s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/410daf138bf845649b2ea2ca77fdcf73 : SUCCESS in 1h 12m 31s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/5e883d718ba648df89ea0a78d42e63bf : SUCCESS in 1h 09m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60ae0041c7d54dc28c6e4ae828327ce2 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3579cd8f49d245b7b39bb75ef26b95dc : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01ba21db063b49d383191be030196227 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7beee6f2e42047a5895320b2f12a1a3f : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da45b4aad18a4badb1cde1760352be05 : SUCCESS in 6m 05s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e639d2bf5553440f8763ce091a2400f8 : SUCCESS in 20m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/415748d07393455da7afb8cbd8813086 : SUCCESS in 57m 18s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0c55dcd9efb4403087ac55ab667aed28 : SUCCESS in 1h 02m 29s","accounts_in_message":[],"_revision_number":3},{"id":"270241fd856c9be48b61415d2dfc601f6977adc1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 19:25:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"fa981762fa3a7db7be69c46fe62ebcf60d0631a6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 19:27:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d2fdd1087de4c21986be3a330e1808c : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c160f5b1000ee14900d7e46b161f97bf5fd47f92","revisions":{"14f13e146c29692c7a01bc7525680978d078cffb":{"kind":"REWORK","_number":1,"created":"2020-11-06 13:43:11.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/43/761743/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/43/761743/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/761743/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/761743/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/761743/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/43/761743/1"}}},"commit":{"parents":[{"commit":"6a933f1858ef921feed123218b2aa20232d45564","subject":"Switch OPENSTACK_RELEASE to back to master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6a933f1858ef921feed123218b2aa20232d45564"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-11-06 13:40:46.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-11-06 13:41:47.000000000","tz":60},"subject":"Revert \"Switch to Victoria stable branches\"","message":"Revert \"Switch to Victoria stable branches\"\n\nWe use master branches again.\n\nThis reverts most of commit 78351df2ccfa7d4ccc20a41fc51837b124e18358\n(curator changes stay).\n\nChange-Id: Ic670f323a646c39a5caa56812cbd185020afa99b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/14f13e146c29692c7a01bc7525680978d078cffb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/14f13e146c29692c7a01bc7525680978d078cffb"}]},"branch":"refs/heads/master"},"9e4a0fcb671287356fc3cc5e1d3f63d2621132bc":{"kind":"NO_CHANGE","_number":2,"created":"2020-11-09 09:38:08.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/43/761743/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/43/761743/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/761743/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/761743/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/761743/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/43/761743/2"}}},"commit":{"parents":[{"commit":"4df47d7c60534bc863c9aa9e3eaeaef6a6c3ee31","subject":"Switch OPENSTACK_RELEASE back to master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4df47d7c60534bc863c9aa9e3eaeaef6a6c3ee31"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-11-06 13:40:46.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-11-09 09:38:08.000000000","tz":0},"subject":"Revert \"Switch to Victoria stable branches\"","message":"Revert \"Switch to Victoria stable branches\"\n\nWe use master branches again.\n\nThis reverts most of commit 78351df2ccfa7d4ccc20a41fc51837b124e18358\n(curator changes stay).\n\nChange-Id: Ic670f323a646c39a5caa56812cbd185020afa99b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9e4a0fcb671287356fc3cc5e1d3f63d2621132bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9e4a0fcb671287356fc3cc5e1d3f63d2621132bc"}]},"branch":"refs/heads/master"},"c160f5b1000ee14900d7e46b161f97bf5fd47f92":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-11-10 11:02:13.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/43/761743/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/43/761743/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/761743/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/761743/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/761743/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/43/761743/3"}}},"commit":{"parents":[{"commit":"4df47d7c60534bc863c9aa9e3eaeaef6a6c3ee31","subject":"Switch OPENSTACK_RELEASE back to master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4df47d7c60534bc863c9aa9e3eaeaef6a6c3ee31"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-11-06 13:40:46.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-10 11:02:10.000000000","tz":0},"subject":"Revert \"Switch to Victoria stable branches\"","message":"Revert \"Switch to Victoria stable branches\"\n\nWe use master branches again.\n\nThis reverts most of commit 78351df2ccfa7d4ccc20a41fc51837b124e18358\n(curator changes stay).\n\nDepends-On: https://review.opendev.org/761835\n\nChange-Id: Ic670f323a646c39a5caa56812cbd185020afa99b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c160f5b1000ee14900d7e46b161f97bf5fd47f92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c160f5b1000ee14900d7e46b161f97bf5fd47f92"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
