)]}'
{"id":"openstack%2Fkolla~761928","triplet_id":"openstack%2Fkolla~master~I2592a736206eaec811290e9fbdbf2540c0518ffe","project":"openstack/kolla","branch":"master","topic":"bug/1902101","hashtags":[],"change_id":"I2592a736206eaec811290e9fbdbf2540c0518ffe","subject":"CI: revert to public package mirrors after build","status":"MERGED","created":"2020-11-09 14:08:03.000000000","updated":"2021-01-13 08:47:37.000000000","submitted":"2021-01-11 11:58:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"761928-bug/1902101","meta_rev_id":"f6fb64ee7a2552b02770b15fd9fa69f99660b73b","_number":761928,"virtual_id_number":761928,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-11 11:58:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-01-11 10:31:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-01-11 10:41:46.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-01-13 08:47:37.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-01-11 10:41:46.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-01-11 10:31:46.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-19 11:06:32.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-11-20 13:17:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-07 18:53:49.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2021-01-11 10:31:46.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-01-13 08:47:37.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"94dd6baaeeeb4ab17bb6ba05fc83c0ac7d4b1e2e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-09 14:08:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5345270d0e35ea781060eb17d16c0272bc72ff9a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-09 14:09:36.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"abf26ce0c9fb3ed9f75669c451db769632573e3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 15:11:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/8f1580bbe0f84dcea39455932a3a567e : FAILURE in 4m 28s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7705efe0764848ddba82579f6ce7a306 : FAILURE in 5m 18s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/888f739f041a4511b6a0e99eceaa0b7a : FAILURE in 2m 59s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/80cdd19d8c1849788809f5a5f7ddc611 : FAILURE in 3m 30s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5199836e14154a20be2cbd567777ae60 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d6e7e94fa644fabb8cc371b22faceca : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1847fd575cf468aada49f656062434a : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/850b6815754945efa20975281317b9ec : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3658f939aaea4737961ef6e4d0cdd16c : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1865419a126548aaa61230e7ffe606bb : SUCCESS in 5m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7f146656ffe8458eba29cafc18c509f0 : FAILURE in 3m 42s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3c17647e7a36480cbd324a1059be24a4 : FAILURE in 3m 53s","accounts_in_message":[],"_revision_number":2},{"id":"3ff0d2367a4dfc86f235cc104235471e4fc680f6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-10 11:53:51.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"ca7141b0cf907c5ed409ffa89fca57aa48c820aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 13:48:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/ed7699a542b04c4984b418a4ff52d031 : FAILURE in 5m 45s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a4e6ada1c5a4482a846c9527657eae51 : FAILURE in 5m 08s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/654b29ab97e04d57aeddd607cb9e1ce6 : SUCCESS in 17m 24s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/10b07f5b399f424a93259b68579bbce5 : SUCCESS in 16m 48s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f2f8f6a600240d39975869a2ccd6921 : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03851d0dbdf045e6ae3a336c676a3fa9 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43a593da887649779558a40badfd4582 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c00437441b0943cb8b7c1a8cbb3b6d46 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab32024b31b34ac1ab6ced7819091e35 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dcf55a521494ef2b5f9873d200a4370 : SUCCESS in 6m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/99f79cdf32c34076bf1f6d06fc90b9aa : SUCCESS in 1h 14m 17s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/16af890fd29a4f57a072ad64983f7167 : SUCCESS in 19m 27s","accounts_in_message":[],"_revision_number":3},{"id":"8a4539dc78729979a6a5f053ecff5c01b67f6b2f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-16 11:32:41.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"f117f2167db0987d5cc325a07e29bb7485714002","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 12:47:20.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/9584dda571744955b927c14aa055f0a6 : FAILURE in 5m 33s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/0f895d807031472dab475d944e8344d8 : FAILURE in 5m 19s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/866d06a372d54c07b294bc8504bc0478 : SUCCESS in 39m 38s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9aaeb97cfef64e8784a8280aeb00f8bd : SUCCESS in 23m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/feee2fdbefa04083a697f4050f426020 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20228bcd39c84477ab1a2951df353541 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f82f53ec950648f9929c4bcc5b7f2bba : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f7ed76271484ab080c8e96e899782b3 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/987fbe2043514b1b9c78ac38f8e35fdf : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e65c56c946ab4b2b895dbd2147fd220d : SUCCESS in 5m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a455c606906f4862ae9bb1cf65206142 : FAILURE in 4m 12s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2569bec6cd4b427c85e29a8e23bff4cc : SUCCESS in 42m 50s","accounts_in_message":[],"_revision_number":4},{"id":"4d95ec45e8e0f63634f1b49b266b2e34ea0fbb31","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-16 12:52:55.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"0a7824d7ba475a492ecb174f79c9bd117cfe2c72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 13:51:17.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/8f65396c5ccb410eae05e8f4d3ab7c6a : FAILURE in 6m 48s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/5842f79777d24681bef02fff8a0e317f : FAILURE in 5m 51s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/472894f5da6d4746b8a272f6e908db6c : SUCCESS in 37m 42s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1670590e7a7544498d74cb3e26297bcb : SUCCESS in 21m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0926bd1062654e90a28a847d1531440d : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4616319e2bda47b8839777205e071051 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d4d78fd6e8e4a9b8e838a4bae93d9a7 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce3d321c1444493c9ccb6e3182491f60 : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d60b4e7f98a4c7db5cf04029ef5fbc4 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c49ce093585f471fb93d8da41062ad96 : SUCCESS in 5m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d7452c4721884f9d80bf42a32ca97e4a : SUCCESS in 22m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4f2d14c39a6847568423b3e20aa69530 : SUCCESS in 35m 31s","accounts_in_message":[],"_revision_number":5},{"id":"a2acf9345e4818da95471cfec4f3386892ff1eeb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-16 13:55:06.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"6a4940640f32825349db2d554bc11fe264ba8eb0","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-11-16 14:06:04.000000000","message":"Patch Set 6: Code-Review-1\n\nWould be nice to have some description why.\n\nIt should rather be done as some kind of footer action which would replace CI mirrors with official ones in every image right after it was built.","accounts_in_message":[],"_revision_number":6},{"id":"10cbe2590fa39579f6741b4b2d088e9488c635ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 14:43:26.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/584a2226b3c9454c90cd87e5b8308057 : SUCCESS in 46m 00s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/52b2eb2df8d44ce5acbf1dbdfc94ee1c : SUCCESS in 46m 26s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ea879a8059384e239833c7593885c36a : SUCCESS in 36m 42s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d1c30f5e3b4e447796bda7e852247ca8 : SUCCESS in 25m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dbd831faee0542e7b8533dc0301ec7e8 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/083a9a8ab3e145c3bb53d3661e6611ba : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f7950712b42491c8c47a2a74c12cd72 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88b5d86e87074935bbedb0e705368a34 : SUCCESS in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b15240480ee4720a38ce3d27bdf15d4 : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/513f153b33ec470b899d3291ba6313fc : SUCCESS in 6m 26s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/df76974bbfb04e099471ff43f7f93706 : SUCCESS in 28m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/220d013b1bce47d18c38a7d64f40a27d : SUCCESS in 38m 41s","accounts_in_message":[],"_revision_number":6},{"id":"7b16e8d2b36f0a1ab270d2b50079bd7979c0c21f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-16 16:22:29.000000000","message":"Patch Set 6:\n\n\u003e Would be nice to have some description why.\n \u003e \n \u003e It should rather be done as some kind of footer action which would\n \u003e replace CI mirrors with official ones in every image right after it\n \u003e was built.\n\nIt\u0027s still WIP. I wanted to test the changes are valid by doing them at the beginning. Will move to footers.","accounts_in_message":[],"_revision_number":6},{"id":"50f066ed47d1197d460aa687659e23464adece1b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-16 16:24:48.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"bc31528cadf04a4032dcb9cff135a18b8cc6dce5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 17:53:13.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/f4a36daf895842dd9d7b2a8fb5113786 : SUCCESS in 32m 49s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/9d905e08ecb140598b57925969c82ddb : SUCCESS in 48m 24s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0b3ea7e9a2754d7d84abc742708b6b51 : SUCCESS in 37m 46s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6c7cc16a4f7144b4981dc8d8534bd1b0 : SUCCESS in 20m 36s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f141c9a780524cb9a4f243ed07c0a785 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e3e7e2d93414107910341dd0407e0f6 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c66a9bde750440883d2debd57ad6719 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55765e0a04b24f0bbbcba58062db6940 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da62ff879f164caba97f0052b9f48a59 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/580b7068b66a4941ad8cdbca969a81b5 : SUCCESS in 5m 50s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/185b3a8164754d94b5470d45a3abefb2 : SUCCESS in 1h 25m 23s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/019987eefbdc406c8099e20616911cd6 : FAILURE in 4m 05s","accounts_in_message":[],"_revision_number":7},{"id":"5c7baa7e01289b9746174bd474bcab6c9635734e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-17 11:42:56.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"5929960a9b7ad5123d236c26835f8c0fee0f9427","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 12:51:56.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e032be4e72b742f59c1dd3791d50be54 : FAILURE in 23m 57s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/5458f4fa82314c79a4f66eb7f54fae3f : FAILURE in 51m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1db05038609d49e7851ac59f72454ee0 : FAILURE in 40m 02s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/96f5da357a684f89a0f63ad3f4519431 : FAILURE in 19m 48s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c81d5b212174ef58daeb0152790387f : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/864e5b80180946d39d501f3c4e4d47e5 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5a2c48834884277a7cc1a883cf8f064 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3efe70ee0d3f446e8904394caee473e3 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e04e31b40f42406ebdae923cf5502f80 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6deeaef23464a4187ddacaa2bbbd33a : SUCCESS in 6m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a8d0fa3351484bed9d566e528f98255e : FAILURE in 24m 42s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0fc2534be16e464db08ae1ca902c0044 : FAILURE in 37m 27s","accounts_in_message":[],"_revision_number":8},{"id":"5674762d61d8b36afe1fcf9e8249fdb295aea19b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-11-17 13:44:12.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"286731cfa981c2165a98b158e32c57cad2c9c9a4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-17 14:42:21.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"6e634577f43925b56c15c0633f7129346990bcd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 15:35:02.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/06e58891d35f42a1ac12a2d27bff317f : FAILURE in 33m 51s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/19579413d127452fbd7e5fce3ca2d134 : FAILURE in 45m 49s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f21d377d96bc44a0b13ec2d260eae6fd : FAILURE in 40m 05s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/eb9b758b717b41969beb0bf1705e98b6 : FAILURE in 20m 11s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2db2ebcb9fe348988e0808f9810fb432 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fae62309bfea4721b14db8f579ea04cc : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94415f0c4ea54610aa68b4ed912dee80 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f57aaf7b3cb42b2888fa5add323a113 : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/afdbc4d15366468eae01200f9869e9fd : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbea3d5d22a44f8aabc5b5302ede65cc : SUCCESS in 5m 41s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e7ef3714f37041b0b8fb960d873d5626 : FAILURE in 21m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/70ac1074eb204bfdac6923adcc1ebfa9 : FAILURE in 39m 33s","accounts_in_message":[],"_revision_number":9},{"id":"52bd7da1ab336d12a64802ec2b55a9079dc8e1e8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-17 15:45:37.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"9a7c6091aa3f541e4187c75a9019005b4b13b5ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 16:40:41.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/9f1c823793cd424aab5fa8e846719918 : FAILURE in 45m 37s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2e4a841b1bec4f9ea23f00fe9ec67a9d : FAILURE in 48m 59s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/dd3be3e017b947cabb777061d2466552 : FAILURE in 40m 35s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/bfe0685e07ac4f3faed373f6c33c502e : FAILURE in 19m 23s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c00864fc74ec496c8da65e1239d73fa8 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5406e6054584d449c68a51c22731648 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/243f0d1e58eb4e53817920b730bfb79c : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b81eb6cbbfb4be79eb4e1a88ffb572f : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19a36505103c433db5ad8c92c553a364 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b478ecbadecc4b7799a7a8ab8bf020cf : SUCCESS in 6m 15s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1226e833e423487eb816062cde862393 : FAILURE in 22m 00s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/84ac143e91264a898d99ec8401ea2803 : FAILURE in 29m 41s","accounts_in_message":[],"_revision_number":10},{"id":"0629f0bcff666b867ad3d9496e97b3c83be0d6d4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-19 11:03:36.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"53f72760ebc0a625bd873183cc879a5f6ab7ebdf","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-19 11:04:15.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"9d5f7d3aa292ea2ce4a9bf34ed21ccfbe7d61ed2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-11-19 11:06:32.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"3f736368578c47ec94ed3966f7bb4c57d6c3f4db","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-19 11:07:08.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"f17edb57871b1d095b1c4a34ff7d6da1fa87972d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 14:33:46.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/358341c9c121444496cf77be2aa7508d : FAILURE in 24m 33s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2231ce670b614808a5ba5583c34d16bc : FAILURE in 38m 25s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f92060ddfcbb41288c32c4c3a8838b31 : TIMED_OUT in 3h 11m 03s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/41724e65937e47fb82b00dddd3820093 : FAILURE in 23m 44s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/733d7cb4e6c54621ac265481051a5d67 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc141eeb0de74208b4d7bfe9dd02770c : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be65e6ac25c54545930618e6922bbc2f : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aec5fb1cb92e47fd81de116ef2959286 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99af1ab4f54f41df89a1efe415f14f14 : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/520b23c9644b403ab9fc7838dd21a61e : SUCCESS in 5m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e736e5fc9f70407fb57b4e8bd38bb22e : FAILURE in 21m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ac8e56fd52b5468b9b90713a253e1d54 : FAILURE in 43m 17s","accounts_in_message":[],"_revision_number":13},{"id":"71c85984c11705f2028e88ad904b82d27df5be76","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-19 15:37:01.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"661ad1b008b91573d8378bc766c400a7f8f0305a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 16:30:40.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/158bc7d747f149e180ab4259c037d3a7 : FAILURE in 21m 20s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/c91f87bfeb4d45cbaabed3f750bfbd0a : FAILURE in 44m 34s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/919bf02c52bb46499c9e26a1d1fa5f5b : FAILURE in 39m 52s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/fe35bc95067442e69d54e1f7c1026b34 : FAILURE in 21m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b2644cc9db6b4aaaac0ab7ca10420c79 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9be918b717214a4bbe26aed7dbde68d2 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/629561425dee496fbf129697c48e779e : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f77a424ee28498ab187622ffe1d3f6e : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b949bc1c920b4374b604d53ad07091cf : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a5ae6c9a68941b0955b611eefef7cfe : SUCCESS in 5m 43s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/10a6bd10ab834c5e85348c45265b5538 : FAILURE in 20m 12s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7b2da7601cbe4a2590d7b93f08cafd8e : FAILURE in 41m 41s","accounts_in_message":[],"_revision_number":14},{"id":"d57919d612954ea0778e1bc845c5c8b92c71d1d2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-20 12:20:35.000000000","message":"Patch Set 15: Patch Set 14 was rebased","accounts_in_message":[],"_revision_number":15},{"id":"af8d812b8215d30136892ecfb8a03835e0d4cfba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 13:17:35.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e6c56633be404071986df2a904591d49 : SUCCESS in 30m 29s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/23da62c6224f4a639d9d14b1a93b99ae : SUCCESS in 47m 41s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e4785cab537041298521d0343d153384 : FAILURE in 25m 40s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ee6af1a6202b413589c11ed6aca7c0c2 : FAILURE in 18m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a210387076a4dedbd2991ed26e89f8b : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed71d97aa16d4889a6bb474350023d84 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff7cada395e948978865781eeb733ec0 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2caea5335c1b48dab46273f27ff92c14 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1b1e84b179f4a21bdd5113c215b5970 : SUCCESS in 3m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c17180d2524349fc87721d0841d4e061 : SUCCESS in 5m 51s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3933e4940d554114839b7ec454971aa5 : FAILURE in 20m 51s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/eda5dd2e2808421a8df14972d0a3d9b0 : FAILURE in 37m 51s","accounts_in_message":[],"_revision_number":15},{"id":"557126d623a845d2f79d7786f4c4621c758f360d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-23 09:42:36.000000000","message":"Patch Set 16: Published edit on patch set 15.","accounts_in_message":[],"_revision_number":16},{"id":"5307dad0b3f06cd819e2185f3996c0ec12d2ea44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 10:42:52.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/26be508bd6094ffebc1bb570bca43ac0 : SUCCESS in 23m 42s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a57f0e5698374ec7a8b6be17a5b588d5 : SUCCESS in 41m 07s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/413de5c02d3e431cb50310428a3165b1 : SUCCESS in 33m 04s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/db7db00994194cb58a8eaab1d58d5dbe : SUCCESS in 19m 19s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4c0629796b446c78dc114a9fdbad2f8 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f4e5ad65bc54e3a861798636ada86b1 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13305b35ee9e4dbeb4e0a1f376401719 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73429b28b16f491eab2f2923dc9956bd : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b4638c2c5104714ad2b1d2e896090f7 : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c89068d75bc94467923bae2d48c84e35 : SUCCESS in 5m 21s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a7fcd98dc0784bcb8facca6815a511b0 : SUCCESS in 24m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a889be31c98646b99cc42eb515c812dd : SUCCESS in 36m 04s","accounts_in_message":[],"_revision_number":16},{"id":"1477e20842420a32a66ff15cc342785e7e5222c7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-12-07 11:19:31.000000000","message":"Patch Set 16: Code-Review+2\n\nCan we drop \"WIP: \" from title, re-add CI and merge?","accounts_in_message":[],"_revision_number":16},{"id":"64c386fd93daccb18a8ea0ceddd2dbb44903ebfc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-07 18:53:49.000000000","message":"Patch Set 16:\n\n(1 comment)\n\nWould it make sense if we run a test script against one final image to verify it contains no references to zuul things in /etc ?","accounts_in_message":[],"_revision_number":16},{"id":"8db369239c1b5d7749790e03f50c5ba512b8fba2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-17 11:20:00.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"a5de14406f92d2de76a5b315fe666891d3b65dfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 12:59:27.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/45e615b5946745659e074744b14098a9 : SUCCESS in 25m 05s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3789bab88b884eeaa0d7ef0d23fdc321 : SUCCESS in 36m 50s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/84c225e4e15b403bb97d989498d68440 : SUCCESS in 1h 04m 14s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/895b60c170a74ec99f0bb53a0d1e3d82 : SUCCESS in 1h 03m 26s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/5ba4cdbabff4451297ea0481b99fd466 : SUCCESS in 1h 32m 11s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bbbee61994e643be9a7df8ba82db8c77 : SUCCESS in 42m 52s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/552ab98bc8764529a3dcccfa65c5db9d : FAILURE in 18m 49s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d299c382093b4a179713d31eff966d98 : SUCCESS in 1h 15m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c8cadba5a744a84b5850bff06c91475 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29b1308044844aa2b71a5bff4cf2dc0d : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66e4c721d11045b49a34d8adb3383dcd : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5921b8c1f1b74d529cb00feb9cbfdfa7 : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48ccf68654304448a305d3e6abc43fdb : SUCCESS in 6m 26s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8354bf3d1af945c182966ed5a18aee4c : SUCCESS in 32m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1f5a764075844af2849436357b184152 : SUCCESS in 45m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d769b3aa03904904ac4f2fdb4faa6e7b : SUCCESS in 50m 54s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cbcbe98c1cca493597ab5fa7a393d6e4 : SUCCESS in 57m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/290a94ceea3f4cd382068ce95bae6998 : SUCCESS in 1h 06m 19s","accounts_in_message":[],"_revision_number":17},{"id":"e8c36d5d97aef40e07a687c27878ea0b41d994c5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-17 13:51:15.000000000","message":"Patch Set 17:\n\n(1 comment)\n\nTested that packages are installable at the end of a run in https://review.opendev.org/c/openstack/kolla/+/763351.","accounts_in_message":[],"_revision_number":17},{"id":"c3d765cf0600934a419f6a052accf9bdaa2539ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 15:05:46.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/ddb58c28627e43418fb9bfdcde1b8a86 : FAILURE in 2h 09m 14s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/af9445af94e24894bb53b21b060a0a58 : SUCCESS in 2h 11m 27s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/64e84e7f41ee464ea6c2647936508a64 : SUCCESS in 1h 47m 09s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e910f3cf9d084cda90a023423d18cc3f : SUCCESS in 2h 10m 34s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"a0448578ca5ce03ae7f6e54f59e327c6da4cf735","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-01-11 10:31:46.000000000","message":"Patch Set 17: Code-Review+2 Review-Priority+1\n\nMarking as important, because rally is unusable with Opendev PIP mirrors in non-CI env :)","accounts_in_message":[],"_revision_number":17},{"id":"48e94f669fc50842ca07d65e7abdd071b885cb6b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-01-11 10:41:46.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":17},{"id":"5785e2248c4cf18dd09ba3683c8ddb824e0df219","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 10:41:59.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"30a748685c97ac979d1cc765af6d37a1ec88ce73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 11:58:48.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/028e2a958e1447b7a334df87f95c1b21 : SUCCESS in 40m 29s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/39c14729208249c683308d8da5adc924 : SUCCESS in 1h 15m 07s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/9dae187ad983447bbe9a261c28eb29d3 : SUCCESS in 1h 11m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b49901629d9d42248abb9b0bcf6d04fa : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7058a17c24534bd3ba3d07fe8c04d5a6 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67eed1c9cdf3435b92768c03ecab4dd9 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aed6cdd7ad664e1b8c7faa508398d969 : SUCCESS in 5m 23s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a7f79263e4b84c30b80ebe59535c930d : SUCCESS in 36m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2c506544bbca4e39803002cbb22208ef : SUCCESS in 59m 18s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/afdbe93845064739a217f87fb44e67bb : SUCCESS in 1h 06m 38s","accounts_in_message":[],"_revision_number":17},{"id":"5542b3470902a8ac63a40b2506902ae149a9ea94","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 11:58:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":17},{"id":"e2e662242069134f1745f85c3985373d97b021ab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 12:00:37.000000000","message":"Patch Set 17:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cb75f5b7b5043d7ba5b773ea599be7d : SUCCESS in 49s","accounts_in_message":[],"_revision_number":17},{"id":"f6fb64ee7a2552b02770b15fd9fa69f99660b73b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-13 08:47:37.000000000","message":"Patch Set 17: Code-Review+2","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"0fc464332c789ce05e9d6442726ada84ce7fc65a","revisions":{"6e400d9010abcf33c82213ef031f605128e7dad2":{"kind":"REWORK","_number":1,"created":"2020-11-09 14:08:03.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/28/761928/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/761928/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/761928/1"}}},"commit":{"parents":[{"commit":"e605a8ae91d5defc2a6844567a64e30a0005721b","subject":"Merge \"Switch OPENSTACK_RELEASE back to master\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e605a8ae91d5defc2a6844567a64e30a0005721b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-09 14:07:37.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-09 14:07:37.000000000","tz":0},"subject":"WIP: revert to public package mirrors after build","message":"WIP: revert to public package mirrors after build\n\nChange-Id: I2592a736206eaec811290e9fbdbf2540c0518ffe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6e400d9010abcf33c82213ef031f605128e7dad2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6e400d9010abcf33c82213ef031f605128e7dad2"}]},"branch":"refs/heads/master"},"a578c05b4fb4cbb2d56b2a01c96281ac2000f778":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-11-09 14:09:36.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/28/761928/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/761928/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/761928/2"}}},"commit":{"parents":[{"commit":"e605a8ae91d5defc2a6844567a64e30a0005721b","subject":"Merge \"Switch OPENSTACK_RELEASE back to master\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e605a8ae91d5defc2a6844567a64e30a0005721b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-09 14:07:37.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-09 14:08:58.000000000","tz":0},"subject":"WIP: revert to public package mirrors after build","message":"WIP: revert to public package mirrors after build\n\nCloses-Bug: #1902101\n\nChange-Id: I2592a736206eaec811290e9fbdbf2540c0518ffe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a578c05b4fb4cbb2d56b2a01c96281ac2000f778"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a578c05b4fb4cbb2d56b2a01c96281ac2000f778"}]},"branch":"refs/heads/master"},"0b0eaad56fb3b224066e6b5c0b87a6a8bdad8d43":{"kind":"REWORK","_number":3,"created":"2020-11-10 11:53:51.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/28/761928/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/761928/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/761928/3"}}},"commit":{"parents":[{"commit":"e605a8ae91d5defc2a6844567a64e30a0005721b","subject":"Merge \"Switch OPENSTACK_RELEASE back to master\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e605a8ae91d5defc2a6844567a64e30a0005721b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-09 14:07:37.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-10 11:53:27.000000000","tz":0},"subject":"WIP: revert to public package mirrors after build","message":"WIP: revert to public package mirrors after build\n\nCloses-Bug: #1902101\n\nChange-Id: I2592a736206eaec811290e9fbdbf2540c0518ffe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b0eaad56fb3b224066e6b5c0b87a6a8bdad8d43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b0eaad56fb3b224066e6b5c0b87a6a8bdad8d43"}]},"branch":"refs/heads/master"},"39ba0fceb91212c3be688e3dd355e8a617f9f5f8":{"kind":"REWORK","_number":4,"created":"2020-11-16 11:32:41.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/28/761928/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/761928/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/761928/4"}}},"commit":{"parents":[{"commit":"e605a8ae91d5defc2a6844567a64e30a0005721b","subject":"Merge \"Switch OPENSTACK_RELEASE back to master\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e605a8ae91d5defc2a6844567a64e30a0005721b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-09 14:07:37.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-16 11:32:38.000000000","tz":0},"subject":"WIP: revert to public package mirrors after build","message":"WIP: revert to public package mirrors after build\n\nCloses-Bug: #1902101\n\nChange-Id: I2592a736206eaec811290e9fbdbf2540c0518ffe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/39ba0fceb91212c3be688e3dd355e8a617f9f5f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/39ba0fceb91212c3be688e3dd355e8a617f9f5f8"}]},"branch":"refs/heads/master"},"a67e853339d6a9a1a5562920b8003fd81f721042":{"kind":"REWORK","_number":5,"created":"2020-11-16 12:52:55.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/28/761928/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/761928/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/761928/5"}}},"commit":{"parents":[{"commit":"e605a8ae91d5defc2a6844567a64e30a0005721b","subject":"Merge \"Switch OPENSTACK_RELEASE back to master\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e605a8ae91d5defc2a6844567a64e30a0005721b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-09 14:07:37.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-16 12:52:51.000000000","tz":0},"subject":"WIP: revert to public package mirrors after build","message":"WIP: revert to public package mirrors after build\n\nCloses-Bug: #1902101\n\nChange-Id: I2592a736206eaec811290e9fbdbf2540c0518ffe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a67e853339d6a9a1a5562920b8003fd81f721042"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a67e853339d6a9a1a5562920b8003fd81f721042"}]},"branch":"refs/heads/master"},"d1c390b66073b82f5b78707416e059c2fa613436":{"kind":"REWORK","_number":6,"created":"2020-11-16 13:55:06.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/28/761928/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/761928/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/761928/6"}}},"commit":{"parents":[{"commit":"e605a8ae91d5defc2a6844567a64e30a0005721b","subject":"Merge \"Switch OPENSTACK_RELEASE back to master\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e605a8ae91d5defc2a6844567a64e30a0005721b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-09 14:07:37.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-16 13:55:03.000000000","tz":0},"subject":"WIP: revert to public package mirrors after build","message":"WIP: revert to public package mirrors after build\n\nCloses-Bug: #1902101\n\nChange-Id: I2592a736206eaec811290e9fbdbf2540c0518ffe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d1c390b66073b82f5b78707416e059c2fa613436"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d1c390b66073b82f5b78707416e059c2fa613436"}]},"branch":"refs/heads/master"},"466d55944d28f71bfa989b22896bd8181762ef26":{"kind":"REWORK","_number":7,"created":"2020-11-16 16:24:48.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/28/761928/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/761928/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/761928/7"}}},"commit":{"parents":[{"commit":"e605a8ae91d5defc2a6844567a64e30a0005721b","subject":"Merge \"Switch OPENSTACK_RELEASE back to master\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e605a8ae91d5defc2a6844567a64e30a0005721b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-09 14:07:37.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-16 16:24:45.000000000","tz":0},"subject":"WIP: revert to public package mirrors after build","message":"WIP: revert to public package mirrors after build\n\nCloses-Bug: #1902101\n\nChange-Id: I2592a736206eaec811290e9fbdbf2540c0518ffe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/466d55944d28f71bfa989b22896bd8181762ef26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/466d55944d28f71bfa989b22896bd8181762ef26"}]},"branch":"refs/heads/master"},"d3b975fafc686d163eeadd443c1e099a6b2b9ce3":{"kind":"REWORK","_number":8,"created":"2020-11-17 11:42:56.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/28/761928/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/761928/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/761928/8"}}},"commit":{"parents":[{"commit":"09c890b12463a4b2c8767231189e7016f337fbc3","subject":"CI: add templated Dockerfiles to build logs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/09c890b12463a4b2c8767231189e7016f337fbc3"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-09 14:07:37.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-17 11:42:42.000000000","tz":0},"subject":"WIP: CI: revert to public package mirrors after build","message":"WIP: CI: revert to public package mirrors after build\n\nWhen building images in CI, we use local package mirrors to reduce\nexternal network traffic. These mirrors are not necessarily reliable nor\nare they guaranteed to be accessible outside of the CI environment\n(although many are).\n\nWhile it is an antipattern to install packages in containers at runtime,\nthe Bifrost container does just that. If the mirrors configured in the\nimage are unavailable, Bifrost deployment can fail.\n\nThis change fixes the issue by reverting to public package mirrors as a\nfinal step of the image build process.\n\nCloses-Bug: #1902101\n\nChange-Id: I2592a736206eaec811290e9fbdbf2540c0518ffe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d3b975fafc686d163eeadd443c1e099a6b2b9ce3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d3b975fafc686d163eeadd443c1e099a6b2b9ce3"}]},"branch":"refs/heads/master"},"ffd1a0ab1d23c3e033cab9eac8a099b028106057":{"kind":"REWORK","_number":9,"created":"2020-11-17 14:42:21.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/28/761928/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/761928/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/761928/9"}}},"commit":{"parents":[{"commit":"762d64e3f56701c4740363e765cc7ebee1174dd8","subject":"Remove footer block from intermediate images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/762d64e3f56701c4740363e765cc7ebee1174dd8"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-09 14:07:37.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-17 14:42:07.000000000","tz":0},"subject":"WIP: CI: revert to public package mirrors after build","message":"WIP: CI: revert to public package mirrors after build\n\nWhen building images in CI, we use local package mirrors to reduce\nexternal network traffic. These mirrors are not necessarily reliable nor\nare they guaranteed to be accessible outside of the CI environment\n(although many are).\n\nWhile it is an antipattern to install packages in containers at runtime,\nthe Bifrost container does just that. If the mirrors configured in the\nimage are unavailable, Bifrost deployment can fail.\n\nThis change fixes the issue by reverting to public package mirrors as a\nfinal step of the image build process.\n\nCloses-Bug: #1902101\n\nChange-Id: I2592a736206eaec811290e9fbdbf2540c0518ffe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ffd1a0ab1d23c3e033cab9eac8a099b028106057"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ffd1a0ab1d23c3e033cab9eac8a099b028106057"}]},"branch":"refs/heads/master"},"1c65a20a71b4aedf4de8cdc3572dd6bb952133ca":{"kind":"REWORK","_number":10,"created":"2020-11-17 15:45:37.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/28/761928/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/761928/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/761928/10"}}},"commit":{"parents":[{"commit":"762d64e3f56701c4740363e765cc7ebee1174dd8","subject":"Remove footer block from intermediate images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/762d64e3f56701c4740363e765cc7ebee1174dd8"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-09 14:07:37.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-17 15:45:29.000000000","tz":0},"subject":"WIP: CI: revert to public package mirrors after build","message":"WIP: CI: revert to public package mirrors after build\n\nWhen building images in CI, we use local package mirrors to reduce\nexternal network traffic. These mirrors are not necessarily reliable nor\nare they guaranteed to be accessible outside of the CI environment\n(although many are).\n\nWhile it is an antipattern to install packages in containers at runtime,\nthe Bifrost container does just that. If the mirrors configured in the\nimage are unavailable, Bifrost deployment can fail.\n\nThis change fixes the issue by reverting to public package mirrors as a\nfinal step of the image build process.\n\nCloses-Bug: #1902101\n\nChange-Id: I2592a736206eaec811290e9fbdbf2540c0518ffe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1c65a20a71b4aedf4de8cdc3572dd6bb952133ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1c65a20a71b4aedf4de8cdc3572dd6bb952133ca"}]},"branch":"refs/heads/master"},"ac28ef7e4faafe804bd166dbdd2dd75203154a3f":{"kind":"NO_CHANGE","_number":11,"created":"2020-11-19 11:03:36.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/28/761928/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/761928/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/761928/11"}}},"commit":{"parents":[{"commit":"31cdf010048a2daa6f6675d56c22f6a459945592","subject":"Remove footer block from intermediate images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/31cdf010048a2daa6f6675d56c22f6a459945592"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-09 14:07:37.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-19 11:03:36.000000000","tz":0},"subject":"WIP: CI: revert to public package mirrors after build","message":"WIP: CI: revert to public package mirrors after build\n\nWhen building images in CI, we use local package mirrors to reduce\nexternal network traffic. These mirrors are not necessarily reliable nor\nare they guaranteed to be accessible outside of the CI environment\n(although many are).\n\nWhile it is an antipattern to install packages in containers at runtime,\nthe Bifrost container does just that. If the mirrors configured in the\nimage are unavailable, Bifrost deployment can fail.\n\nThis change fixes the issue by reverting to public package mirrors as a\nfinal step of the image build process.\n\nCloses-Bug: #1902101\n\nChange-Id: I2592a736206eaec811290e9fbdbf2540c0518ffe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac28ef7e4faafe804bd166dbdd2dd75203154a3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac28ef7e4faafe804bd166dbdd2dd75203154a3f"}]},"branch":"refs/heads/master"},"95ba9d6724aa7f0ebbadd3f61cf8264e14db5611":{"kind":"REWORK","_number":12,"created":"2020-11-19 11:04:15.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/28/761928/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/761928/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/761928/12"}}},"commit":{"parents":[{"commit":"31cdf010048a2daa6f6675d56c22f6a459945592","subject":"Remove footer block from intermediate images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/31cdf010048a2daa6f6675d56c22f6a459945592"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-09 14:07:37.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-19 11:04:03.000000000","tz":0},"subject":"WIP: CI: revert to public package mirrors after build","message":"WIP: CI: revert to public package mirrors after build\n\nWhen building images in CI, we use local package mirrors to reduce\nexternal network traffic. These mirrors are not necessarily reliable nor\nare they guaranteed to be accessible outside of the CI environment\n(although many are).\n\nWhile it is an antipattern to install packages in containers at runtime,\nthe Bifrost container does just that. If the mirrors configured in the\nimage are unavailable, Bifrost deployment can fail.\n\nThis change fixes the issue by reverting to public package mirrors as a\nfinal step of the image build process.\n\nCloses-Bug: #1902101\n\nChange-Id: I2592a736206eaec811290e9fbdbf2540c0518ffe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/95ba9d6724aa7f0ebbadd3f61cf8264e14db5611"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/95ba9d6724aa7f0ebbadd3f61cf8264e14db5611"}]},"branch":"refs/heads/master"},"f3828ecf9a98e4ddcf7018e8e3c4d7a6ea9d18eb":{"kind":"REWORK","_number":13,"created":"2020-11-19 11:07:08.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/28/761928/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/761928/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/761928/13"}}},"commit":{"parents":[{"commit":"31cdf010048a2daa6f6675d56c22f6a459945592","subject":"Remove footer block from intermediate images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/31cdf010048a2daa6f6675d56c22f6a459945592"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-09 14:07:37.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-19 11:07:04.000000000","tz":0},"subject":"WIP: CI: revert to public package mirrors after build","message":"WIP: CI: revert to public package mirrors after build\n\nWhen building images in CI, we use local package mirrors to reduce\nexternal network traffic. These mirrors are not necessarily reliable nor\nare they guaranteed to be accessible outside of the CI environment\n(although many are).\n\nWhile it is an antipattern to install packages in containers at runtime,\nthe Bifrost container does just that. If the mirrors configured in the\nimage are unavailable, Bifrost deployment can fail.\n\nThis change fixes the issue by reverting to public package mirrors as a\nfinal step of the image build process.\n\nCloses-Bug: #1902101\n\nChange-Id: I2592a736206eaec811290e9fbdbf2540c0518ffe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3828ecf9a98e4ddcf7018e8e3c4d7a6ea9d18eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3828ecf9a98e4ddcf7018e8e3c4d7a6ea9d18eb"}]},"branch":"refs/heads/master"},"ea942cb38787e351bba5bc5cdf23af0c5e8b1e18":{"kind":"REWORK","_number":14,"created":"2020-11-19 15:37:01.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/28/761928/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/761928/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/761928/14"}}},"commit":{"parents":[{"commit":"31cdf010048a2daa6f6675d56c22f6a459945592","subject":"Remove footer block from intermediate images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/31cdf010048a2daa6f6675d56c22f6a459945592"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-09 14:07:37.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-19 15:34:52.000000000","tz":0},"subject":"WIP: CI: revert to public package mirrors after build","message":"WIP: CI: revert to public package mirrors after build\n\nWhen building images in CI, we use local package mirrors to reduce\nexternal network traffic. These mirrors are not necessarily reliable nor\nare they guaranteed to be accessible outside of the CI environment\n(although many are).\n\nWhile it is an antipattern to install packages in containers at runtime,\nthe Bifrost container does just that. If the mirrors configured in the\nimage are unavailable, Bifrost deployment can fail.\n\nThis change fixes the issue by reverting to public package mirrors as a\nfinal step of the image build process.\n\nCloses-Bug: #1902101\n\nChange-Id: I2592a736206eaec811290e9fbdbf2540c0518ffe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea942cb38787e351bba5bc5cdf23af0c5e8b1e18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea942cb38787e351bba5bc5cdf23af0c5e8b1e18"}]},"branch":"refs/heads/master"},"03cc818053012702c5f0c11e33ebbb2e84b0f81a":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2020-11-20 12:20:35.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/28/761928/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/761928/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/761928/15"}}},"commit":{"parents":[{"commit":"4b792b9a43bee3bfb499cea14924527550f2bce0","subject":"Remove footer block from intermediate images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4b792b9a43bee3bfb499cea14924527550f2bce0"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-09 14:07:37.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-20 12:20:35.000000000","tz":0},"subject":"WIP: CI: revert to public package mirrors after build","message":"WIP: CI: revert to public package mirrors after build\n\nWhen building images in CI, we use local package mirrors to reduce\nexternal network traffic. These mirrors are not necessarily reliable nor\nare they guaranteed to be accessible outside of the CI environment\n(although many are).\n\nWhile it is an antipattern to install packages in containers at runtime,\nthe Bifrost container does just that. If the mirrors configured in the\nimage are unavailable, Bifrost deployment can fail.\n\nThis change fixes the issue by reverting to public package mirrors as a\nfinal step of the image build process.\n\nCloses-Bug: #1902101\n\nChange-Id: I2592a736206eaec811290e9fbdbf2540c0518ffe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/03cc818053012702c5f0c11e33ebbb2e84b0f81a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/03cc818053012702c5f0c11e33ebbb2e84b0f81a"}]},"branch":"refs/heads/master"},"8718fa0bc10b00202a0ffbcc4225af508bb25cc5":{"kind":"REWORK","_number":16,"created":"2020-11-23 09:42:36.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/28/761928/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/761928/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/761928/16"}}},"commit":{"parents":[{"commit":"4b792b9a43bee3bfb499cea14924527550f2bce0","subject":"Remove footer block from intermediate images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4b792b9a43bee3bfb499cea14924527550f2bce0"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-09 14:07:37.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-23 09:42:26.000000000","tz":0},"subject":"WIP: CI: revert to public package mirrors after build","message":"WIP: CI: revert to public package mirrors after build\n\nWhen building images in CI, we use local package mirrors to reduce\nexternal network traffic. These mirrors are not necessarily reliable nor\nare they guaranteed to be accessible outside of the CI environment\n(although many are).\n\nWhile it is an antipattern to install packages in containers at runtime,\nthe Bifrost container does just that. If the mirrors configured in the\nimage are unavailable, Bifrost deployment can fail.\n\nThis change fixes the issue by reverting to public package mirrors as a\nfinal step of the image build process.\n\nCloses-Bug: #1902101\n\nChange-Id: I2592a736206eaec811290e9fbdbf2540c0518ffe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8718fa0bc10b00202a0ffbcc4225af508bb25cc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8718fa0bc10b00202a0ffbcc4225af508bb25cc5"}]},"branch":"refs/heads/master"},"0fc464332c789ce05e9d6442726ada84ce7fc65a":{"kind":"REWORK","_number":17,"created":"2020-12-17 11:20:00.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/28/761928/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/761928/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/761928/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/761928/17"}}},"commit":{"parents":[{"commit":"0dd767c2de1bfed0edc93df3d0d845b07b6ad153","subject":"Merge \"Add daxio for nvdimm scenario\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0dd767c2de1bfed0edc93df3d0d845b07b6ad153"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-09 14:07:37.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-17 11:13:20.000000000","tz":0},"subject":"CI: revert to public package mirrors after build","message":"CI: revert to public package mirrors after build\n\nWhen building images in CI, we use local package mirrors to reduce\nexternal network traffic. These mirrors are not necessarily reliable nor\nare they guaranteed to be accessible outside of the CI environment\n(although many are).\n\nWhile it is an antipattern to install packages in containers at runtime,\nthe Bifrost container does just that. If the mirrors configured in the\nimage are unavailable, Bifrost deployment can fail.\n\nThis change fixes the issue by reverting to public package mirrors as a\nfinal step of the image build process.\n\nCloses-Bug: #1902101\n\nChange-Id: I2592a736206eaec811290e9fbdbf2540c0518ffe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0fc464332c789ce05e9d6442726ada84ce7fc65a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0fc464332c789ce05e9d6442726ada84ce7fc65a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
