)]}'
{"id":"openstack%2Fkolla~762107","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~I84401e5eb3b86182ff35c782d6125d0871b0b2c9","project":"openstack/kolla","branch":"stable/victoria","topic":"use-victoria-rdo","hashtags":[],"change_id":"I84401e5eb3b86182ff35c782d6125d0871b0b2c9","subject":"Use Victoria RDO","status":"MERGED","created":"2020-11-10 10:18:24.000000000","updated":"2020-11-12 14:43:37.000000000","submitted":"2020-11-12 12:51:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"762107-1605185504415-d7fdaadf","meta_rev_id":"6ede47ee292e6314700c4093859a5b6351bbf8f2","_number":762107,"virtual_id_number":762107,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-12 12:51:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-11-12 11:25:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-11-12 09:37:26.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2020-11-12 11:25:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-10 10:18:24.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-11-12 09:37:26.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-11-12 12:51:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bd0c019801c98b66a8a96ce5a91198abf5bc9217","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-11-10 10:18:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a7c9faa0c48762754bb208c15e08c3f9e18719b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-11-10 10:52:12.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f3733bd18699482cbdf699cf63813c5e1959c435","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-11-10 10:53:32.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"fd421c41b5bb94b3fb614a4c4ba1ec8f296918f9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-11-10 10:55:19.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2662e6ee7751bb3fbd60dd25e97effe4d85dc110","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-11-10 11:36:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d5328296cbe0f16044db27022c00a809f7965e5f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 11:43:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/1b37583273da45e19b8f937c24e6bc1d : SUCCESS in 48m 47s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8965d1833b0c4f1b9c2eef499b18a047 : FAILURE in 41m 54s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7ca55163444b4af9aaaef64171ceb26f : FAILURE in 32m 17s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d1a934bfeafd4bf28cce586946a1095e : SUCCESS in 43m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"96244c4f7c080e4c556fd69dcf8353e0335e40f8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-11-10 13:11:43.000000000","message":"Patch Set 2:\n\nwaiting for https://review.rdoproject.org/r/#/c/30827/ - python3-requests-kerberos is missing.","accounts_in_message":[],"_revision_number":2},{"id":"4c5c237cdc064b7cfea2c17370eb367457d3c20d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 13:55:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/372b344bc00e4e86b4a850da74dbfdc3 : FAILURE in 25m 45s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ceee4b1cd2dc48ddaa0ccabb32d74cbb : SUCCESS in 25m 09s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/054934483cb74bdbb833cc6867e702d2 : SUCCESS in 1h 11m 38s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/316a10e255094d9eb86de25b6f19f3ec : FAILURE in 23m 28s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/444d1c938fb94a5482df597a72c2283d : SUCCESS in 1h 07m 50s\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/6723d75295d344f0b4626a2bb22f91b1 : SUCCESS in 28m 42s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/352ea33a916f4d75ba2ec7c9fd8e3c24 : SUCCESS in 1h 33m 51s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/3e91bf8148924acab849137f1e703514 : SUCCESS in 1h 47m 27s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/14f7baeb873843bd8ad5d1b22d71c012 : SUCCESS in 1h 33m 34s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/7fae1a51ea084d61a2780b4a93992cf1 : FAILURE in 1h 04m 18s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/f56e6155e98643a0b6721a6185410174 : SUCCESS in 1h 07m 31s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/56d5b5c10b9f42dbb4d9b5752568a354 : SUCCESS in 1h 24m 51s (non-voting)\n- kolla-ansible-centos8-source-mariadb https://zuul.opendev.org/t/openstack/build/0ebc85691f914ccaa472da5f1275d3e9 : SUCCESS in 33m 04s (non-voting)\n- kolla-ansible-centos8-source-masakari https://zuul.opendev.org/t/openstack/build/0b9fe8ef0ca14352bf0e9ad7f1e78401 : SUCCESS in 53m 56s (non-voting)\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/46b27f1f0d7247abb238c6490e33a6ab : SUCCESS in 1h 19m 55s (non-voting)\n- kolla-ansible-centos8-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/1a6c7ca1e1e54e8e8178028e898721db : SUCCESS in 28m 42s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3ba6549027d041f1af8cc503eaa44c1a : SUCCESS in 17m 15s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0785a28cf63645268974f4b154dea22e : SUCCESS in 16m 03s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/297ebdcdd5f24fd4945dada5d673405e : SUCCESS in 59m 27s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2df0b977999b4b6fbc68e9b75e21527d : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc8ef512b0b64154b5d2fd4ead75ec59 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c77be425234e4c8e9bed92e2e5de3d55 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0677d65d2be48ddb30dfbb9f8d2c1ad : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da9c73ccc9f54aa99e4a5ff4182e7076 : SUCCESS in 5m 33s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1d11038ef1544294b7a6930619112931 : SUCCESS in 22m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a2e0f8f1aedf4a13a8d70c9b342be65d : SUCCESS in 18m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/baa456e03fb845faaf8e25a561f53817 : SUCCESS in 49m 21s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/25d59f3b99bd48918682ced8bb0c788b : SUCCESS in 50m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c201e676496b49d4892a80363703024c : SUCCESS in 1h 08m 31s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/ba53c872985e4939ab2cc5dc931055b5 : SUCCESS in 1h 11m 40s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/53ddebae081e42b6982030443357077d : SUCCESS in 1h 37m 37s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/729d0153e4644a85add51bf579e8b48e : SUCCESS in 1h 22m 17s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/83fd8ca240b84378af9d4a4922eb7ab0 : SUCCESS in 57m 30s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/aa1c7d78505f42829959de0563ea8d55 : SUCCESS in 1h 13m 07s (non-voting)\n- kolla-ansible-ubuntu-source-mariadb https://zuul.opendev.org/t/openstack/build/f49e79f018b24c4c945e41bc1ba73537 : SUCCESS in 28m 21s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/1eec7cb0dd2f47668adb2a18b02179cf : SUCCESS in 44m 45s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/98e2b6dba6c7455db4b7d8996c1a67cd : SUCCESS in 1h 45m 04s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/b197d982d61e4686924c07cf4aa64525 : DISK_FULL in 1h 06m 53s (non-voting)\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/903d9a6f2f694de5a76c5cbb81fa0d4b : SUCCESS in 24m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9034b11e6813fb2acaefee7bb8404d8a74650375","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-11-11 12:12:33.000000000","message":"Patch Set 2:\n\nrdo patch merged\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"203285b599e57df450f44d14a5c9b9850b1384a8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-11-11 12:27:13.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c2e076ba40f0325f1b3e5a47af226701dd32f7b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 12:59:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/27d0ab249cc1434ca32c6a9873e58003 : SUCCESS in 29m 05s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6a13fb203c174a39ad183fdac317d512 : FAILURE in 28m 18s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9ef3be11de5f4ef0bdc4a3b61846625e : FAILURE in 23m 29s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/bc9549c2c9ce49f18a22c9ab50449c6f : SUCCESS in 27m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"91abe56d56439cb7c124aad43d61fd4d4b4b952c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 16:07:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/3d4edc17e61a4491977e383e4c38aa94 : SUCCESS in 19m 51s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/9d2d3d1fcef94990bf392c626c36a42e : SUCCESS in 22m 06s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/38563bd7418f4cc99b3d38110d40c24a : SUCCESS in 1h 15m 55s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/cb511ecfbdfd49908000e2a44b2e9af8 : FAILURE in 23m 53s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/630abee23ee04669b15622178db2cfe9 : SUCCESS in 1h 17m 56s\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/adc082b85ae24d0e85066d9ce2c52d73 : SUCCESS in 30m 48s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/023fa5ece54644ffb241e83a325eb1bc : SUCCESS in 1h 37m 30s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/806d1806719f4c31b2c040172c2e9f5a : FAILURE in 1h 42m 37s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/e4405fdfbf43417bbba02013e7bd8c93 : SUCCESS in 1h 40m 08s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/d62b37add3554f7a9070df99f2ff68ce : FAILURE in 1h 01m 16s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/2bb825e5553047d8bbdf54f57aacbc48 : SUCCESS in 1h 07m 22s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/911e6ce6aac14dbeb49921a1ba7c5bba : SUCCESS in 1h 32m 39s (non-voting)\n- kolla-ansible-centos8-source-mariadb https://zuul.opendev.org/t/openstack/build/1659365c727649518bf3488937c7d1ec : SUCCESS in 31m 50s (non-voting)\n- kolla-ansible-centos8-source-masakari https://zuul.opendev.org/t/openstack/build/76b991a1d8f34835b07eb6c1a71f34f5 : SUCCESS in 57m 04s (non-voting)\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/4c5b9115741b440790866d5343c1091b : SUCCESS in 1h 15m 03s (non-voting)\n- kolla-ansible-centos8-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/82f3bae27866462db3917bba54ac71ea : SUCCESS in 28m 59s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/358baae74610493abee8caef5df191b2 : SUCCESS in 18m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5980712d69314590a4f2b00077cd7863 : SUCCESS in 21m 10s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/60592c91f75445db8ab6fa33181ec2e6 : SUCCESS in 57m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84c3966f0b6947b48286f84774f0301c : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f47c5afb1eb64323b2dc4c213ab71a72 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af79e21490c14d83bd0f72e87545dd11 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d89e68e3cb594329bc7c2d14219ade9e : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/874cfe10dc1a42ddaced866f2b82fc59 : SUCCESS in 4m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8f0736e160fa4ddba5a6beb06c9cf50b : SUCCESS in 17m 54s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/152cdc241adf4e08bcede6d0b5330f45 : SUCCESS in 21m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/41d2a98c35fa44989fe0e0718aa87263 : SUCCESS in 53m 57s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3e591a0fadb04a58908417cf1692e912 : SUCCESS in 50m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e72c2928a77c4b8a9422749af7f8be17 : SUCCESS in 57m 08s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/f7aceb1bb2304d94b8bc3cfcfe7d3c66 : SUCCESS in 1h 23m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/6513be8b70e14bcb853652f0f209f446 : SUCCESS in 1h 41m 55s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/c50866cd0e5449a994aabfc06ffc159c : DISK_FULL in 1h 24m 01s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/36072d4736cf4c1a88aefb91657ad8f0 : SUCCESS in 53m 51s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/77cf602f6d7a4850b3e4968deac99c23 : SUCCESS in 1h 09m 39s (non-voting)\n- kolla-ansible-ubuntu-source-mariadb https://zuul.opendev.org/t/openstack/build/7a268bb1ae26472f86b3927f34e48275 : SUCCESS in 27m 28s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/82c3fbecfc924aa5856e754b58d12203 : SUCCESS in 52m 45s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/06ba7815548346638f5dfb7ecaed33b9 : SUCCESS in 54m 30s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/bb21876c555f488b900ba2ec6dcaa04d : SUCCESS in 1h 07m 23s (non-voting)\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/71597b01301649a0a2569363e36a491c : SUCCESS in 22m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ca8e793df876664d4cbd55932111ea175dd862e9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-11-12 08:20:08.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6dea3898765cbd24e2f21b0beecd93eddb777917","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 08:59:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/5c44007cf5a246c080030573bee1e47a : SUCCESS in 29m 10s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5dbb1803661d4d88968148be97720af9 : FAILURE in 28m 47s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a62ddd33fbd24e6db7efa839bf30cbbf : FAILURE in 8m 47s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/11080f0f34de4bb194aa797b2fe946bd : SUCCESS in 23m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5e263e7d7a87b8df79f093a30f0c14140ca22eae","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-11-12 09:37:26.000000000","message":"Patch Set 2: Code-Review+2\n\ncentos8-binary built.","accounts_in_message":[],"_revision_number":2},{"id":"ed7564c51b8d795e9da07eab555d4bc78deb7a25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 10:47:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/322800d9cf2c486ca2af783dc92400bc : SUCCESS in 21m 22s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8bef7414861942efa399199679d18cdf : SUCCESS in 29m 57s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/85e09049fac94dbbae16dbdb893ad5f3 : SUCCESS in 1h 05m 05s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/d625677cea7e4304b2ace78d44ca6c29 : SUCCESS in 1h 11m 36s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/a41f1bda722f4a049563a0919fb6fa36 : SUCCESS in 1h 09m 59s\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/ed24da36f8754ce19a1fbd8c482e95fa : SUCCESS in 34m 40s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/03faebcb0c784151913e9ee5ac1ebf75 : SUCCESS in 1h 38m 07s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/f551d142b1944803b0891b36bfe89a9e : SUCCESS in 1h 42m 02s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/6b56046147ab448d8399a9dbadcce237 : SUCCESS in 1h 44m 22s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/1303b76b4af845eda0309c6d0d5ca1ed : FAILURE in 1h 07m 31s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/7d754dd06df3491c8d9e5d640e9c0151 : SUCCESS in 1h 03m 28s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/d5ed82265e8b4716997a9e7c4c53e4ae : SUCCESS in 1h 29m 14s (non-voting)\n- kolla-ansible-centos8-source-mariadb https://zuul.opendev.org/t/openstack/build/6544332356e943f3a30b635e99b329f4 : SUCCESS in 29m 50s (non-voting)\n- kolla-ansible-centos8-source-masakari https://zuul.opendev.org/t/openstack/build/8b0242d4506c42ccbc902cae4b392a4c : SUCCESS in 57m 35s (non-voting)\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/811bd5f21d7143688072742eaf0a4880 : SUCCESS in 1h 14m 32s (non-voting)\n- kolla-ansible-centos8-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/60356a67ad244c8b8915f73c995840c2 : SUCCESS in 28m 56s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d57b25929b6846e6808663f4c0b24651 : SUCCESS in 19m 24s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/950d625692b04b5a9eaea942debbd5fa : SUCCESS in 15m 53s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1bef532a1a4c434f8090d721d6835f55 : SUCCESS in 1h 02m 01s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dba38a9bfcd742f488cbb81da99c47a6 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb93f2e054e04298a0ed716a2aa3b18e : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0a80e3da2ea42b39f8037600e58e76d : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/add6fba3cc974cf68133ba6e465d151d : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a4ff28a873c49e38217c3f943a46d87 : SUCCESS in 5m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0dac98d009b848dfa4234952a278b251 : SUCCESS in 18m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/49c2bd635401419093603420f991ea13 : SUCCESS in 20m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b880c3bc897146f2ae0a05bf591bb22c : SUCCESS in 52m 37s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/617ad450faf8462cb29165f2dceed0f0 : SUCCESS in 58m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d9547406dbc649759a10d79cbf16ded5 : SUCCESS in 1h 00m 08s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/9c9e812e81bb476fac546453165dde0b : SUCCESS in 1h 30m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/1c8f722ddfb24a9fbdeee023207b2479 : POST_FAILURE in 1h 14m 38s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/0410ccddc88640e0a9df104a72c8e5a8 : SUCCESS in 1h 26m 46s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/d49a1419cfb74f7180b418f7b531d9a1 : SUCCESS in 51m 19s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/b2509761acf740d49cf5da68948dc2f0 : TIMED_OUT in 2h 04m 25s (non-voting)\n- kolla-ansible-ubuntu-source-mariadb https://zuul.opendev.org/t/openstack/build/5b4ae54e723740b1b5d9dfb7cab29099 : SUCCESS in 24m 46s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/e74b7f32ef3147f2b96fb6a859ce9295 : SUCCESS in 46m 35s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/6cae7a3cbff84da3983128f091a499d2 : SUCCESS in 59m 25s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/b9b1a192cd764b5fa1cf3e3d311dd575 : DISK_FULL in 57m 25s (non-voting)\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/d8a53df8fe6c46cb9a0c4d6fd70203be : SUCCESS in 23m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ac5289885aeedb5e5ce772db3c98947c04a070e4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-11-12 11:25:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d48e5393553d7479ffa8f33ea93f86c00f7477fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 11:25:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"480fbc51d0c07208e60744548268505405b20137","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 12:51:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7fea5ec14edf449d8ee362864921b72a : SUCCESS in 26m 22s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/02ddb9fe42574b6097ab1900009c448d : SUCCESS in 1h 07m 04s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/7c7c555b7e6843808bb51c73f8cf6501 : SUCCESS in 1h 18m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3fd81f5510bf460e9d47f443f544ddce : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/550250a24fb34263bc9c8ba8a2612314 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3520e5096f95482b8ed01f659209295a : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2f2e1f13da949b9ba246dfeec961469 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14a190eb40e14a69993591144126f9c4 : SUCCESS in 5m 02s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/546a0c155f224bc894e6a304c8cc8c14 : SUCCESS in 19m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/603be0e9347f41baa97852b572fc76ff : SUCCESS in 51m 41s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/44bd4cfd74d94fe9a0ce0e7469786a09 : SUCCESS in 1h 12m 30s","accounts_in_message":[],"_revision_number":2},{"id":"a86b4d2154d76b3af52791e4f5e21cdf3e83fb01","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 12:51:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"dc506054734d2052b5337d40a03d1d93bd8368ef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 12:56:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1df0c22cad854a719e794d5b2a4b1b29 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2},{"id":"1cc04a58e026c8b54b5ddebe56cf5c1d81faf352","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-12 13:37:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ee4faf53053bf2e3a148d277128d67fec029e261","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-11-12 13:57:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"217ebf3a47813334639e3123a516304a0ac7a087","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-12 14:43:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fa5b4b3e9c14ecf29838e76c7f7c9c01823573b4","revisions":{"bb33ab64d2c7a043de45cb00b516abbca22d6ae4":{"kind":"REWORK","_number":1,"created":"2020-11-10 10:18:24.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/762107/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/762107/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/762107/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/762107/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/762107/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/762107/1"}}},"commit":{"parents":[{"commit":"dd87bb9f0720190a558c4249a4c0e782a8c2f580","subject":"[CI] Use local RDO mirror","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dd87bb9f0720190a558c4249a4c0e782a8c2f580"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-10-28 20:05:28.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-11-10 10:18:02.000000000","tz":0},"subject":"Use Victoria RDO","message":"Use Victoria RDO\n\nWe are not using usual centos-release-openstack-victoria\npackage now, because it\u0027s not yet available.\nNeeds a follow-up once it\u0027s published.\n\nChange-Id: I84401e5eb3b86182ff35c782d6125d0871b0b2c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb33ab64d2c7a043de45cb00b516abbca22d6ae4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb33ab64d2c7a043de45cb00b516abbca22d6ae4"}]},"branch":"refs/heads/stable/victoria"},"fa5b4b3e9c14ecf29838e76c7f7c9c01823573b4":{"kind":"REWORK","_number":2,"created":"2020-11-10 10:53:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/762107/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/762107/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/762107/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/762107/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/762107/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/762107/2"}}},"commit":{"parents":[{"commit":"dd87bb9f0720190a558c4249a4c0e782a8c2f580","subject":"[CI] Use local RDO mirror","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dd87bb9f0720190a558c4249a4c0e782a8c2f580"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-10-28 20:05:28.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-11-10 10:53:27.000000000","tz":0},"subject":"Use Victoria RDO","message":"Use Victoria RDO\n\nWe are not using usual centos-release-openstack-victoria\npackage now, because it\u0027s not yet available.\nNeeds a follow-up once it\u0027s published.\n\nChange-Id: I84401e5eb3b86182ff35c782d6125d0871b0b2c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa5b4b3e9c14ecf29838e76c7f7c9c01823573b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa5b4b3e9c14ecf29838e76c7f7c9c01823573b4"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
