)]}'
{"id":"openstack%2Fkolla~763027","triplet_id":"openstack%2Fkolla~master~I95e5bb5583be2a76e289749e8adfc303a4e7197a","project":"openstack/kolla","branch":"master","topic":"bug/1902101","hashtags":[],"change_id":"I95e5bb5583be2a76e289749e8adfc303a4e7197a","subject":"Remove footer block from intermediate images","status":"MERGED","created":"2020-11-17 14:42:21.000000000","updated":"2020-12-12 21:22:15.000000000","submitted":"2020-12-12 21:21:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763027-bug/1902101","meta_rev_id":"dfb61053f349ac28befa388ebd33209af7a32b5a","_number":763027,"virtual_id_number":763027,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-12 21:21:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2020-12-01 19:22:49.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-11-23 09:31:10.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2020-12-12 18:54:04.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-19 11:05:35.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-11-20 13:32:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-23 09:31:10.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"213ad639945feabfe8071ab03e788c77f4815038","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-17 14:42:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07a5b7e99bb29de61cd9e44d2bc2049e2465c438","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 15:52:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/e59f29a881744570a3842aed86e966e7 : FAILURE in 1h 02m 59s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/97ed0a5dba284d9e8a86980e611bbd39 : FAILURE in 40m 14s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a02411e4fa614639880060d81af6b7b6 : FAILURE in 24m 11s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/661a05ca10214bbcb080e88e4755bf8a : FAILURE in 49m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1ea357ccc0bdf562fe0a2c28be8d71699fe1214f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 17:29:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/bfa6fbbc3cdf4b0ca5c1642341d26786 : SUCCESS in 22m 24s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e892703547aa42e1b24c3463fc3ddbaa : SUCCESS in 56m 04s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/85db5853abac4ab0a96d59ab7d22f185 : SUCCESS in 1h 21m 44s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/a2926615a09c4a8d86d2df7511c11097 : SUCCESS in 1h 16m 11s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/0376c6ec6e33455ea374eb8beff6510b : FAILURE in 38m 05s\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/ed1d301681b74938a1985433c5267724 : SUCCESS in 1h 34m 02s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/cfe3f4d8b41d4457b5179124dcd7b942 : TIMED_OUT in 2h 37m 33s (non-voting)\n- kolla-ansible-centos8-source-mariadb https://zuul.opendev.org/t/openstack/build/5ecf87ebe042450cb7ad51d148d17312 : SUCCESS in 47m 18s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/de42bed311464af0bc7187b798213120 : FAILURE in 40m 14s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/594c41739eb045e88a2d9a4b3a4bb0ad : SUCCESS in 18m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5314121d338143e2be43403255c32fe4 : FAILURE in 14m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69cf901b731043249dcc99b12c046135 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df8bc7818f2a420d85b1b6880ad4ac54 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4aa143fced1a40c391e8920cf0694202 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce3233de7c7941908f667fe49cb8b19e : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bebf91e5d9de4e46a6a7bea1becaf456 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8b2f462b176418aa4b2ccf26234e59b : SUCCESS in 6m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/abc58c7652984238866c136950b1bd37 : SUCCESS in 23m 00s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/892dfebb58df45c9a2c54042735f7aea : FAILURE in 33m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/14af501d55674b35b10a6f4bc6d53c5b : FAILURE in 31m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a36176375e7c4f8990ef9e19141f82e0 : SUCCESS in 55m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/91617848ece547f985715f6e3ae6b208 : FAILURE in 31m 12s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/3ee47fa0629e4073b96c9d96d4e90390 : FAILURE in 16m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/6342bfe7c1904b71a338dfd594f922b8 : FAILURE in 36m 55s (non-voting)\n- kolla-ansible-ubuntu-source-mariadb https://zuul.opendev.org/t/openstack/build/65df94def666417a8742290fd7c92b0d : SUCCESS in 30m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a494cdc037c704c2e5604e77cf0031e9a5b51a1e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-19 11:03:12.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e20d592dfa8e69ea87adeed92beb6eeb522b3759","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-11-19 11:05:35.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f54c627fd0ba9131ef7c34ae31c8d0e70ab75de2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 12:11:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/9a696f7bca7942bf8323cf29aa28d1c2 : FAILURE in 1h 06m 27s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cd5bcb1f1e9f4955bb7d80d91d4669b7 : FAILURE in 41m 57s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9e0516623698494e976d7e3215f7b2bc : FAILURE in 23m 51s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e01ce54b1275416783754bdc7f09d28d : FAILURE in 41m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8630117daf589ee86e4e87b7a4fe2306e794e45a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 12:54:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b5e93036818d4111b007a779b1d4fd8a : SUCCESS in 23m 39s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7aba348b97894fa6bca48570b3f1199e : SUCCESS in 41m 49s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/06dfc04bde9c4f4c8a76086c636bb149 : SUCCESS in 1h 25m 23s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b15a08db998c49a2994c961524d38c73 : SUCCESS in 55m 05s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/a1ce9aeafea04cf1910f2f86d9cd268f : SUCCESS in 1h 05m 31s\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/c1d3143033da42509990ba9ed6d7535e : SUCCESS in 1h 46m 05s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/1fd4214d736d490a95de15297724b487 : FAILURE in 1h 37m 37s (non-voting)\n- kolla-ansible-centos8-source-mariadb https://zuul.opendev.org/t/openstack/build/d3f92b44c2d44a09be8aea7492790629 : SUCCESS in 31m 56s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b5e76aa51f6c4ffda68ee67c40945a92 : SUCCESS in 1h 12m 46s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d68d567afc0c4e8c877a68ee37c2c2ef : SUCCESS in 18m 43s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b0855cd1a3bb4c55883adac0b1f736e3 : SUCCESS in 54m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7216349d25c4a529a92d7295fa7283a : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/048df1721cfb4088ae1fd5d4b32b48de : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c4bdfc4ee584618b0e89854da36cb3e : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99b1e0a6ae144650804911549503f8f3 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65282468ae114cc9b1eff037c450631c : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/416242e6f3464254a1349e52c756d010 : SUCCESS in 6m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a65246bc788246398ce9f9ee01881156 : SUCCESS in 22m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c9dfdde08b934cc3b2bbfeb7d269985a : SUCCESS in 42m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cc3c9fcb72314e4cb51d741dd49d9dc3 : SUCCESS in 1h 01m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/568af6d13241478b92bf04daf540c042 : SUCCESS in 54m 26s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/28e56c9c848a44888ed4d879fe3b03ea : SUCCESS in 1h 01m 04s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/cde9b8d809314188b87ca76611068085 : SUCCESS in 1h 16m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/aff07160b3234333a1aa25c6e46b8637 : POST_FAILURE in 1h 29m 24s (non-voting)\n- kolla-ansible-ubuntu-source-mariadb https://zuul.opendev.org/t/openstack/build/de88af042dfd4d298fcaf68e6a068cc0 : SUCCESS in 29m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"956b9fb9c39def58c9c9bee225037686411f443f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-20 12:20:28.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"b8eaa6f2c34fd71d4897c9254063b15776a1f76b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 13:32:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/54756032d7ac475d8a4091e381b844f9 : FAILURE in 1h 08m 00s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9ed07db04d5644ad8d177ea8f2a5bf53 : FAILURE in 42m 47s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b7522196f1d247ec9e1f003ac16f51ca : FAILURE in 29m 38s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/aa61d35c0e274e5a93d93fed059d74df : FAILURE in 40m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"708826607302cf8d95fcc57117d1a1338d092f19","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-23 09:31:10.000000000","message":"Patch Set 3: Code-Review+2\n\nI wonder... you changed infoblox in the latest patch - are you using it? How did you discover this issue in the first place?","accounts_in_message":[],"_revision_number":3},{"id":"b0b3843ce750e02ea89098aa77523e9a13edb82b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-23 09:35:21.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3: Code-Review+2\n\u003e \n\u003e I wonder... you changed infoblox in the latest patch - are you using it? How did you discover this issue in the first place?\n\nIt broke https://review.opendev.org/c/openstack/kolla/+/761928/, since the footer was not run as root.","accounts_in_message":[],"_revision_number":3},{"id":"26f75c68b7d6c9115a9f2071e7c8df995f966771","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-23 17:02:48.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3:\n\u003e \n\u003e \u003e Patch Set 3: Code-Review+2\n\u003e \u003e \n\u003e \u003e I wonder... you changed infoblox in the latest patch - are you using it? How did you discover this issue in the first place?\n\u003e \n\u003e It broke https://review.opendev.org/c/openstack/kolla/+/761928/, since the footer was not run as root.\n\nAh, that makes sense! Thank you!","accounts_in_message":[],"_revision_number":3},{"id":"10224b7d0a60be5c32a6af134066df6801ab02ff","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-12-01 19:22:49.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5bfc4243a6bf7009fbf46b9ed8de9c07e1e54a2a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-06 16:25:19.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"69c0d3d249600c80c8420aaf13b131dd1e8e866b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-06 17:48:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/f2e46ac15e6e475cb7da7dfc35c892fc : FAILURE in 5m 33s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/13bbe7fb984d46c7b57b4b31f3b320bf : FAILURE in 44m 06s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9ac092c7ee874ff6a22da1a2885870a6 : FAILURE in 8m 23s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/49f196d1219643b29ca8d4501f8b4d80 : SUCCESS in 1h 03m 20s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/80332dde65494a66beac439cbbd8f7e7 : SUCCESS in 1h 13m 58s\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/f217dea6456446d1afbab868945d9701 : FAILURE in 9m 11s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/9137b055987c401c92338a64252427dc : FAILURE in 1h 08m 02s (non-voting)\n- kolla-ansible-centos8-source-mariadb https://zuul.opendev.org/t/openstack/build/390e8303d17740e6b5ceeb34cc333178 : SUCCESS in 37m 24s (non-voting)\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/69a522bc87644843a904afc2f67ba4b7 : FAILURE in 59m 16s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9bf8ebbba1ee4fef8aafdb8f723996f0 : SUCCESS in 34m 55s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/223305337fa848738799335b12e46119 : SUCCESS in 19m 27s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e04bf1cd8e4943b5ac063587aa10104a : SUCCESS in 1h 07m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5382b642437492bbbb068c7075a2d8a : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b21af358f0494f43a8fb19e1c63ee66c : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96b09f836f194751860bea5a6cd61bdf : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4df79a7cf5e44fec8700eaf649c487ad : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a6b73ee8db34e6eabbf6ac3db6f92e3 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63edfbac816b4947a512a88328243f57 : SUCCESS in 5m 15s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3ace24d9c77646e1a8fad3431f059bf6 : FAILURE in 28m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6b88189bca224cf1aefc2784dea7a4e7 : FAILURE in 32m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/948feb151582409fae51b05ff05d8632 : SUCCESS in 47m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/89987566fa0f4f52b6206d613e57d543 : SUCCESS in 57m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/dfbcba796a1e491b83623a3baf1c0100 : SUCCESS in 59m 12s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/fab14396c9764d18b5dfbb8099615e9c : SUCCESS in 1h 16m 17s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/2b916639f6b14d9bb8ae0aefd4071253 : DISK_FULL in 1h 15m 44s (non-voting)\n- kolla-ansible-ubuntu-source-mariadb https://zuul.opendev.org/t/openstack/build/d4d012d0d9494436b3e68a8eadf0e122 : SUCCESS in 26m 30s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/0c34cedd9fab4e5d9cd98ff858f9c725 : SUCCESS in 56m 16s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/86bb7501faf04486951bd51a8c8753f3 : FAILURE in 40m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1c4de858496f2035e0634bb54c01653d7613a170","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-06 17:52:00.000000000","message":"Patch Set 3:\n\ngrafana....\n\nINFO:kolla.common.utils.monasca-grafana:\u003e publish-please@5.5.2 postinstall /grafana/src/github.com/grafana/grafana/node_modules/publish-please\nINFO:kolla.common.utils.monasca-grafana:\u003e node lib/post-install.js\nINFO:kolla.common.utils.monasca-grafana:!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!\nINFO:kolla.common.utils.monasca-grafana:!! publish-please hooks were successfully setup for the project. !!\nINFO:kolla.common.utils.monasca-grafana:!! Now follow few simple steps to configure your publishing...   !!\nINFO:kolla.common.utils.monasca-grafana:!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!!\nINFO:kolla.common.utils.monasca-grafana:-- Configuring option \"prePublishScript\":\nINFO:kolla.common.utils.monasca-grafana:? Do you want to run any scripts before publishing (e.g. build steps, tests)? (Y\nINFO:kolla.common.utils.monasca-grafana:/n) \nINFO:kolla.common.utils.monasca-grafana:\u001b[1A\nINFO:kolla.common.utils.monasca-grafana:\u001b[4D\u001b[84C\nINFO:kolla.common.utils.monasca-grafana:\u001b[1B\nINFO:kolla.common.utils.monasca-grafana:Removing intermediate container 9b4f1594d011\nERROR:kolla.common.utils.monasca-grafana:Error\u0027d with the following message\nERROR:kolla.common.utils.monasca-grafana:The command \u0027/bin/sh -c gem install fpm     \u0026\u0026 curl -o /tmp/monasca-grafana.tgz ${monasca_grafana_url}     \u0026\u0026 mkdir -p ${monasca_grafana_build_path}     \u0026\u0026 tar --strip 1 -xvf /tmp/monasca-grafana.tgz -C ${monasca_grafana_build_path}     \u0026\u0026 rm -f /tmp/monasca-grafana.tgz     \u0026\u0026 cd ${monasca_grafana_build_path}     \u0026\u0026 go run build.go setup     \u0026\u0026 npm config set prefix /usr/local -g     \u0026\u0026 npm install npm@6.x -g     \u0026\u0026 /usr/local/bin/npm install     \u0026\u0026 go run build.go build package     \u0026\u0026 mv ${monasca_grafana_build_path}/dist /grafana_pkgs     \u0026\u0026 rm -rf /grafana\u0027 returned a non-zero code: 130","accounts_in_message":[],"_revision_number":3},{"id":"73d4b55d8c97fc2855244d888797a475ca5323ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-06 17:53:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/270afbf89ed743ff8eed1193f1b2b2d4 : SUCCESS in 50m 46s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0f6cc62af9fb481281980edb41c08550 : SUCCESS in 41m 23s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e95bbcf486a84dec8234f5ff9ae97b6d : SUCCESS in 1h 21m 33s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e23f13fa8e51457383a5501fa6b757e8 : SUCCESS in 45m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b666c4f0b5b49adcb236b722d620e8f19de84c56","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-11 14:25:05.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"49ea4c77d30aa8ce5f53abd08862d747ca9b3031","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 15:13:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d3d70ba8134d764370141bb13eb9d29a6957ca19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 16:57:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/d52fe802dfb843149b2ce34668176fb4 : SUCCESS in 23m 02s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/96bf429a008b471396ed0430af93afb3 : SUCCESS in 37m 01s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9eed2bb6e91b4878b01baf1af9a96808 : SUCCESS in 1h 04m 28s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/7778aea275244ab0819f991d292c0640 : SUCCESS in 1h 47m 00s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/381f90f56fcc42e1919ff4e3cdff0195 : SUCCESS in 2h 03m 40s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/31fbf267a94f4c1d94a4e4f66a7aa059 : SUCCESS in 1h 21m 49s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/beb53a501c154406b0d127bfdef8d251 : SUCCESS in 1h 11m 00s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3cd547514d9e44489505f69cb369934c : SUCCESS in 1h 46m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60c9dcb4b36d4c1887a81b572df823d0 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/537d28df0db840beae5f021fc19f8a75 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a6fc97dbc43483dbab06eb499c54e50 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/818d40dbeae84fafb62cf682d36d3f29 : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60cc6f9c3d8946509b45457a6463c9cf : SUCCESS in 13m 02s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f5548a9645de464baecd531764e4bd59 : SUCCESS in 1h 09m 01s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/80a046d2ef50413f964758485febcf6a : SUCCESS in 34m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d72347a62a1e4faab6c8c4100c6c482b : SUCCESS in 1h 34m 27s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8ad0a85197bb4e23908656b02e6de77c : SUCCESS in 49m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6612b96b06e44d918f5d597ac21f456f : SUCCESS in 1h 04m 44s","accounts_in_message":[],"_revision_number":3},{"id":"71a52371d64844e3011e2b7c341f7c6d86dc394d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 16:58:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3059a5fc7b2f234fe7e0dbdb615d332525a7be64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 18:14:18.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8ab7c8e441d94ebe8c11e9790f5da7bd : FAILURE in 5m 52s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d89ff1384dec4541b8c65d0593a60da0 : SUCCESS in 56m 24s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/cd83f6d39d554833ab5508047c854649 : SUCCESS in 1h 14m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9d8f210706840f6bdac5684fbebd718 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11887121a7cd429caa6daf2ac336347d : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc092609ecc54dd4ba0e0d114a7d1167 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d81f17cf645d443a8ea396000d9f6716 : SUCCESS in 6m 05s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d7646644c3e1426baa206c9b2d7971bb : SUCCESS in 37m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e0b4d1f64ce84bff8cc52310cd3c5dfd : SUCCESS in 58m 07s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9c88926e9a8345078c8e4a67adb43ff3 : SUCCESS in 57m 46s","accounts_in_message":[],"_revision_number":3},{"id":"67d8ca7a6149512f11571620487a9b515f095a8a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-11 18:42:24.000000000","message":"Patch Set 3:\n\nERROR:kolla.common.utils.base:toomanyrequests: You have reached your pull rate limit. You may increase the limit by authenticating and upgrading: https://www.docker.com/increase-rate-limit","accounts_in_message":[],"_revision_number":3},{"id":"f0834a84c60438e19c1ceed410428587e718495c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-12-12 18:53:54.000000000","message":"Patch Set 3: -Workflow","accounts_in_message":[],"_revision_number":3},{"id":"54e78bc281a3f587b672587b57910b23f4aed816","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-12-12 18:54:04.000000000","message":"Patch Set 3: Workflow+1\n\nretry","accounts_in_message":[],"_revision_number":3},{"id":"380a7d117901df74a718f44a232e82c79c5e5a13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-12 20:14:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/c9b2a532ad0e4b378dad210e575f5441 : SUCCESS in 23m 33s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7fbc1639152b4f3b8031071952e8a417 : SUCCESS in 33m 12s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c7371895b88d495590664b3821078e14 : SUCCESS in 1h 04m 25s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c6dfb9ace1454455884685ab8ab6fdc9 : SUCCESS in 1h 00m 49s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/f3781502c17b41efa5c5a76a8023ab5a : SUCCESS in 1h 08m 18s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/17d1353827f14b0eb8571576ea468db3 : SUCCESS in 32m 10s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/66a664be69194cf49f40c030f905f3b5 : SUCCESS in 21m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4b33db20171f454f89201b2ebba69913 : SUCCESS in 50m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb4238660f4c4b7fb3d4d071b3a4172b : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e120d3a85154f0281c5e8e7c332a4ab : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c42709a69a4e43c08017d66baf338526 : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a278217a210840069d1697a364937130 : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6e014bda28e495881ac24f58d2e16be : SUCCESS in 6m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/361755c346b943eaa58d0749429dc41b : SUCCESS in 20m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9cba7c6785e247ef85bf64dfc4302dfa : SUCCESS in 35m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9f31f31e9ebe405581f3dbdca3a055d0 : SUCCESS in 1h 08m 57s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d0d5d719953d42ae86b9747b517624fa : SUCCESS in 50m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f9ee19ca7e854ef28cb13a25e2cf04d9 : SUCCESS in 56m 45s","accounts_in_message":[],"_revision_number":3},{"id":"2b0009a94adefac44afd74620a71e514c3e71753","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-12 20:15:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9e41e5dcd50d587f9074c8423dbbdc837ddda7ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-12 21:21:06.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/9a034adf5b294614898ba919eae00f8b : SUCCESS in 40m 45s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7f2bb1ba227b4611957fe1211e45bb95 : SUCCESS in 1h 05m 19s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/3b622323e6b34b839c07d91b0e8700b6 : SUCCESS in 1h 03m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/274e3abf861949f59f4da17ad59a84f5 : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8449181b1e40426f8d0b086f58c11089 : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/138de31813874c16a08e451779a54f62 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e4b7d6d80fe4e388be4c024fd633ef4 : SUCCESS in 4m 51s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fa137e4364a6441eab70d7778a83a89f : SUCCESS in 33m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3f79d1d87f24423a88393c188ee7cc5d : SUCCESS in 48m 30s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/16165dbee463416aad2b49263d40fb5a : SUCCESS in 58m 47s","accounts_in_message":[],"_revision_number":3},{"id":"1e86ff187c65938cbb3bb7a492cfac64bd7a1a0a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-12 21:21:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"dfb61053f349ac28befa388ebd33209af7a32b5a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-12 21:22:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a9fa9aaa8224522b1638c77aeac4f92 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4b792b9a43bee3bfb499cea14924527550f2bce0","revisions":{"762d64e3f56701c4740363e765cc7ebee1174dd8":{"kind":"REWORK","_number":1,"created":"2020-11-17 14:42:21.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/27/763027/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/27/763027/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/763027/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/763027/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/763027/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/27/763027/1"}}},"commit":{"parents":[{"commit":"09c890b12463a4b2c8767231189e7016f337fbc3","subject":"CI: add templated Dockerfiles to build logs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/09c890b12463a4b2c8767231189e7016f337fbc3"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-17 14:40:32.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-17 14:42:07.000000000","tz":0},"subject":"Remove footer block from intermediate images","message":"Remove footer block from intermediate images\n\nOnly the final image in a hierarchy should have the footer block.\n\nChange-Id: I95e5bb5583be2a76e289749e8adfc303a4e7197a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/762d64e3f56701c4740363e765cc7ebee1174dd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/762d64e3f56701c4740363e765cc7ebee1174dd8"}]},"branch":"refs/heads/master"},"31cdf010048a2daa6f6675d56c22f6a459945592":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-11-19 11:03:12.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/27/763027/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/27/763027/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/763027/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/763027/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/763027/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/27/763027/2"}}},"commit":{"parents":[{"commit":"09c890b12463a4b2c8767231189e7016f337fbc3","subject":"CI: add templated Dockerfiles to build logs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/09c890b12463a4b2c8767231189e7016f337fbc3"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-17 14:40:32.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-19 11:03:07.000000000","tz":0},"subject":"Remove footer block from intermediate images","message":"Remove footer block from intermediate images\n\nOnly the final image in a hierarchy should have the footer block.\n\nneutron-infoblox-ipam-agent image still has two footers because it\ninherits from the neutron-server image.\n\nChange-Id: I95e5bb5583be2a76e289749e8adfc303a4e7197a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/31cdf010048a2daa6f6675d56c22f6a459945592"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/31cdf010048a2daa6f6675d56c22f6a459945592"}]},"branch":"refs/heads/master"},"4b792b9a43bee3bfb499cea14924527550f2bce0":{"kind":"REWORK","_number":3,"created":"2020-11-20 12:20:28.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/27/763027/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/27/763027/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/763027/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/763027/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/763027/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/27/763027/3"}}},"commit":{"parents":[{"commit":"09c890b12463a4b2c8767231189e7016f337fbc3","subject":"CI: add templated Dockerfiles to build logs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/09c890b12463a4b2c8767231189e7016f337fbc3"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-17 14:40:32.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-20 12:20:21.000000000","tz":0},"subject":"Remove footer block from intermediate images","message":"Remove footer block from intermediate images\n\nOnly the final image in a hierarchy should have the footer block.\n\nneutron-infoblox-ipam-agent image still has two footers because it\ninherits from the neutron-server image. We set the user to root at the\nbeginning of that Dockerfile, so that headers and footers have the right\ncontext.\n\nChange-Id: I95e5bb5583be2a76e289749e8adfc303a4e7197a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4b792b9a43bee3bfb499cea14924527550f2bce0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4b792b9a43bee3bfb499cea14924527550f2bce0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
