)]}'
{"id":"openstack%2Fkolla~763146","triplet_id":"openstack%2Fkolla~master~I083ce70772edd1abb873ad43f6beb73aae01383f","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I083ce70772edd1abb873ad43f6beb73aae01383f","subject":"CI: Improve publish playbook","status":"MERGED","created":"2020-11-18 09:54:23.000000000","updated":"2020-11-19 13:47:46.000000000","submitted":"2020-11-19 13:46:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763146-1605793572230-81340386","meta_rev_id":"c38a42e3665055ae8a074c9f5de896e1cd887eba","_number":763146,"virtual_id_number":763146,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-11-19 11:21:09.000000000","_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-19 13:46:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-11-19 10:36:39.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-11-19 12:33:43.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-11-19 12:33:43.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-19 10:36:39.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-11-19 12:33:43.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-11-19 13:46:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0fee2ace729d1cc0b2cbacf32a1b82f3261bfef3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-11-18 09:54:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"30ca69e85617fdd7f6dc338098bbd35bab7d299a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 10:38:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/bcc702d7818f42dc8611f91b42261d14 : FAILURE in 43m 46s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/04b02a00d343476d8ef6289df7cad485 : FAILURE in 25m 06s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/930e709706704e9ca7fc2b5a4b2a8ac9 : FAILURE in 24m 20s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/80f795cae89c44eea6ce22039af514be : FAILURE in 37m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"51c4841d03e66db13ac1535cfc8c06756d67bc40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 11:24:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/cf80295bc60a4ff4bed4f663f3c353dd : SUCCESS in 25m 15s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2ea9f66c666a46eaa4bc0929f7d53b37 : SUCCESS in 24m 29s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/da40f0d619d346759b3363dd180e6154 : SUCCESS in 1h 15m 53s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c6c7374f62054b39995fe47b3f472a9c : SUCCESS in 1h 18m 12s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/fc107c940891475eb81bac62584c9bbf : SUCCESS in 1h 15m 36s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3b0a3b7874854a5097d97caf77373677 : SUCCESS in 19m 34s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e6e296eb893749bca94292dac37b23ae : SUCCESS in 18m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/27937b86b3de4614b5df9b952071200d : SUCCESS in 1h 01m 43s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c49ce86a15dc411fadb5c44e02980a8b : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78cec7eb185f4401ba5cfc689b6029eb : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6838b89bcc74a1a9dd3411798ff01ad : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/676663d273694d5e9fc6ae6514a16083 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c06d1046679a48a68ba7f4d460d17233 : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/784c00a6c42343feb2ba8e2d0e3b3fdd : SUCCESS in 5m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9e5a3ad0573a4deb97c4775d6223694b : SUCCESS in 18m 11s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5ae4cf08d353448da411c6883c1ec8e4 : SUCCESS in 22m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/81526a91647447a585ef38b8109be25c : SUCCESS in 1h 01m 12s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/97300013c0964a4eb3b70f862dd7aed1 : SUCCESS in 56m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6b55bba707aa4405b6ab09dfaeaaf841 : SUCCESS in 1h 04m 27s","accounts_in_message":[],"_revision_number":1},{"id":"209eb04bd24e7e350717ab3bc26b72d36a2ca0d5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-19 09:19:24.000000000","message":"Patch Set 1: Code-Review+1\n\nIt looks reasonable. Have you tested it?","accounts_in_message":[],"_revision_number":1},{"id":"20c60ffcb8606b3d1d91f55754264820497ebd48","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-11-19 09:42:29.000000000","message":"Patch Set 1:\n\n\u003e It looks reasonable. Have you tested it?\n\nYes, it would me hard to make it up ;-)","accounts_in_message":[],"_revision_number":1},{"id":"b94608957444e6106c412907d87380a362964529","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-19 09:47:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"26c24ebb30dfe9831c30ae1a4cdeaefa5c3ae455","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-11-19 10:13:37.000000000","message":"Patch Set 1:\n\nmaybe add few words in commit message?","accounts_in_message":[],"_revision_number":1},{"id":"2ac32403cab7566edf947222b84e23ef5845f960","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-11-19 10:34:24.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a27ae5a339fa066fc9376de8eb572cb9ce0bc156","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-11-19 10:36:39.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"754841d23c148667f25114ad3c28f28b502c0cc1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-19 10:47:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"057b9026fe35540ce15909d7bc94ce556eadf705","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-11-19 11:21:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4d4d421a59d76e5265e2bf946fc0b823ed9723ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 11:25:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/7a616e3b61c245769c9f5b4d27c9b3d6 : FAILURE in 49m 35s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/78848be6ed8840c18ff1cd92a8afbeaf : FAILURE in 26m 49s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cf228054a99045ad88a4cbc8e1e2ffa3 : FAILURE in 24m 14s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/62fbdad48dd640beac4b6d776614038e : FAILURE in 34m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8a4790ac058963eebecefe646b4a62c649ca0647","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 11:55:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/0f29342b3ba940ceba15a9dc28160af6 : SUCCESS in 20m 31s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/4ce3ffeb05c24fc8a07f12a3854d29d5 : SUCCESS in 29m 25s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/eb2f665b6cd84afbb093f6c092cb2aca : SUCCESS in 1h 06m 15s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/efce0e7dc07941e19ce65f3bb9d56030 : SUCCESS in 1h 07m 35s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/d10c9bdf6b7641a2ab00cb706f98b0a4 : SUCCESS in 1h 16m 12s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3e275f5301ef48aa8c547a0c8810eb53 : SUCCESS in 17m 15s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9fd583758ff74e1d87adbb8adaf7528e : SUCCESS in 17m 24s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b75e128013984b35b34678822402c1b4 : SUCCESS in 50m 29s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef3d75efdf094236ab12996e4f170a7d : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7eb0d018529c49ddb86ff1cf0fdd056a : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bef1288c50644757be97c6221df97526 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76810322259d47a38ed293c03d0801a5 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c61dd75f21eb47be8a4790d5d45c91d5 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab8597b455024b04af32133fe8c66f99 : SUCCESS in 6m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/18d73f5b9aff4eebabd0cb90c7b4306c : SUCCESS in 22m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/23c343d51a1340b19b7dfd4ef520323d : SUCCESS in 25m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/71ece7f89fee47db8a7b51780dc46d7d : SUCCESS in 56m 04s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cea20467944f40e5a3e80687c36faba0 : SUCCESS in 59m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/edb32c86fd284fe1b3fe2401591fea58 : SUCCESS in 1h 08m 33s","accounts_in_message":[],"_revision_number":2},{"id":"7bac552b921eb4dea4bf726a67a8f6bc111a3421","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-19 12:33:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"be30c6c473f8152c5ef89f58c34ffdcbb3d86e9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 12:33:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a42f1a96d969e53f388b3faa331ee305f73f03fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 13:46:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/90e625e49263465f869a0cb82d44933d : SUCCESS in 23m 36s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/34cca452b512489e86c39d069ecb8ebc : SUCCESS in 1h 06m 35s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/0b8f27c6a5854aa884e34251df644055 : SUCCESS in 54m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2c1966c99f3446987382bfe2faf6bd1 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d0b4c1f0662475a87d43e7b8ace88d8 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de17f44c3a9c47438da41a79360cf62a : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3cc2a08e36b432885f9a8f7d1d1ec90 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d5e98aff8134491a9c5956ca2f5293c : SUCCESS in 5m 31s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/adcf3a7a4ece4b63b251128221754dad : SUCCESS in 34m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7bac90a3e36c4f1fa27280b341f27ac6 : SUCCESS in 51m 34s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/499d70cce7994ea2ae65e25754f3bbcc : SUCCESS in 1h 02m 09s","accounts_in_message":[],"_revision_number":2},{"id":"e9a9520a8e69abda8bf7b0668f360ccba2215872","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 13:46:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"c38a42e3665055ae8a074c9f5de896e1cd887eba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 13:47:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0eee0f77c03d48feb9a019ac7637bc7a : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3ad0c1212978fba797c79cf1017c293a8839b7be","revisions":{"70e3993ae0a4947a58ab135bc8a516927848fa59":{"kind":"REWORK","_number":1,"created":"2020-11-18 09:54:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/763146/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/763146/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/763146/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/763146/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/763146/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/763146/1"}}},"commit":{"parents":[{"commit":"bb12d8ac60cb8ca66dbf13b49a0dd1a844cde6a3","subject":"Merge \"Bump up openstack exporter to 1.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb12d8ac60cb8ca66dbf13b49a0dd1a844cde6a3"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2020-11-18 09:53:46.000000000","tz":60},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2020-11-18 09:54:19.000000000","tz":60},"subject":"CI: Improve publish playbook","message":"CI: Improve publish playbook\n\nChange-Id: I083ce70772edd1abb873ad43f6beb73aae01383f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/70e3993ae0a4947a58ab135bc8a516927848fa59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/70e3993ae0a4947a58ab135bc8a516927848fa59"}]},"branch":"refs/heads/master"},"3ad0c1212978fba797c79cf1017c293a8839b7be":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-11-19 10:34:24.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/763146/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/763146/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/763146/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/763146/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/763146/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/763146/2"}}},"commit":{"parents":[{"commit":"bb12d8ac60cb8ca66dbf13b49a0dd1a844cde6a3","subject":"Merge \"Bump up openstack exporter to 1.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb12d8ac60cb8ca66dbf13b49a0dd1a844cde6a3"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2020-11-18 09:53:46.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-11-19 10:34:20.000000000","tz":0},"subject":"CI: Improve publish playbook","message":"CI: Improve publish playbook\n\nCurrently we use a for loop in shell that performs push of images\nto Docker Hub - that doesn\u0027t really seem to fail properly on push\nerrors (e.g. related to new Docker Hub limits).\n\nThis change introduces usage of Ansible docker_* modules for doing the\nsame job and improves logging (so it\u0027s easier to find out which image\nfailed to publish).\n\nChange-Id: I083ce70772edd1abb873ad43f6beb73aae01383f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3ad0c1212978fba797c79cf1017c293a8839b7be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3ad0c1212978fba797c79cf1017c293a8839b7be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
