)]}'
{"id":"openstack%2Fkolla~763351","triplet_id":"openstack%2Fkolla~master~I47f8611ab74b6f33f74907b5afcacdefd65a4adb","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I47f8611ab74b6f33f74907b5afcacdefd65a4adb","subject":"DNM: testing mirror revert","status":"ABANDONED","created":"2020-11-19 11:11:04.000000000","updated":"2021-02-03 12:04:28.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8afd47e82b893f30604dd2a2fc00eba7a82986dd","_number":763351,"virtual_id_number":763351,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-12-17 12:47:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-19 19:00:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0d4ef74b19c8ffaf4c29b3fd8b74d207109c4442","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-19 11:11:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b89d44152f67c57abb228f26674e4ed00fde6a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 12:04:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/ae79717ab51147a8913b9274bb6bbc52 : FAILURE in 4m 36s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a21109f72d604cce908436d6e9c1d1eb : FAILURE in 4m 24s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a7cdb8211a0d48dc9071a3055216d15d : FAILURE in 3m 59s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/36aa2324ea854159873f4bac4c6bfd66 : FAILURE in 6m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e0deba7eea44ddeacc8f555564e97bb : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0d9c078ef6e47709783e9b724736460 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e034d69e4da40f7bf08158a4f548e35 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5fb5736d33bb4a878c7b36519fabb922 : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a09f69e6a9e54ca1a8adf6b511711627 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db11a4367c564f32a5335aa094d42f8c : SUCCESS in 7m 53s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/31e72ecee382483a8fef2248c3e943e3 : FAILURE in 4m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9939d15b09f847d7be63b78d1b9794fa : FAILURE in 4m 31s","accounts_in_message":[],"_revision_number":1},{"id":"b80522621dc2105f49d6fd4fee57bc34211db0c3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-19 12:36:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"898c6bed1b42d9865cbb01ec44e30854199078c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 12:56:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/adf50eff07794468a3aafca455b46796 : FAILURE in 6m 11s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/335c34fd7f8b422bac04b87c26b81b27 : FAILURE in 4m 35s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ddb5ce45bc5a4a889c7c14183330b6a9 : FAILURE in 4m 37s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a3d05b405be749e38e348c6602d7779f : FAILURE in 5m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c89e14f6112f47b788088b418d5668af : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/290625e6ea11402fb01c1dcdf22da5a3 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd377328c9594953b8dd2bd69ac90d50 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae89df71f72041e1a5ef917de2d51eb4 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d35902e8115b46ebb6c2ed739ea7f995 : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c71a1e7025e7421db161554ae55d0ac8 : SUCCESS in 6m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/011d9f8a9dfb4e469753d7a9ce0b33c4 : FAILURE in 4m 30s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e61036c996f74c97baa36272fd2dcaae : FAILURE in 4m 13s","accounts_in_message":[],"_revision_number":2},{"id":"469bc126d9a154c3ab0c1f1e89d72096c67a21d0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-19 14:35:38.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"dc14e4ec481b99da45eda8464286958084cd0e29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 15:29:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/1e971cda37aa40ada3c48fa224cf1c60 : FAILURE in 22m 58s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/90d34e6b55854e7889e99c53ce90eb34 : FAILURE in 47m 15s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3442d8869f1444cf8a2955f96486c732 : FAILURE in 38m 03s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5cbc483b281548c0b6e928d312c3cab7 : FAILURE in 18m 50s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/487228d31d824d4bb0f9c86109e06bd2 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4eba85ea2fd42cfa90b568f84698a18 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df476fcea67b4797b44583a125dd7230 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13c35ebebd764df2aed9943a9de14e07 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09f3edeadf974c17b39c94bab2d768e1 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7891706258fb4a36bd975f92322596b6 : SUCCESS in 5m 43s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ae16e114348e4fb0970f217aa025412c : FAILURE in 22m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/040110b07c4c464faf90d219fea9962f : FAILURE in 30m 45s","accounts_in_message":[],"_revision_number":3},{"id":"d1621df9ee42cb2f3fc31b308f025c7cbb31d5a5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-19 15:37:40.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"bae6ae5448ca58f534ea93dc727d8efeabfdad2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 19:00:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e329deae48674083b25154c45dc5baba : TIMED_OUT in 3h 01m 31s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/78d30a872a7843099aaca4be7b0fa2da : TIMED_OUT in 3h 16m 15s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/198425ae9be1433384d3919b242e1b68 : FAILURE in 41m 41s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3b9200fa320d45d888dbcb7f66124e1f : FAILURE in 21m 36s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/34c85cedef484d27b526c1339f119024 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fc05e2c9f074793988cf4835909b578 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d7c495878e646538095ee5ce19b427a : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef32aabaa30346728b469a2cc32a4cde : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9fcc0fe297c248e1a8a6fddb00fc4bbc : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07fb565800e54e128a76af9d52c93c83 : SUCCESS in 6m 29s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/052ef24c16e94abc9a34536c0ff2cb52 : FAILURE in 22m 05s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/775577d66ee24f1898834da409fcaef4 : FAILURE in 37m 19s","accounts_in_message":[],"_revision_number":4},{"id":"33579f6f5c628d349bc0b21ac08e07bbb4c353c6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-20 12:20:45.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"4fde4f8dee2a39b7680bfe5bdbebea4cedbc5242","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-23 10:48:58.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"2d839700fffde0b25934e39742d23dab8cfceb2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 14:08:40.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/015172fcbf474ff1b03bd5e9ce8fa017 : TIMED_OUT in 3h 02m 56s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/99eb1eef2dcf41a6b93bdac72de847bb : TIMED_OUT in 3h 13m 50s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e0f1d55bd5b24176bc1e73ecbdc16d14 : FAILURE in 30m 18s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c9b86d86b2404368ac52ba741a49e900 : FAILURE in 21m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60f74c3af09147ee8e0025fdbeec6f04 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dff073388164457bd6b338f929af2d6 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d2b93b6682b4c479eb6cd5d7a7c5fd9 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3399561c3851469d95aa7211a6cab40a : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1d3a07a56c94fdfa360f6df68fa0cdc : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76f2677072a542db8f97149424a7b571 : SUCCESS in 5m 02s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/703758d85b8146afb49fb86947a9a259 : FAILURE in 20m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ffbbf57478a14bf78dbd74db04c5fca9 : FAILURE in 36m 27s","accounts_in_message":[],"_revision_number":6},{"id":"ea92829c8446c34c944db9935c36b5cb09213450","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-23 16:59:30.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"7a6408bf1afe8fd252e00147f13e4c1792f6e10c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 20:21:23.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/dcd7d4778a3643cda56e02cd790ddf77 : TIMED_OUT in 3h 02m 03s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ad5ccb893e344ef88b271b7ccaac4f19 : TIMED_OUT in 3h 14m 25s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/096f7cfac84d4000a967dc5b78af9799 : SUCCESS in 1h 13m 45s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a698e8d26d0241f6914b80570d0f2881 : SUCCESS in 25m 31s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/451b663fc3a04fa18058ce21fa2a57e3 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc0a74e7be8e40589ebc89d72bdc0800 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9ecb236723d427487834edd3d851ee2 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3f0458425b143efac719351ee5b4fd1 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1278724fc6874ff1825aef1d390c8604 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebf304b045f84e739a0eca0622dc6b43 : SUCCESS in 17m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/10c9c0bd90ed4fb3b275b2eb1891e449 : TIMED_OUT in 3h 01m 54s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1e9f783f078748caa3b90121e8336b1d : SUCCESS in 43m 26s","accounts_in_message":[],"_revision_number":7},{"id":"262a7444673ad0dcfe9cbc95a2a952d81543f25c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-24 11:27:46.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"c31747894016aa48f1a22d5e33bbfde0ede6dcf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 14:52:34.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/a7620051e56249d8bf41c086d7a1beb7 : TIMED_OUT in 3h 01m 39s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/23baf6f5826645c1b9dceefd8e9355ff : TIMED_OUT in 3h 17m 47s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/86b824c8810a40b189a20a9103b21932 : SUCCESS in 48m 22s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/133b13b1083e49c9a41bffad51ee9ebe : SUCCESS in 25m 01s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9abda4014d124ff989edc05888525dd2 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0792616e7c14314a5b8b6e5047a5073 : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0593bd40a9a416895e1907efc260e2e : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe6dd43ee33049d4ba80ba4a2fcf6151 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/edf5328194654dc5ab6082717ee52578 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6602342a012d4dad9b1c8a74a3d1ac70 : SUCCESS in 6m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6a46d97f67f8443c9c7bd91dd90f05a9 : SUCCESS in 35m 56s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3906ba55ea6940558fdb8ef7a4340ea6 : SUCCESS in 51m 06s","accounts_in_message":[],"_revision_number":8},{"id":"dca917ac6c144e035f70f5b35624ec5028af780c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-24 16:15:39.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"79833b75bf67148025aea7edf2939fd5ed42f4f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 19:57:17.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/34a284d0ff8c4efba9e4dbfa4600e68e : TIMED_OUT in 3h 02m 58s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/c5bbc3caa28b494082f5ef056e9ea45d : TIMED_OUT in 3h 16m 43s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/078f172edb2c40179f90e01ca7bcefea : SUCCESS in 42m 28s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f3c47e80fa66443f8974e6d8b87e3379 : SUCCESS in 20m 38s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0187711e0a1d453aa48cc6e4bcf3916f : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5306fe8dfe64455a964549e4849575a : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99c61d50faee401aad286be37f6faac8 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24200e6b545148049673789b9623ca8b : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/528fbf9ba5134d58b531d1a407d833db : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d200fb8c3cd40a88a8e2d698ce48564 : SUCCESS in 7m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e0a1f1b3db594b7a90871daeafc6d8c2 : SUCCESS in 24m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d7e0063daee14986b6a8b8f9095c9f3f : SUCCESS in 40m 38s","accounts_in_message":[],"_revision_number":9},{"id":"8c4b594665b27d403564b37daccd618f5dfbed3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 19:57:32.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/34a284d0ff8c4efba9e4dbfa4600e68e : TIMED_OUT in 3h 02m 58s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/c5bbc3caa28b494082f5ef056e9ea45d : TIMED_OUT in 3h 16m 43s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/078f172edb2c40179f90e01ca7bcefea : SUCCESS in 42m 28s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f3c47e80fa66443f8974e6d8b87e3379 : SUCCESS in 20m 38s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0187711e0a1d453aa48cc6e4bcf3916f : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5306fe8dfe64455a964549e4849575a : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99c61d50faee401aad286be37f6faac8 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24200e6b545148049673789b9623ca8b : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/528fbf9ba5134d58b531d1a407d833db : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d200fb8c3cd40a88a8e2d698ce48564 : SUCCESS in 7m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e0a1f1b3db594b7a90871daeafc6d8c2 : SUCCESS in 24m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d7e0063daee14986b6a8b8f9095c9f3f : SUCCESS in 40m 38s","accounts_in_message":[],"_revision_number":9},{"id":"782279c1c3d18066ac58175cc38003333d168075","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-25 14:30:18.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"dd686cf889d4241429978e5c15fbc8b8baac8cd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 17:51:11.000000000","message":"Patch Set 10: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/fafc9f537eec4e12841c47c8bccd6b55 : TIMED_OUT in 3h 02m 16s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/c528eb868992449fb3fa5ddefb6a8e1b : TIMED_OUT in 3h 14m 07s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/52d51e8654ab4244b7d8cb1c0f26c7e9 : SUCCESS in 48m 23s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/63024689e1ba441a88c88b6678a2f9ff : SUCCESS in 24m 56s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f969ea6f168a40f780bdb8443a120c4c : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4271a252d7948e8bc6f160a25b73818 : FAILURE in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f111b83a4694b8fbf5f208a73e95bc5 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6ec65b68e864ebea4c80bbcb5e56dcd : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae6c9ecd43a84c288973d3cdc2a3d0d9 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b733d727dae4a26bdb38f77097b0260 : SUCCESS in 6m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e008ea3279194c5196e7f1efc013ad25 : SUCCESS in 30m 43s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d3a37c49e1c1427e8b9674557861126d : SUCCESS in 49m 46s","accounts_in_message":[],"_revision_number":10},{"id":"4d1da505d6c61716aef160927021d128ed0911bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-17 11:20:00.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"f1a4fa139b5fac0f18b9a9a6d1a444c62cfe852d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 12:47:25.000000000","message":"Patch Set 11: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/72b106412eea46e29b052ba0cf95a8c4 : SUCCESS in 36m 38s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/76b00887f1e742508947aea5ad766675 : SUCCESS in 53m 22s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/2fa7f4fdbb604b08b6a92e6b8a4954e1 : SUCCESS in 1h 05m 02s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/ca800e9470a948aa966af72a6d754eeb : SUCCESS in 1h 07m 37s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/0d9d101e2e454563a48081d08b11040b : SUCCESS in 1h 18m 25s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9148a0f522394502b2e11ff9025fd98f : SUCCESS in 26m 21s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6b686828eae64b639e625d9aae13515d : FAILURE in 25m 35s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/51d325b95ce547f7aa33356ebb7744d0 : SUCCESS in 58m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb05696737d54dadb266d5ff1e5b6b7e : FAILURE in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f09456e7cfc439a8e546f08aaca3353 : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ae4d6909b7a4a5bbabc59b6e931416b : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d8cad3f7d7942c4828da97ed3644b5a : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ebb9da20bbe4843bda46b28d8f75269 : SUCCESS in 5m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9c4c11a5b3e5445c819222aa79568538 : SUCCESS in 22m 15s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c38ce6a03ae64b8293473ea9ca44c929 : SUCCESS in 26m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/92dcdf72d4db4f9cbe9f19806e3395cb : SUCCESS in 53m 24s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/654b3ef164964c8a9b2428414219c9be : SUCCESS in 1h 01m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2405dcd7855f4f5686f9af905bc4e09d : SUCCESS in 1h 04m 00s","accounts_in_message":[],"_revision_number":11},{"id":"f00545267fe38347f509eeeebff5c2df354f66a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 15:13:55.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/1162bb1653b449fa8b089573762ebfe8 : FAILURE in 2h 17m 49s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6704140a078340078ffc5494bf33bc2f : SUCCESS in 2h 25m 34s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1b5243b587b54f11aba311e3add00be1 : SUCCESS in 1h 41m 18s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/6923ee17cb2549caadc6b98eaa883f4c : SUCCESS in 1h 35m 14s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"8afd47e82b893f30604dd2a2fc00eba7a82986dd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-03 12:04:28.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"59c112614169a3dee3fe67d29ca449b816e7bfde","revisions":{"5642addf0377a67bdd62ea525f1635e85d93f80d":{"kind":"REWORK","_number":1,"created":"2020-11-19 11:11:04.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/51/763351/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/763351/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/763351/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/763351/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/763351/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/763351/1"}}},"commit":{"parents":[{"commit":"f3828ecf9a98e4ddcf7018e8e3c4d7a6ea9d18eb","subject":"WIP: CI: revert to public package mirrors after build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3828ecf9a98e4ddcf7018e8e3c4d7a6ea9d18eb"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-19 11:10:48.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-19 11:10:48.000000000","tz":0},"subject":"DNM: testing mirror revert","message":"DNM: testing mirror revert\n\nChange-Id: I47f8611ab74b6f33f74907b5afcacdefd65a4adb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5642addf0377a67bdd62ea525f1635e85d93f80d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5642addf0377a67bdd62ea525f1635e85d93f80d"}]},"branch":"refs/heads/master"},"92afc0a0899006c36fa9aa3214a6fdeafff32c49":{"kind":"REWORK","_number":2,"created":"2020-11-19 12:36:58.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/51/763351/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/763351/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/763351/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/763351/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/763351/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/763351/2"}}},"commit":{"parents":[{"commit":"f3828ecf9a98e4ddcf7018e8e3c4d7a6ea9d18eb","subject":"WIP: CI: revert to public package mirrors after build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3828ecf9a98e4ddcf7018e8e3c4d7a6ea9d18eb"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-19 11:10:48.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-19 12:36:43.000000000","tz":0},"subject":"DNM: testing mirror revert","message":"DNM: testing mirror revert\n\nChange-Id: I47f8611ab74b6f33f74907b5afcacdefd65a4adb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/92afc0a0899006c36fa9aa3214a6fdeafff32c49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/92afc0a0899006c36fa9aa3214a6fdeafff32c49"}]},"branch":"refs/heads/master"},"db347907fb4a458342068b97af4e2964328b4274":{"kind":"REWORK","_number":3,"created":"2020-11-19 14:35:38.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/51/763351/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/763351/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/763351/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/763351/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/763351/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/763351/3"}}},"commit":{"parents":[{"commit":"f3828ecf9a98e4ddcf7018e8e3c4d7a6ea9d18eb","subject":"WIP: CI: revert to public package mirrors after build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3828ecf9a98e4ddcf7018e8e3c4d7a6ea9d18eb"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-19 11:10:48.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-19 14:35:33.000000000","tz":0},"subject":"DNM: testing mirror revert","message":"DNM: testing mirror revert\n\nChange-Id: I47f8611ab74b6f33f74907b5afcacdefd65a4adb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db347907fb4a458342068b97af4e2964328b4274"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db347907fb4a458342068b97af4e2964328b4274"}]},"branch":"refs/heads/master"},"db9ebc929cc25dfd4f91a399f3334abedda769f8":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-11-19 15:37:40.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/51/763351/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/763351/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/763351/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/763351/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/763351/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/763351/4"}}},"commit":{"parents":[{"commit":"ea942cb38787e351bba5bc5cdf23af0c5e8b1e18","subject":"WIP: CI: revert to public package mirrors after build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea942cb38787e351bba5bc5cdf23af0c5e8b1e18"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-19 11:10:48.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-19 15:37:40.000000000","tz":0},"subject":"DNM: testing mirror revert","message":"DNM: testing mirror revert\n\nChange-Id: I47f8611ab74b6f33f74907b5afcacdefd65a4adb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db9ebc929cc25dfd4f91a399f3334abedda769f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db9ebc929cc25dfd4f91a399f3334abedda769f8"}]},"branch":"refs/heads/master"},"f8946eec84bf960527bef124feb37e27dacbfd56":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-11-20 12:20:45.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/51/763351/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/763351/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/763351/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/763351/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/763351/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/763351/5"}}},"commit":{"parents":[{"commit":"03cc818053012702c5f0c11e33ebbb2e84b0f81a","subject":"WIP: CI: revert to public package mirrors after build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/03cc818053012702c5f0c11e33ebbb2e84b0f81a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-19 11:10:48.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-20 12:20:45.000000000","tz":0},"subject":"DNM: testing mirror revert","message":"DNM: testing mirror revert\n\nChange-Id: I47f8611ab74b6f33f74907b5afcacdefd65a4adb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f8946eec84bf960527bef124feb37e27dacbfd56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f8946eec84bf960527bef124feb37e27dacbfd56"}]},"branch":"refs/heads/master"},"48b3e5db7829aa60028a4a0de7118630ddc69cf9":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-11-23 10:48:58.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/51/763351/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/763351/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/763351/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/763351/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/763351/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/763351/6"}}},"commit":{"parents":[{"commit":"8718fa0bc10b00202a0ffbcc4225af508bb25cc5","subject":"WIP: CI: revert to public package mirrors after build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8718fa0bc10b00202a0ffbcc4225af508bb25cc5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-19 11:10:48.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-23 10:48:58.000000000","tz":0},"subject":"DNM: testing mirror revert","message":"DNM: testing mirror revert\n\nChange-Id: I47f8611ab74b6f33f74907b5afcacdefd65a4adb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/48b3e5db7829aa60028a4a0de7118630ddc69cf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/48b3e5db7829aa60028a4a0de7118630ddc69cf9"}]},"branch":"refs/heads/master","description":"Rebase"},"55cfdc4c3c0c09eaf4d1a6e73a5df83ec0a18c85":{"kind":"REWORK","_number":7,"created":"2020-11-23 16:59:30.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/51/763351/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/763351/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/763351/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/763351/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/763351/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/763351/7"}}},"commit":{"parents":[{"commit":"8718fa0bc10b00202a0ffbcc4225af508bb25cc5","subject":"WIP: CI: revert to public package mirrors after build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8718fa0bc10b00202a0ffbcc4225af508bb25cc5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-19 11:10:48.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-23 16:59:17.000000000","tz":0},"subject":"DNM: testing mirror revert","message":"DNM: testing mirror revert\n\nChange-Id: I47f8611ab74b6f33f74907b5afcacdefd65a4adb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/55cfdc4c3c0c09eaf4d1a6e73a5df83ec0a18c85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/55cfdc4c3c0c09eaf4d1a6e73a5df83ec0a18c85"}]},"branch":"refs/heads/master"},"d847835eb2d9058352ea51c670c3a1933eab53d8":{"kind":"REWORK","_number":8,"created":"2020-11-24 11:27:46.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/51/763351/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/763351/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/763351/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/763351/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/763351/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/763351/8"}}},"commit":{"parents":[{"commit":"8718fa0bc10b00202a0ffbcc4225af508bb25cc5","subject":"WIP: CI: revert to public package mirrors after build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8718fa0bc10b00202a0ffbcc4225af508bb25cc5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-19 11:10:48.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-24 11:27:39.000000000","tz":0},"subject":"DNM: testing mirror revert","message":"DNM: testing mirror revert\n\nChange-Id: I47f8611ab74b6f33f74907b5afcacdefd65a4adb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d847835eb2d9058352ea51c670c3a1933eab53d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d847835eb2d9058352ea51c670c3a1933eab53d8"}]},"branch":"refs/heads/master"},"50b3c8987192c5af651d83da95795c5e5aada62d":{"kind":"REWORK","_number":9,"created":"2020-11-24 16:15:39.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/51/763351/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/763351/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/763351/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/763351/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/763351/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/763351/9"}}},"commit":{"parents":[{"commit":"8718fa0bc10b00202a0ffbcc4225af508bb25cc5","subject":"WIP: CI: revert to public package mirrors after build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8718fa0bc10b00202a0ffbcc4225af508bb25cc5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-19 11:10:48.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-24 16:15:26.000000000","tz":0},"subject":"DNM: testing mirror revert","message":"DNM: testing mirror revert\n\nChange-Id: I47f8611ab74b6f33f74907b5afcacdefd65a4adb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/50b3c8987192c5af651d83da95795c5e5aada62d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/50b3c8987192c5af651d83da95795c5e5aada62d"}]},"branch":"refs/heads/master"},"cf30a8fdd443e982fcf47e9a3473435196bbc196":{"kind":"REWORK","_number":10,"created":"2020-11-25 14:30:18.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/51/763351/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/763351/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/763351/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/763351/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/763351/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/763351/10"}}},"commit":{"parents":[{"commit":"8718fa0bc10b00202a0ffbcc4225af508bb25cc5","subject":"WIP: CI: revert to public package mirrors after build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8718fa0bc10b00202a0ffbcc4225af508bb25cc5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-19 11:10:48.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-25 14:28:44.000000000","tz":0},"subject":"DNM: testing mirror revert","message":"DNM: testing mirror revert\n\nChange-Id: I47f8611ab74b6f33f74907b5afcacdefd65a4adb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf30a8fdd443e982fcf47e9a3473435196bbc196"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf30a8fdd443e982fcf47e9a3473435196bbc196"}]},"branch":"refs/heads/master"},"59c112614169a3dee3fe67d29ca449b816e7bfde":{"kind":"REWORK","_number":11,"created":"2020-12-17 11:20:00.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/51/763351/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/763351/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/763351/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/763351/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/763351/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/763351/11"}}},"commit":{"parents":[{"commit":"0fc464332c789ce05e9d6442726ada84ce7fc65a","subject":"CI: revert to public package mirrors after build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0fc464332c789ce05e9d6442726ada84ce7fc65a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-19 11:10:48.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-17 11:19:45.000000000","tz":0},"subject":"DNM: testing mirror revert","message":"DNM: testing mirror revert\n\nChange-Id: I47f8611ab74b6f33f74907b5afcacdefd65a4adb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/59c112614169a3dee3fe67d29ca449b816e7bfde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/59c112614169a3dee3fe67d29ca449b816e7bfde"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
