)]}'
{"id":"openstack%2Fkolla~763491","triplet_id":"openstack%2Fkolla~master~Icae24e3a5f522ec97cc43d6eda1811d5b2231d94","project":"openstack/kolla","branch":"master","topic":"apt","hashtags":[],"change_id":"Icae24e3a5f522ec97cc43d6eda1811d5b2231d94","subject":"Use apt instead of legacy apt-get","status":"ABANDONED","created":"2020-11-20 01:52:50.000000000","updated":"2020-11-30 10:12:41.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"185d12b2efde019dd7f9dc5d8f8765159d11a491","_number":763491,"virtual_id_number":763491,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":-1,"date":"2020-11-30 10:12:41.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-20 08:39:22.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-11-20 10:28:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-20 10:54:26.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"17d85c6975a81651ee2932adc1cb35f1e20e0ca7","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-11-20 01:52:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e79b5977066b12f675bc7cf7c6300d6b6404d02","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 02:43:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/a11d646d01054206b0f8ff3772d0d92a : FAILURE in 49m 40s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a7b5b08e289a4afdaf1b48908935e934 : FAILURE in 31m 42s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/11f2bf675ab445969ef37b004c21039d : FAILURE in 27m 26s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d412ba3c0d1943c5911690384afdd72a : FAILURE in 32m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d3cf1f77f84ce64448661ff8d6b866917965f754","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 05:41:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e9b789b28d934319bfee6dc0328ce53f : SUCCESS in 29m 24s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6bffa566e8554fbdad4462ee729d3527 : SUCCESS in 26m 55s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c582b113bba34354af65db30672573ab : SUCCESS in 1h 17m 58s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/f3f1a561717c4a37bf36e693f5788682 : SUCCESS in 1h 04m 52s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/131c9b0095e64c2cb5b9ab830ea51850 : SUCCESS in 1h 06m 58s\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/5b2c84a25d564b45bfefd586a4b2777e : SUCCESS in 30m 12s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/9f0a07f5237f46508e94551593f79327 : SUCCESS in 1h 39m 31s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/342a9f6762d74c63a6dec35da4b8d21d : SUCCESS in 1h 34m 48s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/0b7efe1a2b6a40c5ba1f9566bbb9909a : SUCCESS in 1h 37m 28s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/ef58bcec09674b40b3d8524155674e77 : TIMED_OUT in 2h 03m 05s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/f1ccc1ed584241a18fa44afe2d4612c0 : SUCCESS in 1h 09m 52s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/02db9e5ca31c4e10943f2822a96d758a : SUCCESS in 1h 32m 22s (non-voting)\n- kolla-ansible-centos8-source-mariadb https://zuul.opendev.org/t/openstack/build/a1217ef7dd174a07aa8237426a11d5d8 : SUCCESS in 30m 54s (non-voting)\n- kolla-ansible-centos8-source-masakari https://zuul.opendev.org/t/openstack/build/cd54dfe5a1684e9686af9bbe62878f6a : SUCCESS in 1h 11m 43s (non-voting)\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/27d9909680aa41f9a23011c773b48694 : FAILURE in 59m 46s (non-voting)\n- kolla-ansible-centos8-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/3811bf5c44454df18e3c777fd3bb2715 : SUCCESS in 29m 46s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ddd0b47a16f04ea2ad1717e629662239 : SUCCESS in 16m 45s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ee19bc61d8fb46a5aea8628ec1900f24 : SUCCESS in 20m 22s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a13d74331ad6441682565c7a8972ef22 : SUCCESS in 1h 15m 27s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ad9ba60ec154d4d973d7da91d7f2d32 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/825252ebf18b4b6bbf29f4de5acfaea1 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2915cc81bd34e8ba87b19a40169d6b3 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15aef33658e24c69bcf12ae17ff10222 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5416f167598e4337b63f6de210397f71 : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/188ba4753b234a93ac96d98748c03f6b : SUCCESS in 5m 58s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/61f8c8ca8ed74291b814b64d15b71e64 : SUCCESS in 20m 56s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a6d4c1a36ae04b1899162dfc7cfc3b1b : FAILURE in 17m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f70fe011ee844a52b8918ce883385cfc : SUCCESS in 52m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/389fd17e455d44f0801f10c7ea4f4fa5 : SUCCESS in 51m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cfb79cd51fc04d33944cbf98df6f8a8e : SUCCESS in 1h 04m 55s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/da8e25dc176c4fb098ba9b5dd6976119 : SUCCESS in 1h 23m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/966d7de4563f47c2a693b974259e209c : POST_FAILURE in 1h 26m 24s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/09e56733c73246c2b253cc891e5a1031 : DISK_FULL in 1h 21m 58s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/d57b78e2aac94198a39bf57a67078f55 : SUCCESS in 52m 36s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/f7063e7d6d23410b834ce8d30211f155 : SUCCESS in 1h 17m 36s (non-voting)\n- kolla-ansible-ubuntu-source-mariadb https://zuul.opendev.org/t/openstack/build/f846b19c11404361b0fe8b406a5fc399 : SUCCESS in 33m 50s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/6602ba17bcda4196871b4e46b11c3e76 : SUCCESS in 46m 31s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/9af60ad7fd4540b59e09cfdd1358615f : SUCCESS in 1h 11m 09s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/f180aeca51ef4095a2a13898e457bbc7 : FAILURE in 43m 32s (non-voting)\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/21a7a10b618141338d5742267c3123ba : SUCCESS in 46m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"54cfe89e357eac2b9ba16f0750a66386213278f6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-20 08:39:22.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7224e5334b827caca2ec6afcd21cfa43584ffb27","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-11-20 09:41:31.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"da5cd4c2575a0ebbda18af55125bd85c8e777386","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 10:28:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/df2ecdf1026b4d04a5a5e021fe82f4d4 : FAILURE in 45m 45s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/04bec52f34c54ddda80f51965cd3f4dc : FAILURE in 27m 46s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2c25e7d15e5748d19849073a2eaf5043 : FAILURE in 25m 43s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/8b1c25d942154f5ea595760bd4e116c2 : FAILURE in 27m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a5a0191e56c7808a8a214a73ff0a565a604555a6","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-11-20 10:54:26.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fcf6504788d09cb9cc6b4af53a320ba9f151358b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-11-27 07:12:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"aa39cd13257442483474a2d6cf1dfba509fa53a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 08:42:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/cfa6df042c0d4b95b939ffb7a654438a : SUCCESS in 50m 28s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/11ef65bcc2fa4378a8729cbe13efb72b : SUCCESS in 46m 10s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8ff07ef747824227b61e7256c63e4db1 : SUCCESS in 1h 23m 30s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/8523c4ea634a4fb9a4614d3f587cb603 : SUCCESS in 42m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2ec3948519b2d922f4f5304c52fb754b28510e53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 10:05:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/3a5342a35e144aeb88023d651e2e3f3e : SUCCESS in 23m 14s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/59fd1caab1eb43889414f657c047dcd5 : SUCCESS in 37m 16s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/60d0620dbc6d400bba7be16c1c4fee66 : SUCCESS in 1h 08m 41s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/efaf95e51ac84dbb96d479be35facc2d : SUCCESS in 1h 01m 47s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/499cb739ca8343ae9888a44335baf830 : SUCCESS in 1h 06m 32s\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/07a2905eeec34436800492b57360cacc : SUCCESS in 28m 36s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/322c75c194ab43509dcc8bf47c691d09 : FAILURE in 1h 13m 34s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/f126ba38dc5a4808a737f79637244a23 : TIMED_OUT in 2h 47m 38s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/51197db3b9f74844a5a1a32024ee87c3 : SUCCESS in 1h 33m 36s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/2e4e7fcee56f4497b385406e47851aa4 : FAILURE in 1h 06m 07s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/b7d1a16147ad42a88ac2fd8a34b2bc58 : SUCCESS in 1h 22m 24s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/f0112e715f4548b693dae2c772d0a2f8 : FAILURE in 1h 28m 49s (non-voting)\n- kolla-ansible-centos8-source-mariadb https://zuul.opendev.org/t/openstack/build/7a1aa742bf0341c6ad7d944997833c06 : SUCCESS in 32m 21s (non-voting)\n- kolla-ansible-centos8-source-masakari https://zuul.opendev.org/t/openstack/build/93c72e018e5747f38c342da02bcf8fe5 : SUCCESS in 57m 59s (non-voting)\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/50ae9d2ee7d54fb386349886d471b851 : SUCCESS in 1h 18m 03s (non-voting)\n- kolla-ansible-centos8-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/a1c30453824d47d0a3254c1bebc70ce4 : SUCCESS in 36m 00s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/dad76a3593e34f4f95db65600db2c111 : SUCCESS in 26m 36s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d1079479859a475eb6f903d0bf71c585 : SUCCESS in 29m 49s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5892f548e5664f63a51ea361ec456a17 : SUCCESS in 52m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a7ede3eb6d64e8a8226a7156ab086c5 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89fd48bff7b84e6ab183a1a92f0a4a38 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/482cc1d6c52845aeb1ae13653c55a4db : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/28a7a00a913d4e62866bb2f8376a8f50 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25f1c8180725473fb2762b67e63e71d4 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58c4106be5b445fdb52d982f998f10d2 : SUCCESS in 6m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9b77676a99ac44c9a74a1287ba3dcb3d : SUCCESS in 21m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cdc90851328943afb92a36bcaf5a490d : FAILURE in 34m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bb646478f139490888d7ede9d9d5a722 : SUCCESS in 54m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/833ca09d892742b68c2ba6a230694c53 : SUCCESS in 53m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5735227b07894dc199b959632497688b : SUCCESS in 1h 22m 57s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/39f66b6f0d704120b8b2b41d0e2f0822 : SUCCESS in 1h 12m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/94452bb0dc8e4fee9be14473189058d9 : POST_FAILURE in 1h 34m 40s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/8dae2485ea234cc897700eb677c70b38 : SUCCESS in 1h 15m 40s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/b36464e57b1c48ab8bc2e1a932fc64ea : SUCCESS in 51m 45s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/a9bd6a08f6a847d9832f5623a342ae67 : DISK_FULL in 1h 08m 35s (non-voting)\n- kolla-ansible-ubuntu-source-mariadb https://zuul.opendev.org/t/openstack/build/543ae59e45024626838b5308a4ec9b8f : SUCCESS in 34m 55s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/c5e930326e4f41df8dce9a9505ecb5d9 : SUCCESS in 1h 03m 16s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/30b039e0dd2b4f8796d6b38ae8920c3e : SUCCESS in 1h 04m 07s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/4e0a677da33b414aa43975f3b64e11ba : SUCCESS in 1h 15m 28s (non-voting)\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/e72d26cb03034c44a0193b9feee310ea : SUCCESS in 22m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dcc7f043e3a1debba0fcca1bf990082659dc3bd5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-11-27 10:49:50.000000000","message":"Patch Set 3: Backport-Candidate-2 Code-Review-1\n\nCommit message is misleading - apt-get -\u003e apt affects also Debian. And \u0027apt\u0027 command was even in Ubuntu 18.04 version.\n\nMove python -\u003e python3 into separate patch.","accounts_in_message":[],"_revision_number":3},{"id":"bc640365b38e0f63b5c65d58ba6e330f3a2ddb0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-11-30 09:30:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8aee44f0af354fa228331ba77c9c7c387df17cb8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-11-30 10:12:41.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"428dea79a2ac99ce2d1fcd4d9e7af98ffabd6f7d","revisions":{"8b6a63585499e061571b1c925cc5f54d0bfa3273":{"kind":"REWORK","_number":1,"created":"2020-11-20 01:52:50.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/91/763491/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/763491/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/763491/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/763491/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/763491/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/763491/1"}}},"commit":{"parents":[{"commit":"01972fc749cdabed36dab1f3020a8db4344086b8","subject":"Merge \"Finish removing heat-all image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/01972fc749cdabed36dab1f3020a8db4344086b8"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-11-20 01:51:43.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-11-20 01:51:43.000000000","tz":480},"subject":"Update docs for Ubuntu 20.04","message":"Update docs for Ubuntu 20.04\n\n* Use \u0027apt install\u0027 command instead of legacy \u0027apt-get\u0027\n\nChange-Id: Icae24e3a5f522ec97cc43d6eda1811d5b2231d94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b6a63585499e061571b1c925cc5f54d0bfa3273"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b6a63585499e061571b1c925cc5f54d0bfa3273"}]},"branch":"refs/heads/master"},"9d6969e3dafeefa0eb7876bfc732bb66418d5bf7":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-11-20 09:41:31.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/91/763491/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/763491/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/763491/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/763491/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/763491/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/763491/2"}}},"commit":{"parents":[{"commit":"01972fc749cdabed36dab1f3020a8db4344086b8","subject":"Merge \"Finish removing heat-all image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/01972fc749cdabed36dab1f3020a8db4344086b8"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-11-20 01:51:43.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-11-20 09:36:43.000000000","tz":480},"subject":"Update dockerfiles for Ubuntu 20.04","message":"Update dockerfiles for Ubuntu 20.04\n\n* Use \u0027apt install\u0027 command instead of legacy \u0027apt-get\u0027\n\nChange-Id: Icae24e3a5f522ec97cc43d6eda1811d5b2231d94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d6969e3dafeefa0eb7876bfc732bb66418d5bf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d6969e3dafeefa0eb7876bfc732bb66418d5bf7"}]},"branch":"refs/heads/master"},"e109bef7b9c4fa2378c51d61d0bda1fad01917a6":{"kind":"REWORK","_number":3,"created":"2020-11-27 07:12:46.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/91/763491/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/763491/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/763491/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/763491/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/763491/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/763491/3"}}},"commit":{"parents":[{"commit":"01972fc749cdabed36dab1f3020a8db4344086b8","subject":"Merge \"Finish removing heat-all image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/01972fc749cdabed36dab1f3020a8db4344086b8"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-11-20 01:51:43.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-11-27 07:09:16.000000000","tz":480},"subject":"Update dockerfiles for Ubuntu 20.04","message":"Update dockerfiles for Ubuntu 20.04\n\n* Use \u0027apt install\u0027 command instead of legacy \u0027apt-get\u0027\n\nChange-Id: Icae24e3a5f522ec97cc43d6eda1811d5b2231d94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e109bef7b9c4fa2378c51d61d0bda1fad01917a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e109bef7b9c4fa2378c51d61d0bda1fad01917a6"}]},"branch":"refs/heads/master"},"428dea79a2ac99ce2d1fcd4d9e7af98ffabd6f7d":{"kind":"REWORK","_number":4,"created":"2020-11-30 09:30:43.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/91/763491/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/763491/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/763491/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/763491/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/763491/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/763491/4"}}},"commit":{"parents":[{"commit":"07dd933043f5ddb7284a1942873d66345127410c","subject":"CI: Ensure docker-py is installed on publish","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07dd933043f5ddb7284a1942873d66345127410c"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-11-20 01:51:43.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-11-30 09:29:04.000000000","tz":480},"subject":"Use apt instead of legacy apt-get","message":"Use apt instead of legacy apt-get\n\nChange-Id: Icae24e3a5f522ec97cc43d6eda1811d5b2231d94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/428dea79a2ac99ce2d1fcd4d9e7af98ffabd6f7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/428dea79a2ac99ce2d1fcd4d9e7af98ffabd6f7d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
