)]}'
{"id":"openstack%2Fkolla~763903","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~I2c82c6269b2c357b0815ea29bc6a7c322f84dac9","project":"openstack/kolla","branch":"stable/victoria","hashtags":[],"change_id":"I2c82c6269b2c357b0815ea29bc6a7c322f84dac9","subject":"base: fix QEMU dependencies for Debian/aarch64","status":"MERGED","created":"2020-11-24 08:10:56.000000000","updated":"2020-11-24 12:29:26.000000000","submitted":"2020-11-24 12:27:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763903","cherry_pick_of_change":763757,"cherry_pick_of_patch_set":1,"meta_rev_id":"ed6144aa3d75c06d6a8112c23a54d2c0c8e05ea8","_number":763903,"virtual_id_number":763903,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-24 12:27:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-11-24 09:42:41.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-11-24 08:42:03.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-11-24 11:10:08.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":-1,"date":"2020-11-24 08:13:38.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2020-11-24 08:13:38.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-24 08:42:03.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-11-24 09:28:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-11-24 09:42:41.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-11-24 10:50:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"89b9829bca9a634d92a8e167ae83256a7728e191","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-11-24 08:10:56.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"22bc21e9fb664ef7f5f4ebc31484661b7e874205","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-11-24 08:13:38.000000000","message":"Patch Set 1: Backport-Candidate-2 Review-Priority+2\n\nMarking important as it fixes AArch64 situation. And is not needed for older versions.","accounts_in_message":[],"_revision_number":1},{"id":"27bc248e49e3902ea94d38ae3c663c4f961c6633","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-24 08:42:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ade645c13f3744ee2bd60e4f086e69551826532e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 09:28:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/a31b0258a869461580f9f6e5d542fc35 : SUCCESS in 31m 12s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fe33188ad10c4d768eefef138dc1a1a0 : SUCCESS in 28m 01s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/592a5c5581ec4627b8561eaa83ab6c90 : SUCCESS in 1h 16m 46s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/bd64b7ca71ed4ed4b813d5f380a5344d : SUCCESS in 29m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e63efa9381ca075853c7d91fdfb77a184c266d66","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-24 09:42:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"10ca3794d2a4fcda0288dc57cf7e8aa4001e3bde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 10:50:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/8a351ca28c824d2a8f3d8a6e04302733 : SUCCESS in 17m 52s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/afae5d0841bd49ccbebcbd5647cb919c : SUCCESS in 21m 12s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/53fd692e862a41df9cd9225b1c196227 : SUCCESS in 1h 04m 16s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/8bdefba8764d42769eb98d68dbf00ebb : SUCCESS in 1h 02m 21s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/43066e6579f44ec6be171544f2bde2f5 : SUCCESS in 1h 07m 34s\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/8ae5765217ea4fdfbde3347f4feec827 : SUCCESS in 27m 00s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/9a06fb49cc9449a48520874a24ddf75a : SUCCESS in 1h 24m 50s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/5513bb877ccb440b96873a8f4c81afea : SUCCESS in 1h 19m 49s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/f297e72e0a3a42898b50ea9325971875 : SUCCESS in 1h 32m 48s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/c3a7db028503411aa82a5be66d0fde96 : FAILURE in 56m 39s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/7040b37b76304088953dcd350d5b835a : SUCCESS in 57m 22s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/81687deadf8a4c69b153046f6508383c : SUCCESS in 1h 30m 20s (non-voting)\n- kolla-ansible-centos8-source-mariadb https://zuul.opendev.org/t/openstack/build/ce795696835c475cab6d801c2bcfbc0c : SUCCESS in 31m 32s (non-voting)\n- kolla-ansible-centos8-source-masakari https://zuul.opendev.org/t/openstack/build/ba1a649115114fec91357eb9092d8d4a : SUCCESS in 55m 32s (non-voting)\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/8ed08860569d4db694661e5388471fdd : SUCCESS in 1h 09m 34s (non-voting)\n- kolla-ansible-centos8-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/22f72f9beebb445facbaaf55d3a8c271 : SUCCESS in 27m 24s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/56275cf131464abea9dc05c34022d785 : SUCCESS in 15m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/befc23025bd34e7e8e9b75eef4142519 : SUCCESS in 14m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ff39626c37c941298e319d8b4382a03a : SUCCESS in 52m 15s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f98d87af34994bb9be2fc47f0d36160c : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/107ad3c9f3fa4569919c13065c63dc7c : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/635fdf18976849abafa374220f7018c2 : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76886ee2377c4ea08659fe0ac2bf6446 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76adb73a8ca64194b2bbf0044d706fec : SUCCESS in 7m 51s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4c38274a80b648e4b10fcef1846bc8e0 : SUCCESS in 2h 33m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9e7c4381a0ca49faad0b72c599c0115b : SUCCESS in 1h 46m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f30b47a442f24b7386412fcef8eed6eb : SUCCESS in 50m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6e428564cdd547d9b50c82962e7dff9f : SUCCESS in 1h 47m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e3069f67028f4297aa34505ec45b33ac : SUCCESS in 2h 01m 10s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/3695107b54054cb1b8fce7016e4109b3 : DISK_FULL in 2h 25m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/48685798b7424cfcbce218de86b589c6 : SUCCESS in 1h 38m 42s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/6f8a3acc0eb34d218a57dc44a73cf838 : SUCCESS in 1h 18m 48s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/a79babe902a447d887e7ecd2ba911de8 : SUCCESS in 54m 47s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/16900bd2db0e4030af9c1e9264810103 : SUCCESS in 1h 15m 30s (non-voting)\n- kolla-ansible-ubuntu-source-mariadb https://zuul.opendev.org/t/openstack/build/232035e713024079a60ce1a3bd1a2e5d : SUCCESS in 26m 40s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/a54f72740d9e4f73b89884ef274eb2df : SUCCESS in 46m 38s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/c1d7a9afe42848c29c601dccf0094fd2 : SUCCESS in 52m 59s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/19026dbcfb4b455fa424e6a72f358d73 : SUCCESS in 1h 08m 24s (non-voting)\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/b664f5b6edc24b51bb1cb60c07d1db16 : SUCCESS in 23m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b5bbff96cd1a6a4b7476fa59c8b70efe1e874329","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-24 11:10:08.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"42da5c2931338388fa3b0eeae28c519bbdba32e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 11:10:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2086996937b1d96d46e550cab2b9290c3208a364","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 12:27:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d0a2eb2802f041a3a1d13792070d55a4 : SUCCESS in 21m 46s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/42aa5f29b9c14b379e8a0418bca013a3 : SUCCESS in 49m 40s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/9bcc1a58f5984168bd37f170682a85f0 : SUCCESS in 1h 09m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63d5d5a26b9a4748a7ae0c74efebb44b : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dff70a42695445fc98884cd8b2cc2ab0 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21d54442fd924b4492ec78ff22096654 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0de3d1ec7104410294b62582ddc79ac2 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b99f9be662284a4ab14cf98c02de18db : SUCCESS in 5m 07s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1c9a5ec839fd4a61aeace31862c95e15 : SUCCESS in 27m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e01be865965b456b9cd4ff28c1f06eee : SUCCESS in 1h 09m 44s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f9e339b8c8484ec396edd2adbcd7d651 : SUCCESS in 1h 00m 39s","accounts_in_message":[],"_revision_number":1},{"id":"f9b62f8aca828aabfbd7e0d1ebeced3a0f130c2a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 12:27:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ed6144aa3d75c06d6a8112c23a54d2c0c8e05ea8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 12:29:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb25eb2a9bec46fca17ddc9c4b8181f0 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"31a94651d4317c259b0cc8cdc126d72754e65f72","revisions":{"31a94651d4317c259b0cc8cdc126d72754e65f72":{"kind":"REWORK","_number":1,"created":"2020-11-24 08:10:56.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/03/763903/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/03/763903/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/763903/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/763903/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/763903/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/03/763903/1"}}},"commit":{"parents":[{"commit":"f99c37996a10753b1b060fc5058246bbb72049ba","subject":"Switch to centos-release-openstack-victoria package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f99c37996a10753b1b060fc5058246bbb72049ba"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-11-23 09:43:55.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-11-24 08:10:56.000000000","tz":0},"subject":"base: fix QEMU dependencies for Debian/aarch64","message":"base: fix QEMU dependencies for Debian/aarch64\n\nChange-Id: I2c82c6269b2c357b0815ea29bc6a7c322f84dac9\n(cherry picked from commit 753623b29bdf66b5bbf50960d44addf9e3855bee)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/31a94651d4317c259b0cc8cdc126d72754e65f72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/31a94651d4317c259b0cc8cdc126d72754e65f72"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
