)]}'
{"id":"openstack%2Fkolla~763947","triplet_id":"openstack%2Fkolla~master~I51c7c6e2f520a582de0409025eb5d1a6da2952a2","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I51c7c6e2f520a582de0409025eb5d1a6da2952a2","subject":"swift: fix group membership in debuntu/binary","status":"MERGED","created":"2020-11-24 09:28:26.000000000","updated":"2020-11-24 16:47:39.000000000","submitted":"2020-11-24 16:45:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763947","meta_rev_id":"a8e3d7410c2f28e752c76a19df899e6fb2373f31","_number":763947,"virtual_id_number":763947,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-24 16:45:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-11-24 14:03:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-11-24 15:22:39.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-11-24 15:22:39.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-24 10:55:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-11-24 11:35:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-24 14:03:10.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-11-24 15:22:39.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"d5dd61afba530b22c782480f68e31cffabcc528a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-11-24 09:28:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7753883e559638e3788c04f4865c3cc759ed200f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 10:55:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/67fa87a9fdc54d6cbab75fab7b7f5f08 : SUCCESS in 33m 31s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/94103a63265c4ba29f08286b384d4755 : SUCCESS in 31m 04s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/699b38e8a4f14ff092746440df32b65f : SUCCESS in 1h 20m 48s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/139a36f344734846a6a3710221d80345 : SUCCESS in 29m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a80f723c720002acd8156323a2b7e42ca8a58e1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 11:35:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/bbc2a7c848b84d288c6432ec03ea91b1 : SUCCESS in 46m 56s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/c367f5a7424942d4b56eb7fdbff5618f : SUCCESS in 23m 19s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d66ffb6dc93a41b7b9a5cdf48e355144 : SUCCESS in 1h 34m 33s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/bf9b2e64c6e945ce8f1b14706af2d8e6 : SUCCESS in 1h 25m 22s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/15d6959fbe2646b5af29b788e8ee6120 : SUCCESS in 1h 53m 35s\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/fd274e5ca74b425689eae9d49ac8b0a2 : FAILURE in 58m 48s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c3f5b784ca9540a4bebe8776fad9e1be : SUCCESS in 16m 25s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ea1dda8288ae4873a7091191318d105f : SUCCESS in 16m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/334b94a6558a4ad182482ed5313d1b49 : SUCCESS in 54m 34s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4686d14f329f459f89c56f1f6c6493f6 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22dc8cac919b4fd68f54495c1b8d9f2c : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/515f1f8c0c084d3cbd6e34cc6f5b611e : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c86bac76c6c4c9e83eda9c010df1308 : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6f7cdf7d0264ce18b21ec582ae6ed79 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3953fad65e0940af8e4cc8c0dc95f969 : SUCCESS in 5m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f08fd98330a40afb1ecc2dcf5c06b6e : SUCCESS in 5m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5be6ff0f6b0b40e18b7310c6c128b27f : SUCCESS in 56m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5c363f58e757427cb68cee6111ce12ea : SUCCESS in 52m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f1e54d8c47c84167982856c4eae1be35 : SUCCESS in 51m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cf1db0b71d2140398ec96692689409e5 : SUCCESS in 51m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7841beb4374444778b9a4bd45ba80082 : SUCCESS in 1h 05m 47s\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/480304ff448e4efab95c595a870470d0 : SUCCESS in 1h 15m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"91de33103cc77261666039c0e428a16d0b812e96","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-11-24 14:03:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d85fb7d90cf9ad21349867dd54b7555a323867fa","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-11-24 15:22:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6c5cae569bc0a90f9feda93e960fa1c8b28adac2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 15:23:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d06225f0256416270fa4eed2821924266fd37627","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 16:45:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8a7682c921954acf813952f58c7ceff6 : SUCCESS in 1h 04m 15s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/255786b23d6048d792254006677bf4e4 : SUCCESS in 1h 08m 48s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/2185344fbed847849dbaf5999ba1c45d : SUCCESS in 1h 10m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e1a03693d0c54a3e87b213987f63f7ca : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b441c0511bb481abf93eb8ea70b7ce7 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a183dd6993948f9bc62f96031ca9a7d : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3631d46563db4f168bf9aaa5b20bf786 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9e67e4a79a24a6da94ba18db5db355f : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/504601e208094b849917307e44c65e88 : SUCCESS in 3m 45s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/68eae38e01ae4e50a4d5211fb3ea81a9 : SUCCESS in 30m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/66c9cc6b4e024725955e6aca2c9e84b4 : SUCCESS in 53m 44s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9718ed839f464d348f329a1e709aad0f : SUCCESS in 1h 07m 05s","accounts_in_message":[],"_revision_number":1},{"id":"4b2d33f84d414dab6b2a8d5c6c43aaf0ba8b9d0c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 16:45:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a8e3d7410c2f28e752c76a19df899e6fb2373f31","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 16:47:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c48defc32f0549be95055f5902c9d749 : SUCCESS in 1m 29s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4191dfe8339d482bbfeb6de7b2359155 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aa3208ea0f10889469f864aaccdb593547412d76","revisions":{"aa3208ea0f10889469f864aaccdb593547412d76":{"kind":"REWORK","_number":1,"created":"2020-11-24 09:28:26.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/47/763947/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/763947/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/763947/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/763947/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/763947/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/763947/1"}}},"commit":{"parents":[{"commit":"3ad0c1212978fba797c79cf1017c293a8839b7be","subject":"CI: Improve publish playbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3ad0c1212978fba797c79cf1017c293a8839b7be"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-24 09:27:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-24 09:28:03.000000000","tz":0},"subject":"swift: fix group membership in debuntu/binary","message":"swift: fix group membership in debuntu/binary\n\nThe UCA packages seem to be removing the swift user from the kolla\ngroup. Explicitly add it after installation.\n\nCloses-Bug: #1905279\n\nChange-Id: I51c7c6e2f520a582de0409025eb5d1a6da2952a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa3208ea0f10889469f864aaccdb593547412d76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa3208ea0f10889469f864aaccdb593547412d76"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
