)]}'
{"id":"openstack%2Fkolla~764662","triplet_id":"openstack%2Fkolla~master~I303ee5367aba0ae114e52b7b559faf06133f2304","project":"openstack/kolla","branch":"master","topic":"feat/horizon-set-default-for-force-generate","hashtags":[],"change_id":"I303ee5367aba0ae114e52b7b559faf06133f2304","subject":"horizon: set FORCE_GENERATE to no if not set","status":"MERGED","created":"2020-11-30 12:48:59.000000000","updated":"2021-01-16 20:13:37.000000000","submitted":"2021-01-16 17:49:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764662-feat/horizon-set-default-for-force-generate","meta_rev_id":"0726d219b06b33db6e08acb1486e583383aca845","_number":764662,"virtual_id_number":764662,"owner":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-16 17:49:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-01-14 09:30:00.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-01-14 10:35:08.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-01-14 10:35:08.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-01-14 10:35:08.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"CC":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-30 14:15:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-11-30 15:29:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-14 10:28:48.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"},{"updated":"2020-12-18 20:01:03.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2020-12-18 20:01:03.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-01-06 09:59:17.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-01-14 10:35:08.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"6de8dda4e14247a540c861d5092449afb103c5f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2020-11-30 12:48:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a7474ec7872a7172588b1a6b8360ea9f2bee0cb4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 14:15:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/8472d7b7590d4533b03b77f9c35fa70e : SUCCESS in 54m 06s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/04cbe47bee1044ce85acab967c5b1592 : SUCCESS in 53m 22s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1210f98b006a4395a2db43a0ffc863df : SUCCESS in 1h 20m 46s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/94bb66580f9f4b289c47b5928efe42fd : SUCCESS in 37m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8383320dd184716bce17edf82a9d8233ea0436bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 15:29:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b6f1c1bc359543769d635b79d565e5e7 : SUCCESS in 18m 53s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/67251d3b2fa44c9a94466bb5f6881215 : SUCCESS in 39m 50s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0ba53b800e9f4f9f96b478e09b468abb : SUCCESS in 59m 56s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/bebf4c96819c4696a19fcccdae99a726 : SUCCESS in 1h 17m 08s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/9d8c108e95444e9089334e621e3b2c63 : TIMED_OUT in 2h 32m 36s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7b4f676bfa3a42b184a0ccc2a981808b : SUCCESS in 36m 55s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9c9b136c16de4c3f978c0737e5a679ae : SUCCESS in 18m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fffa285237e04cf1bae909b501d29ebe : SUCCESS in 55m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a80f83674ecb4442b42474102070ae95 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9916baeb545a449db068008c664618ab : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e7619947a324a3eb6cd0824d2f4fa39 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0225fe339811413bb7e529da474de0e4 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/edb7912f23944ae6b859407f450a5a62 : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d8ff2ff7c8b46c98457c87df5a512af : SUCCESS in 6m 10s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/83d4bf93439342959c459d68ab84ece6 : SUCCESS in 17m 51s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/402219309b1c43c388864353869debbb : SUCCESS in 44m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2b35b850b0df41a8be4579d8aa3feb28 : TIMED_OUT in 2h 03m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d114507ec07f474d8934a5518f659ae2 : SUCCESS in 42m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5740f3ad43304a73873730b9786d3b5b : SUCCESS in 1h 19m 41s","accounts_in_message":[],"_revision_number":1},{"id":"11e3456ece9826a83340fc9d07f313a492857246","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-12-14 10:28:48.000000000","message":"Patch Set 1:\n\nIs it needed? script later checks only for \u0027yes\u0027","accounts_in_message":[],"_revision_number":1},{"id":"ff1edd00add506ca835bcd1869847fb59d8b28bd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-18 20:01:03.000000000","message":"Patch Set 1:\n\nWhen are you hitting this? I assume you are running this outside of K-A which always sets this to either `yes` or `no`.\nI know the Horizon is now a pain the butt that we enabled failing on undefined vars (exactly why we had to hold the K-A changes dropping Searchlight and Qinling until the Kolla changes got published to DockerHub).","accounts_in_message":[],"_revision_number":1},{"id":"052e346bf38fae96e16d409ec9f71ced7ad79043","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-06 09:59:17.000000000","message":"Patch Set 1: Code-Review-1\n\nNeeds more info","accounts_in_message":[],"_revision_number":1},{"id":"e245745841abcb66600927c9ef1ca3d18f68f694","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2021-01-14 08:11:07.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e Needs more info\n\nI think it makes sense to have default values for variables. The default in kolla-ansible is no: ansible/roles/horizon/defaults/main.yml:      FORCE_GENERATE: \"{{ \u0027yes\u0027 if horizon_dev_mode | bool else \u0027no\u0027 }}\". Because of that I set no here as well.","accounts_in_message":[],"_revision_number":1},{"id":"d628c513d83d800341c42b4c44c15305b1be0131","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-14 09:28:55.000000000","message":"Patch Set 1: -Code-Review\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1: Code-Review-1\n\u003e \u003e \n\u003e \u003e Needs more info\n\u003e \n\u003e I think it makes sense to have default values for variables. The default in kolla-ansible is no: ansible/roles/horizon/defaults/main.yml:      FORCE_GENERATE: \"{{ \u0027yes\u0027 if horizon_dev_mode | bool else \u0027no\u0027 }}\". Because of that I set no here as well.\n\nOk. I think we were just wondering if there is a bug that\u0027s being fixed.","accounts_in_message":[],"_revision_number":1},{"id":"ca22b7f9edde8bf937418b6b74bc27aed899f575","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-14 09:30:00.000000000","message":"Patch Set 1: Code-Review+2\n\nMakes sense for non-kolla-ansible users at least.","accounts_in_message":[],"_revision_number":1},{"id":"066ac38b8110cc11a186d59f60950106b54c12df","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-14 09:30:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fa48881397b30d0349a5a72fc7ea274621735581","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-14 10:35:08.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+2 Workflow+1\n\n\u003e Patch Set 1: -Code-Review\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 1: Code-Review-1\n\u003e \u003e \u003e \n\u003e \u003e \u003e Needs more info\n\u003e \u003e \n\u003e \u003e I think it makes sense to have default values for variables. The default in kolla-ansible is no: ansible/roles/horizon/defaults/main.yml:      FORCE_GENERATE: \"{{ \u0027yes\u0027 if horizon_dev_mode | bool else \u0027no\u0027 }}\". Because of that I set no here as well.\n\u003e \n\u003e Ok. I think we were just wondering if there is a bug that\u0027s being fixed.\n\nYes, that\u0027s why I just left no CR score.\n\nLet\u0027s merge and backport to Victoria to be nicer.","accounts_in_message":[],"_revision_number":1},{"id":"d77a0b4968610b5113ff5be42c0a9a08bf8266d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 16:46:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/0cba91acb90545dda3cd6d2c03ffd984 : SUCCESS in 25m 26s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/0bf007ec0ebd4956adaac4e725210afd : SUCCESS in 1h 06m 14s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a7548f6c53e643ce8482524d00053928 : SUCCESS in 1h 33m 37s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/6ef33604630242be8c896c2b0b38d7e2 : SUCCESS in 1h 06m 07s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/9a18d8395c094cd898778d0ebc188dec : SUCCESS in 2h 17m 00s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d48237b11e9b482eae5b46dbe0353ab7 : SUCCESS in 1h 07m 07s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/fc4a728caf26434890564b1368ad30fe : SUCCESS in 17m 48s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/443f3c9d225c48eeb3210e1b092efa0b : SUCCESS in 55m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1ca26fe099a4e6c963393e725a8fafe : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6eac067165744ad181b77d0efbfbe51f : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea967f7fdd3b4f1c8635e6157f37156c : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89b1975ae991459aa750c10252370dce : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e81288eaa19a44d385c5ebe2234ce9a5 : SUCCESS in 5m 20s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e7a71fe36e624c1c8e9aaa88c7faed0c : SUCCESS in 19m 30s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/95d1c5a9a6394ceba11b5faa132df2dd : SUCCESS in 31m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6720ffeb9b1f44e2a37b220663de60cc : SUCCESS in 50m 22s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7b98df554b9d48ec85fd9cd7bcc621dd : SUCCESS in 1h 02m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f896e350922c487d82635ff1a8742509 : SUCCESS in 1h 05m 27s","accounts_in_message":[],"_revision_number":1},{"id":"eb539ea9c8354eb4e4fc41710499cae9b27f2d78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 16:46:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b1c039d24b791cf47d908d006a934d341b71e659","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 17:27:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/5790e565cd404fb6b11edc1348275645 : TIMED_OUT in 3h 10m 16s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/79cdeaabb96c4aeeb1388471a7286c98 : TIMED_OUT in 3h 15m 00s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"96c3080907298775ea072a61882d1817f3d8b935","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 20:13:32.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ee65b65963894f43bd1b76d73f0c649d : SUCCESS in 54m 26s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/987305e616494fb1a2632b62c5f5249f : SUCCESS in 1h 22m 28s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/b42d57ae10e34c0597d9aa586dc55f03 : SUCCESS in 1h 12m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ef0555a5ffb4191bab24c48254c1379 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/245fd581bb744b28bd3b0f12e73efefa : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e9be7d1f5744f38b74f069c87fbadff : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab33adc0b1684bbb94591cf60379db11 : SUCCESS in 6m 30s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a1f4057401d640f98343e26a226613ec : TIMED_OUT in 3h 24m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3ff609f3cb3f40b4bd54bf05a7a99036 : SUCCESS in 52m 08s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3a2d681749274787b766e001684bee1e : SUCCESS in 1h 02m 59s","accounts_in_message":[],"_revision_number":1},{"id":"0853340fe13c06140752eb9a2e141ef158374a8d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-16 15:14:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0f6e22fb4cc1509e6e03bff76c4cf78e2d02c772","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 16:32:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/aef8a8f29eb447a5abfcdd98fd92c62d : SUCCESS in 29m 06s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d759f5d9f2d5403dbd90168d46071317 : SUCCESS in 39m 04s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1c6e93a22a3f4b7397870093f73fd88b : SUCCESS in 1h 07m 51s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/e3dcd0c488044022834614ed6561fb54 : SUCCESS in 1h 05m 30s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/7943153fa95348668b0a2ed2dfce0ffe : SUCCESS in 1h 16m 20s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f3afb72e66a24f72878735ab8af2cb4c : SUCCESS in 35m 42s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ac071dbef5e1443aaaea4b8e5456bb06 : SUCCESS in 20m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a42b87d501a540c68bd2ddd841cd1171 : SUCCESS in 1h 00m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d372422adbd643f59da4f5e6e914dc28 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19cbdf20ef6445b1a7d33db53ab19bf8 : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6bef18e2afcb4c61bfa51b7df9979712 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d88b33e3d136468c9870692da10e1e25 : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3888715cd5b5470581f42e436d032fdf : SUCCESS in 6m 25s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/317de9096d534e1897a75a107a78a4fd : SUCCESS in 19m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f020ea8f77304c12a169a7f300be396d : SUCCESS in 35m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9f4935d1e46b450b9caabbbeb2da3b06 : SUCCESS in 56m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1dfdc67263d748aaa3bc13f452ae243e : SUCCESS in 56m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cc988a9bac694d088ca92e03961b133e : SUCCESS in 1h 09m 51s","accounts_in_message":[],"_revision_number":1},{"id":"aa74fb3724cc94cb6ce2adb73d98fe718a10df76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 16:33:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2b9093d7c2dd1b2ade413fdf4db36ff9e76209ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 17:49:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8baff427ca364509a66f42b724e6d388 : SUCCESS in 37m 23s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/fc9bc47e132e4fc987f86ebacf108458 : SUCCESS in 56m 03s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/17ab153433804d9b9bae577613908314 : SUCCESS in 1h 14m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9514f49b3954c44a13d5cdd6cfbfa4a : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4fa235a5a6940dba07863397ac1dc99 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f83c5edaa494cfd91ff2355d391fa7c : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad5909f647fb474eaecbb4d1f7b6548c : SUCCESS in 6m 31s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/587262c5d42a45c9b6ee8fb0d1c528d0 : SUCCESS in 35m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c65453c2233f454bb48bc5afdfab7a0c : SUCCESS in 51m 50s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/35fad7611c3545eabe5b41fc3f60a43b : SUCCESS in 1h 05m 36s","accounts_in_message":[],"_revision_number":1},{"id":"176f5776e61387bc7151cdf39ae90b56285f2551","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 17:49:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bf12475c39f93970fd575e6052e73477ab68c1f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 17:50:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6723fb5c563541ab84bfdf5bf3e3fd70 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1},{"id":"0726d219b06b33db6e08acb1486e583383aca845","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 20:13:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/fc8b3e7e47b14f5f82b26614b0c608a2 : SUCCESS in 1h 29m 01s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ac4c9fe5dfd94d76b24f2b66481d4f9a : SUCCESS in 1h 33m 22s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a763522017e145878b81d14f52af3f79 : FAILURE in 1h 30m 19s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/7f374af1156b44a698421c4d9341708c : SUCCESS in 1h 27m 28s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b2d888368b0d850609975800d5df17cf5ca78124","revisions":{"b2d888368b0d850609975800d5df17cf5ca78124":{"kind":"REWORK","_number":1,"created":"2020-11-30 12:48:59.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/62/764662/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/764662/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/764662/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/764662/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/764662/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/764662/1"}}},"commit":{"parents":[{"commit":"07dd933043f5ddb7284a1942873d66345127410c","subject":"CI: Ensure docker-py is installed on publish","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07dd933043f5ddb7284a1942873d66345127410c"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2020-11-30 12:48:23.000000000","tz":60},"committer":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2020-11-30 12:48:23.000000000","tz":60},"subject":"horizon: set FORCE_GENERATE to no if not set","message":"horizon: set FORCE_GENERATE to no if not set\n\nSigned-off-by: Christian Berendt \u003cberendt@betacloud-solutions.de\u003e\nChange-Id: I303ee5367aba0ae114e52b7b559faf06133f2304\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b2d888368b0d850609975800d5df17cf5ca78124"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b2d888368b0d850609975800d5df17cf5ca78124"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
