)]}'
{"id":"openstack%2Fkolla~765146","triplet_id":"openstack%2Fkolla~master~I67081a37977dd6de37a67fcd012e002f2e46743f","project":"openstack/kolla","branch":"master","topic":"bug/1906635","hashtags":[],"change_id":"I67081a37977dd6de37a67fcd012e002f2e46743f","subject":"Install iptables in ironic-inspector container","status":"MERGED","created":"2020-12-02 14:22:04.000000000","updated":"2020-12-22 07:56:57.000000000","submitted":"2020-12-22 07:54:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765146-bug/1906635","meta_rev_id":"85cfff2e1822e0f004995ddaffb835380e0f0a94","_number":765146,"virtual_id_number":765146,"owner":{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-22 07:54:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-12-21 10:03:46.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-12-22 06:36:02.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-12-20 02:31:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-12-22 06:36:02.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":32762,"name":"Zahid Hasan","display_name":"akkim31","email":"bdkingid@gmail.com","username":"akkim31","status":"akkim31"}],"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-02 14:48:57.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-12-02 15:47:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-02 17:38:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-03 11:28:12.000000000","updated_by":{"_account_id":32762,"name":"Zahid Hasan","display_name":"akkim31","email":"bdkingid@gmail.com","username":"akkim31","status":"akkim31"},"reviewer":{"_account_id":32762,"name":"Zahid Hasan","display_name":"akkim31","email":"bdkingid@gmail.com","username":"akkim31","status":"akkim31"},"state":"CC"},{"updated":"2020-12-08 08:59:28.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-12-18 19:46:41.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"6f628d777041a236f370599de6ccb20262b22f85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"},"date":"2020-12-02 14:22:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0540b6096f7ef8a74cfe8071cdae08aafd2b2789","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-12-02 14:48:57.000000000","message":"Patch Set 1: Code-Review-1\n\nWhy?","accounts_in_message":[],"_revision_number":1},{"id":"a67ee08a950f4359dde8a9140d6e90f74c0a2e77","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 15:47:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/bda7d58174f54ae9bb4a5bd535b4834c : SUCCESS in 50m 27s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e020710a8a6e46bbbc9c1da2af32fb08 : SUCCESS in 35m 37s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/37afded5bf4d4197a22413d5678a0f1c : SUCCESS in 1h 24m 15s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/92ab1c6511d9438090dcf405e9583285 : SUCCESS in 46m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"58abed1f925295b2ac5c666aea1cd0aa27270a27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 17:38:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/4ef8da7518a54be6b7d671c8b88631ad : SUCCESS in 24m 42s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/0848b5575f67482599eac08b60100706 : SUCCESS in 38m 50s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/239bc16274644a58894dd1bd3a656681 : SUCCESS in 1h 02m 52s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/988abf5b17e045f795bca3f0413dad69 : SUCCESS in 1h 06m 27s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/454df6ce81b04dc19e1006a97e3e977f : SUCCESS in 1h 09m 38s\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/34b8324167cd4d40bfc893ac8e48f2e8 : SUCCESS in 1h 03m 29s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/68f45e92a95a471597eaa7c29225a731 : SUCCESS in 3h 08m 31s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/de775317035a4e48be8c405938b6d7fb : TIMED_OUT in 3h 04m 17s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e8e3dd8edd214b3fbc6c086da2cbcd94 : SUCCESS in 40m 39s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/354e5e6a822d46a0abefa71b717704f8 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3d6d27964a64ec28d1a069c10b9cb14 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/973763dfe1894b08bf2d5bf6df9f92e4 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ba3fee8855b42be89b95cad79404f0f : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44647c542c194f958f9b7b01b2ae7576 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce346f9d05ef438e8d7db0038fac6496 : SUCCESS in 6m 22s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/963b85295067486b923b428f789fbc43 : SUCCESS in 19m 37s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4cb75c35becd4db4a011fc4355fea8a9 : SUCCESS in 35m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5122e0889d3d452db639b460725b535a : SUCCESS in 53m 20s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2a7b47b929f448319b06cec391b8a59e : SUCCESS in 51m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5ae8d58318ec428f943ab32059b540b9 : SUCCESS in 1h 03m 05s\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/b9fdffccd9be49c0a18e4ff2458f6ac9 : SUCCESS in 55m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"85e7d5fc0107d17c61195b7a930a172dc43c4d8e","author":{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"},"date":"2020-12-03 07:21:42.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e Why?\nI use kolla-ansible deploy ironic-inspector container in centos8 have same problem with \nhttps://bugs.launchpad.net/kolla/+bug/1626539\n\nSo i think centos need install iptables rpm package.","accounts_in_message":[],"_revision_number":1},{"id":"6f0cc6f94e9076189a7f651147bf7f8fb1c31395","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-12-03 07:49:45.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1: Code-Review-1\n\u003e \u003e \n\u003e \u003e Why?\n\u003e I use kolla-ansible deploy ironic-inspector container in centos8 have same problem with \n\u003e https://bugs.launchpad.net/kolla/+bug/1626539\n\u003e \n\u003e So i think centos need install iptables rpm package.\n\nUnderstood, please update the commit message with this explanation and reference the bug (contributor guide: https://docs.openstack.org/contributors/common/git.html#commit-messages)","accounts_in_message":[],"_revision_number":1},{"id":"a4a947841252d19d9e6f5d2ba8e58930b1f7874e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"},"date":"2020-12-03 10:33:00.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ec1e888e91e41846bd41e76bc3ac250028df92da","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-12-03 10:36:07.000000000","message":"Patch Set 2:\n\nThat\u0027s a very old bug, supposedly fixed in 2016 - can you please raise a new one with proper outputs in there?","accounts_in_message":[],"_revision_number":2},{"id":"56a4862343c28767f55513cf6f991039f19fdeb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"},"date":"2020-12-03 10:41:27.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"167faa34432b1a6efe51aea15dac7859b46819d0","author":{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"},"date":"2020-12-03 10:45:46.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 2:\n\u003e \n\u003e That\u0027s a very old bug, supposedly fixed in 2016 - can you please raise a new one with proper outputs in there?\n\nOf course,Where should I put the error log?","accounts_in_message":[],"_revision_number":3},{"id":"667da5b9e6f67eaed8ab4e13161b5afedbd3c48a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"},"date":"2020-12-03 11:02:44.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"47f6ee3a162ae08f9b412e0218d7cfad5ec0e5fd","author":{"_account_id":32762,"name":"Zahid Hasan","display_name":"akkim31","email":"bdkingid@gmail.com","username":"akkim31","status":"akkim31"},"date":"2020-12-03 11:28:12.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e Why?","accounts_in_message":[],"_revision_number":4},{"id":"b80d4fc25c3555f1791d8878e781e533e2506c2f","author":{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"},"date":"2020-12-03 11:34:29.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 3:\n\u003e \n\u003e \u003e Patch Set 2:\n\u003e \u003e \n\u003e \u003e That\u0027s a very old bug, supposedly fixed in 2016 - can you please raise a new one with proper outputs in there?\n\u003e \n\u003e Of course,Where should I put the error log?\n\nI add a bug.\nhttps://bugs.launchpad.net/kolla/+bug/1906635","accounts_in_message":[],"_revision_number":4},{"id":"b32979b1907aa7853786057440cc1623fe0a36ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 12:20:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/3414b8cec46145f1b5102aaad5bdbf93 : SUCCESS in 25m 17s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8cf4977892fd41788f304cc00d73823e : SUCCESS in 37m 04s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/35f0599fb7cd4406958c61d309e1cbce : SUCCESS in 1h 12m 57s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/2a7704d792314e2eba6da7df4e2994dc : SUCCESS in 1h 10m 03s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/9e9bb6204f384b19a71ac3cc0ddc17da : SUCCESS in 1h 06m 15s\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/fad0f6e840a14c73a558de3f8ea3d6da : SUCCESS in 1h 08m 11s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/786382d31f5849b4998d362f2a311489 : SUCCESS in 34m 21s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/02076a95d79c4a66951b6e64a4924090 : SUCCESS in 17m 15s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7dce4be87c2e4b10866506a89d581eb1 : SUCCESS in 59m 24s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc317b5cdf1542ed8554bab70bd74edc : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7667562649294327b248e28da9ffdc05 : SUCCESS in 7m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01624a8b44da449888e234eb0459b1a6 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67ed8d4ff0f84f99af6c5f7c0ad90599 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6a72729db7b497c8fe76bf5e28eedc0 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1bf44d7841b42ccb3cb76ef6114c5f8 : SUCCESS in 6m 53s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c126aa4bc5c14426b9519375f108cb14 : SUCCESS in 21m 11s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/463ad6eb66b84de685115e0a00467296 : SUCCESS in 37m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/84c376596c384ec5b7a97da7eca1b92b : SUCCESS in 55m 46s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d6a402216baa4ad1bb94131397f4e487 : SUCCESS in 51m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/03402474aa3e422db58b42f8c0a35a50 : SUCCESS in 1h 10m 51s\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/8485c23519ed4b0a8f87e001f971c7b4 : SUCCESS in 54m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"13aa0a5a961a5bd2b89246a82a23ff7623724bb4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 12:35:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/542030b35e1544238000441865e3c428 : SUCCESS in 49m 53s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4d81b48a68204ac2bf6290ce6ab2b11b : SUCCESS in 36m 49s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c54561cd27044aac83ace74301fd6917 : SUCCESS in 1h 26m 59s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/78958507f708441d9581226aec906dfd : SUCCESS in 45m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"02176159f296c4aa5d0c8aff821476098ea4acac","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-08 08:59:28.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"1b4f5e3d4d61d96b28177686198df111292d3359","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-18 19:46:41.000000000","message":"Patch Set 4: Code-Review-1\n\nThis is missing doing the same for the source branch.","accounts_in_message":[],"_revision_number":4},{"id":"d7a253c3cdffa8ac41eaa5c3fad26248368c46ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"},"date":"2020-12-20 02:31:14.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c8b7334b8accdc86f59d86e25d5010f46978a47e","author":{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"},"date":"2020-12-20 02:31:45.000000000","message":"Patch Set 5: Code-Review+1\n\n\u003e Patch Set 4: Code-Review-1\n\u003e \n\u003e This is missing doing the same for the source branch.\n\nDone","accounts_in_message":[],"_revision_number":5},{"id":"7b309ca958e0f4cc77a8de83e987af003ea27f0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-20 03:53:17.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e7d75b688fc8445eac83c7b35c66c981 : SUCCESS in 25m 26s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/1520a7c493b6441783f11c4f65506e4d : SUCCESS in 51m 22s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/77a07e71ecf74bc5af648619ebd9f974 : SUCCESS in 1h 18m 03s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/1ce62ea7600d4e20aa3865d3933c7c6b : SUCCESS in 56m 02s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/2bcee56bc2464f03aac3ba14cc5c1e4e : SUCCESS in 58m 47s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bc3b9b2504a44c129dbbe4eb8e1bda2f : SUCCESS in 24m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/82204c4fc0c347f589074546844c0a0d : FAILURE in 18m 02s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c54a2acb0714490b833ef884ce2da389 : SUCCESS in 51m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25451dd262454cdead3bcec6d640778f : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be600b5366b241b490c58d99b1538b1c : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cec63cfc8ea24c83a2e8cfa288b1f1c1 : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c600c3cd1244b25ba42b39ef98a9951 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b169433b9fe24023809cbabcf7666586 : SUCCESS in 5m 15s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9719d92367a345129cb80f17b850357b : FAILURE in 18m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/74adb1fafc694d6299cccbc761a221d6 : FAILURE in 29m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e7836e5b8f8644b4abf1782feb149bf0 : SUCCESS in 47m 53s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6806733cfe7d4198af1b157b08756d0c : SUCCESS in 46m 53s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3a1a9d2c0d2b4e60bd596fcdfb907d4b : SUCCESS in 1h 13m 07s","accounts_in_message":[],"_revision_number":5},{"id":"f65acc39f7669dd6a7d9ac52c82e417b53ca977b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-20 03:59:47.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/ee7e9bcc69c0415d886461692460a16c : SUCCESS in 46m 32s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5193b1ecec064d3f95c2f37bad3d04c1 : SUCCESS in 40m 28s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0aa0f6ed8a324e3299f453f45cc64914 : SUCCESS in 1h 22m 58s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/af827584716f4e2682b07eac36c7b0bd : SUCCESS in 32m 16s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c53e7ff41fa22485215f4ec54ce390cc1963ca7b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-21 10:03:46.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"30c37afa8997b56be0e9c5f9edc4af54ceabfcaf","author":{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"},"date":"2020-12-22 01:27:28.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"bc398b52ebe593550162c368754697d031c38135","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 02:49:33.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/8ecf72dc36c840de8b532cc66b271afc : SUCCESS in 23m 09s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/0b0797bc4c3c4346a3cf32a1b068aa80 : SUCCESS in 38m 03s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/46363b95b94f42e4b1a2b96bb9c19af8 : SUCCESS in 59m 43s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/0427b37bca344b50b46142900f5e3bdc : SUCCESS in 1h 02m 15s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/2226d69874b143099c41e60e0d381c95 : SUCCESS in 1h 17m 15s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3b17cbc362884aff9d04fa80d1af5893 : SUCCESS in 32m 27s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f3477e9f92e8401495c0b325bd6d74b4 : SUCCESS in 18m 00s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ad9dd453b3a047d29635150259c98ba9 : SUCCESS in 1h 02m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56ef73f16ad04475a7013e1caa0f5e6f : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da2a017bc01849e399903be458195bdd : SUCCESS in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1dd9c8ebca75470cbd27897662784bda : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cda611b889754ff481dfab77b0c70c79 : SUCCESS in 3m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19849f80eac74d8abd31ff2390870222 : SUCCESS in 5m 20s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b0088ac535e349e193f7638682a8d776 : SUCCESS in 21m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7b5bd3c7d391451fa41b1de8971ff946 : SUCCESS in 27m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8f3898e41b8e49238d0c9a2fddda75d4 : SUCCESS in 58m 20s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/764ef5471f8743b6b4d3390fb2c6bcf3 : SUCCESS in 51m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4ee617b6d39c4a6d83e41bba80188a41 : SUCCESS in 1h 03m 40s","accounts_in_message":[],"_revision_number":5},{"id":"f96f1dacf3e1bec815ab52dbfd123adf8303a11e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 02:55:06.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/1e14e0b50a6f4ef69a8d0c4643b5f439 : SUCCESS in 50m 13s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2c76014bc7044c3a9987fe79d7752690 : SUCCESS in 47m 11s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/81a0c5f9761f4d728dc1410c74b683d9 : SUCCESS in 1h 22m 12s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/83d362190fa244eba204bc1d8b7956bf : SUCCESS in 46m 52s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"571e8993deafeda842803cf2095daf883d6db326","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-22 06:36:02.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nThanks, Dou!","accounts_in_message":[],"_revision_number":5},{"id":"aee9e240618b4152b410f7734aae659ee98191ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 06:36:16.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"94a9977c193683867cf4c9cd1f131084d3d85577","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 07:54:11.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/39b8aeda994a4028856dc61a74f76fc5 : SUCCESS in 34m 18s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/44d1db0da8b242ce98bed9e5b62482ba : SUCCESS in 1h 12m 39s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/8ba92f8ea0e64361b91bf04e9f4f9b17 : SUCCESS in 1h 08m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5d92574f5114c26b84ba6865d51902e : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/648e4c2be6764aa7b9324784acdc9958 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cffa099a9f80435a948f6cd393d79e3c : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6cec47b6d3e4a898f49e46a6ca85148 : SUCCESS in 5m 40s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8c2a1757caac49f59aadffeac34af285 : SUCCESS in 30m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cad70a10b2ec4ec3a8b48aeb43e5b870 : SUCCESS in 48m 24s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7352c5cf040f48cca5b9c79bcdcb55f3 : SUCCESS in 1h 03m 40s","accounts_in_message":[],"_revision_number":5},{"id":"528823a17d1ef86e7526b9b021e54b102c5c10f8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 07:54:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"85cfff2e1822e0f004995ddaffb835380e0f0a94","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 07:56:57.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3336cf9d91f4d249df187e376c95b75 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"ad1b8e20dfc6f743f1945b9e036cf29597b79053","revisions":{"fdb977a8e8eb6f5b231114582f84fea96a378e0c":{"kind":"REWORK","_number":1,"created":"2020-12-02 14:22:04.000000000","uploader":{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"},"ref":"refs/changes/46/765146/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/765146/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/765146/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/765146/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/765146/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/765146/1"}}},"commit":{"parents":[{"commit":"e0835803d7d3dcf5a33ffa2885275292ff5b6173","subject":"Merge \"docs: stable branch lifecycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e0835803d7d3dcf5a33ffa2885275292ff5b6173"}]}],"author":{"name":"douyali","email":"douyali@troila.com","date":"2020-12-02 14:17:48.000000000","tz":0},"committer":{"name":"douyali","email":"douyali@troila.com","date":"2020-12-02 14:22:03.000000000","tz":0},"subject":"ironic-inspector install iptables","message":"ironic-inspector install iptables\n\nChange-Id: I67081a37977dd6de37a67fcd012e002f2e46743f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fdb977a8e8eb6f5b231114582f84fea96a378e0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fdb977a8e8eb6f5b231114582f84fea96a378e0c"}]},"branch":"refs/heads/master"},"d19a8c322f01142a27471e46ce24fd27eba5213c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-12-03 10:33:00.000000000","uploader":{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"},"ref":"refs/changes/46/765146/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/765146/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/765146/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/765146/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/765146/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/765146/2"}}},"commit":{"parents":[{"commit":"e0835803d7d3dcf5a33ffa2885275292ff5b6173","subject":"Merge \"docs: stable branch lifecycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e0835803d7d3dcf5a33ffa2885275292ff5b6173"}]}],"author":{"name":"douyali","email":"douyali@troila.com","date":"2020-12-02 14:17:48.000000000","tz":0},"committer":{"name":"douyali","email":"douyali@troila.com","date":"2020-12-03 10:29:52.000000000","tz":0},"subject":"Install iptables in ironic-inspector container","message":"Install iptables in ironic-inspector container\n\nChange-Id: I67081a37977dd6de37a67fcd012e002f2e46743f\nCloses-bug: #1626539\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d19a8c322f01142a27471e46ce24fd27eba5213c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d19a8c322f01142a27471e46ce24fd27eba5213c"}]},"branch":"refs/heads/master"},"1f763b4d4bdbff2766981fb75cb44b47a23ded6b":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-12-03 10:41:27.000000000","uploader":{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"},"ref":"refs/changes/46/765146/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/765146/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/765146/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/765146/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/765146/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/765146/3"}}},"commit":{"parents":[{"commit":"e0835803d7d3dcf5a33ffa2885275292ff5b6173","subject":"Merge \"docs: stable branch lifecycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e0835803d7d3dcf5a33ffa2885275292ff5b6173"}]}],"author":{"name":"douyali","email":"douyali@troila.com","date":"2020-12-02 14:17:48.000000000","tz":0},"committer":{"name":"douyali","email":"douyali@troila.com","date":"2020-12-03 10:34:19.000000000","tz":0},"subject":"Install iptables in ironic-inspector container","message":"Install iptables in ironic-inspector container\n\nThe centos environment ironic-inspector use iptables filter need\ninstall iptables rpm package.\n\nChange-Id: I67081a37977dd6de37a67fcd012e002f2e46743f\nCloses-bug: #1626539\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1f763b4d4bdbff2766981fb75cb44b47a23ded6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1f763b4d4bdbff2766981fb75cb44b47a23ded6b"}]},"branch":"refs/heads/master"},"3173a7fef92ba75f096c29c992f10e4adb3ba796":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-12-03 11:02:44.000000000","uploader":{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"},"ref":"refs/changes/46/765146/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/765146/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/765146/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/765146/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/765146/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/765146/4"}}},"commit":{"parents":[{"commit":"e0835803d7d3dcf5a33ffa2885275292ff5b6173","subject":"Merge \"docs: stable branch lifecycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e0835803d7d3dcf5a33ffa2885275292ff5b6173"}]}],"author":{"name":"douyali","email":"douyali@troila.com","date":"2020-12-02 14:17:48.000000000","tz":0},"committer":{"name":"douyali","email":"douyali@troila.com","date":"2020-12-03 11:01:40.000000000","tz":0},"subject":"Install iptables in ironic-inspector container","message":"Install iptables in ironic-inspector container\n\nThe centos environment ironic-inspector use iptables filter need\ninstall iptables rpm package.\n\nChange-Id: I67081a37977dd6de37a67fcd012e002f2e46743f\nCloses-bug: #1906635\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3173a7fef92ba75f096c29c992f10e4adb3ba796"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3173a7fef92ba75f096c29c992f10e4adb3ba796"}]},"branch":"refs/heads/master"},"ad1b8e20dfc6f743f1945b9e036cf29597b79053":{"kind":"REWORK","_number":5,"created":"2020-12-20 02:31:14.000000000","uploader":{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"},"ref":"refs/changes/46/765146/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/765146/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/765146/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/765146/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/765146/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/765146/5"}}},"commit":{"parents":[{"commit":"e0835803d7d3dcf5a33ffa2885275292ff5b6173","subject":"Merge \"docs: stable branch lifecycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e0835803d7d3dcf5a33ffa2885275292ff5b6173"}]}],"author":{"name":"douyali","email":"douyali@troila.com","date":"2020-12-02 14:17:48.000000000","tz":0},"committer":{"name":"douyali","email":"douyali@troila.com","date":"2020-12-20 02:31:00.000000000","tz":0},"subject":"Install iptables in ironic-inspector container","message":"Install iptables in ironic-inspector container\n\nThe centos environment ironic-inspector use iptables filter need\ninstall iptables rpm package.\n\nChange-Id: I67081a37977dd6de37a67fcd012e002f2e46743f\nCloses-bug: #1906635\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad1b8e20dfc6f743f1945b9e036cf29597b79053"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad1b8e20dfc6f743f1945b9e036cf29597b79053"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
