)]}'
{"id":"openstack%2Fkolla~765549","triplet_id":"openstack%2Fkolla~master~I79dcda5303a49de585d55c06c55881f9f1ee1412","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I79dcda5303a49de585d55c06c55881f9f1ee1412","subject":"kolla-toolbox: Add openstack.cloud collection","status":"ABANDONED","created":"2020-12-04 15:15:12.000000000","updated":"2021-03-26 09:35:52.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"8f1ac3cdf648d73e62e22d1481262cecd6f56653","_number":765549,"virtual_id_number":765549,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-12-15 18:01:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-04 16:55:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-04 17:28:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-07 08:59:29.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-12-07 08:59:29.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-12-07 08:59:29.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"89d53de7a36cc3da9675b271d3b1001ff1e8e65c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-12-04 15:15:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9c27a2bab1202d26a47df153228df1eb4deaa174","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-12-04 15:16:34.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"cd881479606da5462e0113aa7ccd3c440ba62bf2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 16:55:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/222206fd2e304668b5134e4c19d70d9b : SUCCESS in 46m 16s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0cd0b11ea33d43c082d10727f0ad3591 : SUCCESS in 46m 40s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1e0677ae393b4859b74eca074037f3e5 : SUCCESS in 1h 27m 23s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/893a53acaae84360993f713780fac9d4 : SUCCESS in 47m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a955fbb2845fa26ddc000b198c97580eafa86268","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 17:28:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/dc8a978e6f2247ecb4edfe914511d63e : SUCCESS in 30m 39s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/afccac7006e848aba5ab7fb7f31849dc : SUCCESS in 44m 07s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/e0ed4e6118a840bc87174f425fcfcf82 : SUCCESS in 1h 06m 32s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/9fb0bf4a36994a19afcb5e6b4f7cfb65 : SUCCESS in 59m 51s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/643ee82a54404ac28ba5413d499b2519 : SUCCESS in 1h 23m 58s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8e2eadb639e44a6f8d1a8b292ed80bbd : SUCCESS in 43m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5e5450253d9c4b36be1cd287c008ff3d : SUCCESS in 26m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/234e652e3991455f9df6e6b9b16b5368 : SUCCESS in 1h 08m 04s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e72a30f08a24cbaba529f371e5ea937 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2412724162ab48bdaad9dadbb46854aa : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f2f3eca4c4e4b50ab780da444fd2455 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3467fcf4aad4ebc914523f1a14d1195 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6fe5f1111a64934941c987ce3909efa : SUCCESS in 3m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d09fba2ff14246daa2c7bf3ec36fe97e : SUCCESS in 6m 20s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e6d81f3b30bd434a8e4e2f0ea4fdaa0f : SUCCESS in 21m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/dd6b00ec76c6428d96f501a75572ced5 : SUCCESS in 46m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/adcaddbc81924865a9ffe8c1f1e7ad31 : SUCCESS in 57m 27s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/aed17c6973724701b6ff7426ee048751 : SUCCESS in 55m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0cba1acfbd3d4a51908cb9f337185336 : SUCCESS in 1h 03m 21s","accounts_in_message":[],"_revision_number":2},{"id":"b09e3afae23cb563c385c64dffe866c3890abf0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-12-07 09:00:07.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"27b5e673c778b21d83e399a849e3d221f8783e0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-07 10:20:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/f5d93ebc9c28459d9acc3b43e5fb2871 : SUCCESS in 27m 05s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b7631ee5a1564dd68c11d44fd974be7a : SUCCESS in 43m 36s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0dbeb4497bd547e3b28de6305353566a : SUCCESS in 1h 13m 40s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/0923fbde1560435fa6f2f516afc5a253 : SUCCESS in 1h 01m 54s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/3e4c3f0838724ffc86ed481d30688082 : FAILURE in 23m 22s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/64deec710c6c45fca64306a9e3c9cffd : SUCCESS in 42m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/cf3d192ff1aa435ca4245d58995f11cd : SUCCESS in 21m 25s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d22a6cbe1d014e0396d51d5e891af2c1 : SUCCESS in 1h 00m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/76f03bbda2f54b4dac3c3dc222dbae3b : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/260d74f290074943a5052b0529f48d31 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/338a45ca399c40f3b4067cd24a627976 : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a54c7853b978478887cd109599f5aba6 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ca6d64aa5554f4d8e3d5a73e608650e : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1a200c7d0ce45db94a79cc45badc862 : SUCCESS in 5m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5789e06e1b8f473f8a1c9683572868e3 : SUCCESS in 24m 15s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9789c81a965d4a9cb4236bbd57a030ff : SUCCESS in 27m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/73adf245f7e44df4aa1d5ecddd1c9ebc : SUCCESS in 55m 38s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8501eb7eb1e640e5974affc0f8d17a27 : SUCCESS in 53m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/eaac3e95dacd4b6da8835e09e5c28c3e : SUCCESS in 1h 03m 42s","accounts_in_message":[],"_revision_number":3},{"id":"f45e69aafad5cb538ca5784365122eadbc6ddc33","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-07 10:20:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/eadda61673c040de9b9852160c64b8f2 : SUCCESS in 48m 27s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/41adec442c824b97820d255ab19090e2 : SUCCESS in 39m 00s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/79115b2b61764b2cb997040082e5aa24 : FAILURE in 1h 14m 59s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/54fde572490d4a55a6665707d84d8792 : SUCCESS in 36m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d92a5b13d31ae17777760c3a8b86289664524920","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-12-15 16:17:39.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"e0950f2a36885a8c5b50b9d499bfe71c4c0c086a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 17:52:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/fac14b562a0d4a5da0784d6a37b7df5a : SUCCESS in 52m 00s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ccb0f3a4e3824ae68c16e697cd09997b : SUCCESS in 50m 06s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/12f245678ca44099acc7d90bfa8c15dd : SUCCESS in 1h 33m 25s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/95a9ea4144c44f9b85cd457aa8a79ad9 : SUCCESS in 49m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"05919ce2ce8c0ff174022211abfd56c544924001","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 18:01:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/6e7f1582912e46b69a438641e39a7cda : SUCCESS in 32m 36s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/de20e786999d4d8490337bdf01cc6ef8 : SUCCESS in 33m 30s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0aab6eeb78524414b79d6f2aa20acc24 : SUCCESS in 1h 09m 07s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/e772c1dd1b2549d8baddecda4dc0b4c5 : SUCCESS in 1h 00m 09s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/4a8d0903e819428eba1b55b7232a1d3e : SUCCESS in 1h 33m 26s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f7831e5f07f54c6481cc5b8c3007554d : SUCCESS in 46m 37s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0d47e89569d04f678d593d87fde5830e : FAILURE in 40m 51s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7068c20d3bfe4d079093a8085a36a4b1 : SUCCESS in 1h 18m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/540c5a2987f74e6a99295dbc103ac7e3 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3895b752e9b4aed97509a4d1caaaccb : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/991e7025a2ab459daf5b5c28361a3413 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b4a9405bad941948789c682ea3e1dcd : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a6eaa8f355e4aca963e321942d0cb8a : SUCCESS in 5m 41s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4da596a71be341a2b3dc89f1985b5540 : SUCCESS in 23m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c01fbd34c26043bcb11463f6f7b9de17 : SUCCESS in 36m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/77ddc71cf66c48cba50c1cdfbab98e58 : SUCCESS in 55m 27s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b7819c2ffbe54651827001f7242f23fc : SUCCESS in 41m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7fa6b2a4101e4e78bdb580889f6c0a97 : SUCCESS in 1h 02m 34s","accounts_in_message":[],"_revision_number":3},{"id":"1e0614cfe921114cf7acbf9a7fe3e6cb0674ef37","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-16 19:47:15.000000000","message":"Patch Set 3:\n\nWhat is the plan regarding these?","accounts_in_message":[],"_revision_number":3},{"id":"6c43d91e9e3058e61e281ca58663a2632a4571c8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-17 10:56:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2783cf1ab7d2aefbe992be0507b257c409661d85","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-25 19:41:37.000000000","message":"Patch Set 3:\n\nIs this still required?","accounts_in_message":[],"_revision_number":3},{"id":"6c29f0332dac8f5c5643479ee9ee37063ea7d835","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-26 09:35:47.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3:\n\u003e \n\u003e Is this still required?\n\nI don\u0027t think so, and in X probably we want to fully switch to collections and install only ansible-core - abandoning for now.","accounts_in_message":[],"_revision_number":3},{"id":"8f1ac3cdf648d73e62e22d1481262cecd6f56653","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-26 09:35:52.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a2d9c98a690586a7218e824128fd02d7fa35bd3c","revisions":{"a6d9996dab8c6008e094a9a5f30a18fd029fc881":{"kind":"REWORK","_number":1,"created":"2020-12-04 15:15:12.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/49/765549/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/765549/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/765549/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/765549/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/765549/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/765549/1"}}},"commit":{"parents":[{"commit":"22c32a30e62d42c3631306775c0cd029eb659e9a","subject":"Unify curl options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22c32a30e62d42c3631306775c0cd029eb659e9a"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2020-12-04 15:13:36.000000000","tz":60},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2020-12-04 15:15:07.000000000","tz":60},"subject":"kolla-toolbox: Add openstack.cloud Ansible collection","message":"kolla-toolbox: Add openstack.cloud Ansible collection\n\nInstall openstack.cloud Ansible collection (see [1]) to expose additional\nAnsible modules that can be used with OpenStack.\n\n[1]: https://opendev.org/openstack/ansible-collections-openstack\n\nChange-Id: I79dcda5303a49de585d55c06c55881f9f1ee1412\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a6d9996dab8c6008e094a9a5f30a18fd029fc881"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a6d9996dab8c6008e094a9a5f30a18fd029fc881"}]},"branch":"refs/heads/master"},"27e208c45d2ac1be6d88832c37bbc0cded6edba1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-12-04 15:16:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/49/765549/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/765549/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/765549/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/765549/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/765549/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/765549/2"}}},"commit":{"parents":[{"commit":"22c32a30e62d42c3631306775c0cd029eb659e9a","subject":"Unify curl options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22c32a30e62d42c3631306775c0cd029eb659e9a"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2020-12-04 15:13:36.000000000","tz":60},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2020-12-04 15:16:14.000000000","tz":60},"subject":"kolla-toolbox: Add openstack.cloud collection","message":"kolla-toolbox: Add openstack.cloud collection\n\nInstall openstack.cloud Ansible collection (see [1]) to expose additional\nAnsible modules that can be used with OpenStack.\n\n[1]: https://opendev.org/openstack/ansible-collections-openstack\n\nChange-Id: I79dcda5303a49de585d55c06c55881f9f1ee1412\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/27e208c45d2ac1be6d88832c37bbc0cded6edba1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/27e208c45d2ac1be6d88832c37bbc0cded6edba1"}]},"branch":"refs/heads/master"},"a2d9c98a690586a7218e824128fd02d7fa35bd3c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-12-07 09:00:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/49/765549/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/765549/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/765549/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/765549/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/765549/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/765549/3"}}},"commit":{"parents":[{"commit":"22c32a30e62d42c3631306775c0cd029eb659e9a","subject":"Unify curl options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22c32a30e62d42c3631306775c0cd029eb659e9a"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2020-12-04 15:13:36.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-12-07 09:00:04.000000000","tz":0},"subject":"kolla-toolbox: Add openstack.cloud collection","message":"kolla-toolbox: Add openstack.cloud collection\n\nInstall openstack.cloud Ansible collection (see [1]) to expose\nadditional Ansible modules that can be used with OpenStack.\n\n[1]: https://opendev.org/openstack/ansible-collections-openstack\n\nChange-Id: I79dcda5303a49de585d55c06c55881f9f1ee1412\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a2d9c98a690586a7218e824128fd02d7fa35bd3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a2d9c98a690586a7218e824128fd02d7fa35bd3c"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
