)]}'
{"id":"openstack%2Fkolla~765696","triplet_id":"openstack%2Fkolla~master~Ie5e0e3720980b2d320878d12e47439068c4ec159","project":"openstack/kolla","branch":"master","topic":"doc8","hashtags":[],"change_id":"Ie5e0e3720980b2d320878d12e47439068c4ec159","subject":"Update doc8 version","status":"ABANDONED","created":"2020-12-07 01:21:23.000000000","updated":"2021-12-30 14:29:18.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3475878660e365ab83f62835c0ae7020fa0e7efc","_number":765696,"virtual_id_number":765696,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-12-07 01:33:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-07 01:33:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-09 17:33:16.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"},{"updated":"2020-12-10 08:52:12.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"}],"messages":[{"id":"8f6014f8dfde6e0b9be946cd74bdfcde541ec2e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-12-07 01:21:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb623cf6aaf1c47a246c9bef6d28376d3ebb646e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-07 01:33:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2317a05d2414b9596bf2198230fcaa9 : SUCCESS in 5m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b83f75c277034138b3da07010ede2a0f : SUCCESS in 2m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c74978b2f998454d92fa7e42c13a9760 : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/886646dbd017479eafa016923a2a987c : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":1},{"id":"6f80944e71cfbe8352a13a3d0afc9f3360f9b5f5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-12-09 17:33:16.000000000","message":"Patch Set 1:\n\nLooks like 0.6.0 works fine with Python \u003e3.4:\n\n\n(venv3) 18:31 (1s) hrw@puchatek:kolla$ doc8 --version\n0.6.0\n(venv3) 18:31 (0s) hrw@puchatek:kolla$ python3 --version\nPython 3.9.0\n(venv3) 18:31 (0s) hrw@puchatek:kolla$ doc8 doc/\nScanning...\nValidating...\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nTotal files scanned \u003d 16\nTotal files ignored \u003d 0\nTotal accumulated errors \u003d 0\nDetailed error counts:\n    - CheckCarriageReturn \u003d 0\n    - CheckIndentationNoTab \u003d 0\n    - CheckMaxLineLength \u003d 0\n    - CheckNewlineEndOfFile \u003d 0\n    - CheckTrailingWhitespace \u003d 0\n    - CheckValidity \u003d 0\n(venv3) 18:32 (0s) hrw@puchatek:kolla$","accounts_in_message":[],"_revision_number":1},{"id":"5314d3d0a27e3526fe3d1efa6d0a69977476f5a0","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-12-09 21:01:56.000000000","message":"Patch Set 1:\n\nLooks like 0.6.0 works fine with Python \u003e3.4:\n\n\n(venv3) 18:31 (1s) hrw@puchatek:kolla$ doc8 --version\n0.6.0\n(venv3) 18:31 (0s) hrw@puchatek:kolla$ python3 --version\nPython 3.9.0\n(venv3) 18:31 (0s) hrw@puchatek:kolla$ doc8 doc/\nScanning...\nValidating...\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nTotal files scanned \u003d 16\nTotal files ignored \u003d 0\nTotal accumulated errors \u003d 0\nDetailed error counts:\n    - CheckCarriageReturn \u003d 0\n    - CheckIndentationNoTab \u003d 0\n    - CheckMaxLineLength \u003d 0\n    - CheckNewlineEndOfFile \u003d 0\n    - CheckTrailingWhitespace \u003d 0\n    - CheckValidity \u003d 0\n(venv3) 18:32 (0s) hrw@puchatek:kolla$","accounts_in_message":[],"_revision_number":1},{"id":"f7bff69c00dbdfb15bb61ca1afa85bf3cd674451","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-10 08:52:12.000000000","message":"Patch Set 1:\n\nI don\u0027t mind bumping but this way we could bump all the deps (as you can\u0027t retroactively update the release package metadata). doc8 clearly works nonetheless and this does not affect anyone","accounts_in_message":[],"_revision_number":1},{"id":"3475878660e365ab83f62835c0ae7020fa0e7efc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-30 14:29:18.000000000","message":"Abandoned\n\nno update for more than a year, abandoning","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c17339faddcdd9b44f80d8c78a03431cd9f79a77","revisions":{"c17339faddcdd9b44f80d8c78a03431cd9f79a77":{"kind":"REWORK","_number":1,"created":"2020-12-07 01:21:23.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/96/765696/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/765696/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/765696/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/765696/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/765696/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/765696/1"}}},"commit":{"parents":[{"commit":"22c32a30e62d42c3631306775c0cd029eb659e9a","subject":"Unify curl options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22c32a30e62d42c3631306775c0cd029eb659e9a"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-12-07 01:19:07.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-12-07 01:19:07.000000000","tz":480},"subject":"Update doc8 version","message":"Update doc8 version\n\ndoc8 only supports python version less than 3.4\n\n[1] https://pypi.org/project/doc8/0.8.0/\n\nChange-Id: Ie5e0e3720980b2d320878d12e47439068c4ec159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c17339faddcdd9b44f80d8c78a03431cd9f79a77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c17339faddcdd9b44f80d8c78a03431cd9f79a77"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
