)]}'
{"id":"openstack%2Fkolla~765832","triplet_id":"openstack%2Fkolla~master~Ifb5948653565e2ae902783762e20e33527020efe","project":"openstack/kolla","branch":"master","topic":"1907124","hashtags":[],"change_id":"Ifb5948653565e2ae902783762e20e33527020efe","subject":"Add daxio for nvdimm scenario","status":"MERGED","created":"2020-12-07 17:23:15.000000000","updated":"2020-12-18 12:33:05.000000000","submitted":"2020-12-16 10:34:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765832-1907124","meta_rev_id":"ec1dc73c14c11b6d998011531aeb18f52865ea3f","_number":765832,"virtual_id_number":765832,"owner":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-16 10:34:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":2,"date":"2020-12-15 15:12:49.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-12-15 19:34:34.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2020-12-15 15:00:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-12-15 19:34:34.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"}],"reviewers":{"REVIEWER":[{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-07 17:28:28.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-12-07 18:19:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-07 19:06:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-08 09:07:49.000000000","updated_by":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"reviewer":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"state":"REVIEWER"},{"updated":"2020-12-08 09:07:49.000000000","updated_by":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"reviewer":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"state":"REVIEWER"},{"updated":"2020-12-08 09:07:49.000000000","updated_by":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"reviewer":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"state":"REVIEWER"},{"updated":"2020-12-15 15:12:49.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"ef23d355da3f1ccf67a9c785f6a39ef68b0581e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2020-12-07 17:23:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc128e3f444c1ba6826e1cedcd405aaf78f173c9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-07 17:28:28.000000000","message":"Patch Set 1: Code-Review-1\n\ngenerally fine but needs to fix the order to alphabetic","accounts_in_message":[],"_revision_number":1},{"id":"19da2695ad881224978d5aff77e74bffb58f7355","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-07 17:29:01.000000000","message":"Patch Set 1:\n\nalso, could use a reno","accounts_in_message":[],"_revision_number":1},{"id":"07deee41bc67021fdb46cff1791b10961835ed8b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-07 18:19:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/9274abacc47149cf8b2bc6480d4caab1 : FAILURE in 48m 18s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bcdbb6d8086844a3bcf6c293987c6434 : FAILURE in 42m 23s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/78f26de533ee4020b5eb880b2424187e : FAILURE in 28m 44s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4d783934a8e84fa0b5c9399c87eff84b : SUCCESS in 49m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fc6309f323668e681c1e7a8a6759019d293bf76f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-07 19:06:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/a5d1754b61ae407992a7ee5eb91417e5 : SUCCESS in 50m 03s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b63b29f942944b4ba74154b00c2fd066 : SUCCESS in 58m 04s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a5abb34c2a1f495c895a08c62ec47c4c : SUCCESS in 1h 29m 59s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/55901c4a5d784b229d1e9fec1f6c2d4f : SUCCESS in 1h 22m 47s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/157d24ab128b46c482f09d96c6804293 : SUCCESS in 1h 36m 55s\n- kolla-ansible-centos8-source-kvm https://zuul.opendev.org/t/openstack/build/f03d3fe27400451491199558095aafbf : SUCCESS in 1h 01m 12s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c890d0a06bba49e2aabe0d741848b4dc : FAILURE in 59m 49s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/701208bfd53244848f6dc58b0db475d6 : FAILURE in 46m 35s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9ff263f989734f6aab5640ff85a0e8d8 : FAILURE in 38m 41s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d4c84a197ab48c99030737692ba3dfa : FAILURE in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/547f4f94dcf147138426d399ee871161 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fba359af348a4772a95f501fdfcc20c2 : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53bf76548840407b951d2eaf796223cd : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26a3cddf70a542dba211199ba6ca4e89 : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cad946402b8140e6b43d2f1fcb999f9f : SUCCESS in 14m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d781445fc5574ce5b5d7cc735a2d7ece : FAILURE in 40m 55s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9e811b85074d417996e27f45f8dbebc6 : FAILURE in 34m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0a83fb4b49b54cbd917ccb5e6f177dbc : FAILURE in 16m 04s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cc183f7518264cd2a1070b04f5c3d4b3 : FAILURE in 21m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/41aea93bd29a4854a89829b60963cb3e : FAILURE in 51m 26s\n- kolla-ansible-ubuntu-source-kvm https://zuul.opendev.org/t/openstack/build/2b62519b44f0426d941ca850f78ea5f2 : FAILURE in 16m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"42dee87b88159debc750a1ce80c0f8d1d5fbd7e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2020-12-08 07:40:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6ea37fa9c3decf860d30a07da3b0dff1d7f9f1b6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-08 07:43:17.000000000","message":"Patch Set 2: Code-Review+2\n\nlovely, thanks; don\u0027t worry about the red CI","accounts_in_message":[],"_revision_number":2},{"id":"352feadf445f4d7b2e3abfe6575de6707a834b80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 08:44:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/4c3274fcda164987a1c39b8c3118bfd4 : FAILURE in 7m 04s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/bc3098b1821b4e9abb9c158b7137b22c : FAILURE in 6m 55s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f3c7cfa0bc0349678d21348dbe60f077 : FAILURE in 11m 01s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/66e75e71c2b34b3eaeda9f9d79f857c7 : FAILURE in 10m 40s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/9b3007e85e9f466ead35ba964bb7c269 : FAILURE in 10m 29s\n- kolla-ansible-centos8-source-kvm https://zuul.opendev.org/t/openstack/build/1f1bb03eb5164ce8a0707e702be5e3ee : FAILURE in 13m 08s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/cce3329fd7bd4b8b8d6ebd54fb7b3275 : FAILURE in 34m 19s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8661adc7ca254f7c85638882f9158d58 : FAILURE in 17m 05s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b88294e069d34bd69cf74ab4d743cf6a : FAILURE in 16m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1404c1d9bced48c0acd465fdbd50e5e8 : FAILURE in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/062290c6c9414280ad96fdb52ac63157 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8c4aacb005044088300c9cb13e94b1f : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9820e7d0424844fc8edb13c9101aa60e : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09dbe0590607474d8a008cd07cdc2563 : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19e12e22bcfc41c8a3c90995302a5174 : SUCCESS in 5m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2aad0fd73ac483c8b05c5e272d84ca2 : SUCCESS in 3m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/47f3dcba6d2a49b8b6697b7c9f1ca25f : SUCCESS in 21m 01s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/71a49d07c67942898b074765ca51a63f : SUCCESS in 36m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/68917f196a884050bdbff0c27d1dd08d : SUCCESS in 54m 43s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/00ec37b183354ffeae4d39cb432cb3fd : SUCCESS in 1h 02m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d3145b9f186e485a8259172f7f737eb3 : SUCCESS in 54m 24s\n- kolla-ansible-ubuntu-source-kvm https://zuul.opendev.org/t/openstack/build/baa5dad8634f4a2eb347fb5a090d1a12 : SUCCESS in 48m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b4984f14d05a8d739c075cca0f1b2c581fe46b62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 09:39:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/cb3b09c312ec4992997b21cfd38f9f68 : FAILURE in 12m 08s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6932e889499a4632aa5dad4f9c7ccf93 : SUCCESS in 44m 55s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0cbc845494834f9a900ef40f09bad8ac : SUCCESS in 1h 50m 54s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3e8593259fb94d3ca2a89cb41abd650d : SUCCESS in 49m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a4e50ab18e7dc9d8e1f9e78ec2ffc00c6d7e9d70","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2020-12-08 11:09:50.000000000","message":"Patch Set 2: Code-Review+1\n\nYep; thanks, Martin!","accounts_in_message":[],"_revision_number":2},{"id":"ab9e1b11059ab0814f1e8c5228b4c9f582d2e491","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2020-12-09 07:59:12.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"830a4ffb3a0282753681197003701f3558a32bc7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 08:47:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/68e85cbcd8fd4328a65e83bac4ad7e4a : FAILURE in 10m 28s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/eab4863ba8b34a7bae8c340c1afefa68 : SUCCESS in 44m 06s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/53244a5a6fab4150bdc559fdd1ca8125 : FAILURE in 37m 34s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/381b28965c4e45dca1c932e0da0d0541 : SUCCESS in 41m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"110cbc31f9b0a9d8d4b4989dbca4d6ad69944785","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 10:46:48.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/ca9123e7f95d4096911a11761c10d031 : FAILURE in 9m 20s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b4ba81ec1bc04312858f264487a4a412 : FAILURE in 7m 45s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/44ce7cbaeea84dc8b7c55ba5db235596 : FAILURE in 11m 39s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/498cc93a6c554890b2b21fedf368ce87 : FAILURE in 12m 23s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/862a2b232c784543836478e55e9772f3 : FAILURE in 10m 49s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9e72e2c9350f4a2c880bc9ba0966a7ac : FAILURE in 31m 41s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/cb1a10d225964837b888d8363b99b4ee : FAILURE in 17m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2fbc857a5ee04bf697bd528937f5fffd : FAILURE in 17m 21s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d44acb02b614326aeb4c206539f95eb : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87ef26c6a8cb4a58a1280e0e4332e542 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f7c6adf8f4344a0b7c12cf0acb8e54f : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/18eff26f6b2043579ec60684bc03c471 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f4b1348f48f4ff888b2a170cc3c443e : SUCCESS in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f456d6efd1149cbb5f2eb3df0811c99 : SUCCESS in 4m 15s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4ad84e65d94e41ab89d8257ce8cb1180 : SUCCESS in 2h 45m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bff74e7a8f26473e8aeec89bb10b2a4a : SUCCESS in 31m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/eb922013262d47e4a22b1e578314886e : FAILURE in 15m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f33b73b807e842b4aa57288b0e185d5f : FAILURE in 19m 04s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/abf45ef99d654cc48c84ac00a67eaf9e : FAILURE in 19m 25s","accounts_in_message":[],"_revision_number":2},{"id":"b9b5e878dd23316466e2b8d9c29b6ce65b114e29","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2020-12-11 08:07:16.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"479431d5eb9cacadd3f592641c2a8a456626043b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 09:00:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/c53d835d1db24886a48ff338ec2c873f : FAILURE in 9m 46s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7c514f3535d644eab9e900d9df9bfb45 : SUCCESS in 35m 26s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/eabf7f207f82400b9a3c18a06bf8843f : FAILURE in 31m 19s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e9386f773e0f4a9796d42aface84f385 : SUCCESS in 41m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0305403b25fa7da8d4997cd22fcdf081a94cbbbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 09:00:45.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/82db3e73c4534059afc1a3ef13175760 : FAILURE in 8m 11s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/bc03da4d609449f2b81ebdcfdbce1e0d : FAILURE in 7m 08s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c21a93b105a8477cb10fa2a35b67efb0 : FAILURE in 11m 37s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/69893a98b12841f48d614589c7122573 : FAILURE in 11m 18s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/36bb96e80ec74145bb1e84af7c15a04c : FAILURE in 12m 57s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bb5e10392e6f4fceaba6a4eafe3816e0 : FAILURE in 26m 22s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/460493fd34644aa288d09bf359e7b563 : FAILURE in 23m 53s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/93508b6ef9ac49bf9c27cc8a39538942 : FAILURE in 20m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb98426e6e924a2f8a651fc020fde546 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/144d8f3bdd8d4c3f86ed07fdf0f6ab17 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8866056ac1e40aaac7f6b01fd551e34 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/edb1ddbb4f5b4c268facb16e2d870540 : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4d29e2563744f7495ff01150b6ad934 : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1ddd4db27ba4dda9191fbb25eba815b : SUCCESS in 3m 21s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/43d80cb7eda84a4892bf8cbe26a895af : SUCCESS in 21m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/46b83b5838654f61b247774d376fc092 : SUCCESS in 37m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d43c47e93c044c5f8317e99296d0eac8 : FAILURE in 22m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9194402b234b45d2beafb875a44ab798 : FAILURE in 27m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/29becc51091e443db1d3695ddb566f8f : FAILURE in 44m 17s","accounts_in_message":[],"_revision_number":2},{"id":"e08993fa6987a25922b609bcd37babcfbd46a243","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-11 09:35:02.000000000","message":"Patch Set 2:\n\nplease do not recheck; we will do it for you when the situation improves","accounts_in_message":[],"_revision_number":2},{"id":"381f28c153b267aa25059a4c6820c0a328599062","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2020-12-11 09:36:36.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e please do not recheck; we will do it for you when the situation improves\n\nack, thanks Radoslaw!","accounts_in_message":[],"_revision_number":2},{"id":"653a4f06d32ba37a1f19adb5f56e4258a5e6459a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-11 19:02:33.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3d749748c804337f0a743faae2493dcdf4c55a0d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 19:57:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"82b907562b9768ad5e86c30081e45e5a1b907b96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 20:22:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/0874d57033a14d27b885136aed1e2100 : SUCCESS in 22m 31s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a0d12234bf0d400c9d6e5bc8be27edc1 : SUCCESS in 48m 31s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f47cf2d9f1504d7d95e2c77100d1d7cc : SUCCESS in 1h 05m 59s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/83b5adc952f9440a98105f9f3aa44341 : SUCCESS in 1h 01m 53s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/77feea3110e445adb8c4c0f2389737c9 : SUCCESS in 1h 10m 18s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/59bfc78f5f494e9bb2b03c12f0844048 : FAILURE in 32m 40s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/14fa4591f14d43229e905c8775bfe430 : FAILURE in 19m 55s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/388163a7097f41dabc317538ddceb16d : FAILURE in 13m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76c156c56324437088fb3aae77eec8d4 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/906966843fb64fdf898e95224131577e : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48a0e5ec138b4ebdaea4b7e09af34368 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7b242ad293b4e7c8207c30fc43c6385 : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd116456eb234ae88def80039aba676d : SUCCESS in 8m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e35a6d465d24c4f8cfa138d39e860f1 : SUCCESS in 4m 02s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/aba9cdf03d2a4886997807d2d3ca7867 : SUCCESS in 19m 41s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c70418e96f8a4882835c5bfda14b8f33 : SUCCESS in 37m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b61f2367a2bd46b0877e983d8de659d1 : SUCCESS in 55m 43s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e2887d8acc6542a28c15af5d4c740373 : SUCCESS in 50m 40s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3a9122138aab4d15be8e0190d552f029 : SUCCESS in 1h 03m 00s","accounts_in_message":[],"_revision_number":2},{"id":"1abd9b99f9414a5dfa1f0df52f49ee5195dc1be9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-11 20:28:12.000000000","message":"Patch Set 2: Code-Review-1\n\nbreaks debian","accounts_in_message":[],"_revision_number":2},{"id":"02c2a8cba8a9d7b9675a25a6953983c2069cf985","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2020-12-15 12:19:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"aab1ed3725667bacc9fa41e68653dbf5010fa221","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2020-12-15 12:21:25.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2: Code-Review-1\n\u003e \n\u003e breaks debian\n\nissue is that libpmem1 is mapped to buster-backports and not mapping also pmdk-tools to buster-backports results in the dep issues","accounts_in_message":[],"_revision_number":2},{"id":"d98c140cdc9a8cff93b3dc99e66fea45b27e3249","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-15 12:23:05.000000000","message":"Patch Set 3: Code-Review+1\n\nthanks, that makes sense; let\u0027s see if it builds","accounts_in_message":[],"_revision_number":3},{"id":"699ac1f7f0ae106663fa36948a943cec4db595c8","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2020-12-15 13:05:51.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3: Code-Review+1\n\u003e \n\u003e thanks, that makes sense; let\u0027s see if it builds\n\nI tested it locally, but lets see","accounts_in_message":[],"_revision_number":3},{"id":"9bf249b14503bbb0acab5ec712c6ea110787a199","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2020-12-15 13:21:26.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3:\n\u003e \n\u003e \u003e Patch Set 3: Code-Review+1\n\u003e \u003e \n\u003e \u003e thanks, that makes sense; let\u0027s see if it builds\n\u003e \n\u003e I tested it locally, but lets see\n\nurgs, no I had pinned libpmem1 to 1.5.1 ... we need to pin all libpmem* libs to buster-backports.\n\nwill update it shortly","accounts_in_message":[],"_revision_number":3},{"id":"9a41282288932a4eeb79e7d38843da44a312400a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2020-12-15 13:22:21.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2ef2b63b09bb5209540b9303d1d4a7729b46003c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 14:48:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/1a1f8e6006364c9abf28c4115e52e1a6 : SUCCESS in 30m 34s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/613b3c0ad4724ec7a9c0a718ba38a2b3 : SUCCESS in 33m 03s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/86d717dcf4714325b5ada59403895785 : SUCCESS in 1h 05m 31s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/8baa759abf7a4b9eb217a628b635a62e : SUCCESS in 1h 20m 19s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/fab90869593e43ab9aa6214a4c86a301 : SUCCESS in 1h 19m 10s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5933cb202f1e42e8b6c679cc861f190c : SUCCESS in 35m 03s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/738475f27d014c79b6da57e1c9db6c05 : FAILURE in 25m 20s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/eb9e8fb797f74e65a40ced24ca236473 : SUCCESS in 1h 00m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bf488b41b3d4fd58af4020f0ab36d86 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/165c9e421e3141b695e89f462aa642a7 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19c8da29ebb744aaae24c4b1bf7e2dba : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ae06e9c269047009144fdf58501f2c2 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c252a7353a804f6184a83fedfb772ae9 : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b35df91864e47d2b0a7013153a9d66c : SUCCESS in 3m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f662a7ee46404aea85815e526db0805f : SUCCESS in 19m 37s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e25769e37d1843d2b5114b3667fa485c : SUCCESS in 33m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7e1603ddc9ee4ce1abb2f70eefbdfd99 : SUCCESS in 54m 03s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d97a3d8dfca049588b0fcf957deae903 : SUCCESS in 52m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b93a7adb1b95442b8eb3a5e9473f6387 : SUCCESS in 59m 27s","accounts_in_message":[],"_revision_number":4},{"id":"5c357f4c3fcfcaeffc5dd11d8c6f28e98827844a","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2020-12-15 15:00:52.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"6594cc8c2633d58d7b0f2631b6dd527e64acd68f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-12-15 15:12:49.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"78366a81a85d5c2871cc0fa5375d5aba88997b59","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 15:30:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/709237e0f27d4b04ab793741ee16c021 : FAILURE in 54m 27s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/02be7dea6a2941b9b1a48d2b814d437c : SUCCESS in 43m 25s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ca9bb92b7c47421cb1bac2e6442ec235 : SUCCESS in 2h 02m 10s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/730e528e95664bbd836ed174c5c87705 : SUCCESS in 47m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cdc8474e1876e85622383b558ce6e4984d32c4b4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-15 19:34:34.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"e6cb179ec0b2d8d2a54d042dfd000f433f0aa800","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 19:34:54.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2908964c46564d1f7b0f69096934e32d981bf103","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 22:59:42.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/093648009e98493b86df08d5f4d7bba3 : TIMED_OUT in 3h 22m 02s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c3e642e8a0db48c4901ddb269d71cd64 : SUCCESS in 58m 29s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/b99f40916b0645b3abe31810bb8eaf44 : SUCCESS in 1h 08m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9078d3054944b439b7db3abf27b62fc : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83045cb0cc2e445b82b00a6b215197f3 : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99816f9b68064897878377b9ea8f5ff1 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf8e7d0b0ad14c0f9ee436b9bce033c0 : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c400edf925e648e5b3a1cc56b748558d : SUCCESS in 3m 51s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d28d7d27a1664b5cabf364ffb1cee471 : SUCCESS in 40m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e259a633d0f44b4d9837d42f2eda6bdc : SUCCESS in 53m 31s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/41728f4a9b5341e9acb887cace4a55fb : SUCCESS in 1h 05m 16s","accounts_in_message":[],"_revision_number":4},{"id":"fd8cea575d95640f16fe35e028e82c40b8a1f262","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2020-12-16 06:45:06.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"f878c45c141941f5c64323b8399e586f979ffdf3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 08:57:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/5b1157897f5c495194ee6c2f3e410e8c : FAILURE in 1h 18m 38s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ad10c89f8baf46d5be005412d166924a : SUCCESS in 1h 11m 31s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/427587d253ad4727927871cde44367ff : SUCCESS in 2h 01m 15s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/cdaf107989b6439fb4336ad872cf9050 : SUCCESS in 1h 11m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b9d2a8b7eba886c448ab30128d891f3aa4afbe69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 09:17:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/4654a0d398854cf493b2b61867a42889 : SUCCESS in 21m 43s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d65304c60b184fe0aa81433b3f3b6ef8 : SUCCESS in 39m 45s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/3b52da32ffba497ba53cf733defac09d : SUCCESS in 56m 52s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/a8e8935e6ae34448b04462179e7d3614 : SUCCESS in 1h 02m 40s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/9d1b31108f7a4e7d911a084997e4eb9f : SUCCESS in 1h 23m 38s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/dba40df1a4e348248fbeb3ba19b05ac0 : SUCCESS in 33m 52s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/831a95b206844074af64fe2bcc983f04 : FAILURE in 19m 50s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7d757c6e2ca045b1bec8b33ffc7a3f52 : SUCCESS in 55m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d391dc3b36ca4711ad8d842d0cc36e5a : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8391d837ab804bf5a6b13a32daaf473e : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae7273d120e842068819d883980ea912 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c994dcd450a6490f976a93a18ca236bf : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2508cd6895f14a83a9bb6a44d7ab1d29 : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a62894cd11294376a05c905b80bae763 : SUCCESS in 3m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c5df3fe70e5a4add90c4414dd994c1e6 : SUCCESS in 21m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/129582e045114402969e0ff5981655e6 : SUCCESS in 35m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7523e5869bec4ec5af9961ed250b7ef9 : SUCCESS in 52m 00s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ac8acab4f992494483203a4d2cd4693b : SUCCESS in 53m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e136164c158544db9308402115153f30 : SUCCESS in 56m 08s","accounts_in_message":[],"_revision_number":4},{"id":"cc3aad7033545d7d94247ccef21a36d50db5dad6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 09:17:42.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e6cc8054e020ca34fd1c1354ad69d0549fbe6998","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 10:34:01.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/32bbd67d25604f7593af2d74b6301239 : SUCCESS in 42m 26s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/ad1e98b18666472dafbce0d361c5ad87 : SUCCESS in 1h 10m 57s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/86f3db7c74f948c890c69a54178160dd : SUCCESS in 1h 10m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5342739f28b64bdbb2bd749ae7d1f30a : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/491d32a3308246f29bdb4c2d7850f0e3 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8639354c290943188631ab63ae546c58 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02f10457469f4bab9bb57cc86f571672 : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f5f227c99be4c0aaefb37d4149f47e2 : SUCCESS in 4m 20s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b7c253be6f174ca99aee1d69dbf575b7 : SUCCESS in 1h 05m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a7c982224dba4c86835f7025f4d622aa : SUCCESS in 46m 17s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0b4b5f3a56d94ccc923914e7b0471dc9 : SUCCESS in 1h 06m 24s","accounts_in_message":[],"_revision_number":4},{"id":"191eee25f035c6ee7f31a54ff54707960007f007","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 10:34:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"36eeed7fec4d3d79b04e046b961c0331df7bfa32","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 10:35:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14065c7746094ef891934a675575d413 : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f39938366a16479fa0c4f65db217fd8a : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":4},{"id":"bc7fea8d5ff57b94a329478b038a1f7ab7f62f25","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-12-17 23:16:24.000000000","message":"Patch Set 4:\n\nWill have to add \"if base_arch \u003d\u003d x86-64\" for centos 8...","accounts_in_message":[],"_revision_number":4},{"id":"9d76cbfa97b410b0d2d372d88c3eb5bcb9034317","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-12-18 10:42:13.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 4:\n\u003e \n\u003e Will have to add \"if base_arch \u003d\u003d x86-64\" for centos 8...\n\nhttps://review.opendev.org/c/openstack/kolla/+/767689","accounts_in_message":[],"_revision_number":4},{"id":"ec1dc73c14c11b6d998011531aeb18f52865ea3f","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2020-12-18 12:33:05.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 4:\n\u003e \n\u003e \u003e Patch Set 4:\n\u003e \u003e \n\u003e \u003e Will have to add \"if base_arch \u003d\u003d x86-64\" for centos 8...\n\u003e \n\u003e https://review.opendev.org/c/openstack/kolla/+/767689\n\nThanks, sorry missed that","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6176d54179109e85dee5ebc29a7726bcb05f76a9","revisions":{"3a9721d23d179d6dc24cb804552a58f50e7ab54a":{"kind":"REWORK","_number":1,"created":"2020-12-07 17:23:15.000000000","uploader":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"ref":"refs/changes/32/765832/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/32/765832/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/765832/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/765832/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/765832/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/32/765832/1"}}},"commit":{"parents":[{"commit":"22c32a30e62d42c3631306775c0cd029eb659e9a","subject":"Unify curl options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22c32a30e62d42c3631306775c0cd029eb659e9a"}]}],"author":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2020-12-07 17:16:07.000000000","tz":60},"committer":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2020-12-07 17:16:07.000000000","tz":60},"subject":"Add daxio package for nvdimm scenario","message":"Add daxio package for nvdimm scenario\n\nnova-compute uses daxio to cleanup vpmem backend device on instance\ndelete. If the daxio binary is missing in the nova-compute container\ninstance delete fails.\n\nChange-Id: Ifb5948653565e2ae902783762e20e33527020efe\nCloses-Bug: 1907124\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3a9721d23d179d6dc24cb804552a58f50e7ab54a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3a9721d23d179d6dc24cb804552a58f50e7ab54a"}]},"branch":"refs/heads/master"},"e85641172667a91b201f472170d7dc3946742043":{"kind":"REWORK","_number":2,"created":"2020-12-08 07:40:24.000000000","uploader":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"ref":"refs/changes/32/765832/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/32/765832/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/765832/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/765832/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/765832/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/32/765832/2"}}},"commit":{"parents":[{"commit":"22c32a30e62d42c3631306775c0cd029eb659e9a","subject":"Unify curl options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22c32a30e62d42c3631306775c0cd029eb659e9a"}]}],"author":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2020-12-07 17:16:07.000000000","tz":60},"committer":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2020-12-08 07:40:08.000000000","tz":60},"subject":"Add daxio for nvdimm scenario","message":"Add daxio for nvdimm scenario\n\nnova-compute uses daxio to cleanup vpmem backend device on instance\ndelete. If the daxio binary is missing in the nova-compute container\ninstance delete fails. daxio is provided in centos via daxio, in\nubuntu via the pmdk-tools package.\n\nChange-Id: Ifb5948653565e2ae902783762e20e33527020efe\nCloses-Bug: 1907124\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e85641172667a91b201f472170d7dc3946742043"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e85641172667a91b201f472170d7dc3946742043"}]},"branch":"refs/heads/master"},"bc1fabe41e4998247d605f43c11bc98e29163a18":{"kind":"REWORK","_number":3,"created":"2020-12-15 12:19:52.000000000","uploader":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"ref":"refs/changes/32/765832/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/32/765832/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/765832/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/765832/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/765832/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/32/765832/3"}}},"commit":{"parents":[{"commit":"22c32a30e62d42c3631306775c0cd029eb659e9a","subject":"Unify curl options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22c32a30e62d42c3631306775c0cd029eb659e9a"}]}],"author":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2020-12-07 17:16:07.000000000","tz":60},"committer":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2020-12-15 12:19:37.000000000","tz":60},"subject":"Add daxio for nvdimm scenario","message":"Add daxio for nvdimm scenario\n\nnova-compute uses daxio to cleanup vpmem backend device on instance\ndelete. If the daxio binary is missing in the nova-compute container\ninstance delete fails. daxio is provided in centos via daxio, in\nubuntu via the pmdk-tools package.\n\nChange-Id: Ifb5948653565e2ae902783762e20e33527020efe\nCloses-Bug: 1907124\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc1fabe41e4998247d605f43c11bc98e29163a18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc1fabe41e4998247d605f43c11bc98e29163a18"}]},"branch":"refs/heads/master"},"6176d54179109e85dee5ebc29a7726bcb05f76a9":{"kind":"REWORK","_number":4,"created":"2020-12-15 13:22:21.000000000","uploader":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"ref":"refs/changes/32/765832/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/32/765832/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/765832/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/765832/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/765832/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/32/765832/4"}}},"commit":{"parents":[{"commit":"22c32a30e62d42c3631306775c0cd029eb659e9a","subject":"Unify curl options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22c32a30e62d42c3631306775c0cd029eb659e9a"}]}],"author":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2020-12-07 17:16:07.000000000","tz":60},"committer":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2020-12-15 13:22:12.000000000","tz":60},"subject":"Add daxio for nvdimm scenario","message":"Add daxio for nvdimm scenario\n\nnova-compute uses daxio to cleanup vpmem backend device on instance\ndelete. If the daxio binary is missing in the nova-compute container\ninstance delete fails. daxio is provided in centos via daxio, in\nubuntu via the pmdk-tools package.\n\nChange-Id: Ifb5948653565e2ae902783762e20e33527020efe\nCloses-Bug: 1907124\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6176d54179109e85dee5ebc29a7726bcb05f76a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6176d54179109e85dee5ebc29a7726bcb05f76a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
