)]}'
{"id":"openstack%2Fkolla~765854","triplet_id":"openstack%2Fkolla~master~I2d67b3ed7b7c521bc755ec9a87365c9962a9eeff","project":"openstack/kolla","branch":"master","topic":"bug/1907213","hashtags":[],"change_id":"I2d67b3ed7b7c521bc755ec9a87365c9962a9eeff","subject":"Support CentOS 8.3","status":"MERGED","created":"2020-12-07 19:21:34.000000000","updated":"2020-12-11 12:41:39.000000000","submitted":"2020-12-11 12:40:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"765854-bug/1907213","meta_rev_id":"27579b125b28925cce0d307d1d1b32c77675ab77","_number":765854,"virtual_id_number":765854,"owner":{"_account_id":31425,"name":"Albert Braden","email":"c-albert.braden@charter.com","username":"ozzzo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-11 12:40:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-12-09 18:50:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-12-09 16:57:47.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2020-12-09 18:50:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-12-11 11:25:05.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-12-11 12:41:39.000000000","post_submit":true,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-12-11 12:41:39.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-12-11 12:41:39.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-07 19:39:02.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-12-07 21:15:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-08 09:10:55.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-12-08 09:34:32.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2020-12-08 09:43:14.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2020-12-08 10:09:38.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-12-08 22:06:17.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-12-09 18:50:09.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"49d169fbcf3ac8be12c6c60dd0b2c2b4dd796310","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31425,"name":"Albert Braden","email":"c-albert.braden@charter.com","username":"ozzzo"},"date":"2020-12-07 19:21:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"edb65a8aea7b7119e68a5bb02da8a9219fd66766","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31425,"name":"Albert Braden","email":"c-albert.braden@charter.com","username":"ozzzo"},"date":"2020-12-07 19:29:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"86efe566886705fecd63e18331385fa75325c7f5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-12-07 19:39:02.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2\n\nWe probably will need it in Victoria, Ussuri and Train.","accounts_in_message":[],"_revision_number":2},{"id":"b113d51b2c4014884c2e26c6427f932294579761","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-12-07 20:24:00.000000000","message":"Patch Set 2:\n\nWill discuss with CentOS devs tomorrow about it. Maybe they revert names.","accounts_in_message":[],"_revision_number":2},{"id":"bd746ae94b3a4f51d1e8779080dc86f53f234834","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-07 21:15:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/fe2f6df7840c4169b7855a98a5142025 : FAILURE in 7m 54s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7974b377cb9b4d58b9a987f39c5ee4b3 : FAILURE in 8m 01s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/66057ba856024ab4831167f769309195 : FAILURE in 11m 25s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/7b009312d4524b7786317781361defd6 : FAILURE in 11m 42s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/a8165bbba7e843528c797624571adcde : FAILURE in 12m 03s\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/95f69e969e794d4299885886dd2c3c28 : FAILURE in 10m 57s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/9e5f984a312245c58282179575b5a566 : FAILURE in 12m 08s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/2f8774132bf14fbab9bcbb8f1596c002 : FAILURE in 14m 05s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/6a67f847f568433c929355bb3a43417e : FAILURE in 13m 54s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/d3ba0b0f4a9f4934a3b9673d67cbc52e : FAILURE in 14m 36s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/25f7ba9853e54bcab6b2e8e3fde33cdd : FAILURE in 10m 33s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/6b0a8d1a9d3e4252aaa6090150fdda70 : FAILURE in 12m 10s (non-voting)\n- kolla-ansible-centos8-source-mariadb https://zuul.opendev.org/t/openstack/build/fa3e573cfd264d22b9ad337e4d38add6 : FAILURE in 17m 22s (non-voting)\n- kolla-ansible-centos8-source-masakari https://zuul.opendev.org/t/openstack/build/7586d5edca5845a6a1e5bfcd7b5f7604 : FAILURE in 12m 32s (non-voting)\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/22c8961cd5464cbbb93ab64fb7fe8a29 : FAILURE in 15m 01s (non-voting)\n- kolla-ansible-centos8-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/e72a5c91e8ff449aad23494d432e288a : FAILURE in 11m 36s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/13760ecb7be74a40b135607f320f60a1 : SUCCESS in 35m 39s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/549d92fcc59e4ffc86515a7b73e65954 : SUCCESS in 21m 42s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/574336f8f99e435786b1bc1fc82a7996 : SUCCESS in 57m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9f70156d5f4443e94f1cbb972c06e55 : FAILURE in 2m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8de3336e717b40199d8e2a4bb4b27911 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9546e06b4b124deb88e788319442ef79 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48780f5d4c2c4f50b5257ff463c97aee : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b0ea6927df749ae88d0ff9537d5ad42 : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff5fac3511d04575b7271581df340ff1 : SUCCESS in 7m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4620d79179cd40df8c070acdb71ab707 : SUCCESS in 25m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/18671ed58ed344d599731e19978f1d0e : SUCCESS in 32m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d42364803b464faf9e3947b2946ec87d : SUCCESS in 55m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/30686b6ebb0c4a69ab0c97399e030436 : SUCCESS in 53m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3c88c6922eb54d3c9e92c6f33ced6178 : SUCCESS in 59m 07s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/06c8b65b937540a2a98103903dd79bc3 : SUCCESS in 1h 22m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/5f9701eb8478479d85661d42bfca144d : POST_FAILURE in 1h 30m 40s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/fa4ab9b7e1254af4959fa40810a5e4e6 : DISK_FULL in 1h 12m 06s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/e1049a3fdcfe4ba2b9a90327a58af734 : SUCCESS in 1h 01m 17s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/57f64d04818d490cafa290075edcb429 : SUCCESS in 1h 16m 41s (non-voting)\n- kolla-ansible-ubuntu-source-mariadb https://zuul.opendev.org/t/openstack/build/ce4dc925e3154ae7ad2144434585c9bc : SUCCESS in 27m 32s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/2d7cdca8c0cd42df92bef241fb396a58 : SUCCESS in 43m 44s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/4949e474dd5243b681a6726e2fc700c8 : SUCCESS in 48m 38s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/d5196ab2a7a74822931fc4228995b6c4 : FAILURE in 46m 58s (non-voting)\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/544e8aba6dfe408eafc1d6dca53ddd4b : SUCCESS in 31m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6a5dd0443373039dd1978a42694214e72b8170e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-07 22:16:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/3e5024a1d02c45629b7c0f4d8abeabe4 : FAILURE in 11m 53s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b438b5142263489eab4ac6f5df8756d2 : SUCCESS in 51m 10s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/061cd284867f4a1aae4c24dcc39b1349 : SUCCESS in 1h 41m 22s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/74bbf91d22264a3ba55a718398bdc08c : SUCCESS in 1h 02m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7df583fa4d6c09e8e5b2444bcf478f89c9f2245c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-08 09:10:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"aa60f624ce60160900687a368782dfe58898683f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-08 09:16:43.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"cc6a2db3c2368f9098391f12f48a499f402e8707","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-12-08 09:34:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2859ae77b6760bb4a4b17d604e55819dcb11fac5","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-12-08 09:43:14.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"0a6d5128ba564b8f88ac04d9add55c82afb53b58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-08 09:54:40.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"34052076ded5e3cf97506ff007b08b12faacf7b5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-08 09:55:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9b74f22f0f7f309ff2c4896673ba5c492d8a19f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-08 09:58:33.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b2b67727c428c54c18f8a51e650e22a9e338383a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-08 09:59:00.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"044f1258c035fda3eea8842697ca70b4ef13eeee","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-08 10:09:38.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"8d0822840fe8e0234aeae3b05575fba8c27bc6c2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-12-08 11:57:10.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"199b670d5633025658bd28a0839d75ac72478c7a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 12:00:57.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/5bf0debbf25e4a5abb67ed9968aac410 : SUCCESS in 1h 15m 18s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cc0d32b945584fd2935412ccccbd3cd8 : SUCCESS in 1h 11m 44s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a4e716726c9448b7a99f1a5d34badeb2 : SUCCESS in 1h 55m 53s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/1c6ec56f8c3447b2a785e6d09d25a7d0 : SUCCESS in 1h 14m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"38cb2833e8b72ddda1e56ddeef88cbee3eb3e5e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 12:56:12.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/aadd6d6fee2842c2992326473d652af0 : SUCCESS in 23m 41s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/dfb6eaeca7e948fe93f46fb06e492e3f : SUCCESS in 1h 56m 17s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8c1d7f0c9b37463f8571b79fbbb57be2 : TIMED_OUT in 2h 04m 02s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/d76a676c13a2409691c737ecb2712d69 : SUCCESS in 1h 02m 34s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/e26292b5ca7640e4916c3ccffd491f6f : SUCCESS in 1h 12m 04s\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/9f912c5648c94a24b5db7a019fc82156 : SUCCESS in 30m 43s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/b315752d55d14943ad872a01fcf6495c : SUCCESS in 1h 24m 26s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/14b8cce623a94fe9925e0d38c9fbabbc : SUCCESS in 1h 36m 18s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/394982c827444b30af2121901855ca96 : SUCCESS in 1h 31m 43s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/5677c7f5c4854746af56e1f889e3884c : FAILURE in 1h 05m 55s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/3c1bd121feaf42689a067a95c0b78948 : SUCCESS in 1h 06m 25s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/4e2a079003e0402a97e9ad65ec8e3ed8 : SUCCESS in 1h 26m 28s (non-voting)\n- kolla-ansible-centos8-source-mariadb https://zuul.opendev.org/t/openstack/build/d5c5f4cafc634537b066f7ba51641c2d : SUCCESS in 32m 36s (non-voting)\n- kolla-ansible-centos8-source-masakari https://zuul.opendev.org/t/openstack/build/a177fca57f5b461f916cfe3acda341af : SUCCESS in 58m 28s (non-voting)\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/7d7d7a1971da4f4d8be8fa434db49672 : FAILURE in 53m 38s (non-voting)\n- kolla-ansible-centos8-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/3e0c6f3d3d154be8a4a9f6ccd6265915 : FAILURE in 41m 01s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/158dfdb78f684c37b443a8cf9cbbd171 : SUCCESS in 2h 53m 31s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/129eb6aff4e24ebdb6a2fa5f6c2a6dcb : SUCCESS in 17m 10s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/05d18a2c45a14a699d0bba7b23a7b76c : SUCCESS in 1h 00m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74527f91c50a4ead8a691de0f3c68917 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5f70d4f086e463f8093cfcd86f5994f : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8014122e0f3d4dee8530ede3b028df11 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4bca3498f63c489a9168ae7a34f33142 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8764d2288b7848b792ef9c2a65a8a162 : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/076b4c25f9894a5b91754e10475621b3 : SUCCESS in 5m 48s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5a812f8d435d4f6b935f4afcfcb3081c : SUCCESS in 21m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/adcc5f5213364cc696b740f1c32f4acd : SUCCESS in 39m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/291776b91ec74e62b0e4d0ca309b4ded : SUCCESS in 52m 31s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5390342276c1444b9867fce7aa4945db : SUCCESS in 1h 00m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cc52274ab4f74e04a4b55b768b4834c7 : SUCCESS in 1h 00m 09s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/48498cddfb4d4f55a2d2c8c9b7fa6e69 : SUCCESS in 1h 13m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/e89dc0825ead4a3b89fb8ca807f96fb0 : DISK_FULL in 1h 30m 55s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/047bfca744734adc8199f47d42e528d0 : SUCCESS in 1h 23m 52s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/aa3441b3b98e449bbb68350365993e20 : SUCCESS in 50m 38s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/a9004564e61e4115ac72bb2f53d914ec : DISK_FULL in 1h 13m 11s (non-voting)\n- kolla-ansible-ubuntu-source-mariadb https://zuul.opendev.org/t/openstack/build/322e685504c84efcb4d1bcd4d1f134f0 : SUCCESS in 30m 00s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/fab5f8e2f0944de7afaa3acf207e1db7 : SUCCESS in 48m 49s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/b7420b46062c4aac9e8ba2c326ea7d44 : SUCCESS in 48m 33s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/cc568eaf2f7e4ee7b31908ed0553dfb6 : FAILURE in 46m 08s (non-voting)\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/4cb689a796bc48f28679233b6465ab23 : SUCCESS in 29m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"53417d7de6121d61b8e4b9dd01fb49d4b791bb99","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-08 14:08:46.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"d96613dd6fa4bd760c24fb3278a7927318b51255","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 15:51:31.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/79b87397190b4e308e65394337b326d1 : SUCCESS in 44m 00s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e6cd80969e994ab79b62dd86929ed27b : SUCCESS in 42m 12s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5303ea9cdf4f4b4b89e29a2a7f4a09e7 : SUCCESS in 1h 36m 38s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b468618efda844cdae02fddd22eb9e74 : SUCCESS in 53m 02s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"64e7506feac0f7cc42bbcc2e93f45bcd35132ac3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-08 15:55:51.000000000","message":"Patch Set 5: -Code-Review Workflow-1\n\nit is failing again; I will be slimming the CI down\n\nfirst of all, we can merge the l-c drop independently (thanks to guards on jobs)\ncores, please approve https://review.opendev.org/c/openstack/kolla/+/765995","accounts_in_message":[],"_revision_number":5},{"id":"c8476fa5c1b296909f30f6a0251ce6ce68407499","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 16:26:12.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/97b9007b649c437b9e8dfe798dfe2706 : SUCCESS in 21m 34s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/83af4bc16ac4457fa50d2428aea0c752 : SUCCESS in 37m 52s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/71d069270a1a47c8ac722a1db61a5ef6 : SUCCESS in 1h 03m 26s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/fbe0632f52b148c38aed73eec6a0ae69 : SUCCESS in 1h 02m 47s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/1bc65886ee554a6088a202a2d1e5fe66 : FAILURE in 12m 17s\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/bb32668f67bc4a719bd89bc61f4b6432 : SUCCESS in 40m 33s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/1bf16cd9d2bb40fa9c1a80971eff3e55 : FAILURE in 1h 18m 11s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/62da1a7eef0b47debb65b1424786b8da : FAILURE in 1h 04m 40s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/a1bde9b2aad3421ea016792baa2fe2bb : SUCCESS in 1h 38m 38s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/2f319e70ed9f4c6f9058936908d9b1cf : FAILURE in 1h 14m 58s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/fc208bffadff43ae8ac36a6c659cc4bd : SUCCESS in 1h 03m 51s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/b220c6750e5f4c04841ce949cb090488 : SUCCESS in 1h 32m 07s (non-voting)\n- kolla-ansible-centos8-source-mariadb https://zuul.opendev.org/t/openstack/build/919093aa87834639a7eebb8940f54335 : SUCCESS in 34m 02s (non-voting)\n- kolla-ansible-centos8-source-masakari https://zuul.opendev.org/t/openstack/build/850db42aade34dad93d2d2b310e2ae82 : SUCCESS in 57m 01s (non-voting)\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/924174ed82e74d15a356713f4f816a70 : FAILURE in 57m 34s (non-voting)\n- kolla-ansible-centos8-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/d2f657e61e334fbc8a42ea7e01623968 : SUCCESS in 32m 20s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5ee6c19613cd444aace2012631bd6dea : SUCCESS in 2h 14m 01s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7da143aa07aa4cd28bae4b2e6d7519e8 : SUCCESS in 1h 57m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/da70a3c1123041b590628253f697efd9 : SUCCESS in 1h 00m 13s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11b48c281d2041a39b6f605878d361f4 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab8f36a2bf6e42529aae4e7381ffdad5 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0d1ce54d6a047eca1fb54df54b8d7ce : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98911f00ca6e47daaead939623fafd48 : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5785d9aa5f234979956a81f57214f203 : SUCCESS in 4m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/442fb437783648a5996d5e1351a39db1 : SUCCESS in 7m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2eb563c17c08428c82e3acc25f227ced : SUCCESS in 23m 48s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f28b4da6fe3549248b430f1b852cd449 : SUCCESS in 42m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/26fe5c44da06402bab337b2d6261a27d : SUCCESS in 52m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b6ae58006deb4a6881855015e6e79bf1 : SUCCESS in 55m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/28587bf3bdd0406da35a89cfb352bbef : SUCCESS in 1h 02m 52s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/b6c4d863951b4e98aa2ac2da04c29af5 : SUCCESS in 1h 18m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/22448d3b710146fa85e49191001c6e90 : DISK_FULL in 1h 27m 14s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/485e54fcfe484481b9e89a452f24a618 : SUCCESS in 1h 20m 01s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/e3cf801a71ce4550abf40318468ed8c9 : SUCCESS in 56m 05s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/4cf7b3a35ef54e00b48cdea1d523fed7 : SUCCESS in 1h 14m 15s (non-voting)\n- kolla-ansible-ubuntu-source-mariadb https://zuul.opendev.org/t/openstack/build/cd761b3840044e1dbfa9a9e6798144d0 : SUCCESS in 27m 49s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/bd1dcad004b4443e969062afefb74b2d : SUCCESS in 51m 41s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/fa763e90bcea429bb21695188d0e84b6 : SUCCESS in 50m 37s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/9fe07d4b9bfb42e1bc40776159533e1e : FAILURE in 52m 03s (non-voting)\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/42ba18933b1c430a885a9299cf8badd0 : SUCCESS in 27m 48s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1b047f1572d5f7a8979c407199699b987cfb523f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-08 17:53:34.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"bb249fa9b06894e9c3290b1001c5d1669d945474","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 19:28:53.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/6c904879c9d2464bb65373031389fd60 : SUCCESS in 53m 56s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/563c1f0ca151433094a23347ce6c97d3 : SUCCESS in 54m 13s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8c0113b2b6b0409f8aa5a314386cc051 : SUCCESS in 1h 29m 51s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/67044c5b51aa4489a5aa01bbd5537442 : SUCCESS in 53m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"08d8b40ff932f2cc358ea3e162b509cdcb73e29e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-08 19:40:44.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"3c83f9b689cb72905771c02679ec8990f0b6d5e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-08 19:41:26.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"13eb99189fc3275158d5c10842546468ba4643cd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-08 19:41:37.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"c2eb86aedbb8dcfc2d0e7db1c23b40b326b82efd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-08 19:41:59.000000000","message":"Patch Set 7: Backport-Candidate+2 Review-Priority+2","accounts_in_message":[],"_revision_number":7},{"id":"a0cf687622e1f35193887368de6868925cf18277","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-08 19:42:24.000000000","message":"Patch Set 7:\n\ncheck experimental\n\n(for fun this time)","accounts_in_message":[],"_revision_number":7},{"id":"0a8f42ea9c47dd832715140b9142c7499859d538","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-08 19:51:57.000000000","message":"Patch Set 7:\n\ncheck experimental","accounts_in_message":[],"_revision_number":7},{"id":"547326075a9dd68b4a51a42ca763775b8a3e4928","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-08 19:52:33.000000000","message":"Patch Set 7:\n\noh, this actually cannot handle trailing characters","accounts_in_message":[],"_revision_number":7},{"id":"f451e85cbb214e725c3716b7d670645918ff23bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 21:02:23.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/631b272427e24d00af2d066b301b600f : SUCCESS in 37m 34s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f9f654b5202e46f9a37702c214c324c8 : SUCCESS in 56m 08s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/50fe6c1700a74f698999829dade61614 : SUCCESS in 1h 00m 39s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/a08f5e75a6954b188933e98ea44e4e0a : SUCCESS in 57m 58s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/bd85c546176f4ce090bbe4500a5ea44c : SUCCESS in 1h 17m 36s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/90292286182044a4b8205515953206da : SUCCESS in 30m 41s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1dca5edca8d24c3b8b4490f34efadcdb : SUCCESS in 17m 58s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5a1457c0c7f54d5b83ac43532aaf3689 : SUCCESS in 1h 01m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/121cdc01b4884b51872b0145b4f13dc8 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7903c8710ae04e30807dde8b55c647ab : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/165db43fc2814710a71e2341bfa9c21c : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca72dbfdc0b347d0b9ef43a8aadc7b65 : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c0a58fa1a5349cbb865b91ddfd4858a : SUCCESS in 5m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f1fe254062e491788179c84d433ea45 : SUCCESS in 3m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/340c6f56973a4b37b978f60fec645c01 : SUCCESS in 24m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/077732c908f747c288e03dc4bdb9647f : SUCCESS in 35m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bb736e4637e84baa8076c62f7538939b : SUCCESS in 52m 41s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/84169b75de4648bf821ade7d9a0a710e : SUCCESS in 1h 10m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9b7045040cde466eadcccd9aed226a24 : SUCCESS in 1h 19m 25s","accounts_in_message":[],"_revision_number":7},{"id":"e3960f1e4c35e9efc8cbbb9264dd29c2a841ac02","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 21:28:48.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/87d9860a1a2542dcaf5fd927c179931b : SUCCESS in 1h 08m 46s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b8cbc75d03934ef3bd70116ecf468a47 : SUCCESS in 1h 01m 31s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bffa9c4f1a104fbd8671700eaf10da2e : SUCCESS in 1h 39m 54s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4f04cd527ca441d093df49dd3f862b62 : SUCCESS in 1h 07m 39s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"678b4ebc43321ff7843ea4d35d8555fc01fc4723","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 21:33:48.000000000","message":"Patch Set 7:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-no-infra-wheels-centos8-source https://zuul.opendev.org/t/openstack/build/b5feda24f8be4e209ff28c4d0a0b09ad : SUCCESS in 52m 07s\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/bdb75e942c83470982e521a58ac5500e : SUCCESS in 24m 58s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/d4d8d8bec5964bd5b22fe11d68ec3ce9 : FAILURE in 58m 18s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/fab9d5520f4c446ab2f5849371e4c3c0 : SUCCESS in 1h 38m 08s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/97c98d6a61e74c59b9371d24df17c1c5 : SUCCESS in 1h 28m 44s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/67764d4a2ec9474a9c55878dda9b7f8d : FAILURE in 1h 01m 18s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/2a4c24bcb1ec463fa472af663c7e7296 : SUCCESS in 1h 06m 17s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/541b284cac164d0a96bc0f95d6750603 : SUCCESS in 1h 24m 11s (non-voting)\n- kolla-ansible-centos8-source-mariadb https://zuul.opendev.org/t/openstack/build/d137866d24254083a475818ca086d00f : SUCCESS in 29m 51s (non-voting)\n- kolla-ansible-centos8-source-masakari https://zuul.opendev.org/t/openstack/build/fdeb7194e5444ecdb3d579aa45e07b08 : SUCCESS in 51m 48s (non-voting)\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/08ac63e0c68e43ce8bc0af3f05304c70 : FAILURE in 1h 00m 18s (non-voting)\n- kolla-ansible-centos8-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/2c46057851b7471ab3c6f399e7d32ae6 : SUCCESS in 34m 13s (non-voting)\n- kolla-build-no-infra-wheels-debian-source https://zuul.opendev.org/t/openstack/build/e265938f55574e5498b2116359805c97 : SUCCESS in 31m 23s\n- kolla-build-no-infra-wheels-ubuntu-source https://zuul.opendev.org/t/openstack/build/3a47a782cce14eb2ad55792e186589ef : SUCCESS in 40m 23s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/b8936ba10db942ba971faa764dd6533c : FAILURE in 55m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/29abaaad0ed344e0957508de5656abee : DISK_FULL in 1h 21m 21s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/634cee4ce1e44455826e218d4d6d8a95 : SUCCESS in 1h 23m 12s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/3b6cd6699f6341e08e8bc79f338b62d0 : SUCCESS in 1h 00m 56s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/843b38f06d2b4fcfb151ece0d6c14b61 : SUCCESS in 1h 11m 59s (non-voting)\n- kolla-ansible-ubuntu-source-mariadb https://zuul.opendev.org/t/openstack/build/ce8cba163b2548f69a7800ab676fd624 : SUCCESS in 35m 28s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/28da788692c34e8191072e7b4c05742b : SUCCESS in 49m 42s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/b2b42e7c9cab49e7a835d4f1d541dddd : SUCCESS in 52m 11s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/9d16720b239b40b589c4cdc700619d42 : FAILURE in 46m 03s (non-voting)\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/54abb0ec87c34e8aa0de6924526641d3 : SUCCESS in 25m 25s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"335a197530c6b454c23efa0a8a389c6aacbccb96","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-12-08 22:06:17.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"607a7caf51e36cb76dbfea03c69b089dcab35974","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-09 09:08:27.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"8525bcae5c6f748ff19808f419180d7b3dcdec7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 09:10:14.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"0f9a934ef4d287e278ae340ff28dd65a424eac60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 10:05:14.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a4a6d5be94e04bcfa90fc70526b3ecba : SUCCESS in 51m 48s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/73b36039cfd840e5bfdb520a229d9953 : FAILURE in 19m 24s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/4f132c6cf0284e56a4efb8c3b47b6d10 : FAILURE in 25m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0401a844a8044d198922f92ee57f5b1 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e7b15922a584561a7166d2baefa267d : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/820f119db647439d9b79da0d42136845 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c34ad08dd2494ea3842befef9165abd3 : SUCCESS in 5m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6dd7a6806fd47e59e25a0b1891ddad4 : SUCCESS in 3m 39s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9cd86f5761b44fc4aed1669d29688253 : SUCCESS in 33m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/99f215140be44b00bc33ede381a88337 : FAILURE in 19m 27s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ecbe29cc0b03420d9bca46c81deebdaa : FAILURE in 17m 14s","accounts_in_message":[],"_revision_number":7},{"id":"782ad22d821e1747374d09323fb30a8583731569","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-09 16:17:44.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"fc89e88458ef485e2865d3e63c21f3174ace9a1f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-09 16:25:47.000000000","message":"Patch Set 7:\n\noh, I had this open and mindlessly rechecked but I know it will fail on k-a 😊","accounts_in_message":[],"_revision_number":7},{"id":"9226690b23eb308b888cd6f5deb60e693a60d709","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-09 16:27:20.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"7a56bb207b146d0d3d53771dcfc4e51e1552b033","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-09 16:29:33.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"b5b5451f67c52766219eb70644def0acf7d21b0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 16:36:10.000000000","message":"Patch Set 9: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":9},{"id":"f2a6f815f7ab62283dddac822dd147c706f3dab1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 16:36:21.000000000","message":"Patch Set 9:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":9},{"id":"5223f1caf85fe1ab90460773a9ae966797ff0bd3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-09 16:57:39.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"88faaf00eb938b5bd925b7b548a9540f315ebc23","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-09 16:57:47.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"dc5cd49fb5efcc9f1404162ab3f68ee48bd6298b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 18:38:42.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/3fa9e6b8b8ff4ac7bea842e8c0dc28ba : SUCCESS in 20m 38s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/95ccb82dd3ca40679b2ccfa6e605a76d : SUCCESS in 48m 16s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/bd3143a082ef44eaad32380c9c0cd6ad : SUCCESS in 1h 01m 42s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/d8d6427d3f8741a8a71237f9ccf9442c : SUCCESS in 1h 07m 42s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/887e24b23d074037bb0514a03b953a04 : SUCCESS in 1h 11m 26s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d4d17064bdc7437cb66435c3bd6e5e93 : SUCCESS in 26m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3a0e9ca190504a65810e3a3e90c4b30c : SUCCESS in 18m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7b36927b3d454211a0437716b03971cc : SUCCESS in 50m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3bff4a3c3eb4864b65767a80f19bdb3 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f27ccd93a4f48b0a0cdea616a8b3065 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/304d3e5066a448369dcbc0986595e640 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/88b789bb283c48abb49e48b8363ba97c : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e38c5c49be604d3ab822d0bc4a4d8d09 : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22f6bf7e4fb94b2c9a5e14d6550ce939 : SUCCESS in 4m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a2c07733204b47ee9c77ea6af41055d3 : SUCCESS in 19m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5c4e80f836474afe9dfd6abcf71702fc : SUCCESS in 31m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c4c302e4740e4632a04c83890ccb0446 : SUCCESS in 52m 22s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c09d5fde63d1492eb3c8e55ffb976269 : SUCCESS in 51m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3af94240977644df9a1676cfd279f61b : SUCCESS in 1h 04m 18s","accounts_in_message":[],"_revision_number":9},{"id":"5b77b9bf34842de2bdabea9aa954796a34c11e81","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-12-09 18:50:09.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"f05ca62702fe953889f40a92abbe89c2b45cc1f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 18:50:54.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/8dcdb7f4701e49d0853545c7b97882e0 : SUCCESS in 1h 12m 06s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7086d1a9963b414684266ebca05e2892 : SUCCESS in 56m 15s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5bb1e8db71d447cb9a3b718d640cec75 : SUCCESS in 1h 40m 58s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/fa4827b8fe664ebf827ba2bdc12e41c0 : SUCCESS in 57m 51s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"7b5b1b4601645d20996566d08bcc02338cc6384b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-11 11:24:59.000000000","message":"Patch Set 9: Workflow-1","accounts_in_message":[],"_revision_number":9},{"id":"b935296e2ed6aff5da4e85a3315b60f546c1d31a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-11 11:25:05.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"f687e7fbc700ccf6c3f2121177c26220ff35f3e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 11:25:21.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"39450b90e900c9be9de790da5634aa3b9ab2ae4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 12:40:06.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d2b7a3da8c2549c291402d9789e15359 : SUCCESS in 43m 47s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f3ced40279954092ba6f3e0630c9f91f : SUCCESS in 1h 07m 10s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/dc646182c8594d14872be1c57c454258 : SUCCESS in 1h 09m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4c39b86710b4cbbbd0319c96e648cff : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a3dde33d0394b329ae74b4fcdc5db38 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4c7effc976546f3aaee2836f9e508fa : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cfc55a12ce14e1bb2fa5ac9ebbe1ab0 : SUCCESS in 5m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/819dfa58ecb24351bd9e3eba8adcbdd7 : SUCCESS in 3m 36s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/668294d24a434b08a2fd85abb349ba8e : SUCCESS in 33m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/378849313c9249d0935502e627186e41 : SUCCESS in 52m 23s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e3b824feec7546d3ba59151fdcad0a48 : SUCCESS in 1h 02m 18s","accounts_in_message":[],"_revision_number":9},{"id":"0686bfd199be0f855f54932a4705096fc9898d4c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 12:40:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"47d71291ca4b0a5781807ef9b299602083fc9d66","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 12:41:39.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/433167db47964c33a44f713a75ba24dd : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ecf0f496b4e4a55982046b81e85cfd6 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"17799d6eb6c78c113285c7b654fec1abdd12cfd4","revisions":{"443ff21c185251ebfd0f2b94dc35dbcdc65e77cf":{"kind":"REWORK","_number":1,"created":"2020-12-07 19:21:34.000000000","uploader":{"_account_id":31425,"name":"Albert Braden","email":"c-albert.braden@charter.com","username":"ozzzo"},"ref":"refs/changes/54/765854/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/765854/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/765854/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/765854/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/765854/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/765854/1"}}},"commit":{"parents":[{"commit":"22c32a30e62d42c3631306775c0cd029eb659e9a","subject":"Unify curl options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22c32a30e62d42c3631306775c0cd029eb659e9a"}]}],"author":{"name":"abraden","email":"C-Albert.Braden@charter.com","date":"2020-12-07 19:20:44.000000000","tz":0},"committer":{"name":"abraden","email":"C-Albert.Braden@charter.com","date":"2020-12-07 19:20:44.000000000","tz":0},"subject":"kolla: Lowercased PowerTools and HighAvailability to match Centos8 change","message":"kolla: Lowercased PowerTools and HighAvailability to match Centos8 change\n\nChange-Id: I2d67b3ed7b7c521bc755ec9a87365c9962a9eeff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/443ff21c185251ebfd0f2b94dc35dbcdc65e77cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/443ff21c185251ebfd0f2b94dc35dbcdc65e77cf"}]},"branch":"refs/heads/master"},"a5be0e2c5d3c0bb7c558c9f57acf6dda937f0f14":{"kind":"REWORK","_number":2,"created":"2020-12-07 19:29:35.000000000","uploader":{"_account_id":31425,"name":"Albert Braden","email":"c-albert.braden@charter.com","username":"ozzzo"},"ref":"refs/changes/54/765854/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/765854/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/765854/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/765854/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/765854/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/765854/2"}}},"commit":{"parents":[{"commit":"22c32a30e62d42c3631306775c0cd029eb659e9a","subject":"Unify curl options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22c32a30e62d42c3631306775c0cd029eb659e9a"}]}],"author":{"name":"abraden","email":"C-Albert.Braden@charter.com","date":"2020-12-07 19:20:44.000000000","tz":0},"committer":{"name":"abraden","email":"C-Albert.Braden@charter.com","date":"2020-12-07 19:28:08.000000000","tz":0},"subject":"kolla: Lowercased PowerTools and HighAvailability to match Centos8 change","message":"kolla: Lowercased PowerTools and HighAvailability to match Centos8 change\n\nChange-Id: I2d67b3ed7b7c521bc755ec9a87365c9962a9eeff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a5be0e2c5d3c0bb7c558c9f57acf6dda937f0f14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a5be0e2c5d3c0bb7c558c9f57acf6dda937f0f14"}]},"branch":"refs/heads/master"},"e3d8a780f1dd2bdbc965e1331f64bb66146279d4":{"kind":"REWORK","_number":3,"created":"2020-12-08 09:10:55.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/54/765854/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/765854/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/765854/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/765854/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/765854/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/765854/3"}}},"commit":{"parents":[{"commit":"22c32a30e62d42c3631306775c0cd029eb659e9a","subject":"Unify curl options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22c32a30e62d42c3631306775c0cd029eb659e9a"}]}],"author":{"name":"abraden","email":"C-Albert.Braden@charter.com","date":"2020-12-07 19:20:44.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-12-08 09:06:39.000000000","tz":60},"subject":"Support CentOS 8.3","message":"Support CentOS 8.3\n\nBackport this until Train.\n\nCentOS 8.3 renamed repos.\nThis caused failures when disabling and enabling.\nMore details in the referenced bug report.\n\nNotice CentOS 8.2 and below will no longer work.\n\nAlso, this disables the lower-constraints job as it started\nfailing and we agreed during the PTG to drop it anyway. [1]\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-October/018445.html\n\nChange-Id: I2d67b3ed7b7c521bc755ec9a87365c9962a9eeff\nCloses-Bug: #1907213\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e3d8a780f1dd2bdbc965e1331f64bb66146279d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e3d8a780f1dd2bdbc965e1331f64bb66146279d4"}]},"branch":"refs/heads/master"},"1ad6ee2906ff3522c5a1072a731c5e3f5f3f5234":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-12-08 09:54:40.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/54/765854/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/765854/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/765854/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/765854/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/765854/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/765854/4"}}},"commit":{"parents":[{"commit":"22c32a30e62d42c3631306775c0cd029eb659e9a","subject":"Unify curl options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22c32a30e62d42c3631306775c0cd029eb659e9a"}]}],"author":{"name":"abraden","email":"C-Albert.Braden@charter.com","date":"2020-12-07 19:20:44.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-12-08 09:53:47.000000000","tz":60},"subject":"Support CentOS 8.3","message":"Support CentOS 8.3\n\nBackport this until Train.\n\nCentOS 8.3 renamed system repos. [1]\nThis caused failures when disabling and enabling.\nMore details in the referenced bug report.\n\nNotice CentOS 8.2 and below will no longer work.\n\nAlso, this disables the lower-constraints job as it started\nfailing and we agreed during the PTG to drop it anyway. [2]\n\n[1] https://wiki.centos.org/Manuals/ReleaseNotes/CentOS8.2011#Yum_repo_file_and_repoid_changes\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2020-October/018445.html\n\nChange-Id: I2d67b3ed7b7c521bc755ec9a87365c9962a9eeff\nCloses-Bug: #1907213\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ad6ee2906ff3522c5a1072a731c5e3f5f3f5234"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ad6ee2906ff3522c5a1072a731c5e3f5f3f5234"}]},"branch":"refs/heads/master"},"611a54e584749fa3aca5817861044058cde1cc55":{"kind":"REWORK","_number":5,"created":"2020-12-08 09:58:33.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/54/765854/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/765854/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/765854/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/765854/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/765854/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/765854/5"}}},"commit":{"parents":[{"commit":"22c32a30e62d42c3631306775c0cd029eb659e9a","subject":"Unify curl options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22c32a30e62d42c3631306775c0cd029eb659e9a"}]}],"author":{"name":"abraden","email":"C-Albert.Braden@charter.com","date":"2020-12-07 19:20:44.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-12-08 09:58:23.000000000","tz":60},"subject":"Support CentOS 8.3","message":"Support CentOS 8.3\n\nBackport this until Train.\n\nCentOS 8.3 renamed system repos. [1]\nThis caused failures when disabling and enabling.\nMore details in the referenced bug report.\n\nNotice CentOS 8.2 and below will no longer work.\n\nAlso, this disables the lower-constraints job as it started\nfailing and we agreed during the PTG to drop it anyway. [2]\n\n[1] https://wiki.centos.org/Manuals/ReleaseNotes/CentOS8.2011#Yum_repo_file_and_repoid_changes\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2020-October/018445.html\n\nChange-Id: I2d67b3ed7b7c521bc755ec9a87365c9962a9eeff\nCloses-Bug: #1907213\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/611a54e584749fa3aca5817861044058cde1cc55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/611a54e584749fa3aca5817861044058cde1cc55"}]},"branch":"refs/heads/master"},"3619986d34b0f56f94f4cd2d17c3a78dee810f72":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-12-08 19:40:44.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/54/765854/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/765854/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/765854/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/765854/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/765854/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/765854/6"}}},"commit":{"parents":[{"commit":"5797c3aa1a0402db9dc159df165b1fe97fc3df45","subject":"[CI] Slim down","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5797c3aa1a0402db9dc159df165b1fe97fc3df45"}]}],"author":{"name":"abraden","email":"C-Albert.Braden@charter.com","date":"2020-12-07 19:20:44.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-12-08 19:40:44.000000000","tz":0},"subject":"Support CentOS 8.3","message":"Support CentOS 8.3\n\nBackport this until Train.\n\nCentOS 8.3 renamed system repos. [1]\nThis caused failures when disabling and enabling.\nMore details in the referenced bug report.\n\nNotice CentOS 8.2 and below will no longer work.\n\nAlso, this disables the lower-constraints job as it started\nfailing and we agreed during the PTG to drop it anyway. [2]\n\n[1] https://wiki.centos.org/Manuals/ReleaseNotes/CentOS8.2011#Yum_repo_file_and_repoid_changes\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2020-October/018445.html\n\nChange-Id: I2d67b3ed7b7c521bc755ec9a87365c9962a9eeff\nCloses-Bug: #1907213\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3619986d34b0f56f94f4cd2d17c3a78dee810f72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3619986d34b0f56f94f4cd2d17c3a78dee810f72"}]},"branch":"refs/heads/master","description":"Rebase"},"fbe400905f3ce5a4d182c544e73980e9031e6d64":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2020-12-08 19:41:26.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/54/765854/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/765854/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/765854/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/765854/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/765854/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/765854/7"}}},"commit":{"parents":[{"commit":"5797c3aa1a0402db9dc159df165b1fe97fc3df45","subject":"[CI] Slim down","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5797c3aa1a0402db9dc159df165b1fe97fc3df45"}]}],"author":{"name":"abraden","email":"C-Albert.Braden@charter.com","date":"2020-12-07 19:20:44.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-12-08 19:41:20.000000000","tz":0},"subject":"Support CentOS 8.3","message":"Support CentOS 8.3\n\nBackport this until Train.\n\nCentOS 8.3 renamed system repos. [1]\nThis caused failures when disabling and enabling.\nMore details in the referenced bug report.\n\nNotice CentOS 8.2 and below will no longer work.\n\n[1] https://wiki.centos.org/Manuals/ReleaseNotes/CentOS8.2011#Yum_repo_file_and_repoid_changes\n\nChange-Id: I2d67b3ed7b7c521bc755ec9a87365c9962a9eeff\nCloses-Bug: #1907213\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fbe400905f3ce5a4d182c544e73980e9031e6d64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fbe400905f3ce5a4d182c544e73980e9031e6d64"}]},"branch":"refs/heads/master","description":"Edit commit message"},"837225aa87a9ffbbaaf2c4798c9fbf8c72167e8e":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2020-12-09 16:27:20.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/54/765854/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/765854/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/765854/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/765854/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/765854/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/765854/8"}}},"commit":{"parents":[{"commit":"5797c3aa1a0402db9dc159df165b1fe97fc3df45","subject":"[CI] Slim down","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5797c3aa1a0402db9dc159df165b1fe97fc3df45"}]}],"author":{"name":"abraden","email":"C-Albert.Braden@charter.com","date":"2020-12-07 19:20:44.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-09 16:27:10.000000000","tz":0},"subject":"Support CentOS 8.3","message":"Support CentOS 8.3\n\nBackport this until Train.\n\nCentOS 8.3 renamed system repos. [1]\nThis caused failures when disabling and enabling.\nMore details in the referenced bug report.\n\nNotice CentOS 8.2 and below will no longer work.\n\nDepends-On: https://review.opendev.org/766211\n\n[1] https://wiki.centos.org/Manuals/ReleaseNotes/CentOS8.2011#Yum_repo_file_and_repoid_changes\n\nChange-Id: I2d67b3ed7b7c521bc755ec9a87365c9962a9eeff\nCloses-Bug: #1907213\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/837225aa87a9ffbbaaf2c4798c9fbf8c72167e8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/837225aa87a9ffbbaaf2c4798c9fbf8c72167e8e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"17799d6eb6c78c113285c7b654fec1abdd12cfd4":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2020-12-09 16:29:33.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/54/765854/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/765854/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/765854/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/765854/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/765854/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/765854/9"}}},"commit":{"parents":[{"commit":"5797c3aa1a0402db9dc159df165b1fe97fc3df45","subject":"[CI] Slim down","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5797c3aa1a0402db9dc159df165b1fe97fc3df45"}]}],"author":{"name":"abraden","email":"C-Albert.Braden@charter.com","date":"2020-12-07 19:20:44.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-12-09 16:29:24.000000000","tz":0},"subject":"Support CentOS 8.3","message":"Support CentOS 8.3\n\nBackport this until Train.\n\nCentOS 8.3 renamed system repos. [1]\nThis caused failures when disabling and enabling.\nMore details in the referenced bug report.\n\nNotice CentOS 8.2 and below will no longer work.\n\n[1] https://wiki.centos.org/Manuals/ReleaseNotes/CentOS8.2011#Yum_repo_file_and_repoid_changes\n\nChange-Id: I2d67b3ed7b7c521bc755ec9a87365c9962a9eeff\nCloses-Bug: #1907213\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\nDepends-On: https://review.opendev.org/766211\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/17799d6eb6c78c113285c7b654fec1abdd12cfd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/17799d6eb6c78c113285c7b654fec1abdd12cfd4"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
