)]}'
{"id":"openstack%2Fkolla~765864","triplet_id":"openstack%2Fkolla~master~I97308c74406d2af0e39b77eddef2693554381ab6","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I97308c74406d2af0e39b77eddef2693554381ab6","subject":"centos: clean up repo enable/disable/keys","status":"MERGED","created":"2020-12-07 20:02:25.000000000","updated":"2021-05-17 17:54:51.000000000","submitted":"2021-05-17 16:04:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"765864","meta_rev_id":"435757330522ee952a51f77e1e3d69f1c2b9cabe","_number":765864,"virtual_id_number":765864,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-17 16:04:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-05-17 17:54:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-05-17 12:58:59.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-05-17 12:58:59.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-05-17 12:58:59.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-07 22:19:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-07 23:46:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-13 10:24:29.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2020-12-16 09:50:23.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-04-14 14:34:40.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-05-17 12:58:59.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"04c64a6f4ac9d2664fae3c819f0f1b078667beb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-12-07 20:02:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4c2b9d61bd8d88c5ed01d2eba9641332a7b58fd","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-12-07 20:26:55.000000000","message":"Patch Set 1:\n\nNeed to ask RDO guys to rename centos-nfv-ovs to centos-nfv-openvswitch like it is in CentOS.","accounts_in_message":[],"_revision_number":1},{"id":"af29bfd592a4f879b4d3d0a023df990929d7bbe8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-07 22:19:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/0d0badad54e54945a7b087da27ca9851 : FAILURE in 12m 44s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f515cfa5263e455690bbcd1f711317c5 : SUCCESS in 1h 22m 24s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3979b29f40254306a5fe0d57994f0489 : SUCCESS in 2h 08m 20s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c3e92c332f0d43358f1c4f0680eb7468 : SUCCESS in 1h 31m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"58a0e5cf974741d8925407675fbb0dcdd58be96b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-07 23:46:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/4e163ae4e0da452fb41bbb8ea34fca30 : FAILURE in 7m 07s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/76c2f03ada0c414d9947d9d51c1f746d : FAILURE in 7m 04s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/91cc7db3588e42f09613f4b986897d9e : FAILURE in 11m 59s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/fa0cd98cee414f4ca5488dd29ab4bd72 : FAILURE in 14m 02s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/5616dc08e3104e5193ffced8492ec30d : FAILURE in 14m 03s\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/476402a47dce4375b4d653d849ae8220 : FAILURE in 13m 28s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/afa6c639e46a4ca78e2854e2c639eed8 : FAILURE in 13m 51s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/68b8eab1b28b4755b9324b4ab121f202 : FAILURE in 15m 29s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/7cb2500f1a45463a9f1b8049f60e0141 : FAILURE in 13m 50s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/320a8890184b40898d7284626502f7ac : FAILURE in 17m 31s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/baee5965248b4d03af19fcc08af01263 : FAILURE in 13m 21s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/4e6e0cc54cc947c3ba948407689d415e : FAILURE in 17m 57s (non-voting)\n- kolla-ansible-centos8-source-mariadb https://zuul.opendev.org/t/openstack/build/b70f8bb200b54ce3bc3923ab1a400a17 : FAILURE in 13m 51s (non-voting)\n- kolla-ansible-centos8-source-masakari https://zuul.opendev.org/t/openstack/build/f297a914c2d141f9adf508ea875ccb06 : FAILURE in 13m 43s (non-voting)\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/3dc058dc731a40b897d426861981ebf2 : FAILURE in 14m 19s (non-voting)\n- kolla-ansible-centos8-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/272c486286c545f5bc3a69c3012971d5 : FAILURE in 13m 39s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d2c9848e11f84ebb95caec5ea6c9af08 : SUCCESS in 28m 48s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c5dc6403e64246e8a97b9c8ff0be4b88 : SUCCESS in 18m 49s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a39e79398bd34b3a8cb49ce390221693 : SUCCESS in 57m 10s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19ab6bf33d8740a1a254f498c1b8915d : FAILURE in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a83cc8b962a3420e927c889e9cfe8d93 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d47ce0e502b54ac69975a8ed98e0036b : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba92efb0e57b43c5abcfe10cecd10fdb : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/114c13737743483ea8065360880a6928 : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/106ca2e100c9420ca2f6db1f6b0b1aa6 : SUCCESS in 7m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cd13011560d740a099904cef9b42a2d5 : SUCCESS in 20m 23s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5196e9d2fffd491cbc00e4c4a4ca7a8f : TIMED_OUT in 3h 31m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d89fc48a93fa4a32af1ff9a6137c3baf : SUCCESS in 51m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d3ba0139bb9742d0a121332378b2305a : SUCCESS in 53m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f19d941decb54e599e6c542654520722 : SUCCESS in 1h 04m 44s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/68607602ac5f483c914e800f8579aed0 : SUCCESS in 1h 16m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/69b3f2494b874a2faf6aa5df150db1b2 : FAILURE in 58m 36s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/84cc5b0be74c4d0488bbfdd0b664490b : SUCCESS in 1h 22m 15s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/f119b965932743f2b207882c4d46cba3 : SUCCESS in 51m 28s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/fc5c6fa45584467987cb7ed0a8bc4ec7 : SUCCESS in 1h 14m 40s (non-voting)\n- kolla-ansible-ubuntu-source-mariadb https://zuul.opendev.org/t/openstack/build/41137795c6bb4af2856fdb494790140d : SUCCESS in 25m 54s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/8b25f9cbd487419b878e16dd1687af6b : SUCCESS in 47m 50s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/32e537580af54da28231fa2ebd98f6d1 : TIMED_OUT in 2h 05m 38s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/3b479661cfcc4c28bcb1a8e1cf7f5d8f : FAILURE in 49m 31s (non-voting)\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/cf8dabcf2d3f4af3a1c7a3c41b63d828 : SUCCESS in 29m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b1a6380ec7c25da331b9fce7a615f4d19092d5cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-12-11 12:58:03.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"0db46928ab2e33c2a59158532c7431db5562cb98","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-12-11 12:59:13.000000000","message":"Topic casefix-patch1 removed","accounts_in_message":[],"_revision_number":2},{"id":"97ff7752e56a64cd4ff7d7d7a00f6e0e65e192e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 15:13:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/5ca874d619f348a8a521a9e93b8d35e4 : SUCCESS in 20m 02s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b1359a44854e434e81f08a510202c78d : SUCCESS in 38m 42s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c4f1b46abf9b4824959d8198665d7121 : SUCCESS in 1h 06m 37s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/27304673652a414d82afc019f766a445 : SUCCESS in 59m 53s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/aa6a8d4a12fc4291a4fbbed479605011 : FAILURE in 1h 07m 44s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/77933d93adc04d228beaac891320e323 : SUCCESS in 37m 10s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f22fcf34e0484efd9587c44199e836cf : SUCCESS in 20m 01s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/665cb6b17e9042289f81ed27aa4d9397 : SUCCESS in 1h 33m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18522167065345c69a64a1536bd02289 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74a7edf4c8d34a18ba9dc6a8a2adce1b : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90856a873132469fb179251df33aed7e : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dcd491685a9945d0beb7d03901002a44 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b20464d45ac14a65b142fbd00a5a4dcd : SUCCESS in 7m 33s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7fc405884d034bc780422eb985314575 : SUCCESS in 20m 09s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6c9959a9b32341c5ace2c83253829cc6 : SUCCESS in 40m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2200955b321b4d5f988d73a32166a83b : SUCCESS in 44m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8ee4934c20c54ec3860423f7013a40ef : SUCCESS in 52m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b0baa74cb135484e9f2e03f776e45f2c : SUCCESS in 1h 22m 23s","accounts_in_message":[],"_revision_number":2},{"id":"78c32e2fcc6387bbc04aface07049be2ba9f206b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 16:19:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/02199011454e46db9fda7d46d69d7172 : FAILURE in 2h 00m 22s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cbbc67cf7680406a9c9f23281b096468 : SUCCESS in 3h 01m 32s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c84f85655f90b6f2a1118268531aed8cc4b63eed","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-12-12 18:44:23.000000000","message":"Patch Set 2:\n\nmariadb fails in upgrade test job - any hints?","accounts_in_message":[],"_revision_number":2},{"id":"5c1b252993e9529ce1b2c5b0779377fc9d9fc0f4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-13 10:24:29.000000000","message":"Patch Set 2:\n\nI\u0027d say it was a random issue; not rechecking because gates still unhappy about the dockerhub load","accounts_in_message":[],"_revision_number":2},{"id":"95258df290ca65c65a39ad02ebae006ef2eb5218","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-16 09:50:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bec838fd5e9fd9cf7b27ada911bf0af6953ee369","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-12-16 09:53:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"410626a37ba0e74c5d9225fcdbf32dea0119e1c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-12-17 12:36:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"21b57aee6f2df52e7e35703b57af46a604d3a817","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 13:52:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/80d0584499914612b4ee801e257705bd : SUCCESS in 20m 10s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d6718cc3626642528da677ba2723ab8d : SUCCESS in 27m 42s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/af2bfa0f25184704a4c3a469ed706825 : SUCCESS in 1h 03m 43s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/92e6019a418b4f4087fece3eb94e3ac7 : SUCCESS in 56m 21s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/2c4bb913849d4dadb40755ec86708be2 : FAILURE in 1h 01m 11s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/60ff8b94651949b0a87265c415f00afa : SUCCESS in 27m 50s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/932775518f644fd88fe5cf793f3dd7d3 : FAILURE in 19m 50s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/198922ede060487a889fe01ce74865df : SUCCESS in 58m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56d6ff2b5aef41419f6c46d535bbd41a : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2778f6cc19e94d79b84abdbe7d35dcc3 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1544a57efbed4f61a18439b670d35482 : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/123b5709fa4b4bf7b87802cd6a390b14 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9691b7130bd94c2b93aa077fc0c2ad2a : SUCCESS in 5m 22s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f152e3ab7f6041e18ad325ca566375d3 : SUCCESS in 17m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c9d3e7abfc1a49069278aa583249800a : SUCCESS in 36m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e010e9fa47e847fcad1024b60a9b7356 : SUCCESS in 1h 02m 49s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/78f30c515bf14535b01fcdf2d1b39cbb : SUCCESS in 1h 04m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3a5d6325455f4c949c647a9e916cc5ea : SUCCESS in 59m 29s","accounts_in_message":[],"_revision_number":3},{"id":"dda5f258bf2a27f8ef5b82352b3a54f21826a3d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 15:16:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/f236334f6c624ae098baceb7c55a7974 : FAILURE in 1h 11m 46s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bfd91f6728af4c718a77b642e97fa890 : SUCCESS in 1h 27m 38s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a4c5f6e64db5461ba9bc35a13220940b : SUCCESS in 1h 29m 51s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9d55e3c8978d4b9892a5da66cec67f3d : SUCCESS in 1h 20m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7b2553ca069554e5d78e0bf99f373b418e0d0990","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-01-21 15:37:39.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"5c63a1f13172071819d4f757be0750e90639357a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 17:20:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/f34d1e80230b4bc9ab52593cce73c40b : SUCCESS in 28m 32s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/0d46283c453c43a3b71ab1be9ad13f75 : SUCCESS in 46m 48s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/00a2a1db5f7c40c58fb1e93bbf6c9808 : SUCCESS in 1h 08m 55s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/877503b982b4476a912e43a93e32b197 : SUCCESS in 1h 07m 43s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c3f5558f5a5640839efb8636c1bf5063 : FAILURE in 1h 12m 01s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/118f9dcf1c6b4dd691a1c500b9bea2ec : SUCCESS in 26m 11s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f794fb761f8f4d50bfcfc9f745010bdf : SUCCESS in 18m 16s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/82b64838966c45fe9e33d51def71ce45 : SUCCESS in 54m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcde56c9a24e4aa7934bda454930e403 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3ab7ecd3f9942d5baf82b2902f5d453 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9661240d675c4ec08dc532f20dc84e3b : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e48e1e9896442e489064d11e1940a2c : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26331101eae341c0ae9c578dd6e538fb : SUCCESS in 7m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a4dab2d3a9294779be8f36002c1edc6e : SUCCESS in 23m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/06820d33900b41b599ab68c67070b5c9 : SUCCESS in 38m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0335f1f56f9b4b6ebfd67b1277e7637b : SUCCESS in 54m 43s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f70fb3568cbc493fa0f38a6c2b1eba54 : SUCCESS in 1h 00m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/056d50b38d8647c5bf1d03faafa9e6cc : SUCCESS in 58m 59s","accounts_in_message":[],"_revision_number":4},{"id":"af84c6f3ee51ab583b9a3cff8b715e20e4695dc3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 19:38:09.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3959ac3584150191638e09071fb42bf9d5f4c8e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-01-29 09:37:13.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"b6db372860d6c66b499154201c09160dfaf48d7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 11:09:36.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/fe5a212a410d4391a4cb3c236b98872f : SUCCESS in 19m 17s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/91e1d344c9c64524b5a9c462277d88b8 : SUCCESS in 39m 56s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/96ed14615f3f483990307dd88a2c398b : SUCCESS in 58m 16s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b0c6424e07394a3aa22ef9de2a82bb8c : SUCCESS in 1h 14m 41s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/08f62c99b8d64d188b13b551cc730659 : FAILURE in 1h 15m 41s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6ec2dd820c364c32aa4599ef9b364035 : SUCCESS in 32m 21s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5a7b982a226c40d6a0ebbb0b52be98be : SUCCESS in 17m 54s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7e7ad6aaa1fd42a5bf84c01098b68957 : SUCCESS in 50m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8153321414d427ea89a143f43f37f77 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80c112a10f7b4fc7a44477158b0151fa : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe63b4ae5d15406e8556756fd4545320 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99872f6c53e24eab9bf088bbff37a00d : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/466a5888b00d43d4993022779f611035 : SUCCESS in 5m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cac558df598e4719ab12f6af037831f1 : SUCCESS in 33m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f4b35fb91e9641498a1da715535baa67 : SUCCESS in 32m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0c64d0fd440a42b4946415dbc95ffbb1 : SUCCESS in 48m 43s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/27fcab8ad2d148b88710031e2684f10c : SUCCESS in 47m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a240e7a316b34a178dd00abf855e2f92 : SUCCESS in 1h 04m 34s","accounts_in_message":[],"_revision_number":5},{"id":"75a2f4bf91103f2daefb27d490e065a218106e39","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 11:17:02.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/4db7ec38b41d46f5908efc8ea7ee1c1d : SUCCESS in 55m 46s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1e4c3a7c5f324868948c5635067ad1df : SUCCESS in 47m 55s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b570b8271a0748a9bf7d0fbf999b8349 : SUCCESS in 1h 33m 43s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/831d4f722ab242fea2da59c916be04bf : SUCCESS in 52m 51s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9958d4a3bda8c3a7a853798a46349a7dac217d09","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-03 15:58:41.000000000","message":"Patch Set 5:\n\nFailing CI","accounts_in_message":[],"_revision_number":5},{"id":"48c110e2797de5e389001fd27a05b8e42b47e2ff","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-25 10:49:51.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"546e9a8babef4226b0a260c956df70218a26677b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 12:06:07.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/f60f033a915f4a439bc5de6ee6f48728 : SUCCESS in 35m 16s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/90cfd56a4a484206be3118bb56a95fe8 : SUCCESS in 35m 44s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8c85c31c73c1498b934ddea691abab8c : SUCCESS in 50m 30s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/24313a428e24400d9deb4e33ff3b5d39 : SUCCESS in 1h 01m 38s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/753471b57567487cba077783eadf97d8 : FAILURE in 1h 09m 46s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/053d9e9a3b064db09cb30a9e27acacee : SUCCESS in 44m 10s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4d5227067d5c4c58bc9be3d4bd7358ce : SUCCESS in 22m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4bcbc30d63e44e0e884ffa81816c21d6 : FAILURE in 41m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11943ab995ce49f19fd861b9de766bff : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14a8cc7a63d64df681c645861491964d : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f093d173e2bd44d8aa4a3cc3040c7b40 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f3b6d4aa2ad48a7abefce6ebb85b0d9 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad674e87cbab4a5586d742a3cfcf4d03 : SUCCESS in 6m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/23ac56918cc542a8a1d6f9ee7be9c10a : SUCCESS in 21m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/af4dcafea0a349d7b92123a7e326f782 : SUCCESS in 44m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/569193404fa944af91fab9b4ce0281cc : SUCCESS in 49m 17s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/54c9dea5c5d2441e9f64455f5567fdfd : SUCCESS in 48m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a94e1678a8b94137bf4bfad109a3f2ab : SUCCESS in 1h 08m 47s","accounts_in_message":[],"_revision_number":5},{"id":"95a93198434aec66e3b8719a99dd12777037ee61","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 18:10:47.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f1564a24c7f34ff6a884c4a814ff0cf6 : FAILURE in 1h 00m 20s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3bc058c53746467a8c7ee5ea5277af1e : FAILURE in 33m 21s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b9c74194e66343fa9b021d0362ddb46a : FAILURE in 56m 21s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"588430a5e76c25296075fb6d1c4a383c77936a4a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-25 19:36:52.000000000","message":"Patch Set 5:\n\nupgrade failure seems repeatable","accounts_in_message":[],"_revision_number":5},{"id":"8b4b074fdb72214cb0ee5eb1f845e10a14e8451b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-03-26 15:27:52.000000000","message":"Patch Set 5: Workflow-1\n\nWill look at it in spare time.","accounts_in_message":[],"_revision_number":5},{"id":"3b3e6937ceebf979acf2f4fcaac66ecf07181f49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-12 11:00:55.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"1b19ec33ceacd772abb84da181ed39657f627510","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 12:00:53.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4ac8cf0086e74aa1b3749931ac38321b : SUCCESS in 17m 58s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/0c7289b476b1466b9ef3333372e8ee32 : SUCCESS in 39m 05s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/83ba7f0320214e95af8525860f363c7b : SUCCESS in 35m 23s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8d809daa2dce45f698db7a302e490d51 : SUCCESS in 21m 06s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/53a9154e3a004e8b85c682ab5ba144d0 : SUCCESS in 49m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e503cca955c947a981de4b92a2c3fbdb : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/05c19f28c6e14b1898c1885543aca67b : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/faf62a66c57441c8b6a9ab0f4ac8feda : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f282468ccc4491086ce535f92839120 : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62fcf2c5945b47a787d206bb76762f3a : SUCCESS in 5m 13s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9ba5bb1b41c6422caca5a23a11382aa9 : SUCCESS in 20m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/754dce818ec947beae51b6437762a0db : SUCCESS in 36m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/010692cadb7b44659d2ad76d80345e28 : SUCCESS in 48m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9e535a9114d2428684f4929d6e85f283 : SUCCESS in 40m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/56f6e0e3dff34e319db72e9480487385 : SUCCESS in 58m 25s","accounts_in_message":[],"_revision_number":6},{"id":"2fe256c9eeadc4d5402f0831e28b9ce01f2ec256","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 12:46:52.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/4d824f741494457b934a31b6d1253918 : FAILURE in 1h 04m 07s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6136a3a725a14ef6b51e7c211665ac4b : SUCCESS in 1h 02m 44s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b42e959c5b29434f8aeed961a2bec182 : SUCCESS in 1h 40m 44s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9a4bd6326cec6ff8a2dd538a752d2d198598d90d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-14 14:34:40.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"daa0a7c09ce2e190aa1be24b022a8e6d05117a00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-14 16:39:55.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a985bbbb9582706973e8ff1c887fbc4220267fec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 17:52:46.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/1c88191bdbd14786b4516429d4cbd5a1 : SUCCESS in 23m 27s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/76fd129d2f2b4f6f913ae09b74de0eec : SUCCESS in 36m 52s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/acf83270d5c244b6b36ca8ba76a7d0db : SUCCESS in 28m 43s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0f38bbd158a04dd1b6596ef29f5eb3e3 : SUCCESS in 19m 43s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/530f2b3d3375436ba30ec4a02584d265 : SUCCESS in 51m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4695800548e64dea9264a2bc8033707c : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17793bd8477542809d135fccab053149 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ed8dda75e894f9ea0c90bc950da4017 : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62c6891c94694ac9a12ec6cf71b37c4e : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e3d49f619154c44a66a156e3967de78 : SUCCESS in 6m 11s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d421bbbdb03745ac8b63286a767dc789 : SUCCESS in 35m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2d53cc21693b469e9afa22a1aebff840 : SUCCESS in 47m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/83728d4fe7c84ef89f74597c8db700da : SUCCESS in 59m 35s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7f1cfedd3d604ec2a1962a3bd92100f8 : SUCCESS in 55m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e500a207c22a4af68223e14fb36eef72 : SUCCESS in 1h 08m 22s","accounts_in_message":[],"_revision_number":7},{"id":"95c09e2fbc6ce1af88206f8ef530be3662918d10","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 19:09:05.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/80ac63e1cf3344cf8cf69ed8f776a5a9 : SUCCESS in 56m 50s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/db82f8b0ae6d4f85b2503dde47171b65 : SUCCESS in 58m 05s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3cc93f4548d04e309668891c76f8bb99 : SUCCESS in 2h 07m 53s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"bec3a1a088ee48ae93f5bb71389949f1f7b3e4cc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-28 09:29:26.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"cbe9bc285e8601e86a00cba12bf26f1481b30b1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-28 09:33:18.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"fe6351de8cc10ee9cc670ef47ce627c32dee1559","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 10:36:10.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/25131ccffcb34748ae885b50160e99b4 : FAILURE in 23m 42s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/675d2288548b4e269d5d71a4f6b25a42 : FAILURE in 39m 51s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/efc2e76dfe024fa49a4be8ed5faad2c6 : FAILURE in 18m 45s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/1cde7ac3efbd474896a2f0fb924724b7 : FAILURE in 21m 22s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/146679f605874049ab5a27dfe8bead9c : FAILURE in 20m 00s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e418fd86371d4db4b61e6f289cdd4538 : SUCCESS in 35m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/100f9bd29dc84ca3ae1de97297b1baf0 : SUCCESS in 16m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e38b80f155424c53b3c528edd138ac60 : SUCCESS in 52m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/558710091c464798b778893b1df06c20 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d2e2514d76e4796b99d7863ed0efa77 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1419799f95442c4a9d9ca9ac7780bbe : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/124fde696ac74e31856a066c53cfb54b : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16e5250869ce4b2f9b9fed7cf47a773b : SUCCESS in 7m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c492f5b7cc704405bacb2f593a42437d : SUCCESS in 32m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f2e28e96bb7744e4b17033256440ee4c : SUCCESS in 52m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4cc2676339c048319ec2ac2c3611977e : SUCCESS in 54m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fc2868468cee4bad9f51c7a4733d4786 : SUCCESS in 55m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d5392e8f143b4c07a29d95804232aaa1 : SUCCESS in 1h 01m 01s","accounts_in_message":[],"_revision_number":8},{"id":"7eb5900c3482c386f669a91ce1a28095b3b1e1fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 11:10:14.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ef492fea52ec4ffcb87521614aaaffcd : SUCCESS in 59m 52s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3061997fc0c54ba8a522f545249de7f0 : SUCCESS in 1h 31m 06s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"1ac5c4e97ad71e6ad8b554ff6fc8d0b38f847fb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-17 07:57:47.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"228c7d4744c15586c1ff786b6cd9206a217300ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 09:19:41.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/7a90e5af5bf64afbaf0c063782fc26cf : SUCCESS in 21m 44s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/bebdb02ec3ec4e1fa7ececa7fa884f39 : SUCCESS in 41m 18s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0dae191616a74fb6b1d7ffcd69b508b0 : SUCCESS in 1h 11m 04s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/0644a15950e34511881c6818dd44937d : SUCCESS in 1h 10m 26s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/705221b7cb574e538713404f9883256d : SUCCESS in 1h 21m 25s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a0fca2f2d1a041d19c94b3517a8378a2 : SUCCESS in 34m 05s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4f6a3822cb3e45b0b90aa376f30a9031 : FAILURE in 18m 47s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9868377b03c041b3b3fc1c368c05117d : SUCCESS in 57m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3f84559fbc74d45a622e66bfb699f74 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a67b67a46b874817aa00daee6ed3a224 : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b524abf4e9f45f3aa2eb5332be9bf0c : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c03b87959034e778186013427c16e79 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fea651d918b4ee3a45b862ba9bcbdaa : SUCCESS in 7m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b29248d9c9094652ab183df12f01653a : SUCCESS in 28m 20s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5f58c6c2ef1e401fb33906e2a34e57c2 : SUCCESS in 42m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e13fc2414d454adb8de166b1701ad337 : SUCCESS in 59m 45s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0e42e600b72e4d6f98b9c1efcdc76f4e : SUCCESS in 55m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c4a1bd3779e64d6e8d2c1bbd15604db9 : SUCCESS in 1h 03m 14s","accounts_in_message":[],"_revision_number":9},{"id":"673a1f03965213cfd9b90ba5f3ca667a15eda130","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 10:27:56.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/af41a0ec455043a1b548f9c860c11621 : SUCCESS in 2h 15m 56s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b61d7de2b5794b869d50fc025e887808 : SUCCESS in 1h 11m 37s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/af8108f820ca4664bd0b8bfcb6a0548c : SUCCESS in 2h 28m 09s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/51d5b06c24284b1c8934ca305e9177cd : SUCCESS in 1h 52m 31s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"11d1b50b41c3c602f68f4457cdd84b27f32c6a0f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-05-17 12:45:19.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"a86922ab933d7b9eea2db8ae1174124df4507ece","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-17 12:57:52.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"eeee088ca49ad34d0449a3fd3aa7392c755a9ca6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-17 12:58:59.000000000","message":"Patch Set 10: Backport-Candidate+1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"f000c77942fc4c2b84ae0bf69ef2b31c9947d75c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-17 13:09:20.000000000","message":"Patch Set 10:\n\nI abandoned my https://review.opendev.org/c/openstack/kolla/+/788461 as this includes it but let\u0027s backport it","accounts_in_message":[],"_revision_number":10},{"id":"6f20a72954b620de6eadf6bc61cb7c709d84efbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 14:47:17.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/89c635800a4e4e37bd52055d5cf4880f : SUCCESS in 1h 47m 06s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/1b9a6df22b5c448f959cc98597dad489 : SUCCESS in 37m 16s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c293dc524ff94e999d9ebccf52caac06 : SUCCESS in 1h 07m 53s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/ba9288e29d154bfa9dbd93046c8df71e : SUCCESS in 1h 09m 55s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/19be8adb89554c6abbb2c5b54c803620 : SUCCESS in 1h 14m 20s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6f3720a676fc4292aa4f17b2ba6f8160 : SUCCESS in 45m 49s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c0073440571f4740b3eb35babbee73b0 : FAILURE in 17m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/499cce1edfe941beb71961b99ac07558 : SUCCESS in 1h 07m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/749f2b2c10504a7db2083ff60409b60f : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3b35237d8a54e3985fe2460699f2e4b : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74677295d90641a280373739a0a004a4 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10f56208209142fa8421fed3a5a56912 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4ad10c3239d456582747c47f8322c4c : SUCCESS in 7m 50s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2a08d6ec3be04e4895cd937615024936 : SUCCESS in 23m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/86653e0b77354f9eba206faca9f47fba : SUCCESS in 46m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2addef61f9bb4f0094f6f338a24bd474 : SUCCESS in 41m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/34ef6f10e5da4244aa1695cd1077218e : SUCCESS in 43m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0f915e49a33042e3ae69bada54747717 : SUCCESS in 1h 03m 36s","accounts_in_message":[],"_revision_number":10},{"id":"5c63e04fb6d096c1b7b05569dfda962d2c83c658","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 14:47:54.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"d52d966c3e7721adb7171c3c8dd398975c28b88b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 16:04:25.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/97b301fed4fa4b4ead5d1bcdd32168c5 : SUCCESS in 36m 08s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ffe94cb95d544314b565db05a6fc17b6 : SUCCESS in 1h 09m 07s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/0fa7a20d27724e419ccd748fef0ccdea : SUCCESS in 1h 12m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/417234590c504075a8c45981a42bda77 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/156df5c68d5a463a9ff47bd05776df3d : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e9ff3fa1fb74193995294b24be433e0 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef319fbdcaad4fc9bb27083d8783403e : SUCCESS in 6m 29s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cd29e6bc36994000af3e13b110f2b764 : SUCCESS in 37m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3e6b240c9409473382dfe094abcce2cf : SUCCESS in 54m 25s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/08c227e39af345298892fd1d0e4ec98c : SUCCESS in 1h 11m 05s","accounts_in_message":[],"_revision_number":10},{"id":"28a4581c7af0bb67cd7b53543ff2f3fee0501822","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 16:04:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"77cbb731bc803e6dbc5fe5a8285196e2dcd286c3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 16:05:58.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b763b03686840bea8a9c31f66c1391a : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":10},{"id":"82dd4bd6ed092d31376b348f6f7c31f9e276e1a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 17:54:51.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/43e00010b033422aad7b65a4458e436d : SUCCESS in 2h 44m 36s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/60255bbff1984dbeb06511a4a0b0425c : TIMED_OUT in 3h 07m 35s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0eabd44963c64e8687cb429dcc29580a : SUCCESS in 2h 31m 09s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/ca5e1d94e4944ddb83aa55c4502171fb : SUCCESS in 1h 45m 43s (non-voting)","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"a93f171da3df0491e2bf3d5c1bf7a5a856a6e128","revisions":{"0aa1ee57bc09602236dbeb5ac2aa7ab5abf273b7":{"kind":"REWORK","_number":1,"created":"2020-12-07 20:02:25.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/64/765864/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/64/765864/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/765864/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/765864/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/765864/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/64/765864/1"}}},"commit":{"parents":[{"commit":"32069fd4f95a0052bbd187483caedf208147525f","subject":"centos: disable PowerTools repo by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32069fd4f95a0052bbd187483caedf208147525f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-12-07 19:45:43.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-12-07 20:01:13.000000000","tz":60},"subject":"centos: clean up repo enable/disable/keys","message":"centos: clean up repo enable/disable/keys\n\n- install all \u0027centos-release-*\u0027 packages we need\n- disable all repos in one place\n- do not import CentOS keys - they are trusted by default\n\nChange-Id: I97308c74406d2af0e39b77eddef2693554381ab6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0aa1ee57bc09602236dbeb5ac2aa7ab5abf273b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0aa1ee57bc09602236dbeb5ac2aa7ab5abf273b7"}]},"branch":"refs/heads/master"},"d74d664f48b4160584918240da2f116de57f9510":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-12-11 12:58:03.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/64/765864/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/64/765864/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/765864/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/765864/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/765864/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/64/765864/2"}}},"commit":{"parents":[{"commit":"360d49002d65dfc4a7a3b69991953f4d54fe9ad9","subject":"centos: disable PowerTools repo by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/360d49002d65dfc4a7a3b69991953f4d54fe9ad9"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-12-07 19:45:43.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-12-11 12:57:11.000000000","tz":60},"subject":"centos: clean up repo enable/disable/keys","message":"centos: clean up repo enable/disable/keys\n\n- install all \u0027centos-release-*\u0027 packages we need\n- disable all repos in one place\n- do not import CentOS keys - they are trusted by default\n\nChange-Id: I97308c74406d2af0e39b77eddef2693554381ab6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d74d664f48b4160584918240da2f116de57f9510"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d74d664f48b4160584918240da2f116de57f9510"}]},"branch":"refs/heads/master"},"c9ab41f2425e91ef30c4c3fb0e947ab8cda339db":{"kind":"REWORK","_number":3,"created":"2020-12-17 12:36:26.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/64/765864/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/64/765864/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/765864/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/765864/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/765864/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/64/765864/3"}}},"commit":{"parents":[{"commit":"1d36ae464830061473302e2ddc563e9c229aed5c","subject":"Merge \"centos: do not enable PowerTools repo in base\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d36ae464830061473302e2ddc563e9c229aed5c"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-12-07 19:45:43.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-12-17 12:35:51.000000000","tz":60},"subject":"centos: clean up repo enable/disable/keys","message":"centos: clean up repo enable/disable/keys\n\n- install all \u0027centos-release-*\u0027 packages we need\n- disable all repos in one place\n- do not import CentOS keys - they are trusted by default\n\nChange-Id: I97308c74406d2af0e39b77eddef2693554381ab6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c9ab41f2425e91ef30c4c3fb0e947ab8cda339db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c9ab41f2425e91ef30c4c3fb0e947ab8cda339db"}]},"branch":"refs/heads/master"},"2b24df63e11eecca077cc8a6ce1b289a25500ec4":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-01-21 15:37:39.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/64/765864/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/64/765864/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/765864/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/765864/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/765864/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/64/765864/4"}}},"commit":{"parents":[{"commit":"6fbae18f075a7f92923766be5b980acc892a0e00","subject":"[docs] Add link for project\u0027s release notes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6fbae18f075a7f92923766be5b980acc892a0e00"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-12-07 19:45:43.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-21 15:37:39.000000000","tz":0},"subject":"centos: clean up repo enable/disable/keys","message":"centos: clean up repo enable/disable/keys\n\n- install all \u0027centos-release-*\u0027 packages we need\n- disable all repos in one place\n- do not import CentOS keys - they are trusted by default\n\nChange-Id: I97308c74406d2af0e39b77eddef2693554381ab6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2b24df63e11eecca077cc8a6ce1b289a25500ec4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2b24df63e11eecca077cc8a6ce1b289a25500ec4"}]},"branch":"refs/heads/master","description":"Rebase"},"a8b6868082dee9984d7d0ed458916341e61c2ea4":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-01-29 09:37:13.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/64/765864/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/64/765864/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/765864/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/765864/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/765864/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/64/765864/5"}}},"commit":{"parents":[{"commit":"849b31303f4aad7d6fb190a80cf95514e8755279","subject":"Merge \"Fix Mistral source images to respect upper-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/849b31303f4aad7d6fb190a80cf95514e8755279"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-12-07 19:45:43.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-29 09:37:13.000000000","tz":0},"subject":"centos: clean up repo enable/disable/keys","message":"centos: clean up repo enable/disable/keys\n\n- install all \u0027centos-release-*\u0027 packages we need\n- disable all repos in one place\n- do not import CentOS keys - they are trusted by default\n\nChange-Id: I97308c74406d2af0e39b77eddef2693554381ab6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a8b6868082dee9984d7d0ed458916341e61c2ea4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a8b6868082dee9984d7d0ed458916341e61c2ea4"}]},"branch":"refs/heads/master","description":"Rebase"},"02b64a0c5dad9e53064596832054ff221457ce37":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-04-12 11:00:55.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/64/765864/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/64/765864/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/765864/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/765864/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/765864/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/64/765864/6"}}},"commit":{"parents":[{"commit":"c7dc1f77146e2b226a2cb7e3ec564777ffcf2c34","subject":"Merge \"Add the healthcheck_rabbitmq script to rabbitmq images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c7dc1f77146e2b226a2cb7e3ec564777ffcf2c34"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-12-07 19:45:43.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-12 11:00:55.000000000","tz":0},"subject":"centos: clean up repo enable/disable/keys","message":"centos: clean up repo enable/disable/keys\n\n- install all \u0027centos-release-*\u0027 packages we need\n- disable all repos in one place\n- do not import CentOS keys - they are trusted by default\n\nChange-Id: I97308c74406d2af0e39b77eddef2693554381ab6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02b64a0c5dad9e53064596832054ff221457ce37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02b64a0c5dad9e53064596832054ff221457ce37"}]},"branch":"refs/heads/master","description":"Rebase"},"4fc6e6a1d728f472919f3bcb8bae952c0bf59202":{"kind":"REWORK","_number":7,"created":"2021-04-14 16:39:55.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/64/765864/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/64/765864/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/765864/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/765864/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/765864/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/64/765864/7"}}},"commit":{"parents":[{"commit":"3330f624f0355b6d53d1e4f34b5bad72624f0a1d","subject":"Merge \"deprecate monasca-grafana image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3330f624f0355b6d53d1e4f34b5bad72624f0a1d"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-12-07 19:45:43.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-14 16:39:13.000000000","tz":120},"subject":"centos: clean up repo enable/disable/keys","message":"centos: clean up repo enable/disable/keys\n\n- install all \u0027centos-release-*\u0027 packages we need\n- disable all repos in one place\n- do not import CentOS keys - they are trusted by default\n\nChange-Id: I97308c74406d2af0e39b77eddef2693554381ab6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4fc6e6a1d728f472919f3bcb8bae952c0bf59202"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4fc6e6a1d728f472919f3bcb8bae952c0bf59202"}]},"branch":"refs/heads/master"},"1ff7a1450ebb0f296f544e91c777e03645e292b1":{"kind":"REWORK","_number":8,"created":"2021-04-28 09:33:18.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/64/765864/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/64/765864/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/765864/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/765864/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/765864/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/64/765864/8"}}},"commit":{"parents":[{"commit":"a4f6a8f05b45838338ee1c6a78f278d49486ac03","subject":"neutron: Remove vmware-nsx and networking-mlx from binary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a4f6a8f05b45838338ee1c6a78f278d49486ac03"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-12-07 19:45:43.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-28 09:32:42.000000000","tz":120},"subject":"centos: clean up repo enable/disable/keys","message":"centos: clean up repo enable/disable/keys\n\n- install all \u0027centos-release-*\u0027 packages we need\n- disable all repos in one place\n- do not import CentOS keys - they are trusted by default\n\nChange-Id: I97308c74406d2af0e39b77eddef2693554381ab6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ff7a1450ebb0f296f544e91c777e03645e292b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ff7a1450ebb0f296f544e91c777e03645e292b1"}]},"branch":"refs/heads/master"},"e3676f04bd6699bfd15e6d20bebcc4f98bdecd5d":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-05-17 07:57:47.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/64/765864/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/64/765864/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/765864/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/765864/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/765864/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/64/765864/9"}}},"commit":{"parents":[{"commit":"73c456f85b4e861e4be12282a67ca59f5f319357","subject":"Merge \"docs: update supported OS distros\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/73c456f85b4e861e4be12282a67ca59f5f319357"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-12-07 19:45:43.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-17 07:57:47.000000000","tz":0},"subject":"centos: clean up repo enable/disable/keys","message":"centos: clean up repo enable/disable/keys\n\n- install all \u0027centos-release-*\u0027 packages we need\n- disable all repos in one place\n- do not import CentOS keys - they are trusted by default\n\nChange-Id: I97308c74406d2af0e39b77eddef2693554381ab6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e3676f04bd6699bfd15e6d20bebcc4f98bdecd5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e3676f04bd6699bfd15e6d20bebcc4f98bdecd5d"}]},"branch":"refs/heads/master","description":"Rebase"},"a93f171da3df0491e2bf3d5c1bf7a5a856a6e128":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2021-05-17 12:57:52.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/64/765864/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/64/765864/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/765864/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/765864/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/765864/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/64/765864/10"}}},"commit":{"parents":[{"commit":"0506ede84f97159a8156ad51aea395d0868e6141","subject":"Drop ppc64le support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0506ede84f97159a8156ad51aea395d0868e6141"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-12-07 19:45:43.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-05-17 12:57:52.000000000","tz":0},"subject":"centos: clean up repo enable/disable/keys","message":"centos: clean up repo enable/disable/keys\n\n- install all \u0027centos-release-*\u0027 packages we need\n- disable all repos in one place\n- do not import CentOS keys - they are trusted by default\n\nChange-Id: I97308c74406d2af0e39b77eddef2693554381ab6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a93f171da3df0491e2bf3d5c1bf7a5a856a6e128"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a93f171da3df0491e2bf3d5c1bf7a5a856a6e128"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
