)]}'
{"id":"openstack%2Fkolla~765916","triplet_id":"openstack%2Fkolla~master~I42972ae890a88f01fc34b7775077d68e45ade68a","project":"openstack/kolla","branch":"master","topic":"prometheus-v2","hashtags":[],"change_id":"I42972ae890a88f01fc34b7775077d68e45ade68a","subject":"Add secondary Prometheus image","status":"MERGED","created":"2020-12-08 08:28:35.000000000","updated":"2020-12-18 21:41:49.000000000","submitted":"2020-12-18 21:40:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765916-prometheus-v2","meta_rev_id":"0c0d153b52db09da38eac0d44956e4db745067bb","_number":765916,"virtual_id_number":765916,"owner":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-18 21:40:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-12-15 16:49:01.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-12-18 20:02:39.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-12-18 20:02:39.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-08 09:42:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-08 22:25:16.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-12-09 12:00:04.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2020-12-14 09:36:48.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2020-12-15 16:49:01.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-12-18 20:02:39.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"c82b12b8ef3ee3c662d21c8a02da1dc05f72b4cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2020-12-08 08:28:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ccee5a2fc2a2c61c59d71647a09d49a6916fe986","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2020-12-08 08:32:25.000000000","message":"Topic prometheus_2x removed","accounts_in_message":[],"_revision_number":1},{"id":"d16165df6c56b537d9d8de89c761ae96e16dadfd","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2020-12-08 08:32:33.000000000","message":"Topic set to prometheus-v2","accounts_in_message":[],"_revision_number":1},{"id":"0dbb2827fa0a9ddfafc77494807ead16593b507a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2020-12-08 08:36:58.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"34cf9bc7bfa53776e9d508f61a2dbc9a4d5c1d32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 09:42:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/d4d2d326f8064c9d98457f2c33941337 : FAILURE in 9m 49s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/184a90508ce64a958e673129f5e77153 : FAILURE in 9m 05s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/71f1675cfe2a43f2bef129dc79828e05 : FAILURE in 12m 12s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/1713c907f6f14961a30b809fe2d2b3b5 : FAILURE in 12m 29s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/356b1d9ae157472da8ce9a84ed91262c : FAILURE in 11m 57s\n- kolla-ansible-centos8-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/fd78bd7463c8456689f56a065fb6132b : FAILURE in 12m 14s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4569f6210bd14107b673e3261dd84ad1 : SUCCESS in 28m 02s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f4bf728784ee4bbfa45164cfd3304ac7 : SUCCESS in 16m 16s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/77055f2e86c2429db95f6d288c6e0593 : SUCCESS in 53m 29s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/903703de73ef43d2a3e33055770f2beb : FAILURE in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db2a8176801c4a8b8c2830cfea32c785 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b48856ff7f54667af275478b806f58a : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af6604b6a65a452f9bdcb785b2a7fd8c : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e032c0e9f73841f4b2802f07fe2484d3 : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c24f347ae14a4089936801578d15b300 : SUCCESS in 7m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/165080f46f6a4723912382efe0d0ec12 : SUCCESS in 4m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7f5ec5121da34491bebc8497902cb9b3 : SUCCESS in 23m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9e4f61186b5c4265b97aa7f49af94baa : SUCCESS in 37m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/117fd7aadc834195acd13de857ae9d75 : SUCCESS in 59m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6e34b7be459c4d49a55dc03239493a59 : SUCCESS in 57m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/57cbb95eea234c0285335784cc612874 : SUCCESS in 1h 01m 06s\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/a75c77813ec54af69bedf3bb71e9f772 : SUCCESS in 25m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4e361d71ec00f1f9fc8bca885e08d6caf009e1be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 10:26:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/26f854413d22453683aca2360070f1b3 : FAILURE in 10m 12s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6990d709e6d44c42a8f6bc5e95786411 : SUCCESS in 46m 44s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3f180a3433b74803aa5c99ae291f1fbe : SUCCESS in 1h 42m 06s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a0e08ad5462a4e64a5bd1af77a9357cd : SUCCESS in 54m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7f231dfb291ca06e7c5a3bf452b15372ea01311b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2020-12-08 15:02:24.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"85b774e7df2763b4ccd40fcbfacdb34f245f6c40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 16:10:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/dc380cfea1ed41ccbf2d4dab9b4419ba : FAILURE in 7m 50s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/83e61e87e35c45918bcdd0c9016e8a8a : FAILURE in 7m 37s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/b41fc49e2a8a443f91f93db56a1865b9 : FAILURE in 11m 14s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/d971223229294445a25de5f437b9e066 : FAILURE in 11m 26s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/638197929d674a72ba896325b3c18572 : FAILURE in 12m 15s\n- kolla-ansible-centos8-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/59f47131aac848a4be6ec4bf24a89c48 : FAILURE in 15m 48s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/01fc825118e346a29138158b3352bda7 : SUCCESS in 36m 36s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3c34d6be261044b799ef5f4934dc5523 : SUCCESS in 20m 37s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/826eeb6ef4bd4a04a2ca9e5ff5cad608 : SUCCESS in 58m 56s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af1b690620b04d2d91f98fce03efae33 : FAILURE in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91f0de167a194fc2bd2dcba804f212f2 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f707828a8839405aaa34932ca4a44ab2 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11125a35cf4b434c93214cd12e2ced09 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1504a429411b4317b2567ed68aa5b9db : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab56c31ceb954c80b11232bd9ed64a71 : SUCCESS in 5m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ea57e0982a84543bbf2dab725bb2e60 : SUCCESS in 3m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ee7ddb5edfa64407a266fa4c1d52aac8 : SUCCESS in 21m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cdad1b1d48b74f27959355cfeffd575c : SUCCESS in 34m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/39de5446d4b847e1afd3b31c8c5d157e : SUCCESS in 56m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c2f271e466e84fefb60cbc1fb91eb608 : SUCCESS in 52m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4bc8018f9561437da99a4e94f167a9e9 : SUCCESS in 59m 35s\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/bfb5494056914c49a2b175afd9cfd973 : SUCCESS in 27m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f9be11d6d5c971a659c249b890cae209f1c67a96","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 16:39:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/792f952f61514e708b660d147b902d1a : FAILURE in 8m 28s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7ebbefd438684dbf8b6da983e2d9e20c : SUCCESS in 44m 32s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/607e9b97f37c4dd0a7c9ac75b2fded6c : SUCCESS in 1h 31m 51s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/56cf345741f7411d9f7447ef1c40801d : SUCCESS in 48m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"070413e55ed10ff5db58b26fccd6db0f27f62e9a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-09 12:00:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b699eccd606ebfe26235f1a06c9fef4e3d79e0f6","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2020-12-09 16:16:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9306fb2ece55a303188c59e7964b620227a519a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2020-12-09 16:21:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"efcdd56ac8647522957fe678e0747a0f6ceb6d5b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2020-12-09 16:22:16.000000000","message":"Topic prometheus_2x removed","accounts_in_message":[],"_revision_number":4},{"id":"5db539d1eba8d0012e181381c6ca49473f05031c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2020-12-09 16:22:27.000000000","message":"Topic set to prometheus-v2","accounts_in_message":[],"_revision_number":4},{"id":"75c5d61ccabd595ac6a4111b0cf6b9ba65144fc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 17:15:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/4e0f38557eaa4bfd8c37f26f8e84fdb6 : FAILURE in 7m 00s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/5eacf0ad84d1441d9ae46e0ec95d81d0 : FAILURE in 7m 12s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/b630aeb7f93240cab3543d8fa9b41ce0 : FAILURE in 11m 11s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/324deb94fe924a19b2fc304e8efc64fe : FAILURE in 11m 16s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/8c42a2a86022412b9d9ea1c8b48f6718 : FAILURE in 17m 07s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9a01320ac03b4b168a8308ceac7fe154 : SUCCESS in 36m 28s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0b80998774924033841cae252a15eb3a : SUCCESS in 18m 42s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7fb299968edd48499a04a215bbaec40d : FAILURE in 21m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b692ac2a2c484aeb8e1c4b69578526be : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a3c637a3cf14a7a987759fac832657c : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23b4426c039a46e382067d9ee1bccf47 : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/33ffe9fb79d04e3486d0d78e7f37a8e1 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bac6039fcd504f8bafdf266605e8b860 : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43f1dce6af9f47c6af351d0982e5d5c2 : SUCCESS in 4m 32s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dbe6e25fdae14e14be456fcd9928a98a : SUCCESS in 21m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0c90e401e51b48688fc4a95eecbf7ff2 : SUCCESS in 37m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/713afd8d1d454c14b4c779dce3d1408a : FAILURE in 16m 56s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b491dfd680764ae3bc5a34019af5de78 : FAILURE in 19m 31s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3b245f5ae7114b0e97bba9071424d880 : FAILURE in 17m 36s","accounts_in_message":[],"_revision_number":4},{"id":"73fac7355ec96c12648992936227f4a580991e63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2020-12-14 23:08:54.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"8d10a713ebc8cf90d19c1cb9d86306c7428401a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 00:44:02.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/5e983ef4796e4a0caca0f5e5ad53b2b0 : SUCCESS in 52m 02s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7b79a33dd7e7498192e0aa328762f207 : SUCCESS in 54m 17s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bc6ca1f8d0254ed98f74c4c64b8e2ff1 : SUCCESS in 1h 29m 56s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b1932ddab6894d2d93542d99316f2878 : SUCCESS in 50m 02s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3280cc0a0ef87b3a5cb2671f9ee8d73bf5040b8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 00:55:32.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/90e95d3c914f4917b418fc02cf872e9c : SUCCESS in 19m 09s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6d7bfbb4dbff4b26ba2bf3ed58a3f486 : SUCCESS in 37m 56s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/890118f410544d16ba586ce97255f8c9 : SUCCESS in 1h 07m 26s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/2d0a80943609436cb5f6075565aaceea : SUCCESS in 1h 04m 18s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/2f461bfadb884e3aa6f6adf0f8e9a22a : SUCCESS in 1h 07m 00s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5fdc871386be423facb1be24edb573d0 : SUCCESS in 32m 58s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4a217e703e1c498ba54725428cc38c4b : SUCCESS in 17m 44s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ac516ad33b9a481ea2e3d647ac1f82eb : SUCCESS in 53m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bddd9309082b4d51b3536316444030ba : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6aaaf330f1304451ad22b7805837aa71 : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8943e2bf3f8241baaa39dfeedfe3fdc2 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f447d821d44457f89f22cecdcf821ec : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75c321cf1feb439a963a8b994464fd84 : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ea690d2ede24872ae40054565d162c8 : SUCCESS in 3m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/42f9b4589a164c07ba1ad988c66c87f1 : SUCCESS in 21m 44s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bd827c495e2343da8acb804b3fc9f71f : SUCCESS in 26m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e5c4b1d3f2704450928d20dfbb264ab3 : SUCCESS in 1h 20m 56s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dc2683fae99143f1ac6f25acb54ff1a7 : SUCCESS in 48m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/493df8cf623a4a5dbd386b98e68ac670 : SUCCESS in 57m 03s","accounts_in_message":[],"_revision_number":5},{"id":"d83881fce2d48315a367ed6ddc2441418e5688f7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-15 16:49:01.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"2a8c1532ee58ad080ab528bc1ae3ea7bc8a2faf3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-18 20:02:39.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nThanks!","accounts_in_message":[],"_revision_number":5},{"id":"1ddcef983d0b35600604fd922b4cc3e1e2af3292","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 20:02:56.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"abae19f9cd7689de021435e05952b75deedb5ab5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 21:40:23.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3a6f2620a2af4d9d9a13c3a20d17ec40 : SUCCESS in 58m 17s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c150308066e04cd28698c7bc09388b40 : SUCCESS in 1h 02m 30s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/0eea671273094b03a98bca4e9301ba99 : SUCCESS in 1h 30m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32e385b7c0604b20aacdee689b250d21 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0d9e8d463d944de8d0fce951f5209c4 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0f8a95607e943efa0cb454f54ddc67f : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7d5f758244f4758ba16a1913271d40a : SUCCESS in 10m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71c80bd8970b4f44862e098e7f84d1d0 : SUCCESS in 4m 20s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7f92d4e3615f4323aab973bad830d58a : SUCCESS in 39m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/85e353f3194e46ee8d573625f3ed1c85 : SUCCESS in 49m 09s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1934f18f6ace4053844c2a2b76901236 : SUCCESS in 1h 13m 09s","accounts_in_message":[],"_revision_number":5},{"id":"9c0f490a566040cc70c56a3448b60906f2c15c78","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 21:40:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"0c0d153b52db09da38eac0d44956e4db745067bb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 21:41:49.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebcdae5dd21b412388c5cb4ce91e1c5b : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a84c2ff54b944bba027c43e7363e8ab : SUCCESS in 56s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"ac53ca9b43b15e0a873448b9a2f3d600fa8d1ff5","revisions":{"4ba76d0df80573f465384298fb1e7d490b211bc8":{"kind":"REWORK","_number":1,"created":"2020-12-08 08:28:35.000000000","uploader":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"ref":"refs/changes/16/765916/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/765916/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/765916/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/765916/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/765916/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/765916/1"}}},"commit":{"parents":[{"commit":"22c32a30e62d42c3631306775c0cd029eb659e9a","subject":"Unify curl options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22c32a30e62d42c3631306775c0cd029eb659e9a"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2020-12-08 08:28:10.000000000","tz":60},"committer":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2020-12-08 08:28:10.000000000","tz":60},"subject":"WIP: Add Prometheus 2.x image","message":"WIP: Add Prometheus 2.x image\n\nNon-backwards compatible version of Prometheus added. It is supposed\nto replace 1.x version.\n\nChange-Id: I42972ae890a88f01fc34b7775077d68e45ade68a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ba76d0df80573f465384298fb1e7d490b211bc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ba76d0df80573f465384298fb1e7d490b211bc8"}]},"branch":"refs/heads/master"},"48fb619b62d5fb7f7d97f1ad3234f4d8a1c0e186":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-12-08 08:36:58.000000000","uploader":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"ref":"refs/changes/16/765916/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/765916/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/765916/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/765916/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/765916/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/765916/2"}}},"commit":{"parents":[{"commit":"22c32a30e62d42c3631306775c0cd029eb659e9a","subject":"Unify curl options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22c32a30e62d42c3631306775c0cd029eb659e9a"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2020-12-08 08:28:10.000000000","tz":60},"committer":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2020-12-08 08:36:58.000000000","tz":0},"subject":"WIP: Add Prometheus 2.x image","message":"WIP: Add Prometheus 2.x image\n\nNon-backwards compatible version of Prometheus added. It is supposed\nto replace 1.x version.\n\nCloses-Bug: #1906388\n\nChange-Id: I42972ae890a88f01fc34b7775077d68e45ade68a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/48fb619b62d5fb7f7d97f1ad3234f4d8a1c0e186"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/48fb619b62d5fb7f7d97f1ad3234f4d8a1c0e186"}]},"branch":"refs/heads/master","description":"Edit commit message"},"4e88fc564922772364d56e834c069eadcb26f8c2":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-12-08 15:02:24.000000000","uploader":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"ref":"refs/changes/16/765916/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/765916/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/765916/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/765916/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/765916/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/765916/3"}}},"commit":{"parents":[{"commit":"22c32a30e62d42c3631306775c0cd029eb659e9a","subject":"Unify curl options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22c32a30e62d42c3631306775c0cd029eb659e9a"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2020-12-08 08:28:10.000000000","tz":60},"committer":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2020-12-08 15:02:24.000000000","tz":0},"subject":"Add Prometheus 2.x image","message":"Add Prometheus 2.x image\n\nNon-backward compatible version of Prometheus added. It is supposed\nto replace 1.x version.\n\nCloses-Bug: #1906388\n\nChange-Id: I42972ae890a88f01fc34b7775077d68e45ade68a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e88fc564922772364d56e834c069eadcb26f8c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e88fc564922772364d56e834c069eadcb26f8c2"}]},"branch":"refs/heads/master","description":"Edit commit message"},"cbca07ed54a5aea691cd3a9b89c324dadb6816a8":{"kind":"REWORK","_number":4,"created":"2020-12-09 16:21:38.000000000","uploader":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"ref":"refs/changes/16/765916/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/765916/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/765916/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/765916/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/765916/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/765916/4"}}},"commit":{"parents":[{"commit":"22c32a30e62d42c3631306775c0cd029eb659e9a","subject":"Unify curl options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22c32a30e62d42c3631306775c0cd029eb659e9a"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2020-12-08 08:28:10.000000000","tz":60},"committer":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2020-12-09 16:17:59.000000000","tz":60},"subject":"Add secondary Prometheus image","message":"Add secondary Prometheus image\n\nNon-backward compatible version 2.x of Prometheus added.\nIt is supposed to replace 1.x version.\n\nChange-Id: I42972ae890a88f01fc34b7775077d68e45ade68a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cbca07ed54a5aea691cd3a9b89c324dadb6816a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cbca07ed54a5aea691cd3a9b89c324dadb6816a8"}]},"branch":"refs/heads/master"},"ac53ca9b43b15e0a873448b9a2f3d600fa8d1ff5":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-12-14 23:08:54.000000000","uploader":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"ref":"refs/changes/16/765916/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/765916/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/765916/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/765916/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/765916/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/765916/5"}}},"commit":{"parents":[{"commit":"9465c722f8a84ac07f9d2a99f935df4a1ce90123","subject":"Merge \"freezer: drop EPEL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9465c722f8a84ac07f9d2a99f935df4a1ce90123"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2020-12-08 08:28:10.000000000","tz":60},"committer":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2020-12-14 23:08:54.000000000","tz":0},"subject":"Add secondary Prometheus image","message":"Add secondary Prometheus image\n\nNon-backward compatible version 2.x of Prometheus added.\nIt is supposed to replace 1.x version.\n\nChange-Id: I42972ae890a88f01fc34b7775077d68e45ade68a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac53ca9b43b15e0a873448b9a2f3d600fa8d1ff5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac53ca9b43b15e0a873448b9a2f3d600fa8d1ff5"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
