)]}'
{"id":"openstack%2Fkolla~766410","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I2d67b3ed7b7c521bc755ec9a87365c9962a9eeff","project":"openstack/kolla","branch":"stable/train","hashtags":[],"change_id":"I2d67b3ed7b7c521bc755ec9a87365c9962a9eeff","subject":"Support CentOS 8.3","status":"MERGED","created":"2020-12-10 07:45:17.000000000","updated":"2020-12-16 00:43:02.000000000","submitted":"2020-12-16 00:41:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"766410","meta_rev_id":"1b0b4089186e10f5cbe504c7b0a76e3ec807324a","_number":766410,"virtual_id_number":766410,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-16 00:41:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-12-15 19:48:27.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-12-15 19:48:27.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-10 10:35:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-11 14:17:26.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-12-11 18:55:38.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2020-12-13 10:18:09.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-12-14 09:55:07.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"}],"messages":[{"id":"2b143856e4a04d0aedaf35d3d966b63138ceb0b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-12-10 07:45:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"398cc5ce38a724525f3eded139f633f503e45fcb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-12-10 07:47:15.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c679851db048c147ab6b0458b33894aa8852ef55","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-12-10 08:18:24.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"c70dffd0819cdee20120faa394032395a90e7db7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 10:35:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/2c0413b8473c45aa900107bdac7197a1 : SUCCESS in 37m 18s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/37cb563425c34fb4810a67d4dbbee16f : SUCCESS in 41m 42s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/143b822094f4414aac7b65f00fd51122 : FAILURE in 7m 27s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b402aef37cfd416797f1d8a290aebd57 : FAILURE in 43m 52s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/5be8a3bb886e43d5848b4ecb3cbd87b3 : SUCCESS in 1h 27m 18s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/93166b84e84c4b038a24af20d990ea69 : FAILURE in 45m 34s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/71600dc8e73841e3aacf7819d44d2044 : SUCCESS in 1h 22m 54s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/01f1c353612841f8bae11ba9b0ba1f2e : SUCCESS in 57m 57s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/841d3413cad04e7b9dbd1b274792f93d : SUCCESS in 1h 22m 04s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/b1100fee2ba5472a837be162e5f5b488 : SUCCESS in 52m 02s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/00daa84902724c71b6652993b86cfedd : SUCCESS in 48m 22s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/501270d4e6584ce9bd0b0df3a6657e87 : SUCCESS in 33m 46s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/0fed44e992d942a79bfa3e77b1e3c172 : SUCCESS in 1h 48m 04s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/38dd28c09417476f894cbb78099beecb : FAILURE in 1h 39m 37s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/db6c79feb1bc49358924a0107b860070 : FAILURE in 1h 29m 58s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/d18067dc1a274a04924e8a4966bd68fc : FAILURE in 21m 27s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/42436fe0edc345fd9b136af4cc36a393 : SUCCESS in 1h 55m 44s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/53a63344cbd940d48e66ad0cc14f61d1 : FAILURE in 20m 20s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/dbc9d856321d4a2e8e76160c933347e2 : SUCCESS in 1h 17m 00s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/1477779bcee145f197142f2a3cbae8d5 : FAILURE in 24m 32s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/421bf1552fdd456e8952a439ac76375e : FAILURE in 46m 47s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b795a0e3b14c4e8399c67d752d904497 : FAILURE in 2h 43m 29s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f7423559cb434822ba15bd3fb347fa95 : FAILURE in 22m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/141eb70b97834d6fbc55cc818d02a25e : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c242680e4da14f39b5c58e13a9cf07c2 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c607d5a5c6240b1be62f3bcdc0efab1 : SUCCESS in 3m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7f101a019acf4f5cbfa052b8c28c5323 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80b8bc6786254208938ff2bc23468389 : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/155d4e4a3b3a47a9908e37a7230d92d7 : SUCCESS in 3m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2f546b041e124fd380e432132e843926 : FAILURE in 4m 15s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/303d9853bd84480388874a662c216236 : FAILURE in 23m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e74a4f40aff54d4c8a00c740a3ed33a8 : FAILURE in 17m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d6c10092bff645c5bf068a51258ab690 : SUCCESS in 46m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f0c3d721f61048418f9f254fc036b846 : FAILURE in 16m 18s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/b5419497346a4d5d9387d2542e4a316d : FAILURE in 18m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/5a32c4286ba146448059a41481aeee7c : FAILURE in 18m 13s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/f35994ee11c245b8af3ca7a4a86e9e6f : FAILURE in 26m 31s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/3342a0ef0f614408b68f46a86d0d2d31 : FAILURE in 17m 45s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"11cf8e66c4689eb096d4334ede1b27a6c671bd9f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-12-11 14:17:26.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fe55c3353a76311c0d490e0b76e368951356e10e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-11 18:55:38.000000000","message":"Patch Set 2:\n\nlet\u0027s wait until ussuri merges....","accounts_in_message":[],"_revision_number":2},{"id":"fb3ae06a312b5d9f9e81cbb1b09e46165b8ef480","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-13 10:18:02.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"f0e32a52adea350fcdc10d22cf18d8743bf04e18","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-13 10:18:09.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f36a4d291b9c69a2775c520c066c956f57d118cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-13 11:07:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/6b78fe49eaa54db9b374499c534366ef : FAILURE in 4m 33s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c876d2bdf2cc4153b3d3ffc9c03bef72 : FAILURE in 4m 26s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b3190a5120214aca8e30e39c87105c4a : FAILURE in 4m 52s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7496e39058384cf8bc63bb9619bf8a7f : FAILURE in 3m 50s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/2cb391dafcbe4a85a139938ad243a2f2 : FAILURE in 7m 22s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/248e5c3cab4b42ffa296ea72441f849d : FAILURE in 19m 09s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/77ea5c1c75b04cee8c0d92b063074f2f : FAILURE in 8m 29s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/a460175b278d4dc0a5bfe271c643fe41 : SUCCESS in 48m 34s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/aaa1b922d01c4b658695a777c2fabe1b : FAILURE in 11m 04s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/ee050f778ba145d28dc58acdec6a957f : SUCCESS in 41m 33s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4a2f200842f44d658393d770daa3f4b2 : FAILURE in 2m 55s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2c8df13cf00c42ccaf62fdca9be63525 : FAILURE in 3m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c8ec26f07abb48ea83ef0a1373277c6d : FAILURE in 6m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/822cb68fe7c544b78896f25036bcbd31 : SUCCESS in 4m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/013aa04dfb8a4003a18eea11f948e169 : FAILURE in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0d86599d39c4a39a7baeaba580fae8e : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4e6a98b470bd4ba4ba0319186e72f3bc : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb70e7600d244c5eacab12c7d6c60d4e : SUCCESS in 5m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f32fb3d3536a47019a18ca3969827ac4 : SUCCESS in 4m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f028afdee971456aab5dd3fc1f270cd1 : SUCCESS in 23m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/aeb84e0d93f941e88639fe660207a9a6 : FAILURE in 19m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5b5dfecaedb74619bafd371cc337285a : FAILURE in 6m 50s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2b408ba878e54211b3e1e206a26248c5 : FAILURE in 6m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/17c6768627964ed0a59ae86dda6dfbcc : FAILURE in 7m 43s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":3},{"id":"114dd57b07a91418ed37b864e760953d221c7a72","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-13 15:29:30.000000000","message":"Patch Set 3: -Workflow\n\nmultiple issues. do not re-check","accounts_in_message":[],"_revision_number":3},{"id":"469ec343eb1b63669ce1c19047442b62d69054d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-14 09:54:39.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"96941538e316fab4af1793c6b2fb1e609a6b0a49","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-14 09:55:07.000000000","message":"Patch Set 4:\n\nNeeds more release notes, let\u0027s see how it goes.","accounts_in_message":[],"_revision_number":4},{"id":"fcc1c2dd7c69a7df26c22c3e40866b0c004e30a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 11:55:51.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f4e4e6bc4449475c8c0eaac82a9f5703 : SUCCESS in 39m 43s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/803b3b129ec74439ba6b516f38d6badc : SUCCESS in 43m 07s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/7cfc5a669cfa42e4b301c0eb93db069a : SUCCESS in 19m 37s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/fe9db3387d67402585589f17a9bb79e8 : FAILURE in 24m 28s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/1e955caaf31747be8237fa7b24399407 : SUCCESS in 1h 23m 16s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/5354eeebe6de40af90c6ad194bacf022 : FAILURE in 20m 42s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d548266d3c9c493186cc55dc7d1827f1 : SUCCESS in 1h 04m 55s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/e7a9a1e204644543beae2b5ec153e6f8 : SUCCESS in 54m 45s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/bc1472ed9d654d07857de1bb512a29ea : SUCCESS in 1h 55m 04s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/8602e1d6f6b447adbb3824fe00eb1871 : SUCCESS in 1h 07m 35s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ae391401988742b8b5060d9ddae35f4e : FAILURE in 20m 39s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/680f81bbc91c4e8ba96f1d5982127664 : SUCCESS in 20m 38s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3ef377f6fb5341f2bc7bd3b387f48796 : FAILURE in 19m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7aee5c31dc084bda8c92dc7d39a7414a : FAILURE in 3m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3de4b6bd4c4049c5a7a0f83a21a2a771 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5028a2c20c994ad4b3c035c6fa712bf8 : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f5c3ef1daf844a2cb45fbaf471067c2d : SUCCESS in 4m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f00392f643d6423e95d7e7718d27ddd8 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e656d525f124cb3af9c3e144720b610 : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d628d108c034c3bb8708db25b5f7393 : SUCCESS in 3m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/540635eb4e4c43a8902ba834812eaff4 : SUCCESS in 25m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b5b95e011f374fdd83a20fbabfd17009 : SUCCESS in 27m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e7a20c86c34e458682143c9aa4dacd04 : SUCCESS in 54m 20s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1c9ae9734de34afd90a331988e4abd2f : FAILURE in 3m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/952e672408e443d5a211cd732550074c : SUCCESS in 1h 03m 17s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":4},{"id":"a5637bab7ca352875611deb77cf91d150e9b9b8b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-14 12:04:59.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8ac185615a5653e51f2eaf1b3f994cf347f8e074","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-14 13:42:55.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"bb54761789ebaa9d3e1b6e05e718001e5c0ebf0d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-12-14 13:45:24.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"f39ca685584f49ff861f1c58f58fa9884c60ad19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 15:12:00.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/193193089c58457ab29605e9a7d2db4b : SUCCESS in 34m 34s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/334e58f698f742369824e92236b411ee : SUCCESS in 45m 20s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/17ad6d0f07eb4bb487451ae8f4109be5 : SUCCESS in 23m 37s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b4960f1f3f664185861788eedee1207d : SUCCESS in 31m 58s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/1bfa6a56470d488f9c6e27bca0273ae3 : SUCCESS in 1h 14m 30s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d79e22d3dea7417fa03ab32f954a76ee : SUCCESS in 56m 10s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/64ce18241b724b0b902af548bb82bc98 : SUCCESS in 1h 13m 53s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/e2841b11ccde464d893f04cab4a04d03 : SUCCESS in 50m 30s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/74ccc2d1181f4a118af48febc870c7db : SUCCESS in 1h 12m 56s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/f34077dd4cd44b95b1cee82c73ae538e : SUCCESS in 41m 17s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a9b11ef3743f4a7db12bc0f21e7ab036 : SUCCESS in 18m 14s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f2bd7352dd8945898a88fb7a549ac02f : SUCCESS in 18m 54s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1525fd6bfdf74dd893a74017affef511 : SUCCESS in 57m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/076bcf9f678446d1ba8721431ef3896d : SUCCESS in 6m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2582db743aa74fa896d656c72a34a726 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4a408a3efb9489781f1b719b1524d99 : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2bab5061173b4b799df730650a6ee10a : SUCCESS in 5m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/08f92080a4434ea489a0b5511c7017c9 : SUCCESS in 2m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8cfb0a976be485382998ccd3e1a20e4 : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a83daad440b44d1b9714b68ecf3737cb : SUCCESS in 3m 55s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/57c075f44c9e4125881f6897099bbea5 : SUCCESS in 24m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7328cc3a54964d9da83ed81fef7b66f9 : SUCCESS in 28m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/aab77a4733424dcb945113988e3df395 : FAILURE in 3m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d21e00a5f62e4afa90cccef292a1b98b : FAILURE in 3m 06s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/132ac3ff4e0c4aacb7c304d2e3dbd4da : FAILURE in 3m 17s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":5},{"id":"9b0e85e256c406b3b81f577fce9cd96ca21078ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-15 17:19:22.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"e00a20d2c8b727e0ecba91b9bb73a75804447c2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 19:19:47.000000000","message":"Patch Set 6: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"f86f67c1158c90630653934a3a36592c6668ec4a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-15 19:21:14.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ad3a7a57a15e99a6041691ba331102aed05e1565","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-15 19:21:33.000000000","message":"Patch Set 6:\n\nbtw, it was passing before I stopped gate due to touching k-a change","accounts_in_message":[],"_revision_number":6},{"id":"a504b47ef50eb1b7b90c4bbeaaca3c15b3adb301","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-15 19:22:27.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3be40dc18e10d38478ab663f5a8bfd2de0aefd0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-15 19:43:02.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"57f32816c07983dc64f49b76a319a56f6b6258ba","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-15 19:48:27.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"32d1411cd2ab348e915b42afa15693ebe48570f9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-15 19:49:02.000000000","message":"Patch Set 7:\n\nI keep the list to revert on the whiteboard. Only really care about kuryr-libnetwork as others see long abandoned.","accounts_in_message":[],"_revision_number":7},{"id":"20a33ef76da1a58a86a2170fc7f90cb54df88272","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 21:53:56.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/325a985b656a4711a536aae5059a5359 : SUCCESS in 38m 29s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/cfefcf6f4f1a42edb0fcdf11881b15a3 : SUCCESS in 51m 38s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/cd8a1d8162d3412987297cda15aa8c6f : SUCCESS in 21m 29s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/0a085f22dca64ca2a27b3626f10e082d : SUCCESS in 26m 28s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f4f5067cdf73456eb8c87797563924bb : SUCCESS in 1h 12m 27s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/b45e464b3e2e42c8a6fc8e139628e3bc : SUCCESS in 57m 21s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/94484874ec4e40fab02cc3ef5c82d004 : SUCCESS in 1h 09m 13s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/69a43b0f06c14957b4ed18ab1d1d8ba3 : TIMED_OUT in 2h 02m 19s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/4fff86a6c93045f19bcb9a4e4461fd04 : SUCCESS in 1h 28m 50s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/4a425ba5643c4dbbbacc216625103d0e : SUCCESS in 44m 55s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8a9b6847ccf34dc78c7f5cc172ed96f1 : SUCCESS in 19m 45s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c4cce8934af941e5867e02a26db459a4 : SUCCESS in 17m 16s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/695576b70a094b7b9b424e6a449e6ecf : SUCCESS in 50m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4309741a66c54716b08fe5148906267c : SUCCESS in 4m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f18215358ffd4c429adb6a09633e5ea6 : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de5eca2273c54bafb41c4232edf109cd : SUCCESS in 3m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/53b41db087774e49bf1847b036c0ac56 : SUCCESS in 4m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4949251699794b218a15a47c01a7dd1f : SUCCESS in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9323385a1d734a63b0bf3e8b473f7014 : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e24a6caa972f4235ab8a2ce0c36c5451 : SUCCESS in 3m 47s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fcf6b27a71704cb0b6eb7694516d4930 : SUCCESS in 24m 42s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2f586e14e3e241a9b16e8ce978e4fad4 : SUCCESS in 31m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8dfe3a5ed5044188b67d02b632286b3d : SUCCESS in 59m 12s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a93c9b97b0f84657aefb57aec59c7e8b : SUCCESS in 47m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2db96041e14b49d4b78ba11826469d07 : SUCCESS in 52m 32s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":7},{"id":"995e9e13b34831d6a21bc92360773101296dbe0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 21:54:15.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"83fcb937d0d6bea1f33c5a683dfad1f3b1097d2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 00:41:11.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/5c5959646e7b41c3a6343e7a727298b4 : SUCCESS in 55m 13s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/98cdaa4d22314800a056bd7dad54bc39 : SUCCESS in 31m 27s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ac1df43b36324fcc913c75e972915959 : SUCCESS in 1h 19m 41s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9e6869985131466598b4b4bbc7179d79 : SUCCESS in 58m 38s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c54194d85dcd43f7a53458223d100f71 : SUCCESS in 1h 32m 29s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/7641447cb1d34d86832be844f2e4ffed : SUCCESS in 1h 02m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a44df8af9f224154b4b6f21a21bb4bf9 : SUCCESS in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8f2cdcb1e25d40c0b9be4644aaa58ba6 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a22f736488d94c9a8fee99e3f44cb003 : SUCCESS in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1f809c5c8351490cb61cb7f052962859 : SUCCESS in 3m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2ca10a61e2e849878b19c23db7379ba9 : SUCCESS in 2m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed1398eecbb34ed282dd17c05ea27a84 : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad300bf7be964b88bbbe7fe157ce77df : SUCCESS in 5m 15s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/91108252013c4d7ead57bfd02ce2e36a : SUCCESS in 22m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b64dd989e4be47df9558d11682b7eda5 : SUCCESS in 46m 34s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4da982f760ea4131bc765c52f63f6527 : SUCCESS in 56m 59s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":7},{"id":"5f407a1bb1a9404e14cf00f39181d45142e1636e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 00:41:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"1b0b4089186e10f5cbe504c7b0a76e3ec807324a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 00:43:02.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a648f88d1714506bb5ebb6ce532ff54 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a5c2e2f956c4499a041a7408d6cf649 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"617421fb380f8b26531c4c4ac78b65a3f5a135ec","revisions":{"84fdb16abc378fac9a52a12442cf563c7c0af8c2":{"kind":"REWORK","_number":1,"created":"2020-12-10 07:45:17.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/10/766410/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/766410/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/766410/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/766410/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/766410/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/766410/1"}}},"commit":{"parents":[{"commit":"576ae967e4c503f56db75e943fd5bc5bf87f400a","subject":"[CI] Stop testing lower-constraints and fix py27 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/576ae967e4c503f56db75e943fd5bc5bf87f400a"}]}],"author":{"name":"abraden","email":"C-Albert.Braden@charter.com","date":"2020-12-07 19:20:44.000000000","tz":0},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2020-12-10 07:45:14.000000000","tz":60},"subject":"Support CentOS 8.3","message":"Support CentOS 8.3\n\nBackport this until Train.\n\nCentOS 8.3 renamed system repos. [1]\nThis caused failures when disabling and enabling.\nMore details in the referenced bug report.\n\nNotice CentOS 8.2 and below will no longer work.\n\n[1] https://wiki.centos.org/Manuals/ReleaseNotes/CentOS8.2011#Yum_repo_file_and_repoid_changes\n\nChange-Id: I2d67b3ed7b7c521bc755ec9a87365c9962a9eeff\nCloses-Bug: #1907213\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\nDepends-On: https://review.opendev.org/766211\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/84fdb16abc378fac9a52a12442cf563c7c0af8c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/84fdb16abc378fac9a52a12442cf563c7c0af8c2"}]},"branch":"refs/heads/stable/train"},"99f2fe6ccb26c2720c3f11908e94412ca3853d98":{"kind":"REWORK","_number":2,"created":"2020-12-10 07:47:15.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/10/766410/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/766410/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/766410/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/766410/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/766410/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/766410/2"}}},"commit":{"parents":[{"commit":"576ae967e4c503f56db75e943fd5bc5bf87f400a","subject":"[CI] Stop testing lower-constraints and fix py27 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/576ae967e4c503f56db75e943fd5bc5bf87f400a"}]}],"author":{"name":"abraden","email":"C-Albert.Braden@charter.com","date":"2020-12-07 19:20:44.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-12-10 07:47:08.000000000","tz":0},"subject":"Support CentOS 8.3","message":"Support CentOS 8.3\n\nBackport this until Train.\n\nCentOS 8.3 renamed system repos. [1]\nThis caused failures when disabling and enabling.\nMore details in the referenced bug report.\n\nNotice CentOS 8.2 and below will no longer work.\n\n[1] https://wiki.centos.org/Manuals/ReleaseNotes/CentOS8.2011#Yum_repo_file_and_repoid_changes\n\nChange-Id: I2d67b3ed7b7c521bc755ec9a87365c9962a9eeff\nCloses-Bug: #1907213\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\nDepends-On: https://review.opendev.org/766211\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/99f2fe6ccb26c2720c3f11908e94412ca3853d98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/99f2fe6ccb26c2720c3f11908e94412ca3853d98"}]},"branch":"refs/heads/stable/train"},"bde2914eb1135b268d453bbb83cd32ef8146cc74":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-12-13 10:18:02.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/10/766410/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/766410/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/766410/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/766410/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/766410/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/766410/3"}}},"commit":{"parents":[{"commit":"13e73e983b849cfd9de0e3df67180389f470b603","subject":"Publish weekly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13e73e983b849cfd9de0e3df67180389f470b603"}]}],"author":{"name":"abraden","email":"C-Albert.Braden@charter.com","date":"2020-12-07 19:20:44.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-12-13 10:18:02.000000000","tz":0},"subject":"Support CentOS 8.3","message":"Support CentOS 8.3\n\nBackport this until Train.\n\nCentOS 8.3 renamed system repos. [1]\nThis caused failures when disabling and enabling.\nMore details in the referenced bug report.\n\nNotice CentOS 8.2 and below will no longer work.\n\n[1] https://wiki.centos.org/Manuals/ReleaseNotes/CentOS8.2011#Yum_repo_file_and_repoid_changes\n\nChange-Id: I2d67b3ed7b7c521bc755ec9a87365c9962a9eeff\nCloses-Bug: #1907213\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\nDepends-On: https://review.opendev.org/766211\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bde2914eb1135b268d453bbb83cd32ef8146cc74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bde2914eb1135b268d453bbb83cd32ef8146cc74"}]},"branch":"refs/heads/stable/train","description":"Rebase"},"6b35e385682bcea17e0f4bffb0451d439a6fcb10":{"kind":"REWORK","_number":4,"created":"2020-12-14 09:54:39.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/10/766410/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/766410/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/766410/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/766410/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/766410/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/766410/4"}}},"commit":{"parents":[{"commit":"13e73e983b849cfd9de0e3df67180389f470b603","subject":"Publish weekly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13e73e983b849cfd9de0e3df67180389f470b603"}]}],"author":{"name":"abraden","email":"C-Albert.Braden@charter.com","date":"2020-12-07 19:20:44.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-14 09:54:32.000000000","tz":0},"subject":"Support CentOS 8.3","message":"Support CentOS 8.3\n\nBackport this until Train.\n\nCentOS 8.3 renamed system repos. [1]\nThis caused failures when disabling and enabling.\nMore details in the referenced bug report.\n\nNotice CentOS 8.2 and below will no longer work.\n\nTrain backport also includes some fixes for the\nfallout from the new pip resolver.\n\n* cap bandit\u003c1.6.3, since bandit no longer supports\n  Python 2.\n\n* mark almanach-base, dragonflow-base \u0026\n  kuryr-libnetwork unbuildable for Ubuntu source.\n\n* Don\u0027t install networking-hyperv in Ubuntu source\n  neutron-server image. pip complains of a conflict.\n\n[1] https://wiki.centos.org/Manuals/ReleaseNotes/CentOS8.2011#Yum_repo_file_and_repoid_changes\n\nChange-Id: I2d67b3ed7b7c521bc755ec9a87365c9962a9eeff\nCloses-Bug: #1907213\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\nDepends-On: https://review.opendev.org/766211\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b35e385682bcea17e0f4bffb0451d439a6fcb10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b35e385682bcea17e0f4bffb0451d439a6fcb10"}]},"branch":"refs/heads/stable/train"},"8c96494e22dbe72a871ee629f53e9a119eee2c30":{"kind":"REWORK","_number":5,"created":"2020-12-14 13:42:55.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/10/766410/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/766410/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/766410/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/766410/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/766410/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/766410/5"}}},"commit":{"parents":[{"commit":"13e73e983b849cfd9de0e3df67180389f470b603","subject":"Publish weekly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13e73e983b849cfd9de0e3df67180389f470b603"}]}],"author":{"name":"abraden","email":"C-Albert.Braden@charter.com","date":"2020-12-07 19:20:44.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-14 13:42:46.000000000","tz":0},"subject":"Support CentOS 8.3","message":"Support CentOS 8.3\n\nBackport this until Train.\n\nCentOS 8.3 renamed system repos. [1]\nThis caused failures when disabling and enabling.\nMore details in the referenced bug report.\n\nNotice CentOS 8.2 and below will no longer work.\n\nTrain backport also includes some fixes for the\nfallout from the new pip resolver.\n\n* cap bandit\u003c1.6.3, since bandit no longer supports\n  Python 2.\n\n* mark kuryr-libnetwork unbuildable for source\n  images.\n\n* mark almanach-base \u0026 dragonflow-base unbuildable\n  for Debian \u0026 Ubuntu source images.\n\n* Don\u0027t install networking-hyperv in source\n  neutron-server images. pip complains of a conflict.\n  (https://bugs.launchpad.net/kolla/+bug/1907826)\n\n[1] https://wiki.centos.org/Manuals/ReleaseNotes/CentOS8.2011#Yum_repo_file_and_repoid_changes\n\nChange-Id: I2d67b3ed7b7c521bc755ec9a87365c9962a9eeff\nCloses-Bug: #1907213\nCloses-Bug: #1907826\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\nDepends-On: https://review.opendev.org/766211\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c96494e22dbe72a871ee629f53e9a119eee2c30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c96494e22dbe72a871ee629f53e9a119eee2c30"}]},"branch":"refs/heads/stable/train"},"4382af9c07c400890331ed3ca7edaf1158bfb09b":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2020-12-15 17:19:22.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/10/766410/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/766410/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/766410/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/766410/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/766410/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/766410/6"}}},"commit":{"parents":[{"commit":"13e73e983b849cfd9de0e3df67180389f470b603","subject":"Publish weekly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13e73e983b849cfd9de0e3df67180389f470b603"}]}],"author":{"name":"abraden","email":"C-Albert.Braden@charter.com","date":"2020-12-07 19:20:44.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-15 17:19:15.000000000","tz":0},"subject":"Support CentOS 8.3","message":"Support CentOS 8.3\n\nBackport this until Train.\n\nCentOS 8.3 renamed system repos. [1]\nThis caused failures when disabling and enabling.\nMore details in the referenced bug report.\n\nNotice CentOS 8.2 and below will no longer work.\n\nTrain backport also includes some fixes for the\nfallout from the new pip resolver.\n\n* cap bandit\u003c1.6.3, since bandit no longer supports\n  Python 2.\n\n* mark kuryr-libnetwork unbuildable for source\n  images.\n\n* mark almanach-base \u0026 dragonflow-base unbuildable\n  for Debian \u0026 Ubuntu source images.\n\n* Don\u0027t install networking-hyperv in source\n  neutron-server images. pip complains of a conflict.\n  (https://bugs.launchpad.net/kolla/+bug/1907826)\n\n[1] https://wiki.centos.org/Manuals/ReleaseNotes/CentOS8.2011#Yum_repo_file_and_repoid_changes\n\nChange-Id: I2d67b3ed7b7c521bc755ec9a87365c9962a9eeff\nCloses-Bug: #1907213\nCloses-Bug: #1907826\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\nDepends-On: https://review.opendev.org/766211\nDepends-On: https://review.opendev.org/767157\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4382af9c07c400890331ed3ca7edaf1158bfb09b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4382af9c07c400890331ed3ca7edaf1158bfb09b"}]},"branch":"refs/heads/stable/train","description":"Edit commit message"},"617421fb380f8b26531c4c4ac78b65a3f5a135ec":{"kind":"REWORK","_number":7,"created":"2020-12-15 19:43:02.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/10/766410/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/766410/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/766410/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/766410/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/766410/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/766410/7"}}},"commit":{"parents":[{"commit":"13e73e983b849cfd9de0e3df67180389f470b603","subject":"Publish weekly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13e73e983b849cfd9de0e3df67180389f470b603"}]}],"author":{"name":"abraden","email":"C-Albert.Braden@charter.com","date":"2020-12-07 19:20:44.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-15 19:42:57.000000000","tz":0},"subject":"Support CentOS 8.3","message":"Support CentOS 8.3\n\nBackport this until Train.\n\nCentOS 8.3 renamed system repos. [1]\nThis caused failures when disabling and enabling.\nMore details in the referenced bug report.\n\nNotice CentOS 8.2 and below will no longer work.\n\nTrain backport also includes some fixes for the\nfallout from the new pip resolver.\n\n* cap bandit\u003c1.6.3, since bandit no longer supports\n  Python 2.\n\n* mark kuryr-libnetwork unbuildable for source\n  images.\n\n* mark almanach-base \u0026 dragonflow-base unbuildable\n  for Debian \u0026 Ubuntu source images.\n\n* Don\u0027t install networking-hyperv in source\n  neutron-server images. pip complains of a conflict.\n  (https://bugs.launchpad.net/kolla/+bug/1907826)\n\n[1] https://wiki.centos.org/Manuals/ReleaseNotes/CentOS8.2011#Yum_repo_file_and_repoid_changes\n\nChange-Id: I2d67b3ed7b7c521bc755ec9a87365c9962a9eeff\nCloses-Bug: #1907213\nCloses-Bug: #1907826\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\nDepends-On: https://review.opendev.org/766211\nDepends-On: https://review.opendev.org/767157\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/617421fb380f8b26531c4c4ac78b65a3f5a135ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/617421fb380f8b26531c4c4ac78b65a3f5a135ec"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
