)]}'
{"id":"openstack%2Fkolla~766500","triplet_id":"openstack%2Fkolla~master~Iabe5b9684a7bf0e39a7dca824e6705b5069560bf","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Iabe5b9684a7bf0e39a7dca824e6705b5069560bf","subject":"elasticsearch-curator: drop botocore workaround","status":"MERGED","created":"2020-12-10 15:48:35.000000000","updated":"2020-12-14 12:33:37.000000000","submitted":"2020-12-14 12:32:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"766500","meta_rev_id":"bbf1186d8f970a7907095451cf48c92169350b8f","_number":766500,"virtual_id_number":766500,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-14 12:32:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-12-14 10:14:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-12-11 20:07:45.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2020-12-14 10:14:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-12-13 15:38:34.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-12-13 15:38:34.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-10 16:31:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-11 20:07:45.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-12-14 10:14:32.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"e17793920e302b008cac38faad34ae88e836b122","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-12-10 15:48:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e9b043617403c16a092817d383f5d16c6015721","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 16:31:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/422c1d4a56fb47c4b81a75ba67addeda : FAILURE in 9m 17s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b72031669c674d46a0456d7e2a6ea5a0 : FAILURE in 9m 30s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7396a2fa0bd7404ebfeca6ee85129ebd : FAILURE in 13m 27s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/e3c226e2826a4a60ab5adc2c91b2d25f : FAILURE in 12m 14s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/6ac8dbcf27bf4724a52eaaffa526ed09 : FAILURE in 11m 27s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d5472461f097424db80d7d2ff87e4d8a : SUCCESS in 33m 12s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1b1a6148512e48bbbe8097544e0c2310 : SUCCESS in 19m 44s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8b83340fb94f438f95c4394c5bf34b48 : FAILURE in 20m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/889f26b4af5b4887a35da56f82f43a7e : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e169a9c137c4a75b5052245fc674a1e : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c19dc5f17d0340cc85b325583f2fb0f4 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61e38aaa7ddf41a39d35fd8622c4ae8b : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/203bb3af200e48b0be077900cbe16535 : SUCCESS in 6m 34s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/94253912566b4768914745823df64f97 : SUCCESS in 21m 44s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/90beed03b1554a1db79de768e9b16d43 : SUCCESS in 35m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/19d6a06ee5804391a25547051e6bc157 : FAILURE in 16m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c0eef68b419641fe932a4c6b5e4f6e43 : FAILURE in 18m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/498218a5f6964231861d8273dc4a7d15 : FAILURE in 18m 13s","accounts_in_message":[],"_revision_number":1},{"id":"1e03586743f596e41e3cda74d6261b07841392d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 16:42:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/2c5bdeeccb4849fc940b77b18ad473cc : FAILURE in 10m 27s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e04bd0e2bc9745bea1fe733c11ecff1c : SUCCESS in 47m 07s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b46d83cd36734c588f140160c9b4d16b : FAILURE in 36m 50s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/ad046315b4374c87aca19ccd35078470 : SUCCESS in 46m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3ae100e506bb40aa8bc9e9b269c283173583ad94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-12-11 13:00:14.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"3ea33e94e689ba1d864efaec58b2922697bcbded","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 15:44:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/f0018f9b56054b1d8921dd30fe5cf9d9 : SUCCESS in 21m 08s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d02a0891069149ce99ad0c03d6b6e1ab : SUCCESS in 39m 18s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/2d160935af6e4700b0b3987888c023ee : SUCCESS in 1h 04m 00s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/a1a511731a004df79cf09d56f9f2b01b : SUCCESS in 56m 54s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/1521078156cc4851899aa5648cbfe989 : SUCCESS in 2h 05m 24s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/522853191e334d888d16d6f2146d9238 : SUCCESS in 58m 44s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4022c9de7075442694ec222f207055e8 : SUCCESS in 20m 54s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/de0bae3c3dfe49aea12c95663b4126a3 : SUCCESS in 54m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc4edb1842b044f18391c1e0a44462a8 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f4e9bc4b40b415e89d1f3001202530a : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c43b92b042804193b3cf102c903a66d3 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4789ac9141f046daa202572c1d2eaf89 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c913b05a1f84cbb8b361aa1a14cf13b : SUCCESS in 6m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cf85a9330fd841c5a90fca61777a7ca6 : SUCCESS in 20m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d3d4b676035743f2b7c93ca6afdc1c13 : SUCCESS in 35m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/610342072517437cb913bc564c3b4532 : SUCCESS in 1h 54m 53s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7079cc5243fa46d298234ec672e928a9 : SUCCESS in 52m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9b7197c0b64b47f7951ca511bf2468a1 : SUCCESS in 1h 04m 02s","accounts_in_message":[],"_revision_number":2},{"id":"78fc65b35547ce9eb408ac6db63ee7f64d6a4ea5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 16:12:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/8126a14d072141359d58ba1ae1992262 : FAILURE in 1h 50m 06s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/78bb6352e38c442ca8998060b5d714c6 : FAILURE in 20m 57s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c85dd73a361d4bf3a754af71fa96f73d : FAILURE in 14m 39s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/940f2467d5454a05893e245736c7fd87 : FAILURE in 1h 42m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7f7b9e0022d43e53771f61228bd8d092c664b91e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-11 20:07:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"be7e8803a56259a3c6849d767b6945c1772bd9e7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-13 15:38:34.000000000","message":"Patch Set 2: Backport-Candidate+2 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"52d332c9845bb3c82eb0a6144877f1fbf4622ad6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-12-14 10:14:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a2f358c33807cf9f72f4fae875623e0d73413ea5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 10:14:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"dc7d184afb10c0e542171c2ac5806d62882444bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 12:32:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/5959af1716c34fdc9f0f6028aaf17410 : SUCCESS in 34m 10s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/ff6fae7495214234a3e63530bf525c0b : SUCCESS in 1h 05m 30s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/a60c283411bf4d1195a30e3fd9c077d9 : SUCCESS in 2h 15m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecfb74022ae243909dfa9458ec594354 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9de7fc038fc410c942ef8ea44a36c5a : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed47edf0e9b04c63a03d7bb76293ce3b : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/713f427dce85406ca20747554deff2f1 : SUCCESS in 5m 48s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/aaadb0edfab94e6896c20f28dc7c662d : SUCCESS in 34m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/362e1ce95927487da0697be0973d1f5a : SUCCESS in 50m 52s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2027c8414ab140968e01f0eb9761cda6 : SUCCESS in 1h 08m 44s","accounts_in_message":[],"_revision_number":2},{"id":"4b0ffccb65975aa111e5f334f4db7a5e1bc5a17d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 12:32:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"bbf1186d8f970a7907095451cf48c92169350b8f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 12:33:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d95366998ab84d168b1e67a769d2fe81 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7db2f100195c7aecde449e47d758c784850bd456","revisions":{"9b4f37f3126166c0f45506857dd2f104127a677f":{"kind":"REWORK","_number":1,"created":"2020-12-10 15:48:35.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/00/766500/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/766500/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/766500/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/766500/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/766500/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/766500/1"}}},"commit":{"parents":[{"commit":"c682a0fe84d28f9b5d01c6c6c90ae0338ff331f0","subject":"Merge \"docs: add code reviews to contributor guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c682a0fe84d28f9b5d01c6c6c90ae0338ff331f0"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-12-10 15:46:29.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-12-10 15:47:32.000000000","tz":60},"subject":"elasticsearch-curator: drop botocore workaround","message":"elasticsearch-curator: drop botocore workaround\n\nCurator got requirements sorted out upstream and builds fine now without\nhack.\n\nThis reverts curator part of https://review.opendev.org/c/openstack/kolla/+/755339/\n\nChange-Id: Iabe5b9684a7bf0e39a7dca824e6705b5069560bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b4f37f3126166c0f45506857dd2f104127a677f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b4f37f3126166c0f45506857dd2f104127a677f"}]},"branch":"refs/heads/master"},"7db2f100195c7aecde449e47d758c784850bd456":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-12-11 13:00:14.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/00/766500/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/766500/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/766500/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/766500/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/766500/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/766500/2"}}},"commit":{"parents":[{"commit":"865f55d4ec408f1e0a5fd8ef4f77edc566c8eb75","subject":"Merge \"Support CentOS 8.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/865f55d4ec408f1e0a5fd8ef4f77edc566c8eb75"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-12-10 15:46:29.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-12-11 13:00:14.000000000","tz":0},"subject":"elasticsearch-curator: drop botocore workaround","message":"elasticsearch-curator: drop botocore workaround\n\nCurator got requirements sorted out upstream and builds fine now without\nhack.\n\nThis reverts curator part of https://review.opendev.org/c/openstack/kolla/+/755339/\n\nChange-Id: Iabe5b9684a7bf0e39a7dca824e6705b5069560bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7db2f100195c7aecde449e47d758c784850bd456"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7db2f100195c7aecde449e47d758c784850bd456"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
