)]}'
{"id":"openstack%2Fkolla~767054","triplet_id":"openstack%2Fkolla~stable%2Ftrain~Iabe5b9684a7bf0e39a7dca824e6705b5069560bf","project":"openstack/kolla","branch":"stable/train","hashtags":[],"change_id":"Iabe5b9684a7bf0e39a7dca824e6705b5069560bf","subject":"elasticsearch-curator: drop botocore workaround","status":"MERGED","created":"2020-12-15 11:56:40.000000000","updated":"2020-12-17 20:24:40.000000000","submitted":"2020-12-17 20:23:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"767054","cherry_pick_of_change":767053,"cherry_pick_of_patch_set":1,"meta_rev_id":"002518ad19521e3cb5fc5ededa23216274736f3b","_number":767054,"virtual_id_number":767054,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-17 20:23:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2020-12-15 12:39:50.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-12-17 13:36:17.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-12-17 13:36:17.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-15 11:56:40.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-12-15 13:01:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-17 10:55:47.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2020-12-17 13:36:17.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"387dbd1b21181563b8747ac6f6df052d3fd6faf6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-15 11:56:40.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.\n\nThe following files contain Git conflicts:\n* docker/elasticsearch/elasticsearch-curator/Dockerfile.j2","accounts_in_message":[],"_revision_number":1},{"id":"ed38904482bae37f67336f2beda46dee96ee0e67","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-12-15 11:59:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ca01eca4879d7383a8e2b7061fcefe5c9aea1393","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-15 12:03:20.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ba00213379f58773e4c7d9a14740d061f1409c0b","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-15 12:03:48.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"34ba7d4ad04c5e7a9e9f2dad26b0186c80cd2cd4","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-12-15 12:39:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"db34b38068a4766ea93b401f3561b193407278a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 13:01:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/113f4bd291e441a59e3cb8ad67552614 : FAILURE in 4m 39s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/5d6405e20d434e149824d1b37bd525c9 : FAILURE in 4m 11s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/67fbca3241c94aa4a1b1f0199036a5d3 : FAILURE in 4m 55s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/9c5633429d824613bfac510a7d65dc03 : FAILURE in 4m 15s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d354cbea70554e44bac68393bca02879 : FAILURE in 8m 52s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/771de1655c0e4440a50e24ecf928649c : FAILURE in 14m 36s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/92697d6b80894677a5ed52b9d0a238de : FAILURE in 12m 09s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/14736f9f03b544bf9e6ea7a92851ab45 : FAILURE in 14m 48s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/985dbc12881f49ac9f5234c9342f039c : FAILURE in 11m 58s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/fdbe7262496b416b9b380ebe9aac286e : SUCCESS in 44m 06s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c369bbbb50f343558a4931458972b6fe : FAILURE in 3m 50s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2ca8f4e79cb74de2acea7978b4d6cc8f : FAILURE in 3m 10s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0916488aa0db411dbbc9d2d2c76fbc78 : FAILURE in 6m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b17784013c18475bbea5dd05681635a2 : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/305b11b4fbd84a8ab543ad719039cd50 : FAILURE in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94d23800d6024154bdadbe5a2f0455b5 : SUCCESS in 5m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6111df089bbc47aebb5cb4835188ad5d : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d770727890bd476fb1d142b7f6848619 : SUCCESS in 6m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f087ba82028841aaa6cb4966a6c476eb : SUCCESS in 20m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f48788f7aa504ce1b075d5ab2d7cb6fc : FAILURE in 25m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b7461d6919444bf590fa2b50d8a75cd3 : FAILURE in 8m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/21e49c8342fe403d9ce2a6f484d919d4 : FAILURE in 11m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/240dbb899e1e4cf696084c421e6f7c01 : FAILURE in 11m 19s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"097bf59b338d405080ad7b0b854958f6bd17e262","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-17 10:55:47.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2dff7503b033544efc673ae263e56ba32f64ffcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 12:59:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/54813958ddcd4a48896d7be83649ac8f : SUCCESS in 36m 06s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/746c2d2a03154612b9bd85c8f25a68c8 : SUCCESS in 41m 16s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/eeaeed3b043c4f6f80d3903744d8610c : SUCCESS in 26m 51s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/0974b3f6a98c410ea6238027db8cbc27 : SUCCESS in 25m 36s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/07d9e0c0e2824192b901c1cf456e3218 : SUCCESS in 1h 24m 35s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/38931f57c20f46efa7d555af8d67c667 : SUCCESS in 1h 12m 46s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ca523a67d9594ccabfee56d6d7d71479 : SUCCESS in 1h 07m 25s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b28d1671e72446cab6cf1c755874f47f : SUCCESS in 53m 52s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/89bcc3f6b80c4e1c859387d5f577703b : SUCCESS in 1h 53m 07s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/2aa337ab3c3848b6a5bb8bf7c7250f2b : SUCCESS in 1h 01m 34s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b1762230fea84a6d90a0ced0cc0f6c60 : SUCCESS in 27m 23s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e626c14d9163417aa6e0dba48bb897ee : SUCCESS in 21m 06s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4183ba283e794e07ad84183845347d1b : SUCCESS in 56m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4beded887f9049748e54817d1da56ce1 : SUCCESS in 4m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3339521407ad49049acd9beae1459352 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6bff6c6c9f0947ce81b84e346125b5d8 : SUCCESS in 3m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/12bc33018bf94f91b8d5a0068f255484 : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3a121f5c2204fd7bbce9cce10518e2b : SUCCESS in 5m 21s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2dfd577cea4549208ee27506ec6b076c : SUCCESS in 23m 54s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c7d519d0ec7d46fc9ebe4ca03fed34d2 : SUCCESS in 38m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7c227aa71d5b4d5ca4f2c7ae61d88243 : SUCCESS in 49m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/facff40044054568b59463a644edae30 : SUCCESS in 51m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/35f9cacb5c0b488a8fcc34fb864b1947 : SUCCESS in 1h 00m 25s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"98ab4a0b8376772f47fc8ec3e997436580040145","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-17 13:36:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"18bc3acbf552b456a57255eadf66fcbb03e6e574","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 13:36:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3487c172b1804adac48835088addcbc3472f67e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 15:00:28.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f074cd73ec93412c8f74ea82da6d8759 : SUCCESS in 54m 55s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f780a8fcb34c41b69a32b9803e388f19 : SUCCESS in 26m 36s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/4668714b42e64de3933f188e845e8adc : SUCCESS in 1h 17m 29s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/e23920cc5bc540dfbf3a22be62df7bc7 : SUCCESS in 52m 58s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/5342e9b0cefc41b0867bd278fafc3b87 : SUCCESS in 1h 20m 55s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/66ec9011e58c4e5b8cfc1fd7aff8018c : SUCCESS in 45m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2e4f54d00c64f60a39c3654d0f1b532 : SUCCESS in 4m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2a9148b8ffd545cea08dc5bc693e82b0 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ad69ccd1c3744de9d06c2651a4ce4af : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4eae0c13b78e437cbc4e25f6878daec3 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5338cfb00994d00b53a6d814eee16fb : SUCCESS in 5m 45s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8bca4f86d6c84808b824ada20f61feab : FAILURE in 5m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/acc62b9197a44b9aac47ab34433f1927 : FAILURE in 9m 31s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8de4767568fc4524932ca579636dde24 : FAILURE in 9m 39s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"d3eabfe9f148d2bec4496642b176bd00ccd66ada","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-17 17:07:53.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4ebb4191db97480ba30a7793de57006327aa1b71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 18:42:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/79a1f55a7b5541fb90598a4781dfbe69 : SUCCESS in 37m 34s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/699204e72c344453a0d5cdee0c13c46b : SUCCESS in 43m 52s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/2279cebe64884e6bb6f5de922069078f : SUCCESS in 20m 35s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/9d9ab54f74d74398bdf256b8f7158920 : SUCCESS in 26m 23s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/1ab66cc736c948d7bbea9f615b37ff49 : SUCCESS in 1h 10m 39s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f6c4dc0182b548b7873da7c25b4e256a : SUCCESS in 1h 02m 09s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/37db4da51118435097fe2df9afb9e9f9 : SUCCESS in 1h 04m 26s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/a5797db06e62406c87c34eba57d2ba54 : SUCCESS in 58m 12s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b1a6e167bdf64924926e6479714baf89 : SUCCESS in 1h 29m 55s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/8fecfd5f30804c12b50c0f03fec1d2f7 : SUCCESS in 42m 16s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9a7f6f4c3aae4e71833a21a79284b1be : SUCCESS in 18m 01s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/807c285a04fe4b75b1cdaea5598d6398 : SUCCESS in 16m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ab35b4112b624d778b1671a1216893b5 : SUCCESS in 48m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a62d62178224a8fbdd7ec8a88abd7c0 : SUCCESS in 4m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/23c6ba86211e4148af72acb32d630773 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e654c696b4ee4d99bce3fc01b95affb2 : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/95c9fae7d2b544c79d2c3478c0928a8d : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9b8692e4df4418894fd1ee79d4adc22 : SUCCESS in 5m 13s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6c5713c7507046ad8941a7c08e21bb82 : SUCCESS in 25m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3b7df19018ba4b26b6146fe8e065f48c : SUCCESS in 22m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c2424915851c4f859f6862b771b294a8 : SUCCESS in 48m 31s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d77b139dcc774b4f80ddd82cfff3ee0d : SUCCESS in 51m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2edd72d297084c60aef00b8b68a51b8e : SUCCESS in 59m 22s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"c1e15529e003a11546d4b6adaa864fe2ee2c38fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 18:42:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9f5b199d80db185e280f6f784267f0d11d7467df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 20:23:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/fef7ce9798804c619ec74a585b1cca2b : SUCCESS in 58m 46s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/cbb9b9fe871c464c80654c38ca1c24f8 : SUCCESS in 1h 10m 40s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/558faa150af5483bb175b9d93ee9f92d : SUCCESS in 1h 28m 54s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a708854660374de4ba5de55d89765f5d : SUCCESS in 57m 11s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/48231753c73a436699b8cb92cb43eac7 : SUCCESS in 1h 17m 58s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/59b298eaf05645d1b628934d38a2ab14 : SUCCESS in 38m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06dc61f86294454c8228ae9d6280d7d7 : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0065f2cc48e24c3084c783585de6c620 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f405eba039d41f7a7969ef4d59f7b6a : SUCCESS in 3m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0b8a2b7d8b254ddba8f2561fbf6c4788 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58ff4551e75241fe866bdc90526bda06 : SUCCESS in 5m 21s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/db55af2c46aa4aba8d41ab7cf71900f7 : SUCCESS in 33m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/82515181d0d9498889dd9ce67a1bddca : SUCCESS in 1h 07m 49s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0c1463b751824d1c8f6fe7065affa6a3 : SUCCESS in 1h 35m 27s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"36a6983e32b37487d551e1c588ff9c12592dcfab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 20:23:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"002518ad19521e3cb5fc5ededa23216274736f3b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 20:24:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2093363bbb5416bb045cadd030584ab : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a5a422f6355a0cbd5c48d16fb8c29d881c841c5b","revisions":{"d2649f036021a395f0bf8cf52b28ee2c5bf074c7":{"kind":"REWORK","_number":1,"created":"2020-12-15 11:56:40.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/54/767054/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/767054/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/767054/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/767054/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/767054/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/767054/1"}}},"commit":{"parents":[{"commit":"13e73e983b849cfd9de0e3df67180389f470b603","subject":"Publish weekly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13e73e983b849cfd9de0e3df67180389f470b603"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-12-10 15:46:29.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-12-15 11:56:40.000000000","tz":0},"subject":"elasticsearch-curator: drop botocore workaround","message":"elasticsearch-curator: drop botocore workaround\n\nCurator got requirements sorted out upstream and builds fine now without\nhack.\n\nThis reverts curator part of https://review.opendev.org/c/openstack/kolla/+/755339/\n\nChange-Id: Iabe5b9684a7bf0e39a7dca824e6705b5069560bf\n(cherry picked from commit 7db2f100195c7aecde449e47d758c784850bd456)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d2649f036021a395f0bf8cf52b28ee2c5bf074c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d2649f036021a395f0bf8cf52b28ee2c5bf074c7"}]},"branch":"refs/heads/stable/train"},"a5a422f6355a0cbd5c48d16fb8c29d881c841c5b":{"kind":"REWORK","_number":2,"created":"2020-12-15 12:03:20.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/54/767054/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/767054/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/767054/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/767054/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/767054/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/767054/2"}}},"commit":{"parents":[{"commit":"13e73e983b849cfd9de0e3df67180389f470b603","subject":"Publish weekly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13e73e983b849cfd9de0e3df67180389f470b603"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-12-10 15:46:29.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-12-15 12:03:10.000000000","tz":0},"subject":"elasticsearch-curator: drop botocore workaround","message":"elasticsearch-curator: drop botocore workaround\n\nCurator got requirements sorted out upstream and builds fine now without\nhack.\n\nThis reverts curator part of https://review.opendev.org/c/openstack/kolla/+/755339/\n\nChange-Id: Iabe5b9684a7bf0e39a7dca824e6705b5069560bf\n(cherry picked from commit 7db2f100195c7aecde449e47d758c784850bd456)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a5a422f6355a0cbd5c48d16fb8c29d881c841c5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a5a422f6355a0cbd5c48d16fb8c29d881c841c5b"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
