)]}'
{"id":"openstack%2Fkolla~767111","triplet_id":"openstack%2Fkolla~master~I411f91af927ffa4d04c8021ddb5c3d584d90bdfd","project":"openstack/kolla","branch":"master","topic":"fix-publishing-login","hashtags":[],"change_id":"I411f91af927ffa4d04c8021ddb5c3d584d90bdfd","subject":"Fix publishing login","status":"MERGED","created":"2020-12-15 10:24:27.000000000","updated":"2020-12-15 13:56:24.000000000","submitted":"2020-12-15 13:55:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"767111-fix-publishing-login","meta_rev_id":"acc167c758c34feb1b1a3e8dd46694b42189372a","_number":767111,"virtual_id_number":767111,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-15 13:55:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-12-15 11:32:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-12-15 11:44:09.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-12-15 11:44:09.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-12-15 10:28:28.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-15 11:32:15.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-12-15 11:44:09.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-12-15 12:10:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-15 12:24:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1e05d596d7ec8d53900f1415fca44640f2d37996","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-15 10:24:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"19077ad47b00da625e94bdfa12ce9e858d6e7584","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-15 10:28:28.000000000","message":"Patch Set 1: Review-Priority+2\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"25aea276ea508ced7c6a72062bc0dc969929fd67","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-15 10:29:13.000000000","message":"Patch Set 1:\n\ntested in https://review.opendev.org/c/openstack/kolla/+/767083","accounts_in_message":[],"_revision_number":1},{"id":"cbe9f3b5cd56141bfe44ef5bd84242d6b6c3c147","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-12-15 11:32:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f387fc9c3105a93b20c16f76ca42e1a55fa8d38e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-12-15 11:44:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"86f32ca91ea23a7fd04db70741a3d202042b320f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 12:10:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/668ea10221dd432cbad72dddb9230b88 : SUCCESS in 1h 07m 09s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/49ad289d3f5241079c67294e622e7756 : SUCCESS in 57m 46s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/65645b626ddb496d80ed4a78c9c03260 : SUCCESS in 1h 39m 18s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9cc4e75d553e4218aa845415a55f5fec : SUCCESS in 1h 01m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3ca9b9b99f6f453c392eefc8a54ce5343999fcfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 12:24:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/880fe605e1cc4fd781b0ebc003bf3528 : SUCCESS in 1h 08m 12s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/838760c47d304eb3be65c38a9c94d343 : SUCCESS in 31m 09s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/567f61a133764917857a7c329b976fb4 : SUCCESS in 1h 00m 50s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/163718ad82e2454d90994dcda8dcc7b7 : SUCCESS in 58m 16s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/4ea044631e0a4e5590e96efcd2c9badd : SUCCESS in 1h 53m 13s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e7c38e8bd77e4f148a81a71b94dbb711 : SUCCESS in 1h 21m 30s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/762b25441f7647ff98864d30c238feaa : SUCCESS in 1h 08m 48s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c2c101ad4b5c482cb702e3cbf63e3872 : SUCCESS in 1h 02m 15s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37a7a50cabc74ab7a7727e691634adb7 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68c5ceac0cac4b9fa39eb6d53976a076 : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a5504b08e414ddaa24a39162b492256 : SUCCESS in 6m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b223b75402b49fe94a5a219f85e830b : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8510167dd6954d209099b3c3a8a7f9f1 : SUCCESS in 19m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/58455254790a47e8aaa47d7964911ba9 : SUCCESS in 1h 08m 53s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2ef72cdaab5e464b9290a07fe75fc738 : SUCCESS in 1h 21m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8b48b10e7bfb4dac98b549f7159a048a : SUCCESS in 1h 14m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a7e24851649448129c24b1041c5e62a9 : SUCCESS in 51m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/996b4dea15f44541b80908e0a500d680 : SUCCESS in 1h 05m 12s","accounts_in_message":[],"_revision_number":1},{"id":"30273f23d89823ef0f3669510e6554d8c9b3cf9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 12:24:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b4f53d5d453d958d3e8f83cd519c3aa0fc1fd785","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 13:55:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/45e40c09c4454e93bd49b7b2fe4a1274 : SUCCESS in 36m 02s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/05a4faa0d5d74cb895c5ce82202d1ece : SUCCESS in 1h 24m 38s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/8ec2a5025b3b4202b2006db8f06627b8 : SUCCESS in 1h 15m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e496bf144e445dea15c19e014e90836 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bd829fdcb674f67823d6a071a52f577 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4344048da7c43a19fcfaa62e0590941 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7deb6883b6fc44bd92d00a1681d264d7 : SUCCESS in 12m 37s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7e83a8bf037646059103e4f106a38a1b : SUCCESS in 1h 00m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/08d030c4f0f54968b0b105c26ea790d0 : SUCCESS in 1h 11m 07s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b29deb02dfa5425e99f9166c3a0ded5c : SUCCESS in 1h 12m 18s","accounts_in_message":[],"_revision_number":1},{"id":"17dbc768dbfdf54f7234901e3ac0814306acbd51","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 13:55:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"acc167c758c34feb1b1a3e8dd46694b42189372a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 13:56:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f635e9f3a7d42ed94475ccb4c12a03d : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cdd0eb7914fb01dac969386feffd8a0eb64347c8","revisions":{"cdd0eb7914fb01dac969386feffd8a0eb64347c8":{"kind":"REWORK","_number":1,"created":"2020-12-15 10:24:27.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/11/767111/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/767111/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/767111/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/767111/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/767111/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/767111/1"}}},"commit":{"parents":[{"commit":"9465c722f8a84ac07f9d2a99f935df4a1ce90123","subject":"Merge \"freezer: drop EPEL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9465c722f8a84ac07f9d2a99f935df4a1ce90123"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-12-15 10:22:36.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-12-15 10:22:36.000000000","tz":60},"subject":"Fix publishing login","message":"Fix publishing login\n\nLogin to DockerHub was failing because of trailing newlines in\nthe encrypted credentials.\nThis patch makes the playbook trim them before usage.\n\nChange-Id: I411f91af927ffa4d04c8021ddb5c3d584d90bdfd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cdd0eb7914fb01dac969386feffd8a0eb64347c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cdd0eb7914fb01dac969386feffd8a0eb64347c8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
