)]}'
{"id":"openstack%2Fkolla~767124","triplet_id":"openstack%2Fkolla~stable%2Fstein~If708c819af94a70d4413e7405398753627f7a097","project":"openstack/kolla","branch":"stable/stein","topic":"ci-healthcheck","hashtags":[],"change_id":"If708c819af94a70d4413e7405398753627f7a097","subject":"[DNM] [CI] Healthcheck run","status":"ABANDONED","created":"2020-12-15 12:25:28.000000000","updated":"2021-01-23 13:17:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ce8db029a0fc9b2afe1ab4abcfd5a9bdd2058248","_number":767124,"virtual_id_number":767124,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-12-15 13:16:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-15 13:16:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1353059422982a988ebb9b7bf604dace4b6a1da8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-15 12:25:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87047d22b6b5921c52bb116695bb5eea66874907","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-15 12:27:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"35b4ceaa99e672e2bd5a9a5c602c2daaed6197e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 13:16:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/fc00434225384cbebf3ae0b5bbd4b625 : FAILURE in 3m 55s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/0e071e42c05e40bda1f7e751870ae1d4 : FAILURE in 4m 02s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ab303fdc0fcf45efad52b3c4aa216fb4 : FAILURE in 7m 56s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/8c153c80c3bd4861a6cfb6e4af03fe58 : FAILURE in 7m 51s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/8f8d2c9e5dde43e8841088cf469a1f3f : FAILURE in 7m 49s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/6ea4bbda0c184deeab36847c6957d17e : SUCCESS in 46m 44s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/1c4976c2d6c6456197420efb0d002fca : FAILURE in 5m 03s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/215f45d0b318492298a8581b4273917c : FAILURE in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61e0e7fffe964ff896e6e45084c59358 : SUCCESS in 3m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f38a9aa9bf644526b947fb4a13b6aec5 : FAILURE in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d32414e9b19b41589bcdff0c8349eb9e : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7db7e8f5895a4a2594f0a62eb2802067 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/803ff45f582f4fa2ab237fdf3a9b1a4e : SUCCESS in 3m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/73ce9dd2c31c4b91881efc1e7ad617f4 : SUCCESS in 28m 42s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b713209d02ea45709d1b0ce86a60b8a0 : SUCCESS in 24m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ca8ad04788cc405ea690b3e83c0a443a : FAILURE in 8m 46s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5ef6cb7523364ba6ba9decd96b854a38 : FAILURE in 11m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d69e8baf56ce485ca25ad3766eb64230 : FAILURE in 11m 40s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"5b754f82ffa34eb6dbeb7844353f321668ecc5a9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-23 17:40:20.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5eb97f54168f8d4884940fa5e4b914b594ca8b23","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-23 17:40:30.000000000","message":"Patch Set 2:\n\ntell me how bad it is","accounts_in_message":[],"_revision_number":2},{"id":"c34fc4408e11867599715eae4c526c4f71c01707","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 20:51:10.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/4c5897ecad7d4958936ba8e4b4a0d00a : FAILURE in 48m 01s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/74e33a270a9c4ae99f75518bdd698914 : FAILURE in 52m 16s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/5d845694f39549ab8008083acd99925d : SUCCESS in 1h 15m 05s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/b33cc78239f54678bcdacd9f95846534 : SUCCESS in 1h 05m 20s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/8a4e3376ebaa47799c53124d2350f2d4 : FAILURE in 30m 52s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/1d78059d1e39478fab5689c58b3b1db3 : SUCCESS in 53m 13s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/3ca661f1c6664aeabd71b7e41b541762 : FAILURE in 50m 42s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/e2421971920043db96c2e6c3be06bd4a : TIMED_OUT in 3h 03m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e2b9be27d1747198e07f8e4ad7a27df : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/202ec54e8a6d4bb28810711e907e1dbe : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc0c0cc2f8a24349adb5396945ad60a6 : SUCCESS in 3m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/278a2ccf5ba4429684fc05f51f379745 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21d0e83484b7488a9deed8ca98bfd76e : SUCCESS in 3m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a7f0f9460e6b4e30889105e89306d65f : FAILURE in 29m 21s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2b50f3420e294d7eaca5026d5c6833b1 : FAILURE in 29m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/db8c94bc713848a0a610530e99b6b620 : SUCCESS in 50m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ad724f57be9847ddbba13ef0f56da7b6 : FAILURE in 19m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ab20f5ea037341169e14a5322c2376df : FAILURE in 26m 43s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"ce8db029a0fc9b2afe1ab4abcfd5a9bdd2058248","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-23 13:17:01.000000000","message":"Abandoned\n\nstein out of order","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"04022f60bcd5b4fcdf99f3f9270fdeea34e7373a","revisions":{"3125e395253ec13238b39f818bb7a477ef16c11e":{"kind":"REWORK","_number":1,"created":"2020-12-15 12:25:28.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/24/767124/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/24/767124/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/767124/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/767124/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/767124/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/24/767124/1"}}},"commit":{"parents":[{"commit":"8824c337a1e2f037941a16d347d8f88db11feabb","subject":"[CI] EM - Stop periodics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8824c337a1e2f037941a16d347d8f88db11feabb"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-12-15 12:24:48.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-12-15 12:24:48.000000000","tz":60},"subject":"[DNM] [CI] Healthcheck run","message":"[DNM] [CI] Healthcheck run\n\nChange-Id: If708c819af94a70d4413e7405398753627f7a097\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3125e395253ec13238b39f818bb7a477ef16c11e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3125e395253ec13238b39f818bb7a477ef16c11e"}]},"branch":"refs/heads/stable/stein"},"04022f60bcd5b4fcdf99f3f9270fdeea34e7373a":{"kind":"REWORK","_number":2,"created":"2020-12-15 12:27:01.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/24/767124/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/24/767124/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/767124/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/767124/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/767124/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/24/767124/2"}}},"commit":{"parents":[{"commit":"8824c337a1e2f037941a16d347d8f88db11feabb","subject":"[CI] EM - Stop periodics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8824c337a1e2f037941a16d347d8f88db11feabb"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-12-15 12:24:48.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-12-15 12:26:40.000000000","tz":60},"subject":"[DNM] [CI] Healthcheck run","message":"[DNM] [CI] Healthcheck run\n\nChange-Id: If708c819af94a70d4413e7405398753627f7a097\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/04022f60bcd5b4fcdf99f3f9270fdeea34e7373a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/04022f60bcd5b4fcdf99f3f9270fdeea34e7373a"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
