)]}'
{"id":"openstack%2Fkolla~767482","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~I9345a123838dd5ff5a4f4c5d37f40ac233807104","project":"openstack/kolla","branch":"stable/victoria","hashtags":[],"change_id":"I9345a123838dd5ff5a4f4c5d37f40ac233807104","subject":"use freezer user for freezer_scheduler container","status":"MERGED","created":"2020-12-17 09:34:06.000000000","updated":"2020-12-18 10:03:12.000000000","submitted":"2020-12-18 10:01:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"767482","cherry_pick_of_change":765681,"cherry_pick_of_patch_set":3,"meta_rev_id":"1deca46b103031046898c5867d4b8806f79cdcd2","_number":767482,"virtual_id_number":767482,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-18 10:01:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-12-18 08:50:49.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-12-17 17:07:44.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-12-18 08:50:49.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-17 09:34:06.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2020-12-17 10:58:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-17 17:07:44.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"354f30542489b74c48357e2fdd300cb1b6cd8bad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-17 09:34:06.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"20deeb2127263f9c2cd46005572294e20615bdff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 10:58:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/45af2fbb494e44fea28897ee9052c60c : SUCCESS in 20m 30s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/03c0f245191e468c8d25e82bb00d7607 : SUCCESS in 25m 32s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/b878b29c6a33442f8d04b8cc96c372a4 : SUCCESS in 53m 40s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/d93a8cf32f804199ba8a12becf36b43d : SUCCESS in 1h 02m 09s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/30dd0a2702c24d30bc03482215f03c87 : SUCCESS in 1h 06m 20s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f2f50ad386c84194a120e337afce4c06 : SUCCESS in 16m 13s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e0e8cf1eca2647cf9acde8d74b38f3fa : FAILURE in 18m 14s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6f26ef3e386b4db5b7a561984b0bea2e : SUCCESS in 1h 06m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/388685d50d08402d83eeccc05b543aea : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f1b4074e71c4b4c94e7979e4a457648 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82d2bb87bb0a4e389ea56d6d2319a049 : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee092e1af71147b8a0e7cc107e43c531 : SUCCESS in 5m 22s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0b8296af12de4e69ac016a709d9cd561 : SUCCESS in 18m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3b1a1948239f435597fefb2072933628 : SUCCESS in 18m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/80f47327b93f4bb6966b2a22c3330979 : SUCCESS in 1h 01m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4a8cf75392af41e2bafed7f50e3161ef : SUCCESS in 1h 14m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b0fc759dddc345fd84fe1a274e41df22 : SUCCESS in 1h 01m 37s","accounts_in_message":[],"_revision_number":1},{"id":"0ca52e970ed2330b01b8d5bc0ab5ed73f48719fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 13:51:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/d216a48886bb4910bb65411404edc24f : TIMED_OUT in 3h 01m 00s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/575e160276584a708d3674a5b42ba678 : SUCCESS in 1h 31m 34s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0053a2ba9c154401997473e8d754deb0 : SUCCESS in 2h 57m 03s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/bc451a0b80d44f069b4dc4df406a491b : SUCCESS in 1h 30m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6927877e154abc5db8ae2c770b8b6baaf3f454f3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-17 17:07:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7b3147f745c8c292a68ea77ff6c1af51a87199f2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-18 08:50:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"95d608ba6fe5a332ceb5583c288a53cb830ea3a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 08:51:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f89deec98d68c87989d69192d9a3caacdc0338ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 10:01:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8e1282ae6c0d4d76b4752e853733fb42 : SUCCESS in 26m 15s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/45321a4ccdea4e4c961395dece6d9f88 : SUCCESS in 1h 02m 04s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/8ada37a35b42434ab3eeae6c6af4f1a5 : SUCCESS in 1h 08m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6411c264a714e9e9e46f2fefbf6f3fd : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1dae9e1ee464936bd1dec918b9a5d29 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f95272b97ef48d8a3b9a20279a6ba74 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1804aa6a39794b7889cb736e02047f26 : SUCCESS in 7m 22s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e72ff80322d64cac9fe5f00b12c82ea4 : SUCCESS in 27m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4d18a0af9f35400499df256fd4dd1ec5 : SUCCESS in 52m 05s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e351f5dc1b6b4eb0946850b8603ee452 : SUCCESS in 59m 30s","accounts_in_message":[],"_revision_number":1},{"id":"5da2738c8376e88d17e94a53e0c941db23a3f382","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 10:01:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1deca46b103031046898c5867d4b8806f79cdcd2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 10:03:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f0d4a183e6a4d2c85b2059f7db51b04 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2cf83cc82a96f1c7deb35c440a84dd0a37561c2e","revisions":{"2cf83cc82a96f1c7deb35c440a84dd0a37561c2e":{"kind":"REWORK","_number":1,"created":"2020-12-17 09:34:06.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/82/767482/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/767482/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/767482/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/767482/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/767482/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/767482/1"}}},"commit":{"parents":[{"commit":"40a1c7a6ebabb1554e0ce7b54d6b356c2bac1ce5","subject":"Add daxio for nvdimm scenario","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/40a1c7a6ebabb1554e0ce7b54d6b356c2bac1ce5"}]}],"author":{"name":"wuchunyang","email":"wuchunyang@yovole.com","date":"2020-12-06 07:41:20.000000000","tz":480},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-17 09:34:06.000000000","tz":0},"subject":"use freezer user for freezer_scheduler container","message":"use freezer user for freezer_scheduler container\n\nfreezer scheduler runs with root privilege which is needless\n\nChange-Id: I9345a123838dd5ff5a4f4c5d37f40ac233807104\n(cherry picked from commit 91cb6166129be975afbd2adc3b43e27931788910)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2cf83cc82a96f1c7deb35c440a84dd0a37561c2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2cf83cc82a96f1c7deb35c440a84dd0a37561c2e"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
