)]}'
{"id":"openstack%2Fkolla~767484","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I9345a123838dd5ff5a4f4c5d37f40ac233807104","project":"openstack/kolla","branch":"stable/train","hashtags":[],"change_id":"I9345a123838dd5ff5a4f4c5d37f40ac233807104","subject":"use freezer user for freezer_scheduler container","status":"MERGED","created":"2020-12-17 09:34:33.000000000","updated":"2020-12-23 18:53:04.000000000","submitted":"2020-12-23 18:51:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"767484","cherry_pick_of_change":765681,"cherry_pick_of_patch_set":3,"meta_rev_id":"b75a5a78ff863478e5c10c817795d9d35da50efa","_number":767484,"virtual_id_number":767484,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-23 18:51:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-12-17 12:03:09.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-17 14:16:29.000000000","_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-12-17 14:16:29.000000000","_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-17 09:34:33.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2020-12-17 11:10:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-18 19:48:32.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2020-12-20 22:58:36.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"}],"messages":[{"id":"cbd2da5f857415108d4235d760a2940932fcbdb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-17 09:34:33.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b546d2f783ef43479239184292b1191719bd719b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 11:10:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a5cc18f3ea7342598eecfc40c906240c : SUCCESS in 37m 28s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7074154b01df4d7891a47e641517b02e : SUCCESS in 1h 24m 23s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/749d80365ab64ea5abd4d46bcb768404 : SUCCESS in 26m 56s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/62c127a334e04f7aa4ee148a67045253 : SUCCESS in 53m 06s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ac694fab713a447d9d48da6a82a092e7 : SUCCESS in 59m 19s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/10b768cad910453385ba4b27e6508774 : SUCCESS in 50m 29s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/0df3d1ee3f774bceb3de4553dd974f20 : SUCCESS in 1h 07m 32s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c2b6e7840fbe4820bfd45c2f4921b302 : SUCCESS in 1h 13m 07s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/ababcd2fa6ce4adf9f5a888c3febb501 : SUCCESS in 1h 19m 50s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/754b9114ff20454393693528a0443bd8 : SUCCESS in 1h 10m 38s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/cf28ef6203244b50b79d447f346d7f24 : SUCCESS in 38m 04s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5c73a90475634d41bf35949d7b9dd90e : SUCCESS in 17m 36s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a67bde959c1a4d16831273f9a47f2d01 : SUCCESS in 50m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e20a77bd41745958819e027330ecb54 : SUCCESS in 4m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/df9d8ddf9f4b42c8a850b26856788c44 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8fc1df6ae9a74d5a9389665fda320ea2 : SUCCESS in 3m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6d4b057a72f946aabac7e263db200c6a : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46c979818e3a4f78965906cfd180e9ae : SUCCESS in 5m 49s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fc5f00e6202d41e5810b09b3c11245d4 : SUCCESS in 36m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9768ff3f209240599a73414ec7452791 : SUCCESS in 27m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2a89c4880fc145c99da7aec71949656e : SUCCESS in 1h 04m 56s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/14077e5770e043ae8f316137b53cf1cc : SUCCESS in 1h 02m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/efcb33e9801e4a40b1d945fdbd4e55e1 : SUCCESS in 56m 10s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"7b85a554ae9446554462734be4167014b7d5b21c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-17 12:03:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c8d02ddb5e2b2f440c2d326c50ffb1b7b91a2960","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-12-17 14:16:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7b1f756a07adacfedaa73cfb777432ce18a73e9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 14:18:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ea85e3eee48e6eafd5275e13ee3e17548bdfc6dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 16:20:38.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/587da143bf894698b0a865bbdc48797b : SUCCESS in 1h 30m 12s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a380a85bb3de4887b82be81165b98dc6 : SUCCESS in 32m 41s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/6e16a233b3824029b203e49a75c1a851 : FAILURE in 7m 04s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/21343ab7c86645f88ecb6faeb2dafec2 : FAILURE in 8m 21s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/209f50f8ce664af494954452a5e9a31a : SUCCESS in 1h 22m 19s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/90773abe4d594a778f30015be2b35fc9 : SUCCESS in 49m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b24a27f018443c88687bb7111d28060 : SUCCESS in 4m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6c69cb6fb9ff4376ad1ef24a61e29dec : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7edbe386b7e41ff887a4940e62d7c97 : SUCCESS in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e77c8306fe114f0b92d50ffb47d91a7a : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25f978737f564b10bdf31a60402a641b : SUCCESS in 6m 13s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1929d5605aff46f984873d72345381a1 : SUCCESS in 23m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7934646addb04986a6cb2e2bbde13ed9 : SUCCESS in 47m 57s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f86ac761bd1748e887ae84f1bb4ea6ba : SUCCESS in 1h 00m 43s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"63687e7a004cce6141e859d330a3ade45c6eac52","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-18 19:48:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b9f6b2591578176fa02f0f496313b80e6c6eabcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 21:41:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/7e8dbf5f424144bd95c5f5550e52e9f8 : SUCCESS in 46m 33s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f2a463e69ecf4931b41ab84b5c6817fb : SUCCESS in 46m 13s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/93dc8cd7ff944b4e9bbcaf35a5fd3219 : SUCCESS in 21m 11s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/33ce748c0b344706aee6543d59036590 : SUCCESS in 27m 50s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/db607b20dbef45c7b0b5bf5a0c51d29d : SUCCESS in 1h 23m 46s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d977dfb759ae4828ad44d39ade96e36e : SUCCESS in 54m 12s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ba0f127ffaff47eaa0244f7d61a61dd8 : SUCCESS in 1h 08m 24s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/598d801bb61345b4a6e3b4cdd40ce954 : SUCCESS in 55m 13s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/cc027a003afc4afa80329d18dc06e233 : SUCCESS in 1h 35m 51s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/1d5413a128af47189fb321ca1ed22402 : SUCCESS in 1h 05m 19s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/71ced2109de24ff783a6ecab818305e5 : SUCCESS in 18m 22s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/af7912aba74a45aa898ff03dd378a527 : SUCCESS in 17m 41s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/24f8efdc6c7d4da6ac50576af2a83cb1 : SUCCESS in 48m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f049ef329a7c4b19b51fa218ff3b9771 : SUCCESS in 4m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4e31f53ba17d453ebdd89c1c3126680a : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb052b14f47549fcb2203f05ad645330 : SUCCESS in 6m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f04dfa98b05f458c8b40615b5b358cec : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8538ae25bb22447090d426048a75979c : SUCCESS in 6m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6b41ceeaac274e6fac9101a4ff5e5c6c : SUCCESS in 19m 49s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/336b170aeb7e4b82b3319d8566a43ec7 : SUCCESS in 20m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b43c591717254a30b13afa201b65608f : SUCCESS in 52m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d0391712646643b092ee75647ca16cba : SUCCESS in 51m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/aea2dc294fba4ea6813adf25bfeff710 : SUCCESS in 1h 01m 48s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"92ba253cd4c1003ecb008d88061a8e4bb9a20498","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 21:41:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2de7bbe08b88e6b651f8f1de9538feb85b799e4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 23:22:12.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e9ed382eeaa74e3592bb38cb959c2ccb : SUCCESS in 55m 08s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e13f6c92b37344c3a4327702146f56ad : FAILURE in 5m 06s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/5012add0c6a24132b9c0ce165a41710b : SUCCESS in 1h 16m 27s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/84ecfbca84104b019d0adb9a039b3355 : SUCCESS in 1h 09m 22s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/0c2d2b348bd54f1e81d8930e9c6c5927 : SUCCESS in 1h 17m 45s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/b5d3f9f73902449582179394a04c3bf3 : SUCCESS in 48m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38a11391229540c59c33739fba57abf2 : SUCCESS in 5m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/73d00ceaf17f4b898329d75dc1d3da74 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aecd7284b4d74ddcada06571ecdc7d6d : SUCCESS in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/29eb6f40d5ab4f62b18f0d1c4a5caff3 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3eb327cb53634740b8f630d46fb209f5 : SUCCESS in 7m 06s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6e320c03872c41518aa231fbb8eecb60 : SUCCESS in 23m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6654ba6cd933424785fcb0662c9d17c0 : FAILURE in 48m 50s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6b7ce32cd9034305b15b59c64b912d8a : SUCCESS in 1h 04m 33s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"c00dfb4f52a8b66e6af06b3cd3c815e8513640b2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-12-20 22:58:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e8bfa1424ba239f40ef764e3926f9ef9a8d32c36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 00:33:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/70b02da208c449adb71d197ab1ecc763 : SUCCESS in 47m 52s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/434f4a4f184644cf86c96afb17527cf7 : SUCCESS in 49m 13s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/63f471175e894f879f5b25f2d324073e : SUCCESS in 27m 45s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/95f42ff0a69045feac69b5aaa409958e : SUCCESS in 27m 56s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/60ccb921ed004a6497b86a4309c7c40e : SUCCESS in 1h 21m 40s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/6df5569cc8d74dbaae9085f234014595 : SUCCESS in 57m 21s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/abd9184e7bb84e2f8e206dbe92ae7811 : SUCCESS in 1h 07m 04s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/cb05b33de97648eeb5a57bf54f7be614 : SUCCESS in 51m 17s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/1bcb85ad84154c1eaf90f1a0edaa381b : SUCCESS in 1h 30m 58s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/fc2b2ec43ff043d284498b3bf84a4e39 : SUCCESS in 53m 38s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/40860fa2aaa24f86ab9b0fd7b3cf4ead : SUCCESS in 23m 58s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0f44d6f20872408595db4b2d1f7772d5 : SUCCESS in 18m 35s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fcb4c00790894b089b368e1d709e5e8e : SUCCESS in 52m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ccb976013ad4867b4d2afce78010edf : SUCCESS in 4m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2b79440affce451f87b83c1c99a705f0 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d5b7398c55f40bbb5a4afbb491ea821 : SUCCESS in 3m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/48b11401ca5b41489bdd3660126c469f : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d507627fbb0345faa2fd928e14a4a6f4 : SUCCESS in 5m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/930c688f77f84441803254588234b67b : SUCCESS in 21m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2cde6b313824483ba66d8a7cf4a9b162 : SUCCESS in 20m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4ef4e72b520e407db800d640ecaa0f57 : SUCCESS in 45m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ab4e98ee49174a7c8d715b6a9513ff42 : SUCCESS in 46m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d6d50cf7d262412ba240bf72cbba473a : SUCCESS in 56m 20s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"bc6145e22ff1a39bbc995c363d9676c2702bb5b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 00:34:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1dedd0a3dc072d68a8c1b337946b351b82064ec1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 03:37:36.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d1c3255af9a241d1b2efebfe9ef22271 : TIMED_OUT in 3h 02m 23s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/56321b586eb44ae7bcb04468fbe30c42 : SUCCESS in 2h 05m 55s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7ea6b60254dc45c998dfdc38715fb606 : TIMED_OUT in 2h 02m 28s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/5e7fef20bc3e414b80bcd98a32d48dde : SUCCESS in 51m 24s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/edb779ab0a824763a62160ab66947bde : TIMED_OUT in 2h 31m 40s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/83cceeb744a7458dac02cd22d667ab30 : TIMED_OUT in 2h 05m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be438906e15f4968921da3cf2963a3b2 : SUCCESS in 5m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8381d56453d74466b1a12172774a3f2a : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40b114cebae7438eb0d75a533c2379c0 : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f909fdea74a64924ba03e0979118e0a2 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1361637538a14bb1b9fe4d4aab552f2c : SUCCESS in 12m 19s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/df64e1ac1b5e4c53849a8758167988f0 : SUCCESS in 28m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/13a3a5ec0a134a27afa775008a0fc8d6 : SUCCESS in 1h 39m 48s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/51de0618bc464811b32bcfef9bdd17a1 : SUCCESS in 51m 53s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"829b6dd5ee8a6101c8e82e028f067d641ff5b76d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-12-23 10:24:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5d7ef5fbb8d58bc34d020d47334beffd81bbfce2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 12:09:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/7e64b33c603249a1b29d8275465eb638 : FAILURE in 48m 27s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7fab440680c74bfaad842ea4875ab35e : FAILURE in 58m 23s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/f753ef7f3cbf4f6eb6cdfbf5d37ba7f7 : SUCCESS in 32m 33s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3e64049dde5444aa8a8b875a61dc0106 : SUCCESS in 29m 58s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/6b6a688b995741cd8ab521c860286f0c : SUCCESS in 1h 38m 16s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/40b5817edc244dc68e4935f235463244 : SUCCESS in 57m 51s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/eea6f87828464395b82e5a769e038a35 : SUCCESS in 1h 03m 40s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c0cd8598d59440469852c28d83a45938 : SUCCESS in 56m 20s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/673e5b0ca9764978953bb7f1a39953da : SUCCESS in 1h 17m 25s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/193c948910474bda91d508fb597c9a63 : SUCCESS in 46m 55s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ab47130faf824d97a3d18c77e9e7688a : FAILURE in 23m 02s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ff01011a94c44ed082d56b2b78f902c4 : FAILURE in 18m 59s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/218badb1beef473a91af7c7f254edcd0 : SUCCESS in 56m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a750c134a0a04ab4b65f534dd6b58f81 : SUCCESS in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a47eee6624c848d5a2eccdbe4da0769a : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bda7840ce3614662b4b4a50e56da2da4 : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7ae5096beb384c7c85f6d52822fa930b : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ea4eddfade049a2a5a810715f754a3b : SUCCESS in 5m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d981c9983031455799dbb44784075eb8 : FAILURE in 24m 51s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0958959dd3c14e21a0510d2b03349cf9 : FAILURE in 24m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a884d86e2fad49adb57495568cb8414e : SUCCESS in 50m 18s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/73f5797be2184b00abfa02b0ac44e130 : SUCCESS in 47m 31s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/abe3c831da6140db9a5ae9a329498adf : SUCCESS in 59m 11s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"c1fbbc27627432d51cf94979b75221283932c3d0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-23 15:35:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c73e9a488ed32597dd7fb6a6c2b8c832da3bdbb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 17:15:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ddae5679fcff4c86b0e0befcd4596221 : SUCCESS in 40m 33s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/383ab562fa5b4c21b4b4e26875aa0639 : SUCCESS in 42m 56s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/83eebef9205c4961bd717778e5f62413 : SUCCESS in 32m 22s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3612bf8e19b24ec6bcc5a1fa2393a7ea : SUCCESS in 39m 23s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9250a537793e42c9a3786782039d4dbc : SUCCESS in 1h 07m 02s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/78cedbb0b94448908f357423b8ad6f7a : SUCCESS in 50m 34s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/7c67e3f0174c4af1af5018587910c9e9 : SUCCESS in 1h 07m 00s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/9bafb7fb174441d697103d95a6731092 : SUCCESS in 57m 15s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/5404f17588f8406daf281ee37585af63 : SUCCESS in 1h 36m 17s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/7513fc07642b4cdfb346f7e15ff3984c : SUCCESS in 43m 24s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9f9a7e4689ea490f882d4ae4fd07cbdb : SUCCESS in 21m 20s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1d4cda9fef144c659a60229780d9c41b : SUCCESS in 14m 47s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/eb49933c2dd44ba7877aac3523a7f372 : SUCCESS in 50m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/964b357f5f524a21be6e399a949f1352 : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a4ee7167a60a4405a4c6c6aabaf059d1 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e4627a00d2a471ab94592737946245c : SUCCESS in 3m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0cb1230471b042c0b7d18653c59dbef7 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97badd9a3a154d44826a28aa05c81dda : SUCCESS in 4m 48s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d3ffdf81ef514ea4972f8fe4dacc2c41 : SUCCESS in 22m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1afc45a998b043ce9d5d5fdce9d5902b : SUCCESS in 23m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f18ed75bf3c74ea0b2ccc721d251b24e : SUCCESS in 50m 25s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/28d615f107b348bb909e038712dd2be7 : SUCCESS in 43m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0ab10747e4874ac0a8406d3fc39e2030 : SUCCESS in 55m 17s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"dedc710d2d4e3e4a0bfa805c7680572d53250cec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 17:15:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2fab7c8e2fdce72edc2897851f5cf4f77326ac6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 18:51:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a3c050f252244fcdbecda60c798dda86 : SUCCESS in 49m 11s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a8a57d169c084c8e85ee6a004b2a2b7a : SUCCESS in 25m 26s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/20e6cb2326c64e648b067979ee48752a : SUCCESS in 1h 30m 27s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7ffc8432a02244d6834307fe0cba648e : SUCCESS in 53m 35s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/942efde325904dc2afd415cc0e3157af : SUCCESS in 1h 31m 38s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/676fe64b0a52440dba486ca1fda4c01a : SUCCESS in 43m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce232a4855b0433a8ec79ad8792231b6 : SUCCESS in 4m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4f2f4f0ddb4047699db329c9cef8339d : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d656015a7584cb2aff6ef47dcff73e2 : SUCCESS in 3m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6f65859e8fcc447cafbdbcae2717ca51 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f37456d53a234c3aa273c27e734ede44 : SUCCESS in 5m 44s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2259e21e633f478996d4e018fe123d18 : SUCCESS in 34m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b1fff1a4486840ea84b91511472c6aa5 : SUCCESS in 46m 52s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6ab9574bb09f482ab9efcf69a33f48f0 : SUCCESS in 1h 00m 07s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"968d1c07d5a0f963d68a3b3ad47f582acbaee268","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 18:51:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b75a5a78ff863478e5c10c817795d9d35da50efa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 18:53:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c73e5162ec044fa838425a606aae171 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"078da7ed655bcb8fccad315d94aaa8a04763da1a","revisions":{"078da7ed655bcb8fccad315d94aaa8a04763da1a":{"kind":"REWORK","_number":1,"created":"2020-12-17 09:34:33.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/84/767484/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/84/767484/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/767484/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/767484/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/767484/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/84/767484/1"}}},"commit":{"parents":[{"commit":"617421fb380f8b26531c4c4ac78b65a3f5a135ec","subject":"Support CentOS 8.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/617421fb380f8b26531c4c4ac78b65a3f5a135ec"}]}],"author":{"name":"wuchunyang","email":"wuchunyang@yovole.com","date":"2020-12-06 07:41:20.000000000","tz":480},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-17 09:34:33.000000000","tz":0},"subject":"use freezer user for freezer_scheduler container","message":"use freezer user for freezer_scheduler container\n\nfreezer scheduler runs with root privilege which is needless\n\nChange-Id: I9345a123838dd5ff5a4f4c5d37f40ac233807104\n(cherry picked from commit 91cb6166129be975afbd2adc3b43e27931788910)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/078da7ed655bcb8fccad315d94aaa8a04763da1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/078da7ed655bcb8fccad315d94aaa8a04763da1a"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
