)]}'
{"id":"openstack%2Fkolla~767551","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~I4172b814d5d95ae3debbc1cf3feaeee2e2bec9bd","project":"openstack/kolla","branch":"stable/victoria","topic":"inc-bkup-fix-stable/victoria","hashtags":[],"change_id":"I4172b814d5d95ae3debbc1cf3feaeee2e2bec9bd","subject":"mariadb: Fixes incremental backup failure when full not created today","status":"MERGED","created":"2020-12-18 08:42:19.000000000","updated":"2020-12-18 19:17:13.000000000","submitted":"2020-12-18 19:15:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"767551-inc-bkup-fix-stable/victoria","cherry_pick_of_change":758486,"cherry_pick_of_patch_set":10,"meta_rev_id":"a3431b06cb41a315a0aa0a65645d185ab3df16b3","_number":767551,"virtual_id_number":767551,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-18 19:15:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-12-18 17:25:34.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-12-18 15:25:04.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-12-18 17:25:34.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-18 10:05:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-18 15:25:04.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"908c6ce44580fcd98dca3aa90b27ad09d7f7fd37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-18 08:42:19.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f49b5926d1928f54e201ee40fe3cef9d9ca982ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 10:05:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/632ec58e1ebe414690519a51b60a0f5e : SUCCESS in 23m 17s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/35e39303fe794213bd257bd27b876516 : SUCCESS in 22m 27s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a2099c897305428185359f04b45f944e : SUCCESS in 1h 04m 10s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/03c2f4ce2ae14174aa3205894ef564f0 : SUCCESS in 1h 04m 15s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/430b002b0e31449597d551341ca4d063 : SUCCESS in 1h 13m 19s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/dc52f874f6fc40368ee45c4d15e8e1c8 : SUCCESS in 17m 37s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b31faabad2d54559b44aa4f62ad9f29f : FAILURE in 20m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/eb55d771875b40e391667cf4b1a8a6c3 : SUCCESS in 58m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d8553176d1347e49e065d1112316580 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ddc1efb8e699477ea01b4aa8a5fd4266 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6582baf267b94ab2a817b43355fef2b2 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fcd60d1ea4a42189756fa520285e985 : SUCCESS in 7m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/369a6722e47f4ec79733da3a41962fd4 : SUCCESS in 4m 13s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1ea3c8f2afe949f5a4a6bc97002d9a07 : SUCCESS in 18m 21s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e0f0e99712cb4645992684716468f799 : SUCCESS in 19m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8e92877771464765b49c2c473e9649e0 : SUCCESS in 53m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/eeab4b1e69764f0e821884d40681d71d : SUCCESS in 1h 21m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/aa1ddac7e1b746a3a75130d691db83a1 : FAILURE in 37m 58s","accounts_in_message":[],"_revision_number":1},{"id":"94f3bb2bcd9eaf625816fc9f37a9227aff536a22","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 11:58:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/1f54c8829e0047b5bac46a22327965d0 : FAILURE in 2h 09m 58s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b7a51e5bd1d24706b7c17de418a04203 : SUCCESS in 1h 13m 19s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/aa94e69fc5ff465bb2cc53f8bd3d2d85 : TIMED_OUT in 3h 03m 54s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/ffe08e924b9c497ca127b788c70df26b : SUCCESS in 1h 06m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a06abaf4eb85e74dafee8d38b610bf02a1d9ea7e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-18 15:25:04.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"46d142a263e273549dbbe0bf61c7cd490603ee6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 16:44:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e68012e97951414a913c50de381122fe : SUCCESS in 20m 16s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ff1a89e2068148a6ae6fe390588bddc9 : SUCCESS in 27m 03s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/96323f2b144345b889b59ddba49de865 : SUCCESS in 1h 00m 28s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/2f6647fad9bd4ccda36ea8cbccf925ab : SUCCESS in 59m 09s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/87e0595cf8324f7e9c3486c383b77025 : SUCCESS in 1h 07m 18s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5f9465d3c33f4dc0b93015852a3f5193 : SUCCESS in 16m 32s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3716f26e71fe4e78a6e9e54e3f2bcae2 : FAILURE in 16m 06s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c259ab1865614c98a58c946181d69c1b : SUCCESS in 50m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb70f95ede414553803ca2be1773c487 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e37f69b97f14c01bb6a6af8979f2ec4 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57bfc8d44ed64f388b268cb4d212ca02 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83217165c52a41fca5819b1e2ae75b1e : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ae5698bcf074df998d282ee5e95f2b1 : SUCCESS in 5m 47s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3ff98eb4b9864121baebd18b9cbc9670 : SUCCESS in 21m 22s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e7bcf7110ce8429bb015b6abb530c3a2 : SUCCESS in 19m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e1d93ca603124c8e89984a19da8a14a3 : SUCCESS in 47m 49s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/44b66bae08dc44a3a52bc464cffdaca7 : SUCCESS in 51m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8e854c6efb674fb0883c993ab81df8d9 : SUCCESS in 59m 14s","accounts_in_message":[],"_revision_number":1},{"id":"7c2e2bca3f9a069bf49da36cef9e42ff7846ba5e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 17:10:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/c9b283c3571e4cbca79fda150abe04b1 : FAILURE in 50m 06s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f5743d042b9b44f8ab05904d2bd709eb : SUCCESS in 46m 21s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a08fa2c589aa452aad3cc313afc3e2d4 : SUCCESS in 1h 37m 40s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/2c962fe3da2e4b6bb86ae2bf5d1329f6 : SUCCESS in 44m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3f99bdba0d9b5a63424bf1474330399dd1131b68","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-18 17:25:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6772d12de7204b03f522087b97d852f042e55c7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 17:25:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9404cde58d2aab23c9b59a1472f16ae0aafdc7b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 19:15:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/1d1c48f675bb4cdea16e5d8789afe8f8 : SUCCESS in 21m 41s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/ef8db3dfbc7f4c528c6dcaadf935d5b7 : SUCCESS in 1h 04m 39s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/8c3e826e51f34de88783617b90aac635 : SUCCESS in 1h 43m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dd10d36837b45b2bf62c15401344227 : SUCCESS in 7m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d77bfb2c665f4e6994fda0d8dc548b82 : SUCCESS in 8m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90f5292698eb4480affa098aba45c826 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5857d2c1593b4fb2839cb9101ded5775 : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e05ad0e476142ef9578813f10d9c572 : SUCCESS in 3m 32s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3b32a4af1ad24ddd896f87e0bd8a2397 : SUCCESS in 47m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7a0a6ff8a9564c5caac2a8db1643029d : SUCCESS in 51m 27s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0e8241ea05cd4bc5b52918177c829cb2 : SUCCESS in 56m 12s","accounts_in_message":[],"_revision_number":1},{"id":"d3277430ad995fdbefc29ead365c4c34d785bd0f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 19:15:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a3431b06cb41a315a0aa0a65645d185ab3df16b3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 19:17:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1445b5677e1412a9f5e41b4760d4b3b : SUCCESS in 1m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8232c5c4ca974a68bc48f52e35efa43a : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2d18e197d9acb3cd06f38d0a2488adb0d601337b","revisions":{"2d18e197d9acb3cd06f38d0a2488adb0d601337b":{"kind":"REWORK","_number":1,"created":"2020-12-18 08:42:19.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/51/767551/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/767551/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/767551/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/767551/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/767551/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/767551/1"}}},"commit":{"parents":[{"commit":"40a1c7a6ebabb1554e0ce7b54d6b356c2bac1ce5","subject":"Add daxio for nvdimm scenario","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/40a1c7a6ebabb1554e0ce7b54d6b356c2bac1ce5"}]}],"author":{"name":"abraden","email":"C-Albert.Braden@charter.com","date":"2020-10-15 19:02:43.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-18 08:42:19.000000000","tz":0},"subject":"mariadb: Fixes incremental backup failure when full not created today","message":"mariadb: Fixes incremental backup failure when full not created today\n\nbackup_full() saves the history name in $BACKUP_DIR/last_full_date.\nbackup_incremental() attempts to read from $BACKUP_DIR/last_full_date\ninto $LAST_FULL_DATE. After the read attempt; if $LAST_FULL_DATE is zero\nlength then $(date +%d-%m-%Y) is used.\n\nChange-Id: I4172b814d5d95ae3debbc1cf3feaeee2e2bec9bd\nCloses-bug: 1897948\n(cherry picked from commit ccfe65b5f7fe437141f342421488c176038fddb0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2d18e197d9acb3cd06f38d0a2488adb0d601337b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2d18e197d9acb3cd06f38d0a2488adb0d601337b"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
