)]}'
{"id":"openstack%2Fkolla~767689","triplet_id":"openstack%2Fkolla~master~Iabacaeccdb360544125ad910d5c3f5f9d895dac5","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Iabacaeccdb360544125ad910d5c3f5f9d895dac5","subject":"nova-compute: daxio on CentOS 8 is x86-64 only","status":"MERGED","created":"2020-12-18 09:49:38.000000000","updated":"2020-12-18 19:17:09.000000000","submitted":"2020-12-18 19:15:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"767689","meta_rev_id":"4b9330672ee59c52c889fde3ad0d680fd1291167","_number":767689,"virtual_id_number":767689,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-18 19:15:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-12-18 17:26:21.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-12-18 14:01:20.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-12-18 17:26:21.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2020-12-18 19:17:09.000000000","post_submit":true,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-12-18 19:17:09.000000000","post_submit":true,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-18 09:51:39.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-12-18 11:11:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-18 14:01:20.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"951e866984f651e1dfba74aa00df706075fcc8ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-12-18 09:49:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"45d07286dd55ce8d8b87650f03b8a4251e5e4fb2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-18 09:51:39.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"a3f2dd8508ca0847cf2d6e8a6de6fad9a2c114e9","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-12-18 10:51:17.000000000","message":"Patch Set 1: Backport-Candidate+2\n\nbackport to victoria","accounts_in_message":[],"_revision_number":1},{"id":"da7db156eb4244ef18dcd031684dfd3713811999","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 11:11:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/80c859508e6b4cafbb22f70eefccac3b : SUCCESS in 24m 19s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f5aaca45e5164b249fe14320d4567a0e : SUCCESS in 36m 26s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0d52200cccbc481caff8f67ac458d8d6 : SUCCESS in 57m 21s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/5cd05a1a1c6442cd84007bf23e3773c8 : SUCCESS in 1h 15m 36s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/d95834e4f9384453b964cb465412ee61 : SUCCESS in 1h 13m 54s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d3f6fab8b37247fb925bacde635b2b22 : SUCCESS in 28m 32s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/263ddb028f2146879db246efe7b5ea22 : FAILURE in 17m 49s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/34aaafa18dda4ef789287f7375a3abce : SUCCESS in 1h 03m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86a07fc3b4344429b722ce9606fcf497 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18d6074075fe4cf9befc3be4feeae22f : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/173ff2deda074633a12747c57038474e : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a6c5c0b32e5e4d67ac797d580843d891 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42e9d64e43c44ffead55daa8fdae15a9 : SUCCESS in 6m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d4ca08887e9d490683a56e5c3475dde2 : SUCCESS in 20m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2c2e8abfa99a4d95ba01f524d08e23fe : SUCCESS in 44m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3dfff0b4a6584e6687e19a67f139a285 : SUCCESS in 53m 13s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/82a399a38c9c4e00a33b899e12aa0bac : SUCCESS in 50m 06s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/385e6b3cac064b01acb9e358ee4ededa : SUCCESS in 1h 02m 07s","accounts_in_message":[],"_revision_number":1},{"id":"145c0add10eb44d78558fcde31e47b9eb736ba05","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 12:49:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/0fa2fcefa9344491a4618049ea900129 : FAILURE in 1h 23m 39s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/46eb95c5ecda4c5d8e58d6aa77dcfbee : SUCCESS in 1h 55m 25s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/0b8d04d2f1c143f8889f90d1263f4602 : SUCCESS in 1h 11m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8b0cf7145e70177268c3421fdc52c2185d14f947","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-12-18 13:24:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c741a5a9618422d0efd5db9d74b9bb078affc164","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-12-18 13:32:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6722212347b43bdc43734010435836f354720647","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-18 14:01:20.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"eb9eb7e1028b8f4083c17de4648d702be0010902","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 15:06:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/3c7df1d504184fe5a3b347a773b8e800 : SUCCESS in 27m 45s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/0f16583fcf4245d696e6ec4d03db297d : SUCCESS in 42m 20s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/75278ee50e4e4d638f656de96308e428 : SUCCESS in 1h 16m 02s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c6a860eb2c454fe0817a3a01d9e6bc41 : SUCCESS in 1h 02m 07s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/20db998798e94cd2a0a72fba63ffe020 : SUCCESS in 1h 24m 45s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b5868acb240045009c9142c5adcc9e5f : SUCCESS in 33m 59s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/807c3316138d44578b3b2daaa258c58a : FAILURE in 18m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e5d7d05cfffc48d18e592fea55a24c0c : SUCCESS in 57m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6b79a3770504fbb89b8ff1e12bcc67b : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e23233a2452a49adbfc3895619253430 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/404c78559ba440f0ac323d34f3634636 : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f9de18cf1eb4b659e9a23ea4bf9b2fa : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4330914a29974118b95c983d8bc7129a : SUCCESS in 6m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2bab37392e6d462eb6e96dd642a31bb8 : SUCCESS in 25m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8824ce65843b4bd8ab61dd8c0dd9f23e : SUCCESS in 48m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/69861d682c59425386cb3bbe94920ce8 : SUCCESS in 59m 57s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/68dde86a1ad743a8b087a5d8964ef70c : SUCCESS in 59m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1ef98d6d258c44fba894e9eae141f378 : SUCCESS in 1h 09m 28s","accounts_in_message":[],"_revision_number":3},{"id":"a118b0f51f8d65a0e3dd72acb534b2c5b4272ff4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 15:37:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/41431607e62b4731a787f26565334693 : SUCCESS in 1h 27m 22s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bc12fbafb89548e19b62865e4e7a21cc : SUCCESS in 1h 21m 32s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d18679b9ee834ebcb2005b5911a67f55 : SUCCESS in 1h 45m 59s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b1e949bf8b7e4cc8a476256cd5e95e9e : SUCCESS in 1h 15m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"aa6dc198018cfbd2ec71bc00fa083d0153c9341c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-18 17:26:21.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a51df2a7c0d2f5f7bce0d862773659083c6c4b0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 17:26:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"10d8e5b6dc5f12256e567a62bcf8be2b9e466ce6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 19:15:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f634d2be96854afbb456af88028e1de4 : SUCCESS in 50m 34s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/b462c6943e9140988f9fe817c3f0431e : SUCCESS in 1h 31m 21s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/129d161277114331a05566ce8a0de1e4 : SUCCESS in 1h 20m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81e3b87db9b147fd8edcb405314f7f5c : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac7ca8fcf5ec43dbb7f7e653dff9f2a0 : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12b9bd929ae94a95a77f3624eada2e38 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ed78a1cf1f94b9ca29b893450f1105e : SUCCESS in 5m 23s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/56f38170bde84090abf04987821fab65 : SUCCESS in 56m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1e589da92b734812836cee08689b305c : SUCCESS in 50m 23s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f15f62b85d4f456f91693637b38bb6c7 : SUCCESS in 1h 18m 27s","accounts_in_message":[],"_revision_number":3},{"id":"b456ff3bfb493350a181c6e53a14288b79c1f710","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 19:15:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"4a5b19c858cd8542aedc2c1455e5eceaf4d78ba6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 19:17:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0532c3be7b854bc7a4a2443c642dcd8a : SUCCESS in 56s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"edb6b60a9864012e3427155e909788f5625d126d","revisions":{"a8d9a4f30fc304bb9a93ad5913a86b9d8396b47f":{"kind":"REWORK","_number":1,"created":"2020-12-18 09:49:38.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/89/767689/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/89/767689/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/767689/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/767689/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/767689/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/89/767689/1"}}},"commit":{"parents":[{"commit":"1d36ae464830061473302e2ddc563e9c229aed5c","subject":"Merge \"centos: do not enable PowerTools repo in base\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d36ae464830061473302e2ddc563e9c229aed5c"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-12-18 09:48:51.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-12-18 09:48:51.000000000","tz":60},"subject":"nova-compute: daxio on CentOS 8 is x86-64 only","message":"nova-compute: daxio on CentOS 8 is x86-64 only\n\nI somehow missed that when it was sent for review.\n\nChange-Id: Iabacaeccdb360544125ad910d5c3f5f9d895dac5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a8d9a4f30fc304bb9a93ad5913a86b9d8396b47f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a8d9a4f30fc304bb9a93ad5913a86b9d8396b47f"}]},"branch":"refs/heads/master"},"b1d7707a14f91ed1bfc12be5629ad0564297c6d2":{"kind":"REWORK","_number":2,"created":"2020-12-18 13:24:19.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/89/767689/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/89/767689/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/767689/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/767689/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/767689/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/89/767689/2"}}},"commit":{"parents":[{"commit":"1d36ae464830061473302e2ddc563e9c229aed5c","subject":"Merge \"centos: do not enable PowerTools repo in base\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d36ae464830061473302e2ddc563e9c229aed5c"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-12-18 09:48:51.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-12-18 13:23:55.000000000","tz":60},"subject":"nova-compute: daxio on CentOS 8 is x86-64 only","message":"nova-compute: daxio on CentOS 8 is x86-64 only\n\nI somehow missed that when it was sent for review.\n\nChange-Id: Iabacaeccdb360544125ad910d5c3f5f9d895dac5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b1d7707a14f91ed1bfc12be5629ad0564297c6d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b1d7707a14f91ed1bfc12be5629ad0564297c6d2"}]},"branch":"refs/heads/master"},"edb6b60a9864012e3427155e909788f5625d126d":{"kind":"REWORK","_number":3,"created":"2020-12-18 13:32:54.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/89/767689/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/89/767689/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/767689/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/767689/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/767689/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/89/767689/3"}}},"commit":{"parents":[{"commit":"1d36ae464830061473302e2ddc563e9c229aed5c","subject":"Merge \"centos: do not enable PowerTools repo in base\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d36ae464830061473302e2ddc563e9c229aed5c"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-12-18 09:48:51.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-12-18 13:32:46.000000000","tz":60},"subject":"nova-compute: daxio on CentOS 8 is x86-64 only","message":"nova-compute: daxio on CentOS 8 is x86-64 only\n\nI somehow missed that when it was sent for review.\n\nChange-Id: Iabacaeccdb360544125ad910d5c3f5f9d895dac5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/edb6b60a9864012e3427155e909788f5625d126d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/edb6b60a9864012e3427155e909788f5625d126d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
