)]}'
{"id":"openstack%2Fkolla~768057","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I8690e5543e6b5f37f41fe35f659910fc869c034f","project":"openstack/kolla","branch":"stable/ussuri","topic":"feat/do-not-fail-on-unsupported-database-types-stable/ussuri","hashtags":[],"change_id":"I8690e5543e6b5f37f41fe35f659910fc869c034f","subject":"ceilometer: do not fail when using an unsupported database type","status":"MERGED","created":"2020-12-21 09:44:51.000000000","updated":"2020-12-23 10:30:55.000000000","submitted":"2020-12-23 10:29:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"768057-feat/do-not-fail-on-unsupported-database-types-stable/ussuri","cherry_pick_of_change":764441,"cherry_pick_of_patch_set":1,"meta_rev_id":"29dd297b4632990355ce383b7c1998e4570d9f44","_number":768057,"virtual_id_number":768057,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-23 10:29:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-12-21 09:48:34.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-12-23 08:49:59.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-12-23 08:49:59.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"}],"reviewers":{"REVIEWER":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-21 09:44:51.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"state":"REVIEWER"},{"updated":"2020-12-21 11:30:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-21 12:04:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-23 08:49:59.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"e9194bb0e54f7a4e2d8716f1e6f43d5efd282570","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-21 09:44:51.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ea498ce5da6acf578ac3af8a163e297bb2c9945b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-21 09:48:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9e6f83d61959682f2e0b21d75439f910f8a2a547","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 11:30:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/c9d80ef0fe04431e95f943b2744df2a2 : SUCCESS in 1h 36m 41s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6feeab952c224b86bbbb1cdb97ac5b83 : FAILURE in 1h 32m 00s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f10cffeece3b46f0a456571c4761814e : FAILURE in 49m 16s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/f62205f4d1324076b05a766036fa3d95 : FAILURE in 1h 34m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"651789abfc7983dd1397013448656c6e662e030e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 12:04:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/3b07b577b5ae404fbdde296c15eb3461 : SUCCESS in 52m 44s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2a3890b17f26462e97950a9f3219bfa9 : SUCCESS in 27m 30s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1130bf73982a4763bda1606e3de95126 : SUCCESS in 1h 02m 08s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/f5f3786e57084667bdf11fa4a1edf93a : SUCCESS in 59m 52s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/8fa24577ac9b430aa84c5a6a177072f4 : SUCCESS in 1h 38m 57s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/524122316422490bb75d956c190380bf : FAILURE in 14m 29s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1a83e25b33da4525a28f17ce10b362d8 : SUCCESS in 21m 17s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e9922767bf8f4acaa1a5340b4b3a3f78 : SUCCESS in 14m 36s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b53be82e1c8a4d9a824f0baf5f23e063 : SUCCESS in 48m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3da32da771aa455eb4ac56cf58619792 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25fbf29284fe43b19a381f8faca02bc9 : SUCCESS in 3m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/99e9bdcd1e84420090bb59c317137ab9 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/727b7d4fc7bb4ff3a15d6b6792ccf7e9 : SUCCESS in 6m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3c0d61355b8b4df98c47e4a5c85ff781 : SUCCESS in 47m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2d87234f4b794a16a07b9d4cf4d26ff3 : SUCCESS in 26m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0fff66df544a45c7a07879b859f74553 : SUCCESS in 1h 18m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1992421341d24445908fd03f91816096 : SUCCESS in 56m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fa1bad11b04a45bfa4fb7dfc3c55999b : SUCCESS in 2h 10m 25s","accounts_in_message":[],"_revision_number":1},{"id":"662c1bd31d87587a40e099cb48b48625a7a65734","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-23 08:49:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9cbd430707283b9028fae4924c885b9ec85b0c40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 08:50:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ca9d3a90ece3633b4b49b875650c144f439f4019","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 10:29:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/0e6b8702b4f04dda8b75079aac3a403c : SUCCESS in 47m 40s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/56e08313ff974994914dbf0b65f78845 : SUCCESS in 1h 32m 26s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/d6a22a47b3324054a95463843e0fd7a1 : SUCCESS in 1h 35m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/603b01860d3a46349ed3512899c96a26 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/becc530f70f04c17ad2453a27bdbeab2 : SUCCESS in 3m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/214e930322c64770bae298da8586ce37 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6346beac608f4d4da14f747f1cb2ac0c : SUCCESS in 5m 01s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/87f5a1998f104e5a8b3cdc4008a073c1 : SUCCESS in 50m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f327c5e777604cc8a693d4d4b7353b63 : SUCCESS in 1h 22m 59s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1c7ce84ed305464b8724e470be613ba1 : SUCCESS in 1h 32m 32s","accounts_in_message":[],"_revision_number":1},{"id":"0c516628caa6e57f4c5ce53e36d1d6f1811aa567","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 10:29:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"29dd297b4632990355ce383b7c1998e4570d9f44","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 10:30:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d2219341b3c4bf39691f5a5f9970569 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"84ffe90e2e577fa1583bcac4b6bd5fde7a90993b","revisions":{"84ffe90e2e577fa1583bcac4b6bd5fde7a90993b":{"kind":"REWORK","_number":1,"created":"2020-12-21 09:44:51.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/57/768057/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/768057/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/768057/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/768057/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/768057/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/768057/1"}}},"commit":{"parents":[{"commit":"eace4f421944c234c0134421482e21c0b6dc38fe","subject":"Merge \"Add daxio for nvdimm scenario\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eace4f421944c234c0134421482e21c0b6dc38fe"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2020-11-27 10:35:51.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-21 09:44:51.000000000","tz":0},"subject":"ceilometer: do not fail when using an unsupported database type","message":"ceilometer: do not fail when using an unsupported database type\n\nThis is necessary if e.g. Prometheus is used. Then the\nceilometer-upgrade is not needed.\n\nSigned-off-by: Christian Berendt \u003cberendt@betacloud-solutions.de\u003e\nChange-Id: I8690e5543e6b5f37f41fe35f659910fc869c034f\n(cherry picked from commit ea8592b077ceccfeb3089d901e6f46252f024881)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/84ffe90e2e577fa1583bcac4b6bd5fde7a90993b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/84ffe90e2e577fa1583bcac4b6bd5fde7a90993b"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
