)]}'
{"id":"openstack%2Fkolla~768058","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I8690e5543e6b5f37f41fe35f659910fc869c034f","project":"openstack/kolla","branch":"stable/train","topic":"feat/do-not-fail-on-unsupported-database-types-stable/train","hashtags":[],"change_id":"I8690e5543e6b5f37f41fe35f659910fc869c034f","subject":"ceilometer: do not fail when using an unsupported database type","status":"MERGED","created":"2020-12-21 09:45:16.000000000","updated":"2020-12-23 18:18:07.000000000","submitted":"2020-12-23 18:16:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"768058-feat/do-not-fail-on-unsupported-database-types-stable/train","cherry_pick_of_change":764441,"cherry_pick_of_patch_set":1,"meta_rev_id":"8555e42a3ed09a2edfef2fc79fe06921f3e35203","_number":768058,"virtual_id_number":768058,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-23 18:16:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2020-12-23 10:23:31.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-12-21 09:48:20.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-12-23 10:50:28.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2020-12-23 10:23:31.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"}],"reviewers":{"REVIEWER":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-21 09:45:16.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"state":"REVIEWER"},{"updated":"2020-12-21 11:47:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-23 10:23:31.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-12-23 10:50:28.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"8772001336bda9d184eae075c20835cf00934331","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-21 09:45:16.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* docker/ceilometer/ceilometer-notification/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"5fe6cc55b619921a581da5a7ae77b14e3577ad03","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-21 09:46:54.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"1d361612e065918081cb482a97a4415ba7410d1c","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-21 09:48:12.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"6a57126b99c131640d4a97de5b1839846a7f2021","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-21 09:48:20.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"dcd7743add97cce701ecfb28302c8703af220e45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 11:47:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/6b4f0731d9f645db9f21a35ea11e4414 : SUCCESS in 37m 00s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/79f983cb11674845a8c01eb30d57057c : SUCCESS in 44m 36s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/cd58db20a1684eefb8d5c07c4bbd7651 : SUCCESS in 49m 59s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/06193f7b99a448bf9cb1e9713ff4f6be : SUCCESS in 52m 44s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/e92e12cf717c4aab8ece1f26c91cd53e : SUCCESS in 1h 35m 16s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f95b95e7aae546e5ad683457e50bd1a4 : SUCCESS in 1h 24m 51s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/c00b3cfcd8ab4edb84ae8b384fd64b5e : SUCCESS in 1h 15m 28s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/13688240994547968b691749602a5228 : SUCCESS in 53m 11s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/9feaaad921184da88503b0b2e3ec9c6f : SUCCESS in 1h 52m 54s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/b68dc405c0404116bbc636d06bf04760 : SUCCESS in 1h 07m 44s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a0c53d5d5b414c80bf617051a991331a : SUCCESS in 27m 13s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a9b56ce0f0b848cab42a9b2da916cdad : SUCCESS in 20m 35s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c497fc6a55ba4f32849497f4ef0210ad : SUCCESS in 1h 06m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a19a50f94e774708bff198fd5ca41e2d : SUCCESS in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a85751bbd8ca4a4db540abf17422e683 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ce9097efc2246b8a9e55a0d783dfde7 : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/75a088fe181e4858bfe80dc809cba8cf : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efe8a31f5d374aaf8932d7902ea1515b : SUCCESS in 5m 50s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/71c9e53d9047452ba39b84a6f211d179 : SUCCESS in 51m 48s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2bc27c3a918b4366b6ec188056097d3f : SUCCESS in 22m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7e583a2581934e8a9c9c22ab52cf7e27 : SUCCESS in 1h 12m 28s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2972500e7e7d48aab6247973bce2cca6 : SUCCESS in 46m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/be430467dfc34f4eb47aaba441378b61 : SUCCESS in 51m 34s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"37e75c7a019a67b8f5ccd2381129401a6bb2b858","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-12-23 10:23:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"eb9360f63c231ae34b180cf28cf2aedbc92570d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 10:23:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ad1f5d61fb98703bfbcce950210f67832cd2cef5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-23 10:50:28.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"dd78039714c5ea4f6150fbb54b6c3848abdb5c60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 11:40:47.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/fd5687ecdbba4414aafc7002a5eddf45 : FAILURE in 51m 23s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/0fb621cb4818499f945d6087d324283d : SUCCESS in 28m 21s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/e30d31f8fc0c4934be547fcfe3608934 : SUCCESS in 1h 09m 33s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/348643c1adea4751a6fe341ddbe3897b : SUCCESS in 55m 16s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/9c604614f6de409b8aecd66be017ff32 : SUCCESS in 1h 15m 10s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/670363145ddc461aa2b8f5e1defb5850 : SUCCESS in 44m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/352659a72d08406ebaccf94ccbd08448 : SUCCESS in 4m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/afa4b1a28a8241d0a5ede867ea11d1e4 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69ba1a10cd304fd59ba6e6335a8048cf : SUCCESS in 3m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ebeb67f857294081b6b81bab6303e29d : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/138c0192f8864fbc99e8a7325467d422 : SUCCESS in 5m 11s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bc44c0b2fa3443159963b3ee76404c8f : FAILURE in 37m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8c7968d9b7a44c7bb3b886f88f82e400 : SUCCESS in 49m 05s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3db1e6e5c1714cf4a2ff237b6e00e1b3 : SUCCESS in 1h 11m 32s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"ee103e220531948d3cb34a3a69a0022032f5759b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-23 15:34:53.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8d4a2541b0d21f21ceb9af05fe3335963f41568b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 16:54:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/1112f0f6295b48fb81afc72ca0bf2e2c : SUCCESS in 43m 18s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/dc0b337367144872beeacba9eb313901 : SUCCESS in 49m 30s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e939e8987d6641efaf53f5ed37a8d28d : SUCCESS in 31m 29s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ab6a321a79284c3b81f8a6460ab1f5fb : SUCCESS in 38m 40s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/544b3eec68a143d487cd0224a0f25448 : SUCCESS in 1h 10m 49s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/18582ee193664bbfa52001977de1fd1b : SUCCESS in 51m 20s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/5e5a15aec1a04250ab96cbe6d7a5afe6 : SUCCESS in 1h 05m 44s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/0cf9a8de0d01495cad87037d8d4b4cb6 : SUCCESS in 50m 39s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/2344dbf9aca0423f9c88dc4be26635e1 : SUCCESS in 1h 17m 37s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/f9e155f20412404eb2b1d230a0e287ba : SUCCESS in 42m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5bb57f31335749bcbb6bbc340aca1777 : SUCCESS in 20m 35s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6592b9e838a342e98b48bf82b937e2f8 : SUCCESS in 17m 13s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/24fb2fa899574428864f4f1ab8adedfd : SUCCESS in 51m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9853ce290d1416ba3f56d37550813f8 : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d72c0c256b0c45f5b0517c8021fde7a1 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1edfa4cf328b4fc291f5008c448f9ed3 : SUCCESS in 3m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d5ffe5304f2c4703be60b488ce3414a6 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dd0c2aab1ea41fab18137861b9cc836 : SUCCESS in 4m 51s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a9d66f9e79bd4df6b86fb9f16ebef969 : SUCCESS in 22m 21s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cc1aa2988cb5409c92d2d86103fe8c30 : SUCCESS in 23m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ad5f4cf7037a4dd98af9eef68bbeee73 : SUCCESS in 48m 38s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/85e24f1d36174a8cb0ee167dbf97aa6d : SUCCESS in 44m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7ffaa4246a7747799643ddf2f6fc5802 : SUCCESS in 58m 15s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"8c4acbad548bf646767fdfe72d92169b2346b157","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 16:55:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"82aa03604afebe8956a4c27b6994e77fcfc699e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 18:16:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1cc786e8024847f8a192f7316c8a6ade : SUCCESS in 56m 10s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a85a3349f985466a968a8e7d295d1cd6 : SUCCESS in 35m 17s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/29a2d9af36e64e7e96acd63a5690a247 : SUCCESS in 59m 07s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f6bb7a0beee94d118910459536ca7032 : SUCCESS in 58m 24s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e745518d2bcb425f97e519e5842dd851 : SUCCESS in 1h 20m 54s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/0ec2a36cbad34ccc90f13f137f1ead6c : SUCCESS in 48m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f212594b24b04407bc2e994063d48e56 : SUCCESS in 5m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ecde0f2ae91b4cfbbb8805ca93f758ec : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2bd1c398c90149b688c8954cc0ed4491 : SUCCESS in 3m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/32722a9f24814b7497b783e3f70dcc09 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9c8e6d58b364736b31f895dc0fbeeaf : SUCCESS in 7m 12s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7afd08a319b2462298a9508a35f232f0 : SUCCESS in 28m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/305052a581464358b07a86f11d7b7e93 : SUCCESS in 49m 15s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6ec510d468fa46d58607d18104bcca70 : SUCCESS in 59m 20s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"aa07276872fc920bea683bd805ae0ecf6477074e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 18:16:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"8555e42a3ed09a2edfef2fc79fe06921f3e35203","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 18:18:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6bbfc70bafc4473a9c6826a5ea0d3eb : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1d8af466f4c894d6952828e3ee95509fd271daf1","revisions":{"0b67a07537690dc7c6dcef802868e21bb7ff7f36":{"kind":"REWORK","_number":1,"created":"2020-12-21 09:45:16.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/58/768058/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/768058/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/768058/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/768058/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/768058/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/768058/1"}}},"commit":{"parents":[{"commit":"10e4503e7d2363a97068a74dbdf9d4ac4ec805f1","subject":"Merge \"Add daxio for nvdimm scenario\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/10e4503e7d2363a97068a74dbdf9d4ac4ec805f1"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2020-11-27 10:35:51.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-21 09:45:16.000000000","tz":0},"subject":"ceilometer: do not fail when using an unsupported database type","message":"ceilometer: do not fail when using an unsupported database type\n\nThis is necessary if e.g. Prometheus is used. Then the\nceilometer-upgrade is not needed.\n\nSigned-off-by: Christian Berendt \u003cberendt@betacloud-solutions.de\u003e\nChange-Id: I8690e5543e6b5f37f41fe35f659910fc869c034f\n(cherry picked from commit ea8592b077ceccfeb3089d901e6f46252f024881)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b67a07537690dc7c6dcef802868e21bb7ff7f36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b67a07537690dc7c6dcef802868e21bb7ff7f36"}]},"branch":"refs/heads/stable/train"},"1d8af466f4c894d6952828e3ee95509fd271daf1":{"kind":"REWORK","_number":2,"created":"2020-12-21 09:46:54.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/58/768058/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/768058/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/768058/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/768058/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/768058/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/768058/2"}}},"commit":{"parents":[{"commit":"10e4503e7d2363a97068a74dbdf9d4ac4ec805f1","subject":"Merge \"Add daxio for nvdimm scenario\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/10e4503e7d2363a97068a74dbdf9d4ac4ec805f1"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2020-11-27 10:35:51.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-21 09:46:51.000000000","tz":0},"subject":"ceilometer: do not fail when using an unsupported database type","message":"ceilometer: do not fail when using an unsupported database type\n\nThis is necessary if e.g. Prometheus is used. Then the\nceilometer-upgrade is not needed.\n\nSigned-off-by: Christian Berendt \u003cberendt@betacloud-solutions.de\u003e\nChange-Id: I8690e5543e6b5f37f41fe35f659910fc869c034f\n(cherry picked from commit ea8592b077ceccfeb3089d901e6f46252f024881)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d8af466f4c894d6952828e3ee95509fd271daf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d8af466f4c894d6952828e3ee95509fd271daf1"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
