)]}'
{"id":"openstack%2Fkolla~768067","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I67f93687d70f1c8c92719857cbf74971d9a17ae6","project":"openstack/kolla","branch":"stable/train","hashtags":[],"change_id":"I67f93687d70f1c8c92719857cbf74971d9a17ae6","subject":"Fix build of monasca-grafana","status":"MERGED","created":"2020-12-21 16:43:39.000000000","updated":"2020-12-23 22:03:49.000000000","submitted":"2020-12-23 22:02:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"768067","cherry_pick_of_change":768066,"cherry_pick_of_patch_set":1,"meta_rev_id":"21621792e1e9d35ced3d96d6aa3d5644dd4ef61b","_number":768067,"virtual_id_number":768067,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-23 22:02:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2020-12-21 16:59:13.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-12-21 16:43:53.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2020-12-21 16:59:13.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-12-21 16:43:53.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-21 16:43:39.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2020-12-21 16:59:13.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-12-21 18:14:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2d3f865120f26c41af4b2198465cc39e79b9603e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-21 16:43:39.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"9854b0ebe8ac39c1051ce816b829cddf5828165b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-21 16:43:53.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"d194c894eea6e3c50a61d087398858f96f727927","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-12-21 16:59:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bea36dd11718130a658b4e1cb7f7ff555e25cf2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 18:14:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/196189b47a434b85be1c20ae93f8388e : SUCCESS in 44m 43s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2bf2b4acbc984e2eb354429b9dd74c36 : SUCCESS in 43m 35s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/682401f302504c21a7d2df972a9beeff : SUCCESS in 25m 41s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/78fa536ca1bb4ac8a49ef5cbf8fbe146 : SUCCESS in 34m 01s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/57a30b19a07546b7b54139c95d1583c9 : SUCCESS in 1h 25m 54s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/83c1630e3cd949cf9922ba3d0f230ffd : SUCCESS in 1h 02m 54s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/c2daaa6146674f9f8633c3ba15d0db13 : SUCCESS in 1h 05m 17s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/ff62d4c1385541c483df80205cd0f77f : SUCCESS in 1h 01m 57s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b75d1d5262fe405abb29d13048bbb521 : SUCCESS in 1h 18m 33s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/31e9dd93d3b34fc5a3526369f2b81b1e : SUCCESS in 1h 03m 10s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3f70dd782ee14fc39724dc94bf02e3cd : SUCCESS in 18m 35s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/959648da721d4f14a33194233efd23aa : SUCCESS in 17m 32s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8fedddb3962c4f1ab0e38a7462e19618 : SUCCESS in 54m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca54c1cc0ebb4dc78a4f35888325b4c2 : SUCCESS in 4m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9f2d68cee3a34b0eac9e7c94a4af9d8a : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08fb402fb5714b849874a95d6f8d87e3 : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8d9c839b98c744deb75c15a058e7d4dd : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63460b2121cf425b8e3d6a5dc624cd62 : SUCCESS in 5m 32s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c87dd6d4fb5c4cc687e1fc57e6f33bec : SUCCESS in 30m 21s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9f5fe4ea5dec499584acae20537cfb51 : SUCCESS in 29m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3c833f0f87df4a95b8ec63be262d80e1 : SUCCESS in 53m 41s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9d9c2370fcdc49c3801541689958e703 : SUCCESS in 50m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/be8725e8bda44c62916308c5789f4049 : SUCCESS in 1h 09m 40s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"d140d5c934e23dfce16ba86af15915f42ad5743d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 18:14:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c621ac3c2c968b13927dedd6ecb5468ba8309e89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 19:58:11.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/954c6bb2c8374e4eb7b95a315c93b087 : FAILURE in 42m 52s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/edfe50a23aea423c81e609c0ea561758 : SUCCESS in 30m 07s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/431eec6d5ce24a048fb222b309df78ee : SUCCESS in 1h 07m 08s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/b0c73768241747d59d6600f6a508cd63 : SUCCESS in 1h 02m 48s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/3a516d79240246c0a35127edbefdd6e4 : SUCCESS in 1h 16m 30s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/141aa863ed6443559060a93330c6be7c : SUCCESS in 49m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65dd80a265e04f8f902e646ad15f43e4 : SUCCESS in 3m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c4fda9ad00d64968a809a51ce7b8b25b : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c81abd7dfd364a5992187bd9d8556ea0 : SUCCESS in 3m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7ce43add8190422abcf158c866fddf3e : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/325feddf3baa4da6910111d957c8c2ab : SUCCESS in 6m 38s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e752985ad3ab4ed5aa26ea8f0ff9bc9a : FAILURE in 22m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8d27ecb5ffb64debaba318e00e924531 : SUCCESS in 50m 02s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/63071990408d43e08e9ca79fb25c456a : SUCCESS in 1h 02m 40s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"c199ff20f2d05a710ca63bd8af214f72e4650f29","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-21 20:13:34.000000000","message":"Patch Set 1:\n\nINFO:kolla.common.utils.helm-repository:\u001b[91mError: Looks like \"https://kubernetes-charts.storage.googleapis.com\" is not a valid chart repository or cannot be reached: error unmarshaling JSON: json: cannot unmarshal string into Go value of type repo.IndexFile\n\nheh","accounts_in_message":[],"_revision_number":1},{"id":"16c9dc0b7664d819c9c528c05cf6e13ca5c47716","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-21 20:13:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6f8e1f72a0d2b5da91c12478bbe084c518967326","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 22:20:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/1eee689744bd42f2b34abaf93f71bbc4 : FAILURE in 36m 57s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/500f998c52eb40c79a9e07df6c43114e : FAILURE in 40m 18s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b6b31ed9d656403493b7c39f6836c24f : SUCCESS in 22m 28s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/63eda7811cbb4652ae27c00415f5b127 : SUCCESS in 33m 39s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/894ef91ab2694ea18ee9eea7cb1a1856 : SUCCESS in 1h 04m 11s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/3a0ded3a808041738f75e4e317f93635 : SUCCESS in 55m 48s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/976281445cb843f98ab858e4fa8ddf79 : SUCCESS in 1h 11m 57s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/ca756032b91348eebbd79c2971a3a172 : TIMED_OUT in 2h 01m 23s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/fbb903db1ca14a7e97b13d49946051ce : SUCCESS in 1h 19m 05s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/92496f95c1f24dbc9c12d04963b70617 : SUCCESS in 39m 40s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/82d0bb752b4145fd893a924eccde7f02 : FAILURE in 21m 21s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e269740e644840f0a98e56dd6cc7678b : FAILURE in 19m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f4b1ba4c4eb9420099fcc108766b84da : SUCCESS in 43m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c21e8f0367354fe9ab1d314d21e60d09 : SUCCESS in 4m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4cecd5d835ab44cbad1bce8fc827d6be : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba43775dc55745648029ea4d5a0b5829 : SUCCESS in 3m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e06a0c98e4114b64a85eb0d6707fbdfb : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df7f02ef6e9042168d9d1b88ca88fccd : SUCCESS in 5m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2aa47c6b453b4c39a9a34ff939c9ce44 : FAILURE in 24m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/61c74dfe8d654827952973bfe16a999b : FAILURE in 22m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f4750cb3a72e4854a9c8ddf52cf93a11 : SUCCESS in 39m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/51d589f62fc645779c3cbfbfc3a0e6e4 : SUCCESS in 46m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/032ce202c01941b391b389e24423c166 : SUCCESS in 2h 04m 04s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"1176bd6455c3919e3ed974e136844ced46d83608","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-23 17:38:06.000000000","message":"Patch Set 1:\n\nhmm, this does not seem necessary for ubuntu bionic, oh well","accounts_in_message":[],"_revision_number":1},{"id":"255f535d9699cfb06619117173017c5fbc9d9e45","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-23 17:38:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3cab5176e408e7b17ce2944b368df0d8c6ad9491","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 20:50:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/d5167f1f998044539867917a9fd2cc85 : TIMED_OUT in 3h 07m 16s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/6d893fffa58e459197cc2ed75cdb7397 : SUCCESS in 1h 09m 26s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/1a8c2868d844439cafd71dde0ecc9756 : SUCCESS in 51m 58s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/39eaeb3f716b4ea29e96d5f262a6a7da : SUCCESS in 29m 01s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/10f81ca36bb14e5d822bad39673c524e : SUCCESS in 1h 08m 36s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f00532ba79c34ca9a1af4dcec91cf69e : SUCCESS in 49m 26s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/96cfe52e7785488e9cdb1a1c2fee0b29 : SUCCESS in 1h 12m 45s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/3f4de36545c1491c975869da0e0d9035 : SUCCESS in 51m 06s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c76bce4833a24f8189b6b38e418ae881 : SUCCESS in 1h 12m 02s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/5c3b76997ab646679da5364f286dbafd : SUCCESS in 47m 00s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b6eb629e4551453088594bb5e2aa3e90 : SUCCESS in 21m 02s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d90d31b68835413bbebeee467d446c89 : SUCCESS in 18m 56s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/140d0c9f22954252932bbe6779503d5c : SUCCESS in 51m 25s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/337d3c01d5354fe2865a96ddbbfc7166 : SUCCESS in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5da4754adeae46a4803e74b15c416ad3 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebc5383dcfe94cffb5e22969cf4e6239 : SUCCESS in 3m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b90e6c7f77a846a0bca1b5cb2913a704 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22696e66352542b1b2a61efeee9efa77 : SUCCESS in 7m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/312ef889b67c4e019530e3ee23477d21 : SUCCESS in 24m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a11832ae7f2d4b279bfcb36df741747f : SUCCESS in 23m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/932d7b6891fd4adb8d99fe319b56f7f8 : SUCCESS in 43m 56s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/37984f2710ed4d8c9cc81347fe615d7f : SUCCESS in 49m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4bb2713007a740c787f4408596363c6c : SUCCESS in 49m 57s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"0de2d22367b4845ee7926ffb0cd90fc1fc2bd0c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 20:50:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b648c2db7ae1d9dea2edb3c896823519cfb16c98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 22:02:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a37d13602bed49a9af8890c4fc2468c8 : SUCCESS in 46m 55s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/576ad7f98401472fb26bc6691233cf02 : SUCCESS in 33m 43s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/753882b8e1f646fa836d3a2684e11ff6 : SUCCESS in 1h 02m 07s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8c456765c913429d82f2c330d0ce6c43 : SUCCESS in 59m 06s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c443b13e8b554d41b76d45c206b42c8c : SUCCESS in 1h 11m 03s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/0b922b21d08c46af8a8a28fbc95535d7 : SUCCESS in 51m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03deb4091dce41399e15c997de6ba8ec : SUCCESS in 3m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/41301bafd25d4978b12e6144c67187ee : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/363ae16b0a4f4f10846daf8dbb84489d : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/382b655bfdfd41e1a8deac216d7977cc : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1077e345a38491092d7333d4d947a4b : SUCCESS in 5m 41s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e3bf15480e4c4e1993dbbd4ffc860750 : SUCCESS in 23m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d3bcc0cb66f24184a5a67777b13a803f : SUCCESS in 1h 01m 30s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0ebe8d743be74e83954b18ee10042f8c : SUCCESS in 59m 52s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"7da6cabc9b0f12cd51e0dfd6ca2980a07fb5e9e7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 22:02:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"21621792e1e9d35ced3d96d6aa3d5644dd4ef61b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 22:03:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77698d6875b14597ba1b4d6c04daf896 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dfdfe86e2e487561ba935214c3f990267f739e9d","revisions":{"dfdfe86e2e487561ba935214c3f990267f739e9d":{"kind":"REWORK","_number":1,"created":"2020-12-21 16:43:39.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/67/768067/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/768067/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/768067/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/768067/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/768067/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/768067/1"}}},"commit":{"parents":[{"commit":"10e4503e7d2363a97068a74dbdf9d4ac4ec805f1","subject":"Merge \"Add daxio for nvdimm scenario\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/10e4503e7d2363a97068a74dbdf9d4ac4ec805f1"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2020-12-20 16:18:05.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-12-21 16:43:39.000000000","tz":0},"subject":"Fix build of monasca-grafana","message":"Fix build of monasca-grafana\n\nThis patch is adding libffi-dev to build_packages\nfor Debian/Ubuntu which is fixing monasca-grafana build.\n\nChange-Id: I67f93687d70f1c8c92719857cbf74971d9a17ae6\n(cherry picked from commit 88d48953357c67de14ff83f0361825b05fc71a07)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dfdfe86e2e487561ba935214c3f990267f739e9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dfdfe86e2e487561ba935214c3f990267f739e9d"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
