)]}'
{"id":"openstack%2Fkolla~768157","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I67081a37977dd6de37a67fcd012e002f2e46743f","project":"openstack/kolla","branch":"stable/ussuri","topic":"bug/1906635-stable/ussuri","hashtags":[],"change_id":"I67081a37977dd6de37a67fcd012e002f2e46743f","subject":"Install iptables in ironic-inspector container","status":"MERGED","created":"2020-12-22 08:43:20.000000000","updated":"2020-12-23 09:58:35.000000000","submitted":"2020-12-23 09:57:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"768157-bug/1906635-stable/ussuri","cherry_pick_of_change":765146,"cherry_pick_of_patch_set":5,"meta_rev_id":"e5516483046c5c303de344fa3a407f6e0f0c655f","_number":768157,"virtual_id_number":768157,"owner":{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-23 09:57:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-12-22 09:33:10.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-12-23 08:48:20.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-12-23 08:48:20.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-22 09:33:10.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-12-22 10:04:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-22 11:53:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-23 08:48:20.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"acdacccb5822882dc3ea8ced5a1e217e172be8be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"},"date":"2020-12-22 08:43:20.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5d0bf7036b752fc180334a3377f9ff9a51ff21a9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-22 09:33:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"90fa58d376712634fba189bd2ef1cda31ce5c3e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 10:04:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/2745327ed8284cc48818a20e54dd50b4 : SUCCESS in 1h 13m 08s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b6429b2c8c7d42978b3d7613109c77b2 : FAILURE in 1h 08m 04s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/94ba8c793524491a96a47686f35a0c3d : FAILURE in 40m 24s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/5d63a89c221e4dee88af05dbfefe3e00 : FAILURE in 1h 12m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b1de1166b31127bdbf61fddcd0c48fe974643798","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 11:53:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/5c9d24d80e2442edbacec435bf44aec6 : SUCCESS in 46m 41s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a45c9a04ff204586a3256113ef0c91b7 : TIMED_OUT in 3h 04m 09s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0c3bd2b540c54b5da8498510512fa480 : SUCCESS in 1h 11m 42s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/f8fa579b87c44f2da70d19520e7a00da : SUCCESS in 1h 16m 42s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/7c1c4c10726141d7aceb3dac8a34b67f : SUCCESS in 1h 08m 19s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/518cb5948b7e4ee48b8d059b348ee162 : SUCCESS in 18m 21s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/379cf8b532ff4e928eb001341d5f5d6b : SUCCESS in 24m 49s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4fd435ed997d4cb98622e89c8776e9fb : SUCCESS in 17m 23s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a9db23792b434ed6a1ba9077ee9207a9 : SUCCESS in 54m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bc9a169ffe443ef96311c3de315010c : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/314e119227fd479190903f37a5335845 : SUCCESS in 3m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6a40c476ee494dd384c48a9ab5fa9b19 : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bff6ace55c054a95b01a9119bf777531 : SUCCESS in 5m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ffad69818df4423d9d7b5d9f946638c3 : SUCCESS in 32m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/de298fd3fabe482580c69a50166fa1b9 : SUCCESS in 23m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0231298f673043bdb9ed80b9a17d0bf8 : SUCCESS in 55m 24s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cde738a7f1ef4453bebc2227233622ec : SUCCESS in 51m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4568aa64658346bf86e908a265d2b9f6 : SUCCESS in 1h 00m 01s","accounts_in_message":[],"_revision_number":1},{"id":"dac9091551e7aa01ff03406f7f82ef391a21daa1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-22 12:16:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"720645e7799a889fc23ab04732d007ac938bb002","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 13:05:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/a6d8f503de3c442186d3f2e6280f1048 : SUCCESS in 48m 00s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/db16368df97347c6b2f91342640651f3 : FAILURE in 42m 37s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/747d3fe8619d48c8a5e587900629b73f : FAILURE in 29m 00s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3d040a951cba43b69121b4c8d43c8d8f : FAILURE in 45m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"90b8bc7f716b75768cfd57bf8a6a14082c686cf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 13:41:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/550740cd9ba54baebe039c0c61df8c44 : SUCCESS in 23m 23s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/240ff2e9506e4733b492231516732144 : SUCCESS in 25m 27s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d787a49b02284cd8b20e244b7fae57ed : SUCCESS in 56m 40s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/f370d7c9990d43b295f95ea3e5250209 : SUCCESS in 1h 04m 35s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/ff781449774c45278c9140940e701231 : SUCCESS in 1h 04m 24s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/38c42c09611f4ca18bc013eda47fdcff : SUCCESS in 17m 11s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/09248ceea5054a7fbc54e8409261b7e1 : SUCCESS in 21m 12s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f1a21998c7904d5194b5db78176fac5b : SUCCESS in 16m 27s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/52cbbed151b64e389f1eb2e99259da2f : SUCCESS in 53m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f47306a4e6874446bc42a3833af9ca81 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f955a15c2fd4381af8860cd4ad70313 : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a519b448d07a463787bb9ee2479cd6f2 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8da7c4d79b248fda4310729ad1692c2 : SUCCESS in 5m 25s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/80ca4b41286d466eac47e08f8066d850 : SUCCESS in 40m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/aae504f51425492c80eaa0423405d622 : SUCCESS in 30m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4d8ffa9003214331bc8ed2f562663139 : SUCCESS in 52m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/67f82a116986460cb2d2779c3788f3e8 : SUCCESS in 53m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2052483f74e542aa9092ad468b1aa0cd : SUCCESS in 1h 18m 04s","accounts_in_message":[],"_revision_number":1},{"id":"f2e95cd8f8bbb4170d0be8f604f5d5868f2bef9a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-23 08:48:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a9a16ed5a2c5162e4a60a42aa69aed4c8d952c94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 08:48:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2585c36fd95d0cba9fbded827673e667830e6e89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 09:57:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/153551740646468aaec0b507c0eef252 : SUCCESS in 29m 02s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7a20e5aa6f954b10b94fcddd968f7f5b : SUCCESS in 59m 17s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/61ab346070a64250a7a523e632a2f93f : SUCCESS in 1h 03m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79cad4bbd6184af2a54b0b757e018fb0 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c83b0db07774fc78a4a37b732018f20 : SUCCESS in 3m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/138a10fe5746467aa78335b954a4be5f : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16ff333353074393936e8453895ccd0f : SUCCESS in 5m 20s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/92097d0e74b24b5c9befd49c018152db : SUCCESS in 24m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c5e7db94472749c8a035ccc984def5f8 : SUCCESS in 1h 02m 50s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2cef3a0e43844497ab51a3c76ce40b93 : SUCCESS in 1h 03m 17s","accounts_in_message":[],"_revision_number":1},{"id":"832d976e4df50a9a6fb724d4cee31aad34972417","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 09:57:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e5516483046c5c303de344fa3a407f6e0f0c655f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 09:58:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd70c64dd32746c092acd88727f2356c : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8d6f21a2089e26ce62a9da8a63fbd2abc159917a","revisions":{"8d6f21a2089e26ce62a9da8a63fbd2abc159917a":{"kind":"REWORK","_number":1,"created":"2020-12-22 08:43:20.000000000","uploader":{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"},"ref":"refs/changes/57/768157/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/768157/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/768157/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/768157/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/768157/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/768157/1"}}},"commit":{"parents":[{"commit":"67e5c68f010f6f7f9ab042b0da36b4f77b385a6e","subject":"Fix build of monasca-grafana","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/67e5c68f010f6f7f9ab042b0da36b4f77b385a6e"}]}],"author":{"name":"douyali","email":"douyali@troila.com","date":"2020-12-02 14:17:48.000000000","tz":0},"committer":{"name":"dou","email":"douyali@troila.com","date":"2020-12-22 08:43:20.000000000","tz":0},"subject":"Install iptables in ironic-inspector container","message":"Install iptables in ironic-inspector container\n\nThe centos environment ironic-inspector use iptables filter need\ninstall iptables rpm package.\n\nChange-Id: I67081a37977dd6de37a67fcd012e002f2e46743f\nCloses-bug: #1906635\n(cherry picked from commit ad1b8e20dfc6f743f1945b9e036cf29597b79053)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8d6f21a2089e26ce62a9da8a63fbd2abc159917a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8d6f21a2089e26ce62a9da8a63fbd2abc159917a"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
