)]}'
{"id":"openstack%2Fkolla~768158","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I67081a37977dd6de37a67fcd012e002f2e46743f","project":"openstack/kolla","branch":"stable/train","topic":"bug/1906635-stable/train","hashtags":[],"change_id":"I67081a37977dd6de37a67fcd012e002f2e46743f","subject":"Install iptables in ironic-inspector container","status":"MERGED","created":"2020-12-22 08:44:08.000000000","updated":"2020-12-23 20:38:07.000000000","submitted":"2020-12-23 20:36:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"768158-bug/1906635-stable/train","cherry_pick_of_change":765146,"cherry_pick_of_patch_set":5,"meta_rev_id":"93c5ca2075c727732e0fc13dc6a47b5b95aa0505","_number":768158,"virtual_id_number":768158,"owner":{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-23 20:36:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-12-22 09:33:16.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-12-23 17:39:29.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-12-23 17:39:29.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-22 09:33:16.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-12-22 10:14:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-23 10:50:49.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2020-12-23 17:39:29.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"49d8ff52f92e7f5a12341e0b0aa6ec721c24c2c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"},"date":"2020-12-22 08:44:08.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"847ec2a30ac5272e5729416b24025070f18a093c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-12-22 09:33:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fc91e8f8f64a04894b45b85aba959092259b7951","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 10:14:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/05c05c5a1eb7447e83c20f6605b113ec : FAILURE in 41m 57s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/11d9c4c55af64fb291eec373a739016f : FAILURE in 48m 19s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/7019763339a340488b65fa1f55c95092 : SUCCESS in 20m 39s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/76eb9f1c53ab477299d5b2b5823b0498 : SUCCESS in 32m 05s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/82f530d915b7412cb990fdebfd3baa3c : SUCCESS in 1h 24m 51s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f68280836b0d469997fd966a3e8b289f : SUCCESS in 53m 25s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d3b569856eaa4b4c8378342fa29f2bc0 : SUCCESS in 1h 06m 47s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/afcbd043e9e443c3ac106e4f8936618a : SUCCESS in 54m 00s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/450e169ff8bd4bdb9e428ef09e0ecf91 : SUCCESS in 1h 12m 47s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/2f8478bfae6249338d13d41e325e426f : SUCCESS in 43m 42s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/11226b79a7b2486c81b8169c26f421cd : FAILURE in 18m 13s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2afe4b912c314d96bb9bec5960c9abbd : FAILURE in 16m 36s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7889d73b208c4fb99471134bb6993ab4 : SUCCESS in 49m 21s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9169d5cf7134975a0519fbecceb9d22 : SUCCESS in 4m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9de508c4362749acb6d639c610791038 : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5c7106d1df941deb7553f170c223f92 : SUCCESS in 3m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a1724958bb64216878c3d0dbfa797e9 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32c275ee45c54883b2851e376a0634ff : SUCCESS in 5m 15s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f31dcc55d4e94d9ca03e9e5091bce5ce : FAILURE in 20m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/13a05ca20ccc4ea28918202ae7345760 : FAILURE in 27m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ee00620bbf3d453b8c87152cf0bddf2e : FAILURE in 36m 11s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b3994315b9534b01a2b41d8f1c5ebe80 : SUCCESS in 48m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ca7069e6d1884e66bbbbf98175351cca : SUCCESS in 1h 04m 07s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"956e9653a773262b383a95c92f5bae828d7ae172","author":{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"},"date":"2020-12-22 12:47:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7ce6c5bd361c4cb33e1798c7f45abd81ea20b4b0","author":{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"},"date":"2020-12-22 13:57:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"46395283535cb250cc36f3003c117bc2e9dd1eb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 14:08:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/2831a29261eb4fbf9f752f5983d58d7e : FAILURE in 34m 28s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/312fc633dfbc48d5b41ef677dd3a554e : FAILURE in 44m 01s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/ef151255bca04910bc21429bd24ae4ea : SUCCESS in 21m 01s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/28882af89e1f43649d1dd4b06db83afe : SUCCESS in 29m 48s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/392d1f76baa943c18a3834d006799c2a : SUCCESS in 1h 06m 36s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c8f8d81fbbf9464c892da8c52a85d7f9 : SUCCESS in 1h 00m 36s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/80616eaf1a6848d38e275c6162dfb475 : SUCCESS in 1h 07m 14s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/9d9b3cdfba4041df918bca6b0a8ac983 : SUCCESS in 1h 01m 38s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b8f1321af2d0466890c67a1e5c19756e : SUCCESS in 1h 19m 58s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/59aeebe9e29d49a2bcdeecdfcb263d5a : SUCCESS in 51m 57s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d26c31b120014251b5f09d32ddf5ac5d : FAILURE in 25m 32s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6c17a122991946fda00d7366d43a076b : FAILURE in 19m 35s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/57ef1f59e2a1437aac187062ca95c93a : SUCCESS in 44m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25bec22846434e669250dc9fb2da0cf6 : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b760e85d39b44403b0e2ba1dae2cb9c0 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94dd0d551940461893b0416a846282be : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0d5d13fbdea5471a928609e39bdd265a : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c91edfa378e4445b64fcd184ade1b0f : SUCCESS in 5m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4f16f411dd1c4391af8ec8b475ea0478 : FAILURE in 36m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/59249aa8b6de45cfa0ad9efb871d58ba : FAILURE in 23m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f292ad0e4766451b9b2e891acc120120 : SUCCESS in 49m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c568534720d9487bbbcaea4f4e904fc0 : SUCCESS in 46m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/db2bcdb21ce2458d95e7d79e197608c3 : SUCCESS in 1h 02m 13s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"37098eff9175e670162a9013b09d18166ed784ca","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-23 17:39:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"03f250ef9b0df38aa0378c27f3a478aff40f9f6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 19:08:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/7c18696ca0904df2a4f2e72c1046e9af : SUCCESS in 1h 09m 40s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/869913c805b34207b8d90c8a10582785 : SUCCESS in 1h 05m 35s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/ee0d3bf5e2144f16be91b974736e01a1 : SUCCESS in 49m 58s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8ccb3fc9a9de4cac866f37aca332a965 : SUCCESS in 34m 42s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/07604920551542c9a2fa328821ebb9ba : SUCCESS in 1h 23m 30s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/6e2c6449cbb04847a1ff585809ee8c21 : SUCCESS in 1h 08m 25s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/6eb8253f76d74d96bd800689a8e746d4 : SUCCESS in 1h 11m 20s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/223e6dc288a448a490ac2271e66aaf3b : SUCCESS in 51m 20s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d4a574b2f3ca458eac5257ba11867162 : SUCCESS in 1h 13m 12s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/db0ad7f6ba6f455b9ada36bcd0e92eaf : SUCCESS in 55m 42s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/09f18bf6cf194895b3ef789137307026 : SUCCESS in 17m 05s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3c8e758778aa4a23af042e2880a2d896 : SUCCESS in 22m 47s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/59b9bcb377fb4cdabf0817433ad0708e : SUCCESS in 51m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5459a78dbc644799be3e4e95f94fcf78 : SUCCESS in 5m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2603218c465042edb97adb33d0ee2f4e : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8b917745c9e4fe9a3b44cb1596cefed : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f96d3e25037647879f18bfbbd3cc6506 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/942bbabece0e4a3ebcd8848b9d47481a : SUCCESS in 5m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9f950c67a61e4e2bb3b798743c6affc8 : SUCCESS in 20m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/78293724f4a6483089f50f1f1320f3d1 : SUCCESS in 36m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4c2aa2fc7f004857bbd1c34669a633ef : SUCCESS in 46m 24s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/318473785d4b4a938e8bea55070f36f1 : SUCCESS in 1h 02m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9fcd75ff5013452785962a5d71d6e7cb : SUCCESS in 54m 19s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"a8457aa09bdded46a96acdfc8f6e7e2966ff2764","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 19:09:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"96c258d773883260ab0c5cb507a6187bc38170d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 20:36:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/18b3392f9041465c82a337d1cde36b94 : SUCCESS in 42m 24s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3b7bce43d48b4827b92ae0d51fdc1498 : SUCCESS in 42m 00s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/6a6e21d9d82042f6b40d95e297cdc905 : SUCCESS in 1h 22m 10s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/401796ff7b3040acb0d9d0ce58c939f1 : SUCCESS in 54m 51s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/3f57b421be13490ea5dff60045ba6b64 : SUCCESS in 1h 12m 16s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/3b3f59c100d64e32827a5c04f9fbc2cb : SUCCESS in 52m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fba6dc769fca4fafbc5ab25c900c2eef : SUCCESS in 5m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/66c00004fa86402a8aad4efc53462900 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07ebc097f5134ae694ad39614a6c9eaa : SUCCESS in 4m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5dabdc81b3214d59854488fb147bf0ac : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30cf8aa03a3f4037acf39e17888a6724 : SUCCESS in 5m 23s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b2ad0cf31ce74665aa6bf5421e098a5c : SUCCESS in 30m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/acaed850b6224c0ab21430bbf56ad458 : SUCCESS in 1h 06m 10s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/746be4a19d3d442a9aee5f55033cd5fa : SUCCESS in 1h 05m 42s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"36516e9e4069891e71237c474089c796e6b6b678","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 20:36:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"93c5ca2075c727732e0fc13dc6a47b5b95aa0505","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 20:38:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/914b33350e8f4ff79cc0fe36d2e0faee : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d43c46def644e9bf32ca5c1ceaa0cb077ae3841c","revisions":{"d43c46def644e9bf32ca5c1ceaa0cb077ae3841c":{"kind":"REWORK","_number":1,"created":"2020-12-22 08:44:08.000000000","uploader":{"_account_id":31310,"name":"dou","email":"douyali@troila.com","username":"douyali"},"ref":"refs/changes/58/768158/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/768158/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/768158/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/768158/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/768158/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/768158/1"}}},"commit":{"parents":[{"commit":"10e4503e7d2363a97068a74dbdf9d4ac4ec805f1","subject":"Merge \"Add daxio for nvdimm scenario\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/10e4503e7d2363a97068a74dbdf9d4ac4ec805f1"}]}],"author":{"name":"douyali","email":"douyali@troila.com","date":"2020-12-02 14:17:48.000000000","tz":0},"committer":{"name":"dou","email":"douyali@troila.com","date":"2020-12-22 08:44:08.000000000","tz":0},"subject":"Install iptables in ironic-inspector container","message":"Install iptables in ironic-inspector container\n\nThe centos environment ironic-inspector use iptables filter need\ninstall iptables rpm package.\n\nChange-Id: I67081a37977dd6de37a67fcd012e002f2e46743f\nCloses-bug: #1906635\n(cherry picked from commit ad1b8e20dfc6f743f1945b9e036cf29597b79053)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d43c46def644e9bf32ca5c1ceaa0cb077ae3841c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d43c46def644e9bf32ca5c1ceaa0cb077ae3841c"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
