)]}'
{"id":"openstack%2Fkolla~769117","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I88e5bcf35207670b8611aeee3072215a7d7eb6e5","project":"openstack/kolla","branch":"stable/ussuri","hashtags":[],"change_id":"I88e5bcf35207670b8611aeee3072215a7d7eb6e5","subject":"Fix Freezer \u0026 Cyborg API startup on CentOS","status":"MERGED","created":"2021-01-04 09:59:42.000000000","updated":"2021-01-04 16:10:19.000000000","submitted":"2021-01-04 16:08:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"769117","meta_rev_id":"72dcc42d4e1e0aef29a0fd6cadbd8441e6cbf5bc","_number":769117,"virtual_id_number":769117,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-04 16:08:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-01-04 14:54:45.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-01-04 14:18:19.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2021-01-04 14:54:45.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-04 09:59:42.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-01-04 10:51:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-04 11:22:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-04 14:54:45.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"40a5c8000f011f396176000b9a93216e041be2f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-04 09:59:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49030c843b16d49b70eb8d518ccd30850a941713","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 10:51:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/2168529746a040d49cdb21d5b173c74b : SUCCESS in 51m 25s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8c41d023bb8c4cfcb663e402ba41111c : FAILURE in 45m 41s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9d9fbe7705fc4c3ab003a5a26b5aa026 : FAILURE in 38m 42s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/14149ab055874caf9e4d51ad6a8f1eac : FAILURE in 47m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4602f5f922092f77bdb023486bbec2227b1715d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 11:22:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/7200e41238aa4a4889fc9c2e55934cf1 : SUCCESS in 28m 21s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f28fb5bf8b7449b695394d9061fc8d96 : SUCCESS in 29m 17s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7004f21093d54fa6bc97dee2900b5dcd : SUCCESS in 58m 35s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/6b1bf22c61644459be77f8688ecfb010 : SUCCESS in 1h 01m 55s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/3e6b180b161e48ebb7a43ee107b8e3b5 : SUCCESS in 1h 04m 34s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/49c75ec053b142e4afc857882e04eb44 : SUCCESS in 13m 05s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3a103d62749d4397ac171375dfa5ced3 : SUCCESS in 17m 56s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e273632dfd604641b00d819ac1baead7 : SUCCESS in 15m 58s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5edea8fe0aac4ac59c2ae34e59f274c2 : SUCCESS in 50m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19099ae256ba4f71bb8e199d1a573243 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77f579106f0f4e3da131644129f9fdc8 : SUCCESS in 3m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ed6e655f44da45f583a01e96f8c6649f : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/814c49d462af44d081ee138296f49629 : SUCCESS in 7m 02s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/484a527e9f444b41a3c4142fc8519bdb : SUCCESS in 25m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/49c6bbafab19468ba7a5f935a52243a7 : SUCCESS in 33m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/210cf39dd01140808bd2564072ec7912 : SUCCESS in 59m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/533a04c39f934524871765879b14ff27 : SUCCESS in 1h 16m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/888ec385f74a4958a3cc423772905bf5 : SUCCESS in 1h 05m 27s","accounts_in_message":[],"_revision_number":1},{"id":"4e2e58d35407b0c0ddd52876ff5085e927590dd3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-04 14:18:19.000000000","message":"Patch Set 1: Code-Review+2\n\nthanks, Mark","accounts_in_message":[],"_revision_number":1},{"id":"61f4aa7e4207a9b4e7a9de0333ba43f519de2e1f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-01-04 14:54:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"03e8721ab26bcf0e292e30622f04134acb8713ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 14:54:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fd31a87d4745cac793eac79940f5079c262a727b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 16:08:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/483fd493ca104065bfbdc60859972915 : SUCCESS in 30m 12s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/98f2e71bdfb24ba7811a3ac93c22004b : SUCCESS in 58m 40s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/12abf0ef30cd4da9a0a48ad2669a7378 : SUCCESS in 1h 11m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2af4b2c1a63949e7bf7ac1fe64358f65 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79fb5fe919bc46d68194f1c50ce5ff8c : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2264b223516846bc9df516901be20b7e : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ede0e8bbd2e40c98d0ab81cd74c3041 : SUCCESS in 4m 44s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7038939d83a64ca4a825adc22b50f9cd : SUCCESS in 28m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9da58d1a8a6242dea7b54f1f2d451216 : SUCCESS in 59m 28s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e2a9e70b5a0041138e75c23f6b7f3bc7 : SUCCESS in 1h 05m 59s","accounts_in_message":[],"_revision_number":1},{"id":"346c004b7d44ee87f0a54c9a5e756ca9ddeae5d2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 16:08:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"72dcc42d4e1e0aef29a0fd6cadbd8441e6cbf5bc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 16:10:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5df89752545445429359bf25c7d30b88 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e0d8772efa25b478467afcd3e2ecf2643ea471ff","revisions":{"e0d8772efa25b478467afcd3e2ecf2643ea471ff":{"kind":"REWORK","_number":1,"created":"2021-01-04 09:59:42.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/17/769117/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/769117/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/769117/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/769117/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/769117/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/769117/1"}}},"commit":{"parents":[{"commit":"9afa9aa108ba04a16eefbf5a67b6585fdbb33f8d","subject":"Merge \"ceilometer: do not fail when using an unsupported database type\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9afa9aa108ba04a16eefbf5a67b6585fdbb33f8d"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-10-04 09:26:41.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-04 09:57:33.000000000","tz":0},"subject":"Fix Freezer \u0026 Cyborg API startup on CentOS","message":"Fix Freezer \u0026 Cyborg API startup on CentOS\n\nAffects: Ussuri, Train\n\nAs part of the CentOS 8 support, we added a kolla_httpd_setup script,\nhowever this was not used for Cyborg or Freezer. This has been fixed\nsince Victoria as part of commit\n032804e5a0ddf89f7726880d1b57dd492d7d5c07, but this was not backported to\nUssuri.\n\nCloses-Bug: #1909981\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n\nChange-Id: I88e5bcf35207670b8611aeee3072215a7d7eb6e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e0d8772efa25b478467afcd3e2ecf2643ea471ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e0d8772efa25b478467afcd3e2ecf2643ea471ff"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
