)]}'
{"id":"openstack%2Fkolla~769177","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I88e5bcf35207670b8611aeee3072215a7d7eb6e5","project":"openstack/kolla","branch":"stable/train","hashtags":[],"change_id":"I88e5bcf35207670b8611aeee3072215a7d7eb6e5","subject":"Fix Freezer \u0026 Cyborg API startup on CentOS","status":"MERGED","created":"2021-01-04 18:58:09.000000000","updated":"2021-01-05 11:04:11.000000000","submitted":"2021-01-05 11:02:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"769177","cherry_pick_of_change":769117,"cherry_pick_of_patch_set":1,"meta_rev_id":"250171a2ed91f75cef38d227381e3c51c39a2bcd","_number":769177,"virtual_id_number":769177,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-05 11:02:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-01-05 09:29:19.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-01-04 18:58:25.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-01-05 09:29:19.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-04 20:25:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-05 09:29:19.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"9995490f263b14918733372c00f92e348a467b98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-04 18:58:09.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"3d1e276ced8158a73a6ceb4fda7fcaf37ae92e80","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-04 18:58:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0574d76721fe501342d01e9a6e9d56fe4f7b993a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 20:25:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/78109ca241494c4eb44060488166117e : SUCCESS in 43m 20s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7f89b76ac7de40b5ab0ae59675262b7c : SUCCESS in 50m 16s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/730d34f4d7f94ea6861fc5104ba5bdb2 : SUCCESS in 30m 01s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f3a72c2e183a4a309f4d53d459d3896c : SUCCESS in 38m 42s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/6da741d934c045a7a4ca51b6ccd03105 : SUCCESS in 1h 08m 06s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a38b8aa9c4c44440a0d6a16842b4a8ae : SUCCESS in 1h 01m 14s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/89f9e772b2204e28a8c310985ed6f021 : SUCCESS in 1h 09m 36s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/ea0610a2c2f54d87835d8a76d27d7ee2 : SUCCESS in 58m 18s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/9f9883b9fd38469b9e288186ae3c6d02 : SUCCESS in 1h 24m 47s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/cf012726d09d48bba815fe7274188522 : SUCCESS in 53m 31s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2427c85d9b1d4edaa94d9783e21e2ac1 : SUCCESS in 29m 28s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f1eeb760fb334b98b4423fc72e97f635 : SUCCESS in 17m 54s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2caea6caffae45749030478b27bd05b3 : SUCCESS in 1h 02m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a99d9768b9d4466bad9bb7b73bc0eaad : SUCCESS in 5m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0ea59ed37a9c4b9d87380d7e79386fbd : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8257a82825e7467b8eb1f2ab6f953dc4 : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7e50d2348f0945a4bb7938fd5746772a : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/242b09824b5a4b489dc62021d23da932 : SUCCESS in 6m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/17f700952238478aa52ceb25614e410d : SUCCESS in 24m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f474d49aa58f4a1ba7334e21d0a59bd8 : SUCCESS in 25m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2b0ec8ac6bdd489c9987bd060e7e2167 : SUCCESS in 47m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/973615bb60ab411bbc56e2f95c60aafa : SUCCESS in 52m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e97542c0f33e434f90e3ea238eaa9aeb : SUCCESS in 1h 00m 33s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"7288a5d601409eaa2f21c5aac34cf71d7da61ae8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-05 09:29:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aca8e3b24194a2117da8cb671f46f3061c4d76a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 09:29:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5ed0d864b1d37ac2ef34e7b306b723b5274c9d87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 11:02:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/de8da295bcb146759e80fab43b977225 : SUCCESS in 43m 53s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f73626705c724727ab7d8f8809d6ddcd : SUCCESS in 34m 12s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/6b29688e50cf4c94ad54969a407f0432 : SUCCESS in 1h 06m 47s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/5c3d3de1af134a70ae1591c921e392e2 : SUCCESS in 1h 02m 39s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/759d5911ec1141e292b72545c2f6368b : SUCCESS in 1h 27m 50s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/e43bcf7d896a4f0c8b12c4c293d53045 : SUCCESS in 1h 04m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c2e9a57025c4b55b49f23cbdfa1be14 : SUCCESS in 5m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aae1a7ca28174ad3a082461c339532d2 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/018cafb77821463aab8749848bbe17b3 : SUCCESS in 3m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6cea51a73d4b4927b45c2d58e830e29c : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c1b6c27bdf249bf8ccfeda201c52bde : SUCCESS in 7m 51s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4401d059b4194b8cb55af45c6a2fe6c1 : SUCCESS in 28m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5d8ee3ff30914ad9b97c51eb9676202c : SUCCESS in 55m 30s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ab46627e41b74fcc8e1af182f877e446 : SUCCESS in 1h 13m 04s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"f6a271c310d2330c9e051e400d2aceaff3f15470","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 11:02:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"250171a2ed91f75cef38d227381e3c51c39a2bcd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 11:04:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cc4403aa9b84667953505e44f83a496 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3e020fdf2a26ad640a56134c25a5060c48bb190b","revisions":{"3e020fdf2a26ad640a56134c25a5060c48bb190b":{"kind":"REWORK","_number":1,"created":"2021-01-04 18:58:09.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/77/769177/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/769177/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/769177/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/769177/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/769177/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/769177/1"}}},"commit":{"parents":[{"commit":"abeb5eaf4144284c22b0da6f3161d0b6550c3bbc","subject":"Merge \"Fix build of monasca-grafana\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abeb5eaf4144284c22b0da6f3161d0b6550c3bbc"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-10-04 09:26:41.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-01-04 18:58:09.000000000","tz":0},"subject":"Fix Freezer \u0026 Cyborg API startup on CentOS","message":"Fix Freezer \u0026 Cyborg API startup on CentOS\n\nAffects: Ussuri, Train\n\nAs part of the CentOS 8 support, we added a kolla_httpd_setup script,\nhowever this was not used for Cyborg or Freezer. This has been fixed\nsince Victoria as part of commit\n032804e5a0ddf89f7726880d1b57dd492d7d5c07, but this was not backported to\nUssuri.\n\nCloses-Bug: #1909981\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n\nChange-Id: I88e5bcf35207670b8611aeee3072215a7d7eb6e5\n(cherry picked from commit e0d8772efa25b478467afcd3e2ecf2643ea471ff)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e020fdf2a26ad640a56134c25a5060c48bb190b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e020fdf2a26ad640a56134c25a5060c48bb190b"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
