)]}'
{"id":"openstack%2Fkolla~770069","triplet_id":"openstack%2Fkolla~stable%2Fussuri~Ie54703e6ecf6a3332ef6f769fa500091068ea155","project":"openstack/kolla","branch":"stable/ussuri","hashtags":[],"change_id":"Ie54703e6ecf6a3332ef6f769fa500091068ea155","subject":"Bump prometheus-openstack-exporter version to 1.3.0","status":"MERGED","created":"2021-01-11 09:11:00.000000000","updated":"2021-01-11 12:34:17.000000000","submitted":"2021-01-11 12:32:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"770069","meta_rev_id":"1eb73ecbc01486c239d229630bdfd1702b15fc68","_number":770069,"virtual_id_number":770069,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-11 12:32:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-01-11 10:42:44.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-01-11 10:02:45.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2021-01-11 10:42:44.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-11 09:11:00.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"state":"REVIEWER"},{"updated":"2021-01-11 09:12:55.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2021-01-11 10:02:45.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-01-11 10:28:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-11 10:42:44.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-01-11 11:22:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"24d7af42537d965de31523094151595bd1aae5f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-01-11 09:11:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"571759bfb09aff03ceb4ed1b5aa1a20d4cf63639","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-11 10:02:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"518aaa75bacdb899ce99ec3a2b22cd3b6695c7e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 10:28:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/27fd846b8c2f439c8db1cca02d9e0fc8 : SUCCESS in 1h 16m 35s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7c69ccad5dbf4dd5bad4ee23d83054be : FAILURE in 1h 13m 19s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b8518ede6f394a1ca029248aa6fc138d : FAILURE in 47m 38s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3f732e4dc2f54dcd99efce5a6cab862c : FAILURE in 1h 11m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1a5cacd4e65857994f72148f40ab2cd6706ffd1e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-01-11 10:42:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"658f12ca3b0a3de20f79464400cf91877821a3b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 11:22:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/25ec1175a7884803a519d224cab059cf : SUCCESS in 21m 47s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3d27aa097c2f41bc81442352a2dbc23b : SUCCESS in 24m 51s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/79067c4fbe6144c58a65a8a3c9390417 : SUCCESS in 1h 00m 21s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c93581eecbb1415da8fba4ac038fd05e : TIMED_OUT in 2h 03m 48s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/60aacf3fa6284d13bc468ea0cc1032be : SUCCESS in 1h 16m 05s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/46193d1092aa4cb88d8271bd5505d672 : SUCCESS in 18m 45s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bd44e4e34d1443128bf1e1b04a5ec8f7 : SUCCESS in 18m 15s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f8b16ef9c4874cd798c77365d0f6eb3b : SUCCESS in 19m 35s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/01ec37a48ad14b7688b7f30d799740ef : SUCCESS in 55m 25s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fee07da27eb43599c4041b10b6596be : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42a9072af3c640d5b8c469a085513656 : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a5d6a61d86a440fba060b4e3511618c7 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d8bc24d01f0458a88d12136678b42a8 : SUCCESS in 6m 20s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ec83f83734d047b1baacd0495d34d214 : SUCCESS in 21m 22s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0340bf4032164ff282c24631d29d20f3 : SUCCESS in 25m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/17c9a5810e424acca9e25a83ce9432fb : SUCCESS in 52m 32s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/423f24aebbd049aea3617603cddd459a : SUCCESS in 52m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2999c62ae7944c7d92d607ab660189eb : SUCCESS in 1h 02m 01s","accounts_in_message":[],"_revision_number":1},{"id":"42cf45ddcb2acd58c19bcdd4f2aa5570c41aed83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 11:22:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c7ce684a4d7f8cc5e0f7f79c992c593ae3b9b4f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 12:32:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3085edca138a4a9888a3a3ed9eed26af : SUCCESS in 25m 12s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1d7b4bf1587d4c21af11d326e6ff2132 : SUCCESS in 1h 00m 32s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/52eb08b376064974bf628b78f3d140f0 : SUCCESS in 1h 02m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/631a785c36c94490be1c2c5e319306a9 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20b46294f5e8481d916e41a1450c09d4 : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4b5a770481934252a9146d5ad48c4da4 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e718e13747fd40c08184d770e401c59b : SUCCESS in 6m 46s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e3ff72e280984a1097212731894b4433 : SUCCESS in 27m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/53d06078c07446549ca3bf9dd6e3fc0d : SUCCESS in 57m 56s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/91eb655cf7da4ac48da91c3da6354ae4 : SUCCESS in 1h 04m 41s","accounts_in_message":[],"_revision_number":1},{"id":"efc2ecd8e6234cccbadc833a438dcfaa5f3217bc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 12:32:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1eb73ecbc01486c239d229630bdfd1702b15fc68","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 12:34:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c74f613b19948aeac93510d84bc591f : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb5f9c1f6dddcd87350024b94d67e99df7d81e70","revisions":{"eb5f9c1f6dddcd87350024b94d67e99df7d81e70":{"kind":"REWORK","_number":1,"created":"2021-01-11 09:11:00.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/69/770069/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/770069/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/770069/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/770069/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/770069/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/770069/1"}}},"commit":{"parents":[{"commit":"e0d8772efa25b478467afcd3e2ecf2643ea471ff","subject":"Fix Freezer \u0026 Cyborg API startup on CentOS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e0d8772efa25b478467afcd3e2ecf2643ea471ff"}]}],"author":{"name":"Jorge Niedbalski","email":"jnr@metaklass.org","date":"2020-12-17 14:12:50.000000000","tz":-180},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-01-11 09:10:56.000000000","tz":60},"subject":"Bump prometheus-openstack-exporter version to 1.3.0","message":"Bump prometheus-openstack-exporter version to 1.3.0\n\nChange-Id: Ie54703e6ecf6a3332ef6f769fa500091068ea155\nSigned-off-by: Jorge Niedbalski \u003cjnr@metaklass.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb5f9c1f6dddcd87350024b94d67e99df7d81e70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb5f9c1f6dddcd87350024b94d67e99df7d81e70"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
