)]}'
{"id":"openstack%2Fkolla~770070","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~Ie54703e6ecf6a3332ef6f769fa500091068ea155","project":"openstack/kolla","branch":"stable/victoria","hashtags":[],"change_id":"Ie54703e6ecf6a3332ef6f769fa500091068ea155","subject":"Bump prometheus-openstack-exporter version to 1.3.0","status":"MERGED","created":"2021-01-11 09:15:15.000000000","updated":"2021-01-11 12:33:25.000000000","submitted":"2021-01-11 12:32:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"770070","meta_rev_id":"fa12c8b734838b25901c77832e800b306d3e2cfb","_number":770070,"virtual_id_number":770070,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-11 12:32:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-01-11 10:42:40.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-01-11 10:02:18.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2021-01-11 10:42:40.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-11 09:15:15.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"state":"REVIEWER"},{"updated":"2021-01-11 09:30:35.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2021-01-11 10:02:18.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-01-11 10:42:40.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-01-11 11:03:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"89cf1157a84aeafbd7078c2b5928ff6b061e67eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-01-11 09:15:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"023da238b4efdad932c5f8f5e287ce8f4fdb82d7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-11 10:02:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3e1083746e1e679b4f9b9b236f608bba14b53be4","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-01-11 10:42:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"624b9ddf0c9b2e31a50e83fee512b23615e5da3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 11:03:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/f64f99c09fd94530bbeca8d6800acf45 : SUCCESS in 19m 24s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7047322c5f0f45958c812b6e2d7f2765 : SUCCESS in 24m 54s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/5938d80054234572b714a2e5fc9ad752 : SUCCESS in 1h 00m 35s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/13d3f435b9f34d61ae1e214ad24674fd : SUCCESS in 1h 31m 48s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/bc941bc51a054ae5b9ca22a08e6cf803 : SUCCESS in 1h 19m 39s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/dc4eda82216e4292bb7375926be30255 : SUCCESS in 19m 11s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2ba248f65f8c4e04809fa506db4e4ce6 : SUCCESS in 19m 03s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7cafc2b9a8444289a5da7b5a01ad2698 : SUCCESS in 1h 02m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aeebbd8d23634f458126bd401394e05e : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/268da5237d3c44a39123259a20faa766 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac6e79c197c547a28acb3179ac310bae : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6509731ed3449a19148dea2ac0d9076 : SUCCESS in 5m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dd0134ac22cc48d2bc83889c72833486 : SUCCESS in 18m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c7af6a71358f499489aeba6cfe70ef74 : SUCCESS in 20m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/73e565c8cd06414ca1e6297c15771eef : SUCCESS in 52m 41s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/872095b859624ef9ae7ea340eb0e94a7 : SUCCESS in 51m 06s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/39f8c0534037493987c02c67a90e4330 : SUCCESS in 57m 43s","accounts_in_message":[],"_revision_number":1},{"id":"2ab387432b38e0ebbdc388505919f180edfcca6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 11:04:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eaf709e2d35350bbf5b6c954b8405a3235e3a581","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 11:15:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/5e8296ad90f54c4d96daf07ce4e90dc8 : SUCCESS in 1h 10m 49s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/869016cc63284484a8316973df8f505c : SUCCESS in 1h 07m 32s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/50b9caffed924f7c989fda61aefe9f67 : SUCCESS in 1h 53m 36s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/47f6a32aa85a4c6786cce869dcf565b7 : SUCCESS in 1h 07m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6c4bc56ed8fd59595e24b6afcef02dc6583ceadc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 12:32:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2c9e8efcc0a64a758ae1a17d8ec489a2 : SUCCESS in 28m 49s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/fbab556f8db44808a83007ef2d7a31a0 : SUCCESS in 1h 02m 44s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/9256c17426114d8582b6874c3e0b950e : SUCCESS in 1h 23m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a596acdad4084f25809d9e20c3641d9b : SUCCESS in 7m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/044b1b06524844f6b1f244f774e559e0 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/066a94568856463c885dd142673dcde4 : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c50b4e8617f041eaa86b588e79e773bc : SUCCESS in 6m 23s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a6f7a1beab3d4c4680faee7af0659107 : SUCCESS in 30m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/215ae8bf701340ac897d55be26c99c29 : SUCCESS in 53m 43s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5d63f2929cba4980b7bab4aa12d89408 : SUCCESS in 1h 07m 16s","accounts_in_message":[],"_revision_number":1},{"id":"66b73fb2ad59c53d62f07b524acd841c0a222269","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 12:32:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fa12c8b734838b25901c77832e800b306d3e2cfb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 12:33:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b60bfd11c77f4dc4979aa6f7e85e4d0b : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cf795f3003066be53664a6b2709d20b3f1af07f7","revisions":{"cf795f3003066be53664a6b2709d20b3f1af07f7":{"kind":"REWORK","_number":1,"created":"2021-01-11 09:15:15.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/70/770070/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/770070/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/770070/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/770070/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/770070/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/770070/1"}}},"commit":{"parents":[{"commit":"fee9edb2a371a5ea75371ea28ab0979afca70444","subject":"Merge \"nova-compute: daxio on CentOS 8 is x86-64 only\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fee9edb2a371a5ea75371ea28ab0979afca70444"}]}],"author":{"name":"Jorge Niedbalski","email":"jnr@metaklass.org","date":"2020-12-17 14:12:50.000000000","tz":-180},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-01-11 09:15:12.000000000","tz":60},"subject":"Bump prometheus-openstack-exporter version to 1.3.0","message":"Bump prometheus-openstack-exporter version to 1.3.0\n\nChange-Id: Ie54703e6ecf6a3332ef6f769fa500091068ea155\nSigned-off-by: Jorge Niedbalski \u003cjnr@metaklass.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf795f3003066be53664a6b2709d20b3f1af07f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf795f3003066be53664a6b2709d20b3f1af07f7"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
