)]}'
{"id":"openstack%2Fkolla~770086","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I95e5bb5583be2a76e289749e8adfc303a4e7197a","project":"openstack/kolla","branch":"stable/ussuri","topic":"bug/1902101-stable/ussuri","hashtags":[],"change_id":"I95e5bb5583be2a76e289749e8adfc303a4e7197a","subject":"Remove footer block from intermediate images","status":"MERGED","created":"2021-01-11 12:28:34.000000000","updated":"2021-01-13 10:31:38.000000000","submitted":"2021-01-13 10:30:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"770086-bug/1902101-stable/ussuri","cherry_pick_of_change":763027,"cherry_pick_of_patch_set":3,"meta_rev_id":"cabe1e299a8af33f1e50dd9cf46d7df6cfa3e041","_number":770086,"virtual_id_number":770086,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-13 10:30:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-01-13 08:51:55.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-01-13 08:54:25.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-12 11:23:51.000000000","_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-01-13 08:51:55.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-11 13:40:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-12 11:23:51.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2021-01-13 08:54:25.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"b589da74ea256ae32af806d6a59350eccd103d08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-11 12:28:34.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e9102952a8459f173556d235e849eef72413bcac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 13:40:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b9a2b28fcf4e4fdca7b7b9252b36bf0c : SUCCESS in 36m 52s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f60a0aeea6834be1af3da494c60ba7a8 : SUCCESS in 41m 20s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7e91928b93aa453982932c9ad0bccd82 : SUCCESS in 57m 54s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b499199f9a3d4a66be5bf7e7b15c38bd : SUCCESS in 1h 02m 12s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/056856807dfa45d9964b3bb1f57830df : SUCCESS in 1h 03m 10s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/b310137dbe804003862b9778bf12c81a : SUCCESS in 18m 09s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a2d8b51bf41543b4929164647183428b : SUCCESS in 28m 39s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1acbbab8ce794f6bb40e1a793bc05b9a : SUCCESS in 19m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c0cd9df7e6d6450aade2c8798a94b20b : SUCCESS in 54m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0136ff6ddfc544a985cfc8d1ac6650e6 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fed010a8c5c41849bd7f12d1a76d459 : SUCCESS in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/92d2881143eb44429e70063a45fdc374 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46808bdd80614cf7869466ec3ecb77be : SUCCESS in 5m 02s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/267e8e50a762418a963dd0dee5a286f5 : SUCCESS in 20m 12s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/393335d57dee470f87f68aaa706b3a8a : SUCCESS in 26m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bfc0d0cf73eb48388cd183d10c11d7d1 : SUCCESS in 54m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e803633467294808b331e6df4a1b686d : SUCCESS in 54m 04s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/11ed722857cf40d5a4edc7b8f2cd51f0 : SUCCESS in 1h 00m 06s","accounts_in_message":[],"_revision_number":1},{"id":"62517e1b243e0d91a99353aad442a9521e9a4489","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 16:54:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/e6ea76a0ba6e4e71b4c12f96775ee31d : TIMED_OUT in 3h 01m 06s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cecf57ba93eb40eabecde1b603280634 : TIMED_OUT in 3h 01m 02s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/db1391edc3e444c39ce05df15ecf7285 : TIMED_OUT in 3h 01m 30s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a837ba2df6474f5a82fa95cb5cf0a958 : TIMED_OUT in 3h 01m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d26d5bf7e7b8582ba305c07fc3ab3fb1032b8ef4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-12 11:09:57.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"013c48b33b905d97e5be03f040b832cf923fd317","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2021-01-12 11:23:51.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"ea1b797b79f11897365c60582473dfdeb74ed7a7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 12:42:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b6b95040eec44d2382505e1b45f6c848 : RETRY_LIMIT in 2m 01s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7806a1f0f3bd48ad9512dc0de63215e9 : RETRY_LIMIT in 2m 14s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/47b41a106da94fa4bce5f85053ea5ca8 : RETRY_LIMIT in 1m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f848d2ac32a071a8743ccb37265694ffdf4b48b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 14:37:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/8e9845f0c22a4ec8acdca804e0e4d5b3 : SUCCESS in 22m 49s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f795f83c922d48ac936423aaa7adf740 : SUCCESS in 27m 43s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1a132546437a49bb96d289aeb5d0b277 : SUCCESS in 1h 02m 31s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/822a033bca5b4f858d62b4788872d657 : SUCCESS in 1h 00m 57s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/5e5e1fe8a6694f2082c8c99560c455e5 : SUCCESS in 1h 06m 44s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/c1b96c5c98ac42cab6ba895998c7c5ec : SUCCESS in 24m 57s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/65ccf4c772c949ea89f8eecb07b59884 : SUCCESS in 23m 08s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0dfc198cb2084668a165db97cc91c9f9 : SUCCESS in 1h 25m 27s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/50cf5762d80a435ea784b8487336e548 : SUCCESS in 1h 01m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f95d999a0d9b4fc7ae760b80a92f1a96 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2ce9e0870044aa4ad132b4e7d29de0a : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9376665156264744a39721983fc23988 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e1fdcf4234f4499a37ba89e79d81a70 : SUCCESS in 7m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/713da01587894d54b744dd0217a40e28 : SUCCESS in 21m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2bdccbba50044360ad72e4329e1ccacd : SUCCESS in 23m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/44dc745cb65c456ca99901a98edb259e : SUCCESS in 58m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1f63810fc2234428b6f59fbfcc2145a5 : SUCCESS in 57m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/14f4950d216f4092b5beaaaa472c31d7 : SUCCESS in 1h 20m 05s","accounts_in_message":[],"_revision_number":2},{"id":"7f11ade8519424cd0f7e307ac0b35817cfab30cd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-13 08:51:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8f4036c3677ef0b8e15e62c2a67b9e98af25b14c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 08:54:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c676176563d1565b4f2f703391cce02041e01efe","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-13 08:54:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8ff411fa4266cc5aa7004e21df8e4236ec998dbc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-13 08:55:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f11b31f3c4dc72d3d941ed1ee1c92cd52947c53c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 10:30:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/4e94006ef38a4c18b7a416d586f73d8a : SUCCESS in 32m 35s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/726fe986ba9b4d5ebca1f893728f7545 : SUCCESS in 1h 01m 55s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/9121bbf027394750b94404a26b801e0b : SUCCESS in 1h 11m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3309529c36b440afa8def0ba9d72e6c4 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cfbedc52cdc14281ada2a5e01098c346 : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ee67dbd23ff94157a61507684daf44c8 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab0ae31f8f414af0808d05b17a8456e6 : SUCCESS in 6m 55s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/36c57b57b78d4a42ad0b8e02ed21d6d9 : SUCCESS in 24m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/00b508c8184149688b944c129364b3a0 : SUCCESS in 1h 01m 51s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5cb693c68a0744598558ffbd8eed805d : SUCCESS in 1h 26m 18s","accounts_in_message":[],"_revision_number":2},{"id":"de5fad8d89ae7baa91bbcce29e609229bf7e1761","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 10:30:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"cabe1e299a8af33f1e50dd9cf46d7df6cfa3e041","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 10:31:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd8994f3b60e43b9a031743a2b5e5a61 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a11b99c85d18065cae36d0d6ab636c4475274f6f","revisions":{"e8ddf18b6110f03b042d740bf88eb8554726bdc3":{"kind":"REWORK","_number":1,"created":"2021-01-11 12:28:34.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/86/770086/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/770086/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/770086/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/770086/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/770086/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/770086/1"}}},"commit":{"parents":[{"commit":"e0d8772efa25b478467afcd3e2ecf2643ea471ff","subject":"Fix Freezer \u0026 Cyborg API startup on CentOS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e0d8772efa25b478467afcd3e2ecf2643ea471ff"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-17 14:40:32.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-11 12:28:34.000000000","tz":0},"subject":"Remove footer block from intermediate images","message":"Remove footer block from intermediate images\n\nOnly the final image in a hierarchy should have the footer block.\n\nneutron-infoblox-ipam-agent image still has two footers because it\ninherits from the neutron-server image. We set the user to root at the\nbeginning of that Dockerfile, so that headers and footers have the right\ncontext.\n\nChange-Id: I95e5bb5583be2a76e289749e8adfc303a4e7197a\n(cherry picked from commit 4b792b9a43bee3bfb499cea14924527550f2bce0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e8ddf18b6110f03b042d740bf88eb8554726bdc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e8ddf18b6110f03b042d740bf88eb8554726bdc3"}]},"branch":"refs/heads/stable/ussuri"},"a11b99c85d18065cae36d0d6ab636c4475274f6f":{"kind":"REWORK","_number":2,"created":"2021-01-12 11:09:57.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/86/770086/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/770086/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/770086/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/770086/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/770086/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/770086/2"}}},"commit":{"parents":[{"commit":"e0d8772efa25b478467afcd3e2ecf2643ea471ff","subject":"Fix Freezer \u0026 Cyborg API startup on CentOS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e0d8772efa25b478467afcd3e2ecf2643ea471ff"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-17 14:40:32.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-12 11:09:53.000000000","tz":0},"subject":"Remove footer block from intermediate images","message":"Remove footer block from intermediate images\n\nOnly the final image in a hierarchy should have the footer block.\n\nneutron-infoblox-ipam-agent image still has two footers because it\ninherits from the neutron-server image. We set the user to root at the\nbeginning of that Dockerfile, so that headers and footers have the right\ncontext.\n\nChange-Id: I95e5bb5583be2a76e289749e8adfc303a4e7197a\n(cherry picked from commit 4b792b9a43bee3bfb499cea14924527550f2bce0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a11b99c85d18065cae36d0d6ab636c4475274f6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a11b99c85d18065cae36d0d6ab636c4475274f6f"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
