)]}'
{"id":"openstack%2Fkolla~770087","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I95e5bb5583be2a76e289749e8adfc303a4e7197a","project":"openstack/kolla","branch":"stable/train","topic":"bug/1902101-stable/train","hashtags":[],"change_id":"I95e5bb5583be2a76e289749e8adfc303a4e7197a","subject":"Remove footer block from intermediate images","status":"MERGED","created":"2021-01-11 12:28:53.000000000","updated":"2021-01-23 13:08:28.000000000","submitted":"2021-01-23 13:07:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"770087-bug/1902101-stable/train","cherry_pick_of_change":763027,"cherry_pick_of_patch_set":3,"meta_rev_id":"1dced395915fea7041748b252af2711a7a68df11","_number":770087,"virtual_id_number":770087,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-23 13:07:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-01-23 13:08:28.000000000","post_submit":true,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-01-23 13:08:28.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-01-23 13:08:28.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-11 14:23:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-13 09:02:12.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2021-01-19 11:38:22.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-01-20 10:23:23.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"df6404b098dbba8732feadc857ac14fbf2aa2565","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-11 12:28:53.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* docker/mariadb/mariadb-base/Dockerfile.j2","accounts_in_message":[],"_revision_number":1},{"id":"8c47eda16b6d50c341474cd8fc16720eedc307ac","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-11 12:35:49.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"e661b7a8e059e78d8647b1b2c9d88945bde6177f","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-11 12:35:56.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"69e82a7f2c56747ac5f47c072db549146a3cfb52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 14:23:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/1c1a3a1892f24bd6bca49a881dff0204 : SUCCESS in 1h 09m 24s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/762f7c1579844734b849d34a9f8d58dc : SUCCESS in 1h 40m 22s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b493c6cbfa044a6186fe63737d7397cf : SUCCESS in 21m 43s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7317dc7a72d643daa28bcb7c0ca39653 : SUCCESS in 26m 44s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/49051b09011943e1a02a11fe86a950ec : SUCCESS in 1h 21m 58s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a200191ecd5e4735a4ff34ad23e9bff4 : SUCCESS in 55m 11s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/009b85b55072462b8d61478c58ad1e44 : SUCCESS in 1h 31m 34s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/2129f3da27224af0a85667d2f4080ed0 : SUCCESS in 1h 11m 28s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/426d5ad8b60e4bbc9c7cc7efd5a39988 : SUCCESS in 1h 26m 48s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/1ef7d452a6a24328b3c4c7548f3dc478 : SUCCESS in 52m 10s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/822198a8f683437db0aa94faba1fcc84 : SUCCESS in 20m 48s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/486af1fa12a64c689ee69042946fe371 : SUCCESS in 20m 45s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8695b571cb8d414c92a28add96cc7396 : SUCCESS in 51m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b94935f3d634655a7db888f9d1157ca : SUCCESS in 4m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4608a6b4ccf849d087cbf83ab6038f91 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3da74ed081f443586ea6c98f5c7e30f : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8afa327aa32b4fbbbb56c7131081bcaf : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7aed516c3894bf1baea389e684f1d9d : SUCCESS in 5m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/abec606b6e5c43dfb932e33e40b66c37 : SUCCESS in 19m 17s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/396fdbb02e7247b5bb2830b4d44cbe84 : SUCCESS in 41m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3958741fa502413faf257d203ad81b8c : SUCCESS in 1h 24m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/be99d62a17cf4783888011ab7c23e51b : SUCCESS in 50m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5ac9b8ced6f84ff8850957fe54d882da : SUCCESS in 59m 41s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"ed068a70ed84ea64a24b678966754fb3c831c507","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-12 15:52:31.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"18f46721cc548697a363e093204b3b0a89c206e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 18:13:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ccca9fd486294f7c8451bce0d060ddb3 : SUCCESS in 50m 17s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d65b62269cd5412fbe4037a8cfbcf026 : SUCCESS in 50m 49s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/9ccac209f4644d838ab43e455b6b8328 : SUCCESS in 24m 55s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/23967f3d033846caab043616fd2b1be2 : SUCCESS in 32m 48s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/139daca5c549458bb5204d0f3454c86b : SUCCESS in 1h 16m 36s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/215aece71cbc4e7c89de61471654e478 : SUCCESS in 57m 09s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/4c9612eba04d4660af2d7e235b1c0a40 : SUCCESS in 1h 17m 28s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/5363a22afb0449c8bcfdcdbc99e78c4f : SUCCESS in 58m 13s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/1e81bbe9f3db408cbb80120a36d133f1 : SUCCESS in 1h 58m 02s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/716ec7dd408943ef941238e32d2417ee : SUCCESS in 43m 27s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d134ca1672704f94a9015bd9aa85558f : FAILURE in 22m 24s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e5a3e9e8c2854276988095f93d5ba756 : FAILURE in 17m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/191846c3b8714299ab37cb720bd3e697 : FAILURE in 19m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc41bc85ac68405f9a99c87099196fc1 : SUCCESS in 4m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/24e3b21f53b945249bb0ef246ede366d : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/831a6237a812404fa2cef5365e9dbe9d : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/66db36b42b2f42edb4cae0c5ed571767 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbd4c4db28c744c7baf39935b7c5f65d : SUCCESS in 7m 49s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d5f325677c6e443a8478fdf2b17703b8 : FAILURE in 27m 11s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6696b1d1e4a94aad9de746bbba37554a : FAILURE in 27m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ed5e6e0d63cb4facbb9463680d5e317d : FAILURE in 24m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bb3e5ae0614e4cc8babd0fd13825d12c : FAILURE in 23m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3f1fb210e35c4e99a707a4f07e275c40 : FAILURE in 15m 59s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":3},{"id":"a2437f4258dc00477babb37b94d26947c9978f12","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-13 09:02:12.000000000","message":"Patch Set 3:\n\nthere is no shell expansion here","accounts_in_message":[],"_revision_number":3},{"id":"19f9fa016ee150dedf7dd1c4134e6a0130b3861a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-13 09:03:48.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3:\n\u003e \n\u003e there is no shell expansion here\n\nsorry, ignore me, still waking up","accounts_in_message":[],"_revision_number":3},{"id":"de2c95d51f341ed102fac5f60806ec4094d345ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-13 09:10:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"65659e49937288d4b89d070f63e79efd3522fa6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 10:57:54.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/985433be1bc049d681db3046da555c1b : SUCCESS in 51m 32s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/0382490525354affacb507e9700850b0 : SUCCESS in 1h 03m 34s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/a6ee7510674841aaa4e9f378c7f40d0f : SUCCESS in 22m 47s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d71371bfb13244618c16a34c3c71bc2f : SUCCESS in 47m 20s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/6a9000de09a54cc7afac481066d93bd1 : SUCCESS in 1h 35m 51s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/abad82f423684e7bb2f2c7bef6e93bda : SUCCESS in 1h 17m 37s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/3da15c049c3e40ec8184d1fe08e3ba72 : SUCCESS in 1h 19m 58s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/4b6157db201343a0b10e750d882f2169 : SUCCESS in 51m 46s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/8f4901ecbe704203b2eedeee9ec00de2 : SUCCESS in 1h 15m 49s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/39a5b3bb6a6345d7bbcf6a1a09dd5871 : RETRY_LIMIT in 4m 58s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d8ac8e4e2b534d6282013bd39320f8f9 : SUCCESS in 20m 07s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c8f7f69502394826b20f434cd3383778 : SUCCESS in 31m 18s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/029571c191654096a1e42d2f9bdc6a61 : SUCCESS in 1h 07m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/361b9bb6e0854900b1793cde8bd2da13 : SUCCESS in 5m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9e821a1b201c41dab432d6559d3e3e83 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76c4348f705e4def95e7d068cc9ddbb0 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f5898b998c364237aed35d169703e271 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/603bb979abb448aa90bc1459bfebeb6e : SUCCESS in 8m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e1b72a9cf111467ebb9aa151b2fd92f8 : SUCCESS in 26m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6101497005ac4de2add503e19742cbde : SUCCESS in 22m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7e510c9221274343b900815e7ff1b99c : RETRY_LIMIT in 1m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/89e0a7e66a4d4196bbfd88e1833c5d0d : SUCCESS in 54m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/32b5071d4d2045f6b25a463a8b33deeb : SUCCESS in 1h 09m 52s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":4},{"id":"e6677e2404038ed969e20a08812f8e9dc277447f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-01-19 11:38:22.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"8cf732c0a37b3251154f6ce945c6a4f31d3008fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-20 10:22:42.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"1264ae4f53cc3dc7205a8810bc1add64f0d29dea","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-20 10:23:23.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"ecca7af134c71a7834623eaa3cefbedc60fbae21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-20 11:58:03.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"e0422776fe8c43c68a1b562d98dffaa31524a3ef","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-20 11:58:09.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"21145350e6c51fd03b05cec080fb646cf9209648","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 13:23:14.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/68a8d668a6ef46edaf8015822ac670ea : SUCCESS in 33m 51s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/06451860b34b489d8184c205eefec1eb : SUCCESS in 39m 26s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b3133d71bf5d423899b484ead2a45a3d : SUCCESS in 22m 54s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/424cd7347ce8406ca88f57ad79dd7e78 : SUCCESS in 33m 52s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ca00a223d36b4025b226563822607482 : FAILURE in 5m 36s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/4c1b6cedcd2142e085737d64762d6451 : SUCCESS in 52m 54s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/7f796d583a4347ffa42b60d964da35fb : FAILURE in 5m 31s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/fd8ecfcd9d84407eb22355f9f6e366da : SUCCESS in 1h 01m 38s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/63d5bfbf10bc49eea96a54ec42fa80b5 : FAILURE in 5m 39s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/2b3a88f6eadc4827829df7399796a488 : SUCCESS in 50m 12s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/108f6c8da660456b92207a7337ebb936 : SUCCESS in 21m 51s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/56dedfc5b3f44ba7b3c573f1e64dfcc0 : SUCCESS in 22m 05s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/403e13aeb022451b85229c68549a1fcc : FAILURE in 5m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66f2a8be481e4a53bb2c3da2611b507b : SUCCESS in 5m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/706877eba4234feaab6c0a9adb94ff39 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f783559ab49a4bdf91677ed4003e8293 : SUCCESS in 5m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8fd721ec82f94a85a70a17ef2f64cce0 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d689e622c4ef476fb8e368b19e785f0e : SUCCESS in 5m 29s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3fd6c2c7a76f48a094673d5edca81c4e : SUCCESS in 25m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ed623e0c4d2349e5b417a2a6460f1968 : SUCCESS in 24m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5cd94992b5394f649403cad91d649f08 : SUCCESS in 42m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5669fd1e249b4a5c8903b65d88227727 : SUCCESS in 48m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/18b9253cf8db484a84329961b81c5461 : SUCCESS in 57m 43s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":6},{"id":"27fae50eb23f5bf2ec11b2120d4299fe35d5dc15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-23 08:51:55.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"0a6f047787e54341e83d8e404ff55eca312a1162","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-23 10:59:23.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/57838ad6fd5b4653acad46eadca39ec9 : SUCCESS in 47m 12s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/aecb8733b7814d878b42d4bfd4b30d3d : SUCCESS in 38m 30s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/9c1c4c346fa24799927e129104399054 : SUCCESS in 26m 03s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/100f9d7c0a794dadbe00158fd95e28d8 : SUCCESS in 33m 21s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/aeaa9c67c4ab438cb37a301f8989d3f7 : SUCCESS in 1h 28m 10s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/4d738d441db2416d8bbc7edb54729793 : SUCCESS in 58m 43s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/8a0d9f5059114f62b748c77a5d4c5250 : SUCCESS in 1h 26m 03s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c94c1aae8ada4defb96c55934140a9b2 : SUCCESS in 1h 04m 05s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c4e8980fc67142e491f5ffff255e6c8a : SUCCESS in 1h 29m 13s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/4d7b2f85be684067a3272d6c8f71848e : SUCCESS in 48m 20s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0c20c923de9c4b3aaac83918ad86ae79 : SUCCESS in 21m 59s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d33f4c64459a44669786386cca7890a7 : SUCCESS in 19m 36s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d29a2d72b7e4442ab9875f469b48f3de : SUCCESS in 52m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23abbd6945164eceaf31e1d2b857843e : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/091dbd5ca9114a999f4e3f3f059d8a5b : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d80c02b49bf34f159616c37f48b9c069 : SUCCESS in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f86ef33980734609adb5ceb51f9cc099 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6118837a337a44f4993195295ca7295c : SUCCESS in 5m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ebd5834585cf4bfa849797e874ddf775 : SUCCESS in 29m 00s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a0b9a0be89b14281a515864715cfbbe4 : SUCCESS in 41m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4ea0fa84a74d48d89a82489c52bf4caa : SUCCESS in 48m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b7bf5d11ee364d32ade9c4f90b11b8eb : SUCCESS in 57m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/71653f2ebede46ed84ed3616bfc4d9e8 : SUCCESS in 59m 30s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":7},{"id":"dab55d3eb6d7fdd25e019e09dbfb662eb8f8694a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-23 10:59:39.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"024814b714eebf3bb6335318f688d90b597f298a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-23 13:07:12.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/569653360a404b9aa65ebb2ceec64508 : SUCCESS in 43m 12s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/baef72efa6784dd290be91c3c5dd5983 : SUCCESS in 24m 55s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/0e1e8357da684d3badca415c86fd2949 : SUCCESS in 1h 05m 06s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/942b09c9dc304dc7b295718936d4ca04 : SUCCESS in 51m 18s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/1c349848381045198f76d6f8aed2a3a3 : SUCCESS in 1h 18m 51s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/03cf44c6907b4dd391499312d9149cb3 : SUCCESS in 39m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8c2d8c665e341c990eea6f40c2e658d : SUCCESS in 6m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eaf34bc6bad347d5adf4d23136cbbf8a : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5cd29633a3e2455eade489ef6fc0151b : SUCCESS in 3m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9df62dde531f4810b6c7f16487043a4b : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5dfd6528b6a48efab80c3a03199303a : SUCCESS in 6m 15s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5d8bba48cb714ee3af7c810012da7f1c : SUCCESS in 24m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2bf98b82413d45118be0c6d166fe5f06 : SUCCESS in 49m 15s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2468bc1413fd47aa80dfb03068d0c86d : SUCCESS in 1h 13m 49s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":7},{"id":"d86da23669f0fe26f9129070c080328c1b6afa15","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-23 13:07:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"dca22bd5cad623635fcb8be51ceb4d995491247b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-23 13:08:28.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c40eacd1b90e4fe896983b6af4eb73ae : SUCCESS in 50s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"297c15de4b92425286cba2eb3977ec5d877a7e9a","revisions":{"73d763f0523ea66d54ae85cde6da3a748aa885a7":{"kind":"REWORK","_number":1,"created":"2021-01-11 12:28:53.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/87/770087/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/770087/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/770087/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/770087/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/770087/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/770087/1"}}},"commit":{"parents":[{"commit":"3e020fdf2a26ad640a56134c25a5060c48bb190b","subject":"Fix Freezer \u0026 Cyborg API startup on CentOS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e020fdf2a26ad640a56134c25a5060c48bb190b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-17 14:40:32.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-11 12:28:53.000000000","tz":0},"subject":"Remove footer block from intermediate images","message":"Remove footer block from intermediate images\n\nOnly the final image in a hierarchy should have the footer block.\n\nneutron-infoblox-ipam-agent image still has two footers because it\ninherits from the neutron-server image. We set the user to root at the\nbeginning of that Dockerfile, so that headers and footers have the right\ncontext.\n\nChange-Id: I95e5bb5583be2a76e289749e8adfc303a4e7197a\n(cherry picked from commit 4b792b9a43bee3bfb499cea14924527550f2bce0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/73d763f0523ea66d54ae85cde6da3a748aa885a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/73d763f0523ea66d54ae85cde6da3a748aa885a7"}]},"branch":"refs/heads/stable/train"},"128e4d32258d29a60820fcfb85a00bc91851e367":{"kind":"REWORK","_number":2,"created":"2021-01-11 12:35:49.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/87/770087/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/770087/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/770087/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/770087/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/770087/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/770087/2"}}},"commit":{"parents":[{"commit":"3e020fdf2a26ad640a56134c25a5060c48bb190b","subject":"Fix Freezer \u0026 Cyborg API startup on CentOS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e020fdf2a26ad640a56134c25a5060c48bb190b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-17 14:40:32.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-11 12:35:44.000000000","tz":0},"subject":"Remove footer block from intermediate images","message":"Remove footer block from intermediate images\n\nOnly the final image in a hierarchy should have the footer block.\n\nneutron-infoblox-ipam-agent image still has two footers because it\ninherits from the neutron-server image. We set the user to root at the\nbeginning of that Dockerfile, so that headers and footers have the right\ncontext.\n\nChange-Id: I95e5bb5583be2a76e289749e8adfc303a4e7197a\n(cherry picked from commit 4b792b9a43bee3bfb499cea14924527550f2bce0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/128e4d32258d29a60820fcfb85a00bc91851e367"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/128e4d32258d29a60820fcfb85a00bc91851e367"}]},"branch":"refs/heads/stable/train"},"ea55ccbf1622822e78197f77a06a36b62bbe9dba":{"kind":"REWORK","_number":3,"created":"2021-01-12 15:52:31.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/87/770087/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/770087/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/770087/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/770087/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/770087/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/770087/3"}}},"commit":{"parents":[{"commit":"3e020fdf2a26ad640a56134c25a5060c48bb190b","subject":"Fix Freezer \u0026 Cyborg API startup on CentOS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e020fdf2a26ad640a56134c25a5060c48bb190b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-17 14:40:32.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-12 15:51:49.000000000","tz":0},"subject":"Remove footer block from intermediate images","message":"Remove footer block from intermediate images\n\nOnly the final image in a hierarchy should have the footer block.\n\nneutron-infoblox-ipam-agent image still has two footers because it\ninherits from the neutron-server image. We set the user to root at the\nbeginning of that Dockerfile, so that headers and footers have the right\ncontext.\n\nChange-Id: I95e5bb5583be2a76e289749e8adfc303a4e7197a\n(cherry picked from commit 4b792b9a43bee3bfb499cea14924527550f2bce0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea55ccbf1622822e78197f77a06a36b62bbe9dba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea55ccbf1622822e78197f77a06a36b62bbe9dba"}]},"branch":"refs/heads/stable/train"},"177cb550626f633b483641dfa82f6703c9f04772":{"kind":"REWORK","_number":4,"created":"2021-01-13 09:10:23.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/87/770087/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/770087/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/770087/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/770087/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/770087/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/770087/4"}}},"commit":{"parents":[{"commit":"3e020fdf2a26ad640a56134c25a5060c48bb190b","subject":"Fix Freezer \u0026 Cyborg API startup on CentOS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e020fdf2a26ad640a56134c25a5060c48bb190b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-17 14:40:32.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-13 09:09:42.000000000","tz":0},"subject":"Remove footer block from intermediate images","message":"Remove footer block from intermediate images\n\nOnly the final image in a hierarchy should have the footer block.\n\nneutron-infoblox-ipam-agent image still has two footers because it\ninherits from the neutron-server image. We set the user to root at the\nbeginning of that Dockerfile, so that headers and footers have the right\ncontext.\n\nChange-Id: I95e5bb5583be2a76e289749e8adfc303a4e7197a\n(cherry picked from commit 4b792b9a43bee3bfb499cea14924527550f2bce0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/177cb550626f633b483641dfa82f6703c9f04772"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/177cb550626f633b483641dfa82f6703c9f04772"}]},"branch":"refs/heads/stable/train"},"02860bf9eb7bde215606d7c7b54d0eaaf64cfa61":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-01-20 10:22:42.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/87/770087/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/770087/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/770087/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/770087/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/770087/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/770087/5"}}},"commit":{"parents":[{"commit":"c1e599ebc44250cafab0347c756bac5b4b7b2160","subject":"CI: Use Docker proxy for pulls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c1e599ebc44250cafab0347c756bac5b4b7b2160"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-17 14:40:32.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-01-20 10:22:42.000000000","tz":0},"subject":"Remove footer block from intermediate images","message":"Remove footer block from intermediate images\n\nOnly the final image in a hierarchy should have the footer block.\n\nneutron-infoblox-ipam-agent image still has two footers because it\ninherits from the neutron-server image. We set the user to root at the\nbeginning of that Dockerfile, so that headers and footers have the right\ncontext.\n\nChange-Id: I95e5bb5583be2a76e289749e8adfc303a4e7197a\n(cherry picked from commit 4b792b9a43bee3bfb499cea14924527550f2bce0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02860bf9eb7bde215606d7c7b54d0eaaf64cfa61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02860bf9eb7bde215606d7c7b54d0eaaf64cfa61"}]},"branch":"refs/heads/stable/train","description":"Rebase"},"4e6b60ea8ddaeb9cf82f491bc4254a57986b84d4":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-01-20 11:58:03.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/87/770087/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/770087/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/770087/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/770087/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/770087/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/770087/6"}}},"commit":{"parents":[{"commit":"71a6eab4e7ff1758f32c186cc55f2bf2e4a7daa0","subject":"CI: Use Docker proxy for pulls + Ruby fixes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/71a6eab4e7ff1758f32c186cc55f2bf2e4a7daa0"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-17 14:40:32.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-01-20 11:58:03.000000000","tz":0},"subject":"Remove footer block from intermediate images","message":"Remove footer block from intermediate images\n\nOnly the final image in a hierarchy should have the footer block.\n\nneutron-infoblox-ipam-agent image still has two footers because it\ninherits from the neutron-server image. We set the user to root at the\nbeginning of that Dockerfile, so that headers and footers have the right\ncontext.\n\nChange-Id: I95e5bb5583be2a76e289749e8adfc303a4e7197a\n(cherry picked from commit 4b792b9a43bee3bfb499cea14924527550f2bce0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e6b60ea8ddaeb9cf82f491bc4254a57986b84d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e6b60ea8ddaeb9cf82f491bc4254a57986b84d4"}]},"branch":"refs/heads/stable/train","description":"Rebase"},"297c15de4b92425286cba2eb3977ec5d877a7e9a":{"kind":"NO_CHANGE","_number":7,"created":"2021-01-23 08:51:55.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/87/770087/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/770087/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/770087/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/770087/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/770087/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/770087/7"}}},"commit":{"parents":[{"commit":"36126c08255fcf8a201167b5d63f23f6bfdeceec","subject":"CI: Use Docker proxy for pulls + Ruby fixes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/36126c08255fcf8a201167b5d63f23f6bfdeceec"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-17 14:40:32.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-01-23 08:51:55.000000000","tz":0},"subject":"Remove footer block from intermediate images","message":"Remove footer block from intermediate images\n\nOnly the final image in a hierarchy should have the footer block.\n\nneutron-infoblox-ipam-agent image still has two footers because it\ninherits from the neutron-server image. We set the user to root at the\nbeginning of that Dockerfile, so that headers and footers have the right\ncontext.\n\nChange-Id: I95e5bb5583be2a76e289749e8adfc303a4e7197a\n(cherry picked from commit 4b792b9a43bee3bfb499cea14924527550f2bce0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/297c15de4b92425286cba2eb3977ec5d877a7e9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/297c15de4b92425286cba2eb3977ec5d877a7e9a"}]},"branch":"refs/heads/stable/train","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
