)]}'
{"id":"openstack%2Fkolla~770088","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~I2592a736206eaec811290e9fbdbf2540c0518ffe","project":"openstack/kolla","branch":"stable/victoria","topic":"bug/1902101-stable/victoria","hashtags":[],"change_id":"I2592a736206eaec811290e9fbdbf2540c0518ffe","subject":"CI: revert to public package mirrors after build","status":"MERGED","created":"2021-01-11 12:36:35.000000000","updated":"2021-01-20 13:44:53.000000000","submitted":"2021-01-20 13:43:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"770088-bug/1902101-stable/victoria","cherry_pick_of_change":761928,"cherry_pick_of_patch_set":17,"meta_rev_id":"a3ae46150237f38a96716ca1461afeb88d258c94","_number":770088,"virtual_id_number":770088,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-20 13:43:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-01-12 12:08:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-01-12 09:33:52.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-01-13 08:48:40.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-01-12 12:08:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-11 14:22:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-12 09:33:52.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-01-12 12:08:14.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-01-13 08:48:40.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"7048edb2783ba2aa834b6108cd9b0ac1a1f8be3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-11 12:36:35.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ec24425a2e93b9f544e8a5154f5211422d58b30d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 14:22:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/065e9a6f4845484f8f38c35d4bd2d88b : FAILURE in 22m 01s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/efa4a92ebde5430490c25f5cd4ca571c : FAILURE in 45m 32s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/ee8bc386e0b346ffaaa40aaf51a1ec89 : FAILURE in 18m 12s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/a24c2a0565b0453fb5201cc11a47c229 : FAILURE in 17m 46s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/a303bd6d38ba4c77af02063da93da06e : FAILURE in 17m 10s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c73a3b30b6e64a0cb60554b9f4e1a120 : SUCCESS in 17m 11s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e16b1e015e1844f589a107d3f2d4e2ac : SUCCESS in 17m 36s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9fd98e0426f34106802d87efd1285230 : SUCCESS in 1h 34m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65d0514dfc144ac1943858ae911c3228 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c7bda87c3f94c40a28928a0f59641b6 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/649257ab61a64752b31560093e8f9737 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3644af799a1d4a0e8dcd31a1273688ec : SUCCESS in 6m 33s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9b1e65f57ab84bb98734dea3b830497c : SUCCESS in 21m 42s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9113560a574a4891b5a6dc540f7e417e : SUCCESS in 42m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e32df70ed4b4489592aa20f5d2b1f016 : SUCCESS in 1h 11m 20s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c16f1fa579da41e088a13242e0fe49c1 : SUCCESS in 56m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/57fe73bea0ec47b0825c6bf8f8d13705 : SUCCESS in 1h 06m 13s","accounts_in_message":[],"_revision_number":1},{"id":"5cec7cc135182c04e0beb13f077f175926897e76","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 15:58:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/34760529a9ca466d90532f3d42f6277c : TIMED_OUT in 3h 01m 08s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9acb29bc04e045d18710080c4e8b5852 : FAILURE in 2h 46m 09s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d530671d4d334a5cad185dbfff5c02f8 : TIMED_OUT in 3h 05m 14s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/05c94c30cdf642b38432c4387f6f9627 : FAILURE in 2h 41m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"604e17b284e2112189cef528a3b548f3a7c25cc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-12 09:33:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"61e7cf0ebbbe7bd288f4e94f9d352a3ed1dda94f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-01-12 09:33:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"49ea155b4f32c015de615024b06b4b7609c8b391","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 10:23:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/0ef71913f37746c5b07232810d49366c : RETRY_LIMIT in 2m 32s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a8f1458f5ed34a0d8946f1697d13034f : RETRY_LIMIT in 2m 26s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0a06c7be2fa74443b88b848c21ff2d90 : RETRY_LIMIT in 2m 45s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d199a9f81caf4755b093e05fe2387339 : RETRY_LIMIT in 2m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a04cbaab4f7c56ddf2b2a981d6be7812cf28179a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 10:57:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/4b054f3da7d84b51bccc1fb1eee2527e : SUCCESS in 28m 29s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/cd2ad65ce10a442fb8773c54ab1a1ff4 : SUCCESS in 30m 02s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c8015b36eca04086a55055ecb2c6a097 : SUCCESS in 1h 13m 24s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/0dfe00b15c5e49fd9543a670e7f5870c : SUCCESS in 1h 02m 20s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/d949f6dbf5bf4813ba60bb23a121cf85 : SUCCESS in 1h 18m 41s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/14f1c4eb21c04717af8f117854adb23a : SUCCESS in 31m 19s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/250757312fbd4727888ed33e322aab1f : SUCCESS in 16m 24s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d71e606ad6974e96a7d602f2206f0d01 : SUCCESS in 55m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a186d83bffd14c988db6eff4fe563e1d : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/502fe8354557420aa832c453e6038a50 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44b0fb4702f74fa39a709754e517ae71 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de72604c4dbb40c48eeb29a67d12eacb : SUCCESS in 5m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/18318a26196d4bc081e3ba6b5b78ac07 : SUCCESS in 21m 50s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1e737e3ac8c54af1b314bdf9aa3d3f24 : SUCCESS in 19m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bdbb95d294b246d58407fbeab8029ff1 : SUCCESS in 1h 13m 12s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/321f13e434c3459fa21d156f81cd16a3 : SUCCESS in 1h 10m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a4b111a7edcb47ac8244f216327754b6 : SUCCESS in 1h 14m 29s","accounts_in_message":[],"_revision_number":2},{"id":"b7be1702694f51b7824e1b6492fb09d86ef7360b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-01-12 12:08:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ce0b36480e47a670e124d5067e882a1dc1d26b9c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-13 08:48:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4453c5f755bfbfbc82606f12bc8294faed391913","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 11:41:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a365a6fd25ddff8f0e82574dba7b235a3d16a63b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 13:11:30.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"577b24b937539a52aeb7da53bf768809c955754e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-20 10:22:08.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f9fee39a2c2fe9e3ec44fae1cef34ce0d03e4570","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 11:29:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/3d51fa2f61ee4fe4886664de252202f9 : SUCCESS in 46m 12s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c3e138bc7e734dc3a473c6a3a08a94b3 : SUCCESS in 43m 31s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6fc133a809904df58222fba6bd4c8506 : FAILURE in 54m 55s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/7a5722ae804c4a9fa393312c433246b1 : SUCCESS in 35m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0ea75758d13d96f925950e38855134bd4bf9810e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 12:22:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/97dbc47a2ebd4b61aa860ddef5c4187a : SUCCESS in 20m 33s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ac0de9eb75074d93a88651db2efb2ab8 : SUCCESS in 26m 14s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7163418fa31347379c7dfa8fb3ad3f8e : SUCCESS in 1h 00m 51s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/769d94e33dc94c9696d79f4a4a31956b : SUCCESS in 58m 45s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/b92e3501319a400dbcb2d62e608c4b00 : SUCCESS in 1h 16m 40s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8c786893657e4ff1a8701d87bc3298c5 : SUCCESS in 17m 28s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7c62d80aecb1448198c42112a73b2dfb : SUCCESS in 21m 20s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/42f8423620b044958c10d86381da3cbf : SUCCESS in 55m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4664ce709f3549f7a03206ae21edb125 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3dbebe27fa934835b36c0436a015babc : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1be80ca4fa544268b6f64dc54777ad21 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3cf58771a3448ad8b00183c3d0b1536 : SUCCESS in 5m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2c6567cda0794679843a49b3bf2aca5d : SUCCESS in 20m 00s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/23b2e27669964473956854d7ace70920 : SUCCESS in 19m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/97d082c4be8b4eb58e55e46233693c03 : SUCCESS in 1h 07m 17s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9d842df7cf084ea88623ef96f6d4fc7c : SUCCESS in 50m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4050dd0dc534425a95ba30d3c36a7671 : SUCCESS in 1h 07m 32s","accounts_in_message":[],"_revision_number":2},{"id":"1f28a2dc36fa4129626e33bf6e3bf8f23fd591b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 12:22:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"011e897554b03c6c4cd2a11c1f41837d71657d7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 13:43:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/c60abae5e8004cac82b6b40494cee13f : SUCCESS in 31m 30s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/99286c08ce234471ae6f5335cb334bf7 : SUCCESS in 1h 05m 23s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/df3ea8116c5746358ca80daa66c0efe5 : SUCCESS in 1h 14m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/257e38d3a20f4f7b84e06c42c0607aba : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51eb6292d1c149b3b130177b82fe9ce1 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f218fc5d99af45738653a02e7e4d1e3c : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e13f180fbd8d41b3ad4eb0d3a6fe70fc : SUCCESS in 6m 48s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/410f13e9938a4d0e9804c83e7202adc9 : SUCCESS in 20m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b7244f53b43d4fc2b804ec275a2955fb : SUCCESS in 53m 04s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/de8bf9dc79d94ab7b6c547ff052d09c9 : SUCCESS in 1h 06m 08s","accounts_in_message":[],"_revision_number":2},{"id":"98296c414f51c7161a35c796b31ceea94bb74edb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 13:43:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a3ae46150237f38a96716ca1461afeb88d258c94","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 13:44:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5639211ce1b24f51a6241dad5a6a2388 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"556b9382a5f8b2e72c65dd3176877bd53ce3f7d7","revisions":{"adcf72abe8f0eeac8b69e28dbb78fd24f944116f":{"kind":"REWORK","_number":1,"created":"2021-01-11 12:36:35.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/88/770088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/770088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/770088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/770088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/770088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/770088/1"}}},"commit":{"parents":[{"commit":"790f96133f442807d7facbc654ef2f14ea13452c","subject":"Remove footer block from intermediate images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/790f96133f442807d7facbc654ef2f14ea13452c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-09 14:07:37.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-11 12:36:35.000000000","tz":0},"subject":"CI: revert to public package mirrors after build","message":"CI: revert to public package mirrors after build\n\nWhen building images in CI, we use local package mirrors to reduce\nexternal network traffic. These mirrors are not necessarily reliable nor\nare they guaranteed to be accessible outside of the CI environment\n(although many are).\n\nWhile it is an antipattern to install packages in containers at runtime,\nthe Bifrost container does just that. If the mirrors configured in the\nimage are unavailable, Bifrost deployment can fail.\n\nThis change fixes the issue by reverting to public package mirrors as a\nfinal step of the image build process.\n\nCloses-Bug: #1902101\n\nChange-Id: I2592a736206eaec811290e9fbdbf2540c0518ffe\n(cherry picked from commit 0fc464332c789ce05e9d6442726ada84ce7fc65a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/adcf72abe8f0eeac8b69e28dbb78fd24f944116f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/adcf72abe8f0eeac8b69e28dbb78fd24f944116f"}]},"branch":"refs/heads/stable/victoria"},"556b9382a5f8b2e72c65dd3176877bd53ce3f7d7":{"kind":"REWORK","_number":2,"created":"2021-01-12 09:33:19.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/88/770088/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/770088/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/770088/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/770088/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/770088/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/770088/2"}}},"commit":{"parents":[{"commit":"790f96133f442807d7facbc654ef2f14ea13452c","subject":"Remove footer block from intermediate images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/790f96133f442807d7facbc654ef2f14ea13452c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-09 14:07:37.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-12 09:33:07.000000000","tz":0},"subject":"CI: revert to public package mirrors after build","message":"CI: revert to public package mirrors after build\n\nWhen building images in CI, we use local package mirrors to reduce\nexternal network traffic. These mirrors are not necessarily reliable nor\nare they guaranteed to be accessible outside of the CI environment\n(although many are).\n\nWhile it is an antipattern to install packages in containers at runtime,\nthe Bifrost container does just that. If the mirrors configured in the\nimage are unavailable, Bifrost deployment can fail.\n\nThis change fixes the issue by reverting to public package mirrors as a\nfinal step of the image build process.\n\nCloses-Bug: #1902101\n\nChange-Id: I2592a736206eaec811290e9fbdbf2540c0518ffe\n(cherry picked from commit 0fc464332c789ce05e9d6442726ada84ce7fc65a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/556b9382a5f8b2e72c65dd3176877bd53ce3f7d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/556b9382a5f8b2e72c65dd3176877bd53ce3f7d7"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
